)]}'
{"id":"openstack%2Fglance~774315","triplet_id":"openstack%2Fglance~master~Ic999c45fd051c3786c12a1e9d61274288cadc5e7","project":"openstack/glance","branch":"master","topic":"bug/1914826","hashtags":[],"change_id":"Ic999c45fd051c3786c12a1e9d61274288cadc5e7","subject":"DNM: Test web-download tempest negative test","status":"ABANDONED","created":"2021-02-05 22:24:41.000000000","updated":"2021-03-23 14:34:52.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0b2dbe6658fb0b191b9f951fe0fcfce40be35d09","_number":774315,"virtual_id_number":774315,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-08 17:01:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-05 23:58:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-09 17:04:39.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"}],"messages":[{"id":"19e721cacc7e7b7f67f8b132b51b5303a90f273f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-05 22:24:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e2b1ed1e9364783a46a0049b940b0be25305545f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-05 22:36:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ec8ff517732ff22a2fdd7ebd2887608b4edd9258","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 23:58:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/019d2b102ec64ef7adb7c9fbda55dd8e : SUCCESS in 52m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a5339fa5c1c346ec9338ad64ae3711e3 : FAILURE in 1h 04m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb34735b89bd4f068955848c6d0a0471 : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d70f9004c5b44ef38b63aed02aa55cb8 : SUCCESS in 12m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eece7b2a9fb24175a07b2207b0efd763 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b60a19078a444bc5b972bce60212c1e5 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cedb39e796364e8e88ed4d6994d21ec4 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/508b880078a2444fa70c71696d51bf05 : SUCCESS in 5m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/484e59dfa77344adb8f5699cf13ef0c7 : TIMED_OUT in 31m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e56954e976c546f58e25dc5f85df57ac : SUCCESS in 12m 58s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a8aff99e7b21434d8339ff4fcfd70823 : SUCCESS in 4m 17s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a0aa290e69c24769b5ab35a4fe22f774 : FAILURE in 56m 30s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ea43f9d6878440ad8ab375af1a848065 : SUCCESS in 1h 05m 21s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/83c14f9134874e0981a2449bdeaf8fd4 : SUCCESS in 1h 19m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0602ddc3c36a43b583755b5e087bffa0 : SUCCESS in 1h 18m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0529225d5af74334b672c4b8da34f244 : SUCCESS in 48m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a4defaf6ec9749548f3d281af6b9af84 : SUCCESS in 1h 11m 29s","accounts_in_message":[],"_revision_number":2},{"id":"2fcd138754b12386e97a0a2e5dda12ab6970c867","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-06 00:31:27.000000000","message":"Patch Set 2:\n\nrecheck tempest changed","accounts_in_message":[],"_revision_number":2},{"id":"239b7e4c477a0335c84f1abfbfed5d175f145209","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 01:50:34.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/e30fd41e7c664aed8490e360fc4349f9 : SUCCESS in 1h 01m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9af0ce1c0669403e842dd8f853dfe934 : FAILURE in 1h 06m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc791a6c5f7b41b4b654571662acfa64 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afae0bf702674a02943481d1eba5e2d3 : SUCCESS in 4m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b149c118b864e0b9e8c045cd1970043 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3813b3711edd4ee1b3db85bfeead05c6 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f508e6a21d3844b99ffed30ca122feba : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d73b3212f8b4aa3a892df4160c52594 : SUCCESS in 8m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6f1efad4c5744e2db427eff238edfb93 : SUCCESS in 6m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ac7768560964440c96f1b2a6671b8de0 : SUCCESS in 7m 45s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/118d4a6420614812b88c3386f2e30b63 : SUCCESS in 4m 25s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/16743d36b32c49d7a70f6959d2c84e09 : FAILURE in 1h 00m 41s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0c30e591bd244a7caa3f7dabd9206f98 : SUCCESS in 1h 04m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/86a7ea3896274f42bdd64206f097dbec : SUCCESS in 1h 16m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d4fc7b99c9594435b3c364e62355e27e : FAILURE in 1h 09m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3abc0e9d3884eaf9b981d258f22be95 : SUCCESS in 51m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c8a58e6f39604400bdab2ed1041ef4b7 : SUCCESS in 53m 14s","accounts_in_message":[],"_revision_number":2},{"id":"31fcd87ce0c22dd500094bf721b81503d8ec05c5","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-06 02:15:08.000000000","message":"Patch Set 2:\n\nrecheck had to tweak tempest config again","accounts_in_message":[],"_revision_number":2},{"id":"f5d2d4e914bdf0ba02299d6fb644acd383093c31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 03:42:00.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/e9c3550c78db48dcb4acb8e62a88c143 : SUCCESS in 47m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/001f34787e99491c90dfce67bd0116be : FAILURE in 52m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2644743f33864497bc646efafed8dc51 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29f9caf04eb0447d8a7444f1230b091a : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/112b46793c494b228cf598df49140739 : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d297510b5f4b47d99ab88f9415a23069 : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3eef7132668546298938a9d2cdc4a665 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5587404ccc39428c9e6672188451692f : SUCCESS in 7m 47s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/be52fbf727c94e90b34156e57a9c0fea : SUCCESS in 8m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/46dbd2185c474df9b957003eb1a293a9 : SUCCESS in 9m 37s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/084c40f62625434796e65c009b3253da : SUCCESS in 5m 34s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7531ed78c276435c8e565896521b07a7 : FAILURE in 1h 00m 51s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6bec416001d24e5d9c040f4aa6887b78 : FAILURE in 1h 00m 26s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/173ab28d977e4e57950ffd54bf2b9524 : FAILURE in 59m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0481c681ebb84be18f893b4286aa7b04 : FAILURE in 1h 24m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce420ebd721a489482386285e5d453f6 : SUCCESS in 1h 00m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ed32e54d412b434e8c2c70a387bd1b82 : SUCCESS in 1h 16m 56s","accounts_in_message":[],"_revision_number":2},{"id":"0c934ffa4dbace1208ecc33804b00cbfe0c48e9d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-06 03:44:39.000000000","message":"Patch Set 2:\n\nrecheck still trying to get tempest to only run the negative test when multistore and import is enabled","accounts_in_message":[],"_revision_number":2},{"id":"bad8175b0d5c2e106d6c9685b16fa1c61545498e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 05:24:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/d29560b94c3f49e481f7a628ba032bd3 : SUCCESS in 52m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/23e3a4e505c144728434fec05f98dfb7 : SUCCESS in 1h 25m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4de8d098d694104a87f33c2284113b4 : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f518ad514494e69b7fb704fa8a8b495 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c9cf0400d7848f78811cce808a91396 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/281311f61f774500a3831a739f7da7bf : SUCCESS in 7m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67dcf1d7133747088609fb30366a9705 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aee5f2dee6ad4041ad81adae53b15225 : SUCCESS in 6m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/17c27a8e3e4d465c8de878b5e05453a2 : SUCCESS in 7m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fe439a4371e24e80b7fb66117c72be32 : SUCCESS in 7m 38s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3d2163fde89f412e95079845924e4fa5 : SUCCESS in 5m 42s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ee98c29707b346758d8d68994acc384c : SUCCESS in 1h 08m 33s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e722141d00894d769802c5a11f5752d3 : SUCCESS in 1h 17m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/429417301e394bc7b5326102eed24ae1 : SUCCESS in 1h 30m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3f5f2e31f0194e5f98e2943278afc621 : SUCCESS in 1h 29m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e81a53ecb26a40e38802df82ffe7c394 : SUCCESS in 1h 11m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f4c1372667a848138f7f29161357d588 : SUCCESS in 1h 09m 51s","accounts_in_message":[],"_revision_number":2},{"id":"bf72770336acf14305ebb5b053b47250fce744d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-08 16:46:28.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"0c3c6d337fdcf2e9d5c1120e1a3fc69ca61b835d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-08 17:01:23.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"db151d1d760b6c318ce9166d4c49753584e0c521","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-03-09 17:04:39.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"59b1acbdd164d7ed9b0e00d6b79b67552a185bfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 18:46:17.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/b8f79b0a4eca45238032c8cae2680cae : SUCCESS in 1h 14m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fd54e969c2274cebbcf5ab435d3b5653 : SUCCESS in 1h 24m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/24b7ac5fd05f4ed991af6158b87f9947 : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/163a8fbccdd44ec6b44337f772ed09f2 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fb28eb7137c42b3a37da1069cbff129 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50e1a3584fe04b1db7ddf1629df3772c : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a35d34e6f82d4f8e87c959a44e27c4df : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d03c8030f18e4e49a2dd6249b9ac6d37 : SUCCESS in 6m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3f45762e4a684a529e2b111cb4fb2d6f : SUCCESS in 8m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bdd52d4916fb4dd2a81938b57eb0e3ea : FAILURE in 8m 16s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/000a19c118c645a1b7094aebaadb3c6f : SUCCESS in 7m 10s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/290356a448ed42adbe6c28d0f5419b6d : SUCCESS in 1h 05m 07s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/18370b0a7e5f417484d5e2d72a3fb4b7 : SUCCESS in 1h 14m 09s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fa388bedc0b54ad6a8edf1f5ae62851e : SUCCESS in 1h 11m 46s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a424d8e451664d5b9478ee0bff78d7d0 : SUCCESS in 1h 14m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc2df13b299d48acb1b4a4ad6d9d655e : SUCCESS in 1h 00m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e64527ad38374fd8b159380fce4db3c3 : FAILURE in 1h 29m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/24c669a1d5b54952b7e46fbca3df4f4e : FAILURE in 38m 55s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f4de9e071f454414a87c1b3ee8819805 : SUCCESS in 37m 17s","accounts_in_message":[],"_revision_number":3},{"id":"e87392a1ebc45953879ef62644773efeb9bf375e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-09 18:54:52.000000000","message":"Patch Set 3:\n\nFailed a cinder test, but passed the one we\u0027re checking:\n\ntempest.api.image.v2.test_images_negative.ImportImagesNegativeTest.test_image_web_download_import_with_bad_url [1.281245s] ... ok","accounts_in_message":[],"_revision_number":3},{"id":"0b2dbe6658fb0b191b9f951fe0fcfce40be35d09","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-23 14:34:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d33a00208aa1230aa685ff105ddcd25d5b4680d7","revisions":{"c58cd8dc87c917d05218ff084b253f5cb247a143":{"kind":"REWORK","_number":1,"created":"2021-02-05 22:24:41.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/15/774315/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/15/774315/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/774315/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/774315/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/774315/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/15/774315/1"}}},"commit":{"parents":[{"commit":"82f529e31a4df58163b390216228aef5c996353b","subject":"Make web-download revert all stores on fail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/82f529e31a4df58163b390216228aef5c996353b"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-05 23:22:07.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-05 23:22:07.000000000","tz":-480},"subject":"DNM: Test web-download tempest negative test","message":"DNM: Test web-download tempest negative test\n\nThis just checks that the patch below this fixes the tempest test\nfor web-download fail.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/774303\nChange-Id: Ic999c45fd051c3786c12a1e9d61274288cadc5e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c58cd8dc87c917d05218ff084b253f5cb247a143"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c58cd8dc87c917d05218ff084b253f5cb247a143"}]},"branch":"refs/heads/master"},"390b45372c6ff331bc7f1396083da17a663a2b7b":{"kind":"REWORK","_number":2,"created":"2021-02-05 22:36:42.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/15/774315/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/15/774315/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/774315/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/774315/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/774315/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/15/774315/2"}}},"commit":{"parents":[{"commit":"82f529e31a4df58163b390216228aef5c996353b","subject":"Make web-download revert all stores on fail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/82f529e31a4df58163b390216228aef5c996353b"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-05 23:22:07.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-05 23:35:01.000000000","tz":-480},"subject":"DNM: Test web-download tempest negative test","message":"DNM: Test web-download tempest negative test\n\nThis just checks that the patch below this fixes the tempest test\nfor web-download fail.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/774303\nChange-Id: Ic999c45fd051c3786c12a1e9d61274288cadc5e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/390b45372c6ff331bc7f1396083da17a663a2b7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/390b45372c6ff331bc7f1396083da17a663a2b7b"}]},"branch":"refs/heads/master"},"d33a00208aa1230aa685ff105ddcd25d5b4680d7":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-02-08 16:46:28.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/15/774315/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/15/774315/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/774315/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/774315/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/15/774315/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/15/774315/3"}}},"commit":{"parents":[{"commit":"cbb2af6e329c1bf6c1c39cb59f95aaad2c0d7a07","subject":"Make web-download revert all stores on fail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cbb2af6e329c1bf6c1c39cb59f95aaad2c0d7a07"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-05 23:22:07.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-08 17:44:35.000000000","tz":-480},"subject":"DNM: Test web-download tempest negative test","message":"DNM: Test web-download tempest negative test\n\nThis just checks that the patch below this fixes the tempest test\nfor web-download fail.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/774303\nChange-Id: Ic999c45fd051c3786c12a1e9d61274288cadc5e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d33a00208aa1230aa685ff105ddcd25d5b4680d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d33a00208aa1230aa685ff105ddcd25d5b4680d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
