)]}'
{"id":"openstack%2Fglance~774317","triplet_id":"openstack%2Fglance~master~I13d9a8a1c875a2a34ec1f5505fbccb50a90a6011","project":"openstack/glance","branch":"master","topic":"repro-bug-1914826","hashtags":[],"change_id":"I13d9a8a1c875a2a34ec1f5505fbccb50a90a6011","subject":"DNM Reproduce failure for bug 1914826","status":"ABANDONED","created":"2021-02-05 22:40:13.000000000","updated":"2022-05-05 14:21:43.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4625061e9e5ee0c3d0a69faf2840ec9f74faa49f","_number":774317,"virtual_id_number":774317,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-06 03:40:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-06 00:04:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d618192876a1f551cd682f3ca2dcc796d36b2a48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-05 22:40:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25431a698cecb1ac54c228880e6c66b99c569c94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 00:04:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/81664152a7b741d6a0a6eb4d68f286f9 : SUCCESS in 59m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/78421f17b18940ab9d62a50047e1a079 : SUCCESS in 1h 06m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ad2b90b82463463b8372f3279a0b6f42 : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/375d53e388a64af19d0bd4b3ae321ac2 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9be1b586a8214902a0f1424ec7b27c8b : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b07886fc3eaa4816a5e94e2bde35a16b : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ef297f6d60346dc9c0bfcdc137dc3e2 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f9ee97d5ac549a9ab1cca1d35cd00ea : SUCCESS in 8m 09s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/30a7bc548fc04ff5966a9c5bd304062c : SUCCESS in 1h 16m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6e353a0968de4096960f52936e941afa : SUCCESS in 1h 03m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/90ee777f6ee14af78d1704ef03dc8edf : SUCCESS in 1h 22m 09s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9239c0056b094784aa8f1d432f23897b : SUCCESS in 1h 05m 52s","accounts_in_message":[],"_revision_number":1},{"id":"81e199208d3736582b5d9be7a7e3c4963c988f60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-06 02:16:45.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"72c55951f1408ac4aa339d3fa031a9f3a601a1f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 03:40:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/5d489ee99ebb4befabfbb3c56045c0fb : SUCCESS in 1h 12m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4bfc5b7bc05e426391d0081408938b48 : FAILURE in 1h 18m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69df58a4d7b24ced94e2629944793983 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9ba2bd7548a451287612b3024f64ae1 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9859d39a5bf24a4aa8324b74b1302db3 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c9355e2559c49d6af3347ae74cfc217 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/464053286f7f4f2697146f1427425187 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cfc2fb77c35426395f308833531f80b : SUCCESS in 6m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6e715b7297404bd291ffe70765ca497e : FAILURE in 1h 01m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fb1a84497afd47268b24c87c5c78c75b : FAILURE in 1h 02m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8f954b824b8b42cb92b72701e5414f82 : FAILURE in 1h 09m 12s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1d212093148540aa8589310767e1a7a3 : FAILURE in 1h 08m 02s","accounts_in_message":[],"_revision_number":2},{"id":"4029b1f720f3864efd8f0c863e0dd5197fdedbd8","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-06 03:43:26.000000000","message":"Patch Set 2:\n\nrecheck tempest changed","accounts_in_message":[],"_revision_number":2},{"id":"f4a921e8df40ce9aef6e260b6ae948e0f9bfbb7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-06 05:13:41.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/c31a0f08b3b84451ad148a84939a43b1 : FAILURE in 14m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e0166e5242804618af3bf679cbe7f96d : SUCCESS in 1h 19m 56s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2d649e832ce46cd8efa0efc88f5ebc5 : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08f055cdd9144d5285365d04cd000a66 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69929d5f2c1c4b11a3897d550779731a : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d88098c4559f43268d04e56d7ad71a8d : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4740a344a3714b26b090d3ee3db07012 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a4ce4e290f8415d846ad16f45e4ef2f : SUCCESS in 7m 31s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4a501df27af94036874d0bf232476fb6 : SUCCESS in 1h 13m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0bf0455915cd46e487dc97f2c0debe86 : SUCCESS in 1h 23m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7b493510a4ce45afa8699cb00662a41e : FAILURE in 1h 25m 17s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8c509ec6145940929ace2962c549aee2 : FAILURE in 1h 08m 50s","accounts_in_message":[],"_revision_number":2},{"id":"4625061e9e5ee0c3d0a69faf2840ec9f74faa49f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-05-05 14:21:43.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"46708fb415e5a754fe7c1738769954e00b2075ae","revisions":{"733a0bd0cd8773171139125d5b5ae868e0202f13":{"kind":"REWORK","_number":1,"created":"2021-02-05 22:40:13.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/17/774317/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/17/774317/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/774317/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/774317/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/774317/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/17/774317/1"}}},"commit":{"parents":[{"commit":"f2452863e74b563fdb4b1001af76ea492c63c484","subject":"Merge \"[goal] Deprecate the JSON formatted policy file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f2452863e74b563fdb4b1001af76ea492c63c484"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-05 23:38:25.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-05 23:38:25.000000000","tz":-480},"subject":"DNM Reproduce failure for bug 1914826","message":"DNM Reproduce failure for bug 1914826\n\nChange-Id: I13d9a8a1c875a2a34ec1f5505fbccb50a90a6011\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/733a0bd0cd8773171139125d5b5ae868e0202f13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/733a0bd0cd8773171139125d5b5ae868e0202f13"}]},"branch":"refs/heads/master"},"46708fb415e5a754fe7c1738769954e00b2075ae":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-02-06 02:16:45.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/17/774317/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/17/774317/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/774317/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/774317/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/17/774317/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/17/774317/2"}}},"commit":{"parents":[{"commit":"f2452863e74b563fdb4b1001af76ea492c63c484","subject":"Merge \"[goal] Deprecate the JSON formatted policy file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f2452863e74b563fdb4b1001af76ea492c63c484"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-05 23:38:25.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dms@danplanet.com","date":"2021-02-06 02:16:45.000000000","tz":0},"subject":"DNM Reproduce failure for bug 1914826","message":"DNM Reproduce failure for bug 1914826\n\nChange-Id: I13d9a8a1c875a2a34ec1f5505fbccb50a90a6011\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/774303\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/46708fb415e5a754fe7c1738769954e00b2075ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/46708fb415e5a754fe7c1738769954e00b2075ae"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
