)]}'
{"id":"openstack%2Fglance~774724","triplet_id":"openstack%2Fglance~master~I0574ee3daf08d59b4547e353c921451e756e09f6","project":"openstack/glance","branch":"master","topic":"bug/1914826","hashtags":[],"change_id":"I0574ee3daf08d59b4547e353c921451e756e09f6","subject":"Allow plugins to mutate image extra_properties","status":"MERGED","created":"2021-02-09 20:14:25.000000000","updated":"2021-03-07 09:49:48.000000000","submitted":"2021-03-07 09:48:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"774724-bug/1914826","meta_rev_id":"2717c84604d371ef34a9667c1fbfcccb12602129","_number":774724,"virtual_id_number":774724,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-07 09:48:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-02-10 14:50:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-03-05 18:30:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-02-10 14:50:22.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-10 00:36:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-10 06:01:59.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-02-10 15:25:17.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"940f27c4675b831fd6b4455414cdebc56d781b48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-09 20:14:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d235107330b5aa6b7553ecb584340a58d05ff5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 00:36:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/91bca1b083c74ca4aab209d170874b9d : SUCCESS in 1h 03m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5d41068c842347a7be89c68591875857 : SUCCESS in 1h 29m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6092fbd444c4fb3a1fddca5abc231d5 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1ba4c6b24d84e64bf5e64e506a8ab4a : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1122f5058aa48e9bc6602a8bec48a0c : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42c5fa5bab3747ae8cfba2a6e6d765d2 : SUCCESS in 8m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49480a2fec4043cd8a02f5b66edfe336 : SUCCESS in 8m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b72ca5c31c04b79bdaa3f4da6f06ecf : SUCCESS in 6m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e4b0dd5e684a4ea1a343a2dedfc28c39 : SUCCESS in 7m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6d2de92fc5fb4649bbe4b8c4f711e309 : SUCCESS in 10m 02s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4ea6d250b5fc496a9d307ba50fd99248 : SUCCESS in 5m 26s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ca1aa533cbfc41b5a07cdc64c88c9bfe : SUCCESS in 1h 01m 46s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cadb8eddcab04d2888c242fde209b571 : SUCCESS in 1h 15m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5c5cf66e3be24c6f8ec7a9d9e5f816e2 : SUCCESS in 1h 17m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f7dd60d108d84dee9858fc1fae25c7da : SUCCESS in 1h 25m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d25ff23970bf4e1a93fee4ca7dbb0d8f : SUCCESS in 1h 01m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6619d20b49a04b6895278d67656d8821 : SUCCESS in 1h 33m 09s","accounts_in_message":[],"_revision_number":1},{"id":"14fb18b7d74bbb152d30f29798894276cadc52b2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-10 06:01:59.000000000","message":"Patch Set 1: Code-Review+1 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"10542e041187e8997fcf37e4fff4f39c2ed95c4a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-10 14:22:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d170d451df5a88ca69d3f9a28fe55cc837697edd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-10 14:42:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"97d1332951aee58ac2ddb2a23eb148d567f484cd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-10 14:50:22.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1\n\nThank you for additional checks.","accounts_in_message":[],"_revision_number":2},{"id":"40e8997a4f01574e8c67533bab08c229e930e0da","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-10 15:25:17.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/774724/774724/2/check-not-vote/ext-glance-zuul/171db04 : SUCCESS in 40m 11s","accounts_in_message":[],"_revision_number":2},{"id":"9803d3b905236b4c63ea3cecd5c6f6a2bc638370","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 16:18:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/970a8ef1286e43818cf129e9c23751eb : SUCCESS in 1h 19m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f3dc3ff257134e0e81d3e70f827d5e26 : SUCCESS in 1h 14m 04s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d98029ac2b9c4409a0ba252e22207955 : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/197375a0110e4544826e6e1f68c472e1 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/421873043a5c4b26afc24389ca61494c : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8af848432d164a9a97dc992e3df628b6 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbd470d2524a44eca2bdff26a19d36e6 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7865375a7f84b97b6b165d4d1934eb1 : SUCCESS in 5m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b19f14ef9867417197a664928f981f6f : SUCCESS in 8m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/af1e1f058b094d2a856b0dcf8a098ac9 : SUCCESS in 14m 28s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/801c188c1d274972b911de5bcd4b11a3 : SUCCESS in 6m 50s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/193eb84f913c4beca2d6e426af317d3d : SUCCESS in 1h 26m 39s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b582a55c0cb842ecb5e8c972b8a0d34e : SUCCESS in 1h 27m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/87ce2a378aae40a08e3d167b20990dd4 : SUCCESS in 1h 10m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/62823615e65b4f978c587b2164a2c7c8 : SUCCESS in 1h 19m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17b689283bab4950b7baf6d5782c5ea2 : SUCCESS in 59m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/37c9aab9176d472a86e28c497849d2e5 : SUCCESS in 1h 19m 27s","accounts_in_message":[],"_revision_number":2},{"id":"e33de8fbcab45d017af017a2070af0c858781a98","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-05 18:30:02.000000000","message":"Patch Set 2: Workflow+1\n\nNinja approving, sitting open for long, good to restrict addition of reserved properties (os_glance_*) using import workflow.\nThank you!","accounts_in_message":[],"_revision_number":2},{"id":"6437179d9b894453aa4e24f7e48faa30cfc1a9f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 01:55:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6f40897463d7fb34f109ced7bc4a8153405a5246","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 11:41:44.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"1d2aef0865439635bae25185638d98f59cfee6a6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-06 11:48:51.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7548032cc676060ed0c8cb2266c1f12d7a7be11e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 13:14:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/f16d82c0976049be87ffb80b37f89177 : SUCCESS in 45m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8daec2b26431485db4524b69def353f3 : SUCCESS in 1h 22m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fdf4c126aa544fdc9c109baddee2412d : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37e66d7d7432494899b68581d4926acb : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e276afc047c142ca83125ef63389ac7b : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5abdba0ed9de421da396e101d8c2303e : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a17db19634ed4a069d7beda46ddfcccb : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/421cfa4ee69543ab97b3e61865119351 : SUCCESS in 6m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/40c467769c544c3489d2457f75a5a7f7 : SUCCESS in 8m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c02afda0e6f841f19f03243110dfc127 : FAILURE in 9m 24s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/be37185108004b9d9dd14349f02310ee : SUCCESS in 5m 20s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/474b241e7d9a4c34a00c4da523d3d7f6 : FAILURE in 1h 21m 12s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3f6e1b9ecd6947d7abcaaf3c563643d6 : SUCCESS in 1h 01m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/47894d7ee78b4089aed8886c11fe3954 : SUCCESS in 1h 04m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/aa855d393c784c3e935081dafaf84d8f : SUCCESS in 1h 07m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/759991e910344210934b23df91652119 : SUCCESS in 51m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d05d9eb20e4b49b3bbb96099ec76c70e : SUCCESS in 1h 02m 14s","accounts_in_message":[],"_revision_number":2},{"id":"4acb403eba36489222649acf3127077ed5f9d156","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-06 13:29:08.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a0aabe297eafc602746201a12c084e75df167c63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 15:02:09.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/64497bb5dda84b1897fc008fcde61756 : SUCCESS in 1h 06m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e48178202dae48a7a7d853ce57fa7588 : FAILURE in 1h 06m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/779679b03db442d0bba91fab6b976de5 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24a67cb3cda74982a8363196d7aaee85 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c08874b657b0463388bbf096aedd5834 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/403b21e299564e34bd85a8e9d79eb995 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2af1338f309d4c00a9a639594797868b : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/740ed4216c9e491d9ddd3fb3ff126857 : SUCCESS in 6m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/84777a5c41794880ac8df9205e767d2c : SUCCESS in 8m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/926483bd302444f0aaa6ab4ff3481924 : SUCCESS in 8m 22s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/fe6d2ce9aded4b01a25e862a70699c50 : SUCCESS in 4m 48s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9efbd8cc9d5a4507b51dd2d43be6b628 : SUCCESS in 1h 15m 11s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/497490ee4fcb4ce9ae2a84fd5dfccfd9 : SUCCESS in 1h 17m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a2ecbaa5c866450ab303835744e49875 : SUCCESS in 1h 26m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3bc2159ec0f942cdbaed819da1e03b9d : SUCCESS in 1h 26m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/237ad7fdee14414dbcae145e517169eb : SUCCESS in 58m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d9c4f313e6e748f5a1d9a56f6d707a51 : SUCCESS in 1h 16m 19s","accounts_in_message":[],"_revision_number":2},{"id":"ba806cd8fe58c7c2d8847364fbbefda2dd575292","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-06 16:51:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"dec9b983b7593391f548f1653cc97af96c675db5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-06 17:41:29.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1cc2bcd472faff57223dbf77cbe5b24a1ae06feb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 18:18:36.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/be7d9131b71f422c99a65f353316b505 : SUCCESS in 43m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a060ddb8c5c142be8e2d71985594e8ac : FAILURE in 1h 24m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/225a280f9197454b9106f4df9e38b639 : SUCCESS in 8m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6dc695cdcd9e46f69c57562bd2043639 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e7063c50b674aa4815a5e1fd91166d0 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/208ee38d0ec04597b05a84529874a00c : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/baeeb37f7d4449278c388a37d99306e0 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2560f6fd1dee4b479b5691ddd1910b40 : SUCCESS in 8m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b5f1604a50894af18e1ba174ed9de356 : SUCCESS in 9m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4bc57e077bb14c0699ca4f095e9366e9 : TIMED_OUT in 31m 02s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d2a1049e07274f6c912592f8dd128713 : SUCCESS in 5m 13s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/60f3f63e3e184a55a6753a9f79704103 : SUCCESS in 1h 23m 12s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/43185d99decd485fac968a12b26ba0b4 : SUCCESS in 1h 07m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f32533a20ceb49f396b42dc56bc62905 : SUCCESS in 1h 07m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f95ec8c6f115401fb0edc5c417649bbc : SUCCESS in 1h 13m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b608e9d7b5d747c9a3aad5cd3d809ba5 : SUCCESS in 58m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/79bd16a0658a43db842622c63ae82ffc : SUCCESS in 1h 05m 50s","accounts_in_message":[],"_revision_number":2},{"id":"c336590e9bcb2667e6abf1dab14e05da26da0cd0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-06 18:21:24.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8fa4026b90a5d96c13b336ae2da94fe9aa6c384a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 20:25:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/a7a877b843504baa9a7b7ab6cba4ce2d : SUCCESS in 1h 08m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b2529724e1684bd99f19e0822f336812 : SUCCESS in 1h 16m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/42a885a518024007b9bd1be051a0a700 : SUCCESS in 8m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d33bb6b697744dc9a50dbea7372eadb : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/238753691ca4475f8bb95e8a556153fe : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5430f3125866429292b771cd9d2a3b01 : SUCCESS in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48dcb44645534a74acb9ff08a9625d87 : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa3016d0150c47f083914ecfffb2460f : SUCCESS in 8m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1908eddca38248a2be85ff516cd1ca0d : SUCCESS in 11m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d61c35ae1e9d4f00b60e2c57f4497cb4 : SUCCESS in 10m 52s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/db7bb8ba92754e5c8ccc73d27a6553ad : SUCCESS in 6m 36s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e2ce37a035bd4007a6fdb47412c9bb26 : SUCCESS in 1h 12m 11s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ed0220efb92241688b4b7e6226ac3306 : SUCCESS in 1h 06m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8e7bf920934b41c79cd9dfe3dd2417dd : SUCCESS in 53m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fd45a90c24904daa9858fa89a754f109 : SUCCESS in 52m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf21cdd9f7c54b6fada08216d63d4c2c : SUCCESS in 48m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3e8ac1f3fc384dc4889c26c4be9c49a6 : SUCCESS in 1h 03m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a843ce40d5064f43bcb7ff255fa2f786 : FAILURE in 49m 33s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7585559608334956823fc889e6a53730 : SUCCESS in 49m 20s","accounts_in_message":[],"_revision_number":2},{"id":"8fe26d421869b74b5afbf1b62993b7b8cbafbf4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 01:56:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"46feb810985ebfe54e6fa08b9424b32047a2749a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 03:26:10.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/bcc51d89999c4475aed0e726c438ce31 : SUCCESS in 56m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dfa7c1a5e2e94ed8b4d0a6cee3e975a1 : SUCCESS in 1h 12m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02c17217148549ce9016bdf7689ce8a0 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1ad0ebe000b4d4b82a75235f5026961 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/467b3512e7e5467597ce231e0f7bb783 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1f69837c7b94e99b319c1cc2ee09d08 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5186b72c56c74974ac1b172d0359bac6 : SUCCESS in 6m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a779ff3116a34346beb9cc9fa377655f : SUCCESS in 9m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d27e83c46e09487b9125e3a4684a6ddd : SUCCESS in 9m 11s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f28462e1358c43afad23c6a3088e3112 : FAILURE in 1h 03m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/728e05ce37b64f3893a69a4379513bb5 : SUCCESS in 1h 15m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2c572b2a15e4b40957e16668e131c84 : SUCCESS in 55m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e1a8ceee64bd4827b456f480347ffee4 : SUCCESS in 1h 17m 29s\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/55db60091cd447d2aee54c602fc2f3c8 : SUCCESS in 32m 15s","accounts_in_message":[],"_revision_number":2},{"id":"7c709cfb3a96cc97723ec5b6a50bd10a483445fd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-07 03:33:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"b88745ad0fe4df7287f85043d25d9c293414b03c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 04:55:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/4da133d3eeee478aab511ff0024a0a80 : SUCCESS in 53m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9fd0e0107df14b959c2fa9009c7a3125 : SUCCESS in 1h 15m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eab71b88fb0548da98e86c694bb2b967 : FAILURE in 10m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3b113ab81a8417d938a9b735084b023 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d61841f7f5cc46d5ac7d51a2dfbf93ca : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/311b1d72989e47978d6706587706e439 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72e3321d53d4410e8446b5abf23a6fc7 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44474608a7594aee9db9bba32a22c16b : SUCCESS in 5m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/af6d8ed1efd0491196d1c0b406f06222 : SUCCESS in 11m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4a5ff71aecb44bff989a5e5309f8818a : SUCCESS in 8m 21s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a42bd2f14f6945bcacdd9a80d8d36ae1 : SUCCESS in 4m 48s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5bc7b3e76f1d46e69bbaf0541361de59 : SUCCESS in 1h 15m 01s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/dad924ca928047fc8058e632a444dfd7 : SUCCESS in 1h 13m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9a1038e0921d46f1893699ddec8814b7 : FAILURE in 56m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fc0a81d801994bdda83aa9466dd8f3d2 : SUCCESS in 1h 03m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4333b01204d6431fb02e0394f7e9eb8d : SUCCESS in 54m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0424870284054d818d54739374944bc9 : SUCCESS in 1h 02m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cb119483eab947208ad8594b09687887 : FAILURE in 33m 26s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b6245c47b9de4131bfc601573e3a9a4b : SUCCESS in 30m 05s","accounts_in_message":[],"_revision_number":2},{"id":"16343ca98c83a0fe66cd4c5787ad7fd961d50ab5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-07 05:04:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a5636f0e4e888db2ededed2997a967692258d5f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 06:38:55.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/4c053c7aff8a4bf1b6aba7abdaad4685 : SUCCESS in 53m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/441ffea440354a09ae13106c2b2fd1ba : FAILURE in 1h 06m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c685ffbd45e408fbf198e3311fdfc49 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20381ae1216f4b86b9809d08d0d8385b : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7cb6a5809c14e9680ca072aac741696 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11b218ab4e244d569cba4ff9eb19d021 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/546bed3a952d4f47a9195a766a6615fb : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5ae3ad2698540cd9f40fd0edc55476d : SUCCESS in 6m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a2bb7133fe0f4fec95f8d89f5e867c47 : SUCCESS in 8m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6d9e992ba8cb4e0ea5f49f4aa4a1a5c8 : SUCCESS in 9m 14s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f35e2527013f48feb2afe6485d161c6d : SUCCESS in 5m 06s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7423d39942724b5c913adf4b54a6a167 : SUCCESS in 1h 22m 56s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d666e32542344aeca0db0febdf6c19b2 : SUCCESS in 1h 22m 38s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1ee30eff921a408aafad5388f7e50619 : SUCCESS in 52m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/aee304f082394f0a8a5eb42fe851651e : FAILURE in 1h 15m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bbf92c445b440dc966c8f3038c279d8 : SUCCESS in 50m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/90857fdaf9684d8191eddd3938cacaa6 : SUCCESS in 1h 26m 22s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1146120ee5864f6fbe71709ac5efa706 : FAILURE in 40m 34s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0986149f183b425691d299cdcccb2e6e : SUCCESS in 33m 57s","accounts_in_message":[],"_revision_number":2},{"id":"d95a2230cad63446f90368e50ce70696acae578a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-07 07:07:12.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"996f36d81c71436a9b2b96dc892aea00ddaab2c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 08:25:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/ee9cc2469e794e84a994bc531fb97174 : SUCCESS in 46m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cb54cccfe1b2439d8af11b85fd93e813 : SUCCESS in 1h 10m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab9b250d32064fbebf559ead33d5e9a5 : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12871669721a4121add200a1e948e761 : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4e64ff390b7c4a06bdc656d25f1125ba : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f06a48241c24f5ca83e491c4a5e1c8b : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ddc8d0e696b42fca4f3a6acdcc14f3d : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dd74c5d218249339872b84ef64b376e : SUCCESS in 7m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7c4ce240055a4a52801832eb3e6a81f3 : SUCCESS in 9m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d27e6446040b4e7d83565a69d783e906 : SUCCESS in 11m 08s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/96dd8fb3336a4b63af04e21eec86b013 : SUCCESS in 5m 23s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d901b3f04b784c7cb282c34d0c06c6c9 : SUCCESS in 1h 16m 19s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0297a82b9e424f279f7b08aa32fab677 : SUCCESS in 52m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ae698de8754340bf89e84ce9d41ea2fc : SUCCESS in 54m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ca02449cd17f47d98eabf16d6dfaf2e6 : SUCCESS in 1h 07m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1b113e84b5b4b3d96a392c8df6a281f : SUCCESS in 49m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9ad9dccf5fa042f0b22ecf4726b6b6af : SUCCESS in 50m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/376a9b217fdc4881a7975bb6e88e584a : FAILURE in 28m 51s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c398a08f579d49c8966b9fd2c034026e : SUCCESS in 30m 09s","accounts_in_message":[],"_revision_number":2},{"id":"899d4af4ec8e11247f0940d5d82f515843fe9575","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 08:25:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2d90823440c57813825bb0e6ebda7ed3f1315cd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 09:48:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/b059b6ab5b9846d4b84611f4d2ced670 : SUCCESS in 1h 01m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3cdd41275fdd4c19af7794cc58fe77fc : SUCCESS in 1h 21m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ef1249c151d42d7976a882893e6e817 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c2db9a493734cd49c80f7a75d1522be : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abbefe1e614c467fa80fd821e4ea1ae7 : SUCCESS in 8m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ddb2f73a41e4363b5d752d880dc6291 : SUCCESS in 7m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a5f23f686ed487fa2e93b5f11f11c5c : SUCCESS in 7m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/44ac721027bc4813b7bc66cf8558fe3f : SUCCESS in 11m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0b4ec1ce529b4f73a7d3a9a2f3c577f0 : SUCCESS in 12m 18s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/89983b028d014726bbbb66638ea31abc : SUCCESS in 1h 06m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/60a170b266dc405a8b126b0d64c525d8 : SUCCESS in 1h 07m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8090e15e146c48bab6001d8d57aeaba6 : SUCCESS in 54m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c77eebabd0b04a4ca2c56f833acea983 : SUCCESS in 1h 05m 06s\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2c908e5875694cc88688607ba97e6605 : SUCCESS in 33m 30s","accounts_in_message":[],"_revision_number":2},{"id":"e5fd47a207ededf4e0aad106eaa23c096d5df5f2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 09:48:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2717c84604d371ef34a9667c1fbfcccb12602129","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 09:49:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c40687d5f6b54f2991eb3ec4730affa0 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e54a1a52ec246e199d2a9731a390111ae3ba61da","revisions":{"9cf5c1769cd05212d58211ee254fa9c95a65a605":{"kind":"REWORK","_number":1,"created":"2021-02-09 20:14:25.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/24/774724/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/24/774724/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/774724/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/774724/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/774724/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/24/774724/1"}}},"commit":{"parents":[{"commit":"d2bb1252e6192773cdb7960c25f8031ae896fd71","subject":"Make image_conversion use action wrapper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d2bb1252e6192773cdb7960c25f8031ae896fd71"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-09 19:55:46.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-09 20:33:05.000000000","tz":-480},"subject":"Allow plugins to mutate image extra_properties","message":"Allow plugins to mutate image extra_properties\n\nThis adds a set_image_extra_properties() method to the action wrapper.\nIt specifically disallows setting os_glance_* properties, for two\nreasons. First, several (such as the task lock and the store lists)\nneed special handling to be atomic. Second, setting os_glance_*\nproperties from image metadata injection is almost definitely\ndestined for failure, now or later, as it would muck with our\ninternals. As discussed during the work to formally reserve that\nnamespace from the API, we also need to make sure operators do not\nset these keys during injection.\n\nThis makes us drop any such keys, with appropriate logging. The\nnext patch will make the metadata injection task use this and is\nthe point at which we will actually change that behavior.\n\nChange-Id: I0574ee3daf08d59b4547e353c921451e756e09f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9cf5c1769cd05212d58211ee254fa9c95a65a605"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9cf5c1769cd05212d58211ee254fa9c95a65a605"}]},"branch":"refs/heads/master"},"e54a1a52ec246e199d2a9731a390111ae3ba61da":{"kind":"REWORK","_number":2,"created":"2021-02-10 14:42:45.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/24/774724/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/24/774724/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/774724/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/774724/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/774724/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/24/774724/2"}}},"commit":{"parents":[{"commit":"d2bb1252e6192773cdb7960c25f8031ae896fd71","subject":"Make image_conversion use action wrapper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d2bb1252e6192773cdb7960c25f8031ae896fd71"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-09 19:55:46.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-10 15:36:21.000000000","tz":-480},"subject":"Allow plugins to mutate image extra_properties","message":"Allow plugins to mutate image extra_properties\n\nThis adds a set_image_extra_properties() method to the action wrapper.\nIt specifically disallows setting os_glance_* properties, for two\nreasons. First, several (such as the task lock and the store lists)\nneed special handling to be atomic. Second, setting os_glance_*\nproperties from image metadata injection is almost definitely\ndestined for failure, now or later, as it would muck with our\ninternals. As discussed during the work to formally reserve that\nnamespace from the API, we also need to make sure operators do not\nset these keys during injection.\n\nThis makes us drop any such keys, with appropriate logging. The\nnext patch will make the metadata injection task use this and is\nthe point at which we will actually change that behavior.\n\nChange-Id: I0574ee3daf08d59b4547e353c921451e756e09f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e54a1a52ec246e199d2a9731a390111ae3ba61da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e54a1a52ec246e199d2a9731a390111ae3ba61da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
