)]}'
{"id":"openstack%2Fglance~774726","triplet_id":"openstack%2Fglance~master~Ib1138ee2cd0383f14c710716f3c103a146b0c4dd","project":"openstack/glance","branch":"master","topic":"bug/1914826","hashtags":[],"change_id":"Ib1138ee2cd0383f14c710716f3c103a146b0c4dd","subject":"Make inject_image_metadata use action wrapper","status":"MERGED","created":"2021-02-09 20:14:25.000000000","updated":"2021-03-07 16:06:54.000000000","submitted":"2021-03-07 16:05:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774726-bug/1914826","meta_rev_id":"8a05a011642831ef97e107e89b26c6cee477c706","_number":774726,"virtual_id_number":774726,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-07 16:05:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-03-07 16:06:54.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-03-05 18:31:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-03-07 16:06:54.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-10 00:27:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-10 06:34:05.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"6971e2c44768b290aedbba5718ff79610fe93e0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-09 20:14:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"30e5e8f1d3e1d290ead54ff752b1c1255c470fa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 00:27:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/277f0f7a7981425887364e4e21ae10b5 : SUCCESS in 1h 04m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d2ceba4f709b4b528fd0dde0838bf57f : SUCCESS in 1h 17m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4e8a80e49ec43a98874e3931b4b6ebf : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b842d57fbb0a41ca92228896e3931776 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2ceced7b4934cf2a929a682c6193a18 : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/435652f416bd44c4ac586e6ebb6a0ef0 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e70f2c45eea49dabdad79b7deba26b4 : SUCCESS in 8m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b63f400f26e4241bcc65fbf7915f1e2 : SUCCESS in 7m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/379e4471bf4448a1a480492c75dc0c08 : SUCCESS in 5m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6a8a1c1e410c40d086cb5876a6e87bf1 : SUCCESS in 9m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3900c797427744f6b47e44764f5a46ab : SUCCESS in 8m 49s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a80b9bc1930d4ad1a2abe83e56353ead : SUCCESS in 4m 28s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f69b343242194ce8911d0d902b12d70f : SUCCESS in 1h 14m 26s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/85b4b6a0146f4bdb8674f96ecf47ff75 : SUCCESS in 1h 12m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e060a6245fff45fd825146d0dc582783 : FAILURE in 1h 05m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/df1f0b151d0b43a2b4e118f04997fb01 : FAILURE in 1h 18m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b141efc0a2e84d25b13a5c2d9f5172f3 : SUCCESS in 1h 00m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fb6daf5015004cf88ba91b446cfc3d80 : SUCCESS in 1h 05m 13s","accounts_in_message":[],"_revision_number":1},{"id":"bbe2d0a4735be7a4562582b5978f4221555ac55a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-10 01:00:10.000000000","message":"Patch Set 1:\n\nrecheck yet another cinder fail","accounts_in_message":[],"_revision_number":1},{"id":"548fe61a84db107a400c5a022d97d427fcb027cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 03:01:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/15a4e2d789d847b2bdc641ec71ebb433 : SUCCESS in 59m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2c8a14411b704a7bb63465f012506954 : SUCCESS in 1h 18m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa89a9643db343a9b7a14c91475b242b : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/570bb74504c64690b5cd777f004cd074 : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3481fe91e38b47f699a3a5e7f25fc436 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffa7acb9d199409d940462d70e71b894 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/756cc70ce1084ccdbeb434e2d47b343e : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cae203a4b1be4c37b53b444fcf8c4aa8 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e97b9e0f1b6e4be280411358efd0f40f : SUCCESS in 7m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c7c0d6862c3d48df927a32116bfa58a0 : SUCCESS in 8m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d4f96398599047e69e2cdc6e9c5ce307 : SUCCESS in 9m 20s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/476be7980da649aebe22642ca685800a : SUCCESS in 6m 01s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/efeb49a85c414ae18c2f44412bc9ce03 : SUCCESS in 1h 11m 09s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c04769478f1a4b44a436a365f335cd2e : SUCCESS in 1h 19m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ca2d8d5961734078bbe75f574813cc8d : SUCCESS in 1h 12m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/709426ebdf674fc893bd78bf76ba4a68 : SUCCESS in 1h 17m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/85a23a0094ba474ab8d317d237e7ec12 : SUCCESS in 1h 05m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9a42885e125d42dba9e55991944465f0 : SUCCESS in 1h 21m 54s","accounts_in_message":[],"_revision_number":1},{"id":"535cc3a892697b79570101376db309a450a29f48","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-10 06:34:05.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"dec3a52745c358d13ee792c4a7a0e377e48cb590","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-10 14:42:45.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"d4b3641d9f03d0cb01593d007f13a9d029c1629b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 16:21:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/ba4e84e540a841c191456c14b05ec227 : SUCCESS in 52m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/62beb2a00b954d4b97a5e790c8f6422b : SUCCESS in 1h 01m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67d25e999097423f84a15033f5727a43 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04bce24a68894e71bc22f8176f8909fb : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42759d053cfa48c88fdc672d38011014 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e46826056d64836ba47a2f021242500 : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/883878f5ae844a92b9cbd7f4090cd062 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6155b79fb75b4c2c942f4c01342d579a : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/526679e9cde5469590967a0c2eb3b7b6 : SUCCESS in 5m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/77c3b0602e0c422e9914fd781b36295c : SUCCESS in 8m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/26eb5a7f16a4455baf127487c1678e50 : SUCCESS in 7m 32s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f3b07ea6ccfb480087b915c3d1771850 : SUCCESS in 5m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9ef72fd2ece440d6ab869e9c8232f9f5 : SUCCESS in 1h 16m 08s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1aca98c03c084ba2bf352227765917ed : SUCCESS in 1h 28m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c55b7bf084f647b0b2eb0ea22e8e1237 : FAILURE in 59m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e4bd301de5fc4319be97b06e527f75e9 : SUCCESS in 1h 25m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4bfe2f3bf7564f2bba844c01abdeb36c : SUCCESS in 48m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/55c07588917147dabd8c24843ceb3718 : SUCCESS in 1h 09m 52s","accounts_in_message":[],"_revision_number":2},{"id":"761a48c8264e2e9a2321898ad8d7ddc48f2e9554","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-11 05:29:48.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d31002bf50e6ccb304b03b835117d840689977f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 07:00:01.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/d173c624abb049a8bded4cfbdbc59105 : SUCCESS in 59m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fb614a271196444da51f8a45f400cdcf : SUCCESS in 1h 23m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/361e226d420143998d10f6af9e276739 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ba4c9fb84ac432a8457b09c7cca0613 : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16cb17606f4e4d10b5f98f273129d497 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e49d538253d46d0bf0d7d800bb0801b : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c677dc579e92420bba2b587ee8e0a792 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0d2c5f706264c50a96ad869fb5f090c : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4cf3da1c2d14196a300651c37dc72ec : SUCCESS in 6m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/31c77139fa054c40ab736d5b9a3f71c4 : SUCCESS in 7m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b3a414c8b9cb448aa1df671b8e000385 : SUCCESS in 8m 32s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ac96aac7b20d49c085bdfef489e4b3ac : SUCCESS in 6m 31s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/dcc549bc059544e9b100c9186d58556d : SUCCESS in 1h 09m 31s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/05525ed575064146a698fcfcb72a7668 : SUCCESS in 1h 11m 33s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5a57f75fc3154aea96dc595dee9c943b : FAILURE in 1h 19m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c50cabc8b48a427cb04f73d38bee7f39 : SUCCESS in 1h 27m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2d7ec4567fa45f5a6d885ec618227dd : SUCCESS in 52m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b62526d1788e44b3ac30ebc33557e38b : SUCCESS in 1h 04m 24s","accounts_in_message":[],"_revision_number":2},{"id":"9aece5a913930da330db16ca05093f1d62c54a6e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-11 07:32:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"1c1af02b25f541312fb3204952a1b4d7bb037c75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 08:57:56.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/5eee56ea8d3e45e1990614db1a68f882 : SUCCESS in 1h 22m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/17d023932fb143aebb2a3cef883142cb : SUCCESS in 1h 03m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/28a3a69ed2c448a0b25eec6e1ff0beab : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/715280e20943411fb750b1280d31cc44 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/547c6d3a624d47068db0bee0e9eb0343 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3831177c17aa4fc7b68ed31790640fb1 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/afb887b983014359adb300fecf5cbc02 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1b558c6fa9048dbb3eb6c2912fe86dc : SUCCESS in 7m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1fcbfa154fa54438a942ca424108fd94 : SUCCESS in 5m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/181a72927692499cb04d156619ef65d1 : SUCCESS in 7m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5b2e095a23094b27bb5d284f0d95c700 : SUCCESS in 7m 26s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3917fb6a5de84f61af67f7ea057becc1 : SUCCESS in 4m 29s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6fe41dac67ce4e4581d2408c057e6d5f : SUCCESS in 58m 43s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ebeec201f49f4f1e9bbf8e22724a4513 : SUCCESS in 1h 09m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b6bcbe6fb90849a0acf93f24027b6f75 : FAILURE in 58m 57s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/047a28866f02482db6da5affacf828ab : FAILURE in 1h 03m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11a5e318f99b4fe4bba4c21e5e252fdf : SUCCESS in 50m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/54b4bf6702ef49b5b21009d4d2d1098e : SUCCESS in 1h 01m 44s","accounts_in_message":[],"_revision_number":2},{"id":"df8442ad725301917c11bd8cab4c8eff4ca5bedf","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-11 13:56:16.000000000","message":"Patch Set 2:\n\nrecheck volume snap fail","accounts_in_message":[],"_revision_number":2},{"id":"bf578b0ff7b9f8726a28eeefb8314e9fcfd099de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 19:37:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/85afe1fa856646fca5a97eec2068e6a1 : SUCCESS in 1h 00m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a792b15c9ac04f1581a2f230c3d99d6f : SUCCESS in 1h 02m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca6bbce4c6c94d8784abc4ad090fb081 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/674b2ac9d5864675b6a1b67c70971007 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec8e752af9b74dbd8ee36dd9ea051932 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ae7aca866544447a7a4fd89f6664b76 : SUCCESS in 8m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf712f78cce84333b5db41294eb2d4dc : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f37df2ea7a0f4f53b6821548f598817b : SUCCESS in 7m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/924dd109561b4b15afc711d3e7952f2e : SUCCESS in 6m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4ab4263b089e4bf8ba46f3d13c1fc64c : SUCCESS in 10m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/337ef1aa053c486cbc4697eda5dd7815 : SUCCESS in 8m 31s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/bcd58f268d48475cbede23de7c972d64 : SUCCESS in 5m 06s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ea96b3b08c8b4f14a883ab30cafb26f7 : SUCCESS in 1h 17m 08s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/22ae51805d134334969a0f4f81c0e12f : SUCCESS in 1h 09m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f0dfa54dc3b9473285e92cdb40681283 : SUCCESS in 1h 28m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e4a264bfd0e3437eb8babcf627f4af5b : FAILURE in 1h 04m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe0ad2d83b4a40f08cc85f59614fadd4 : SUCCESS in 58m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4e960891f6de4e608fc308ea4462211f : SUCCESS in 1h 04m 59s","accounts_in_message":[],"_revision_number":2},{"id":"6ba2ce14e6a531724e5817c127f036aaec21b55d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-05 18:31:32.000000000","message":"Patch Set 2: Workflow+1\n\nNinja approving, sitting open for long, good improvement for import and plugin workflow.\nThank you!","accounts_in_message":[],"_revision_number":2},{"id":"da34a4c228d7c8c9680094a071b67b9ea21d9f39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 02:09:37.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9a98fa0b77e56b606b0559722fb81eb4dfbcf969","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 11:41:47.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"6ea52fd1e9207f7ff8382b9967048f79967826c8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-06 11:49:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2d684c98248d42add6b069f1429a24be3e37480c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 13:26:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/f85e93af31bd4ed7b126ade3c6a7a056 : SUCCESS in 1h 04m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/886aebcae66a4eee85b42415bf69e7c1 : SUCCESS in 1h 25m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53020e4f6a1146e194c75ee760fea89b : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c275ff5ff7d4d788fde6d7596e56d60 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87341b73ba34431cad1093fda439d75e : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48082dcf210b4c3f9626398e0ec852db : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5f94d3563744530b4a01cf89f0523a9 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79d9e93002f44f5894880b7ec760c9d4 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4831160fc4354273857a3c4dd9983105 : SUCCESS in 6m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/516b7bec51de4ea19216f685cbf76e5c : SUCCESS in 8m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0ab3417c8fd546f99f4f8195409024a4 : SUCCESS in 8m 08s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/68aad83dd5174aa1bfb2df62d183cd6b : SUCCESS in 4m 48s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ac1dc19dea0a4d89a277b57259ff4cfd : SUCCESS in 1h 29m 05s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d48edae563d24aa7a8bf573c9e83880b : SUCCESS in 1h 10m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ae77e1a065ea4de0b12d9a4b6bc9d9ff : SUCCESS in 1h 19m 01s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c6b203b83e924708b0f20272cfbada2e : SUCCESS in 1h 19m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/55c4f811b8c046ac973d5a0797614734 : SUCCESS in 1h 21m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/381b465163f249a49902b9064a32f1f1 : SUCCESS in 1h 18m 33s","accounts_in_message":[],"_revision_number":2},{"id":"8df6bf5f6a8a11a355549c1c6a7f473e28740946","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 01:56:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b5e8b2aac213dafa7c7197985c91bf94b24ecd12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 03:26:13.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"b240cf4799f019b7fbc7446544f79614f14291cb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-07 12:14:18.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"31604efab05d0cdcc74f94d63ed7ddec0fb68b8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 13:37:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/3766df3106864ebe9a0e97b1e2ebae07 : SUCCESS in 1h 00m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7374810a97da4af997d6156e0e07c7df : SUCCESS in 1h 14m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1490a85be9c4a9995a2e2fd940e79d3 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35535c7b233f4f07b4afc1e678e3d0ee : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/180447a2b5c641958efda4ffeac60b96 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b77c3ffc1794288a5f0a89f3341db9f : SUCCESS in 5m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68cc846ad4f440d6a9d8ff0a4d0e146a : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98a0b9d360eb40e19e8eefae1021e8dc : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4da56aba2ed2498f8dcd422006c48136 : SUCCESS in 5m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c009cba52f034a8385d34a7954e0f3d8 : SUCCESS in 9m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b29d5dd927ae466098c1710eebbef090 : SUCCESS in 8m 33s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/58344bea5b1842db9f5645ef46ed456f : SUCCESS in 5m 09s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b0893aa4c05449d49b749774ad886d25 : SUCCESS in 1h 09m 50s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e766a9c7a3bc4a75956aef0a578a6d60 : SUCCESS in 1h 10m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ee6685639d7c447face68d208deb8326 : SUCCESS in 1h 06m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3dad5972d8b54b788590864f558e1dda : SUCCESS in 1h 15m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/027143cd197a4c7b9f81fb23b1b0f002 : SUCCESS in 1h 02m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5be706443d964878b3652c1b65c2fe15 : SUCCESS in 1h 08m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/24420539a7084d14acd99e9a8a89cc8f : FAILURE in 31m 58s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f3fbefe7f80840e897a63d235142952d : SUCCESS in 32m 55s","accounts_in_message":[],"_revision_number":2},{"id":"5a7f255bec15f709ce0fc7595ef22a031b4c5298","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 13:37:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1eec115aa901e418f431e4b9181831b4dd744e4f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-07 13:43:30.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2a8f996aa8ac7dd65e52c5f2a100f38e122e305a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 15:24:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/3ee6bc452a4c45f38dbb1e722d51c735 : SUCCESS in 1h 07m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5a71d278af9742668344fb7b4a877349 : FAILURE in 1h 12m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/234d8eca47b84b99932eb15c305c011f : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f96d457eac24d8cba619ea8a99e2816 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/766852a71f794a91a69782f551cdaac4 : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29ba74b6285940d3bfab242a6d3df74d : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91430c1402e449dbabaad154a23ca07f : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd43eba6db064da09aa717fb2ae4831f : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2eb54fad5a3c42c7a4fc486db20cd2d6 : SUCCESS in 5m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/58f23b93922a4466b611cb0ae7f7bf12 : SUCCESS in 9m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/179081c5163945dab8df253abab8eff0 : SUCCESS in 12m 59s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b4b7b50b7c784a4eb0aa7a74068867da : SUCCESS in 5m 06s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/62332dc133a7462a93a77b224b4e0229 : SUCCESS in 1h 20m 59s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/672ffdf1f76f4ff89adced310c91b333 : SUCCESS in 1h 15m 12s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/87ed191c5eaa417eb2c2570f15c0a90c : SUCCESS in 1h 15m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bcc42638c5164d7d90e391e05c81af7f : SUCCESS in 1h 25m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bae4bf8f783845eb96e164edbfcbd299 : SUCCESS in 1h 02m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/375ae085c6494ca9ad6c73aea8afb848 : SUCCESS in 1h 39m 55s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cc8d3142ffa74117ab93ff60c2f0e112 : FAILURE in 28m 59s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d4c6bfa4d01041e393dfc44668b88a44 : SUCCESS in 29m 22s","accounts_in_message":[],"_revision_number":2},{"id":"1798c91de967d5095673d8a08ae7795b98e8814c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-07 15:56:13.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d1797f7a97990787d34430782bf719d70dc78e88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 16:05:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/305e617311824842806cd1eb2e759dda : SUCCESS in 47m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/26a9235c4ca04c0dbec3f9475b74167b : SUCCESS in 1h 10m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ca5f4c62c3c4d128379cb668454aa4c : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/462f3a5925f242c7adbbfdbbd89b09a4 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/06e2e97f9b16436bb2bb3fb7b37e92ff : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a13a259ee374bdaad637d4b3a2ca73e : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8060d9257e2243efae88111cf4c1373a : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ddd6cb93bac847f5b637bd5b9dc985f8 : SUCCESS in 5m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4b0cd176a24d44d299cc1a2407318f2a : SUCCESS in 10m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/78c7271afae94bb89370d4bf83d5758f : SUCCESS in 12m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/91f178fc3b4640fa9174e10ea5404ce5 : SUCCESS in 58m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/62d900263d804cf1b13b23e548f85799 : SUCCESS in 1h 08m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68ad42a6224f4bb39571942400be7b8c : SUCCESS in 46m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/78711e5a803c436fae976da5aaa6f60c : SUCCESS in 1h 03m 43s\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2641fe2874714860803536b44976aa94 : SUCCESS in 26m 59s","accounts_in_message":[],"_revision_number":2},{"id":"2d92c4a99511cfabc19cb1e8bc6afd14f9b0bf4f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 16:05:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"d5d27b75d9c85d80beab799f97373b49865e34cc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 16:06:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1446a6b3b39d4425b207f061b8fc0160 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa02446dae564bbab75038b50bd8af07 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2e06864b2c0126aabd57df84a0830789663f2119","revisions":{"5adc1682449a8eb5e040448314a4337f842b2cfb":{"kind":"REWORK","_number":1,"created":"2021-02-09 20:14:25.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/774726/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/774726/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/774726/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/774726/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/774726/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/774726/1"}}},"commit":{"parents":[{"commit":"fb328ba07ba60791684382c8f9c79d8d56fc59aa","subject":"Fix nonsensical test mocks and assertions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fb328ba07ba60791684382c8f9c79d8d56fc59aa"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-09 20:27:17.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-09 20:33:05.000000000","tz":-480},"subject":"Make inject_image_metadata use action wrapper","message":"Make inject_image_metadata use action wrapper\n\nThis makes the inject_image_metadata task use the ImportActionWrapper.\nBecause of the inbuilt rejection of internal os_glance_* property\nmanipulation, this also brings an operator-visible change to the\ninjection behavior. Specifically, os_glance_* keys can no longer be\ninjected via config (nor should they ever have been).\n\nChange-Id: Ib1138ee2cd0383f14c710716f3c103a146b0c4dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5adc1682449a8eb5e040448314a4337f842b2cfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5adc1682449a8eb5e040448314a4337f842b2cfb"}]},"branch":"refs/heads/master"},"2e06864b2c0126aabd57df84a0830789663f2119":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-10 14:42:45.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/774726/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/774726/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/774726/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/774726/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/774726/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/774726/2"}}},"commit":{"parents":[{"commit":"b1e757884aa1a10c939337ba83caf9bf8d615305","subject":"Fix nonsensical test mocks and assertions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b1e757884aa1a10c939337ba83caf9bf8d615305"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-09 20:27:17.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-10 15:36:24.000000000","tz":-480},"subject":"Make inject_image_metadata use action wrapper","message":"Make inject_image_metadata use action wrapper\n\nThis makes the inject_image_metadata task use the ImportActionWrapper.\nBecause of the inbuilt rejection of internal os_glance_* property\nmanipulation, this also brings an operator-visible change to the\ninjection behavior. Specifically, os_glance_* keys can no longer be\ninjected via config (nor should they ever have been).\n\nChange-Id: Ib1138ee2cd0383f14c710716f3c103a146b0c4dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2e06864b2c0126aabd57df84a0830789663f2119"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2e06864b2c0126aabd57df84a0830789663f2119"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
