)]}'
{"id":"openstack%2Fglance~774727","triplet_id":"openstack%2Fglance~master~Id5251b77fe1671594912e0dcfef7296007dab9b5","project":"openstack/glance","branch":"master","topic":"bug/1914826","hashtags":[],"change_id":"Id5251b77fe1671594912e0dcfef7296007dab9b5","subject":"Make copy_image plugin use action wrapper","status":"MERGED","created":"2021-02-09 20:14:25.000000000","updated":"2021-03-07 22:00:14.000000000","submitted":"2021-03-07 21:58:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"774727-bug/1914826","meta_rev_id":"e46038c807451ae146994a366afad40f40ea6f7c","_number":774727,"virtual_id_number":774727,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-07 21:58:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-03-07 22:00:14.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-03-05 18:32:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-03-07 22:00:14.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-10 00:50:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-10 06:36:08.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"ce4a715d8afe409fa37cfed21a8ea65b8459a4da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-09 20:14:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"920982165b4002a72c703b9a419c89384cfe7c6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 00:50:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/b18e87611c4b464d94f5c96a43de856b : SUCCESS in 1h 08m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/154b7ae068694881acdf772d85d0b0dd : SUCCESS in 1h 08m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9561f7e4c85e401e88846b47ca3f3377 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/297d026cb87c49799232f94258a08fb1 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f9df31b09b045c4b206ac74bbfc91bd : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/596daeb19a944d6f85edc7c44f7e804e : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b40423e380404668be4ed98a536c47c3 : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ec8b8bb02e44f46b3cfccf654d1b6ba : SUCCESS in 7m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ef79646b2111427c931b177e88ef8c41 : SUCCESS in 8m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cb3276aa023c408388803cce56f72b91 : SUCCESS in 9m 24s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c4e4d97750d444b49303abcc56950717 : SUCCESS in 7m 31s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/41b0a0ac33c544a3b9133eac2bd8242c : SUCCESS in 1h 12m 07s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fe3142c610434f05bd0c9c384ad2b047 : SUCCESS in 1h 12m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f35674b10d814ef9b6c8b64985e8db5f : SUCCESS in 1h 15m 18s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/25506d8068494cb5bab322337b63a093 : SUCCESS in 1h 42m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a74c6665b1454d4c909e7056f9fad868 : SUCCESS in 59m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/489efe09db164ee6b76e1cc9f25afac0 : SUCCESS in 1h 05m 25s","accounts_in_message":[],"_revision_number":1},{"id":"045aeb58c1fa4565a84a47fe43938de5ae1c9e61","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-10 06:36:08.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"19e219039cc0c4704ede8880658486f094ba1a2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-10 14:42:45.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"83644b95273171790c91e042f56e1fd71f99783b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 16:36:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/034da265570443739b4574a25dc50433 : SUCCESS in 1h 02m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/87a491e9171c4f5d93daf4f43d7b05c0 : SUCCESS in 1h 15m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b8f3a2e61514792854ee6bcab6ac367 : SUCCESS in 7m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d88e390f40044bca624913976b6d34d : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b4f8fbb094e4952881f267a69ce62ca : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99c91531c3fd4f77a958a701230bb24b : SUCCESS in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9827f2fd5bd74f09b9dc596b8cc4cfea : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eaf3cffa05a4647a1886e979837aa5f : SUCCESS in 6m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2bbf524791b545ffa9a9ccbd1568712c : SUCCESS in 7m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3d75f19fc16442699d3340fcfb22dbda : SUCCESS in 8m 47s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/be1dec2dd35b417fb49f603954c57576 : SUCCESS in 4m 49s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/dd31f5ae7b53456f9479c7ea43f8658a : SUCCESS in 1h 10m 37s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a9064c2d9c924a559c64675a14c0d945 : SUCCESS in 1h 44m 00s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4145b1a33d934f109788824b61f195ed : SUCCESS in 1h 15m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/81983365528f4a5da220b747077c89c3 : SUCCESS in 1h 35m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a31450194c043589a6677a56a7377f4 : SUCCESS in 56m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ee1005acf3d842f18d1273f4d7051bc8 : SUCCESS in 1h 37m 44s","accounts_in_message":[],"_revision_number":2},{"id":"a8d79f83f6fb60af3edb179f99c27c33943bbac5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-05 18:32:55.000000000","message":"Patch Set 2: Workflow+1\n\nNinja approving, sitting open for long, good improvement for import and plugin workflow.\nThank you!","accounts_in_message":[],"_revision_number":2},{"id":"dd35e6ec6129ec36182cbce6b3b4b6b6cfb2b606","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 02:09:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8c80c10c7f2cae50bf8f7ec4cc106452d955a5a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 11:41:54.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"4d2df2f87b868c68c84f47780d8ddc6846b0d806","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-06 11:50:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a8b3378505fa9d983eba84254057d3f77d5c5033","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 13:16:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/19d5e8bd7cc64640916a5bdd4789b3b5 : SUCCESS in 1h 11m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/04ba01ab1c894c148a119d1515d8d055 : SUCCESS in 1h 20m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/55ac643c6e7d4100aa522d77f0931625 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e838f82d19ab4727ad64a0d1ccb621c0 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9b61acee61614b1da1020f1ef3c52111 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ca4ffe8bccb4c9b8aad536942158303 : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6ecfd84c3c54f3db70e0a93dcc7a686 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d75b1cc3808842578c20078223e47335 : SUCCESS in 6m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/45d4186bca31499d8e7ea6f0bb6e76c3 : SUCCESS in 8m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/14e28be0e9a64347887ddcb1c92b73cf : SUCCESS in 8m 34s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/1ff2bf112ee14b4ca7aedf283f7c3fd9 : SUCCESS in 5m 34s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c60b59eed05a44be8546ba5a6572cff9 : SUCCESS in 1h 15m 37s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cdbf67920bb04ca199da349d904580d9 : SUCCESS in 1h 01m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/16fd1ec2c76c4261b8eba2b53767a098 : SUCCESS in 50m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f60fb9f606a34fde8c06b2a0eb6f13b1 : FAILURE in 1h 03m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af316b92b75c463abeba5edd4922e308 : SUCCESS in 46m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ee729da240e2440f809a816e118d76ee : SUCCESS in 57m 25s","accounts_in_message":[],"_revision_number":2},{"id":"dba5a8710c1f8df8fa390eac5d2234b9cd19d93f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 01:56:24.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"28efa738d47d735a069a6b0f9bb289ce103add2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 03:26:15.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"4001f64f7c7fdbe393579b11324bd163eb381667","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-07 16:06:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"da2cd95f8e125c46045d27a4329f97fbda8cd46d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 18:01:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/f91b85ae5f9b4f3ea7048e3df89f50ae : SUCCESS in 49m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8c1642c702fa4bbd97a33c32ad7c5d3a : SUCCESS in 1h 20m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9d0cbdc30b6499e8ac96c768abcef35 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da8a61fe71584c8d8109c29894271be0 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/161f652248be480fb0665ee6856d70d8 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23e47cb3824e46e895aeb85e152d357f : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0eabb33083d9483ebfca786ae4e7d074 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9beabab77b874c65bee99ab0884a2aa7 : SUCCESS in 7m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cfdb1e5807294d659d667052e2e3b3dd : SUCCESS in 7m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5c07375fe6a64adf929300ea70888869 : SUCCESS in 8m 14s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d065d9f44a72481f966be25999f303c4 : SUCCESS in 6m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5549c07a3d6241dca6d14443d7df616f : POST_FAILURE in 1h 03m 02s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f736118f3d82437ebdab2b24aa99dc87 : SUCCESS in 1h 07m 09s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/111259dd2b0c474ca0e325a68df7e929 : SUCCESS in 1h 14m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a40d560bfe584bd5babd1744513289e2 : FAILURE in 1h 19m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4987546f55df4e5199600bc044ec1a33 : SUCCESS in 44m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8c7d32c7cd19433a9e901b17887ee949 : POST_FAILURE in 1h 51m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/eb51570399224e68b44f18ea2c28cf8a : FAILURE in 39m 10s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/535dda665aa6424cabb3975f396c3792 : SUCCESS in 30m 11s","accounts_in_message":[],"_revision_number":2},{"id":"98974b9fbb33dd4d9fc66f16bf4119dfdf5ea7b0","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-07 18:34:14.000000000","message":"Patch Set 2:\n\nrecheck tons of fail and no log collection, so unrelated fail I think","accounts_in_message":[],"_revision_number":2},{"id":"b1ec90092a4713e1618b959d47df67562eb141a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 20:01:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/59ea5bb8166744a19bf4a83b65b3d4c7 : SUCCESS in 44m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/10e3f8adb00a4f519ca2683e4b50aed5 : SUCCESS in 1h 06m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c9d6116acdf5485086d178bf7328ef21 : SUCCESS in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22204a0f1a61485e9ba8fb36295567f5 : SUCCESS in 7m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6beec0477ac847c79e55a1554a7011c6 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb43e38761164b53841c8f9d69173ee1 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a098768b08a24d02b69d38fcaadfe453 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6ca22bb7ba646cba676eef1d5332b6c : SUCCESS in 8m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/437e166ca9a048509534def5f4188830 : SUCCESS in 7m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/42ca39fbd25d4d0a9c645b5075d72440 : SUCCESS in 13m 00s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9159f932a5194c27a98cae52673c60fd : SUCCESS in 4m 41s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/42c6aa857c5a4b5389bd3bf410b45112 : SUCCESS in 1h 02m 40s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/eec6cafc6e5442f6b98b0742f010a8a2 : SUCCESS in 1h 13m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c06d52402e204d77928758df79a9974e : SUCCESS in 1h 12m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d6014491a7e244b4a1997300199c05a3 : SUCCESS in 1h 19m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ab284c7ea5448d68d04e47538fcd7b7 : SUCCESS in 41m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4a5e7e85946f445b82834089a852e0a0 : SUCCESS in 1h 22m 35s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cf18469e3ffd4fd8934e7a4d68f49d5a : FAILURE in 26m 36s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8f9d3e8b71cd4fb5bc3d87ae4b621676 : SUCCESS in 35m 56s","accounts_in_message":[],"_revision_number":2},{"id":"d71facbf61c9dfd6a105af0257d22b977ec9f020","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 20:01:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"89cf9c311e4180efb9cf799c2fb71a4fc21c0280","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 21:58:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/1102db5157ac4579841c8b1f1503a5a3 : SUCCESS in 58m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/911581ba8d5544daa7702b40f4a0a13e : SUCCESS in 1h 16m 14s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/711bd09047d043438ffa04e28c4ef2a7 : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c29d307b41842e4a38fa3e22d7b0fe2 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96c3c042e4784262ae81e9b8d76c1f84 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20f77ca2e9e54724b0d179792ddf372d : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12f79d4c09c54aa492259f2a4b020d00 : SUCCESS in 6m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/619dd6b44c72467e866d94d9a645cd28 : SUCCESS in 8m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2e0c9706ef2d49a7b2383a98c1646e1e : SUCCESS in 9m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7d77ccf6cf204ff2a0f7f6171b8cf6c7 : SUCCESS in 1h 13m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/135784f918264e2ab0fc5e253791ff37 : SUCCESS in 1h 07m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e3534556f84460a9a2fc0961b7a4a71 : SUCCESS in 57m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d5595aab1c984a8795e39391ae18b069 : SUCCESS in 1h 19m 41s\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/60054d98c1aa4331acab3070521f8368 : SUCCESS in 33m 23s","accounts_in_message":[],"_revision_number":2},{"id":"a3a7a5c4d39c0b388670fcf5bbf87ac59e6ccff3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 21:58:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"9db61e19cca1f7d42909077bc1ea5ae21bad80dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 22:00:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8564ba84b5248aca955377c25bd1252 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a3af0719975bd38cd6abdf17aa3a014adf3b46ac","revisions":{"9a19d9c4ce67311ef224c0bf332a2b3a760cc0b1":{"kind":"REWORK","_number":1,"created":"2021-02-09 20:14:25.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/774727/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/774727/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/774727/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/774727/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/774727/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/774727/1"}}},"commit":{"parents":[{"commit":"5adc1682449a8eb5e040448314a4337f842b2cfb","subject":"Make inject_image_metadata use action wrapper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5adc1682449a8eb5e040448314a4337f842b2cfb"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-09 21:04:45.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-09 21:08:48.000000000","tz":-480},"subject":"Make copy_image plugin use action wrapper","message":"Make copy_image plugin use action wrapper\n\nThis makes the copy_image plugin use the action wrapper for its\nmanipulation of the image, in line with all the others. This adds\nimage_size and image_locations properties to the wrapper, which\ncopy-image needs.\n\nChange-Id: Id5251b77fe1671594912e0dcfef7296007dab9b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9a19d9c4ce67311ef224c0bf332a2b3a760cc0b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9a19d9c4ce67311ef224c0bf332a2b3a760cc0b1"}]},"branch":"refs/heads/master"},"a3af0719975bd38cd6abdf17aa3a014adf3b46ac":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-10 14:42:45.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/774727/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/774727/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/774727/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/774727/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/774727/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/774727/2"}}},"commit":{"parents":[{"commit":"2e06864b2c0126aabd57df84a0830789663f2119","subject":"Make inject_image_metadata use action wrapper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2e06864b2c0126aabd57df84a0830789663f2119"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-09 21:04:45.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-10 15:36:24.000000000","tz":-480},"subject":"Make copy_image plugin use action wrapper","message":"Make copy_image plugin use action wrapper\n\nThis makes the copy_image plugin use the action wrapper for its\nmanipulation of the image, in line with all the others. This adds\nimage_size and image_locations properties to the wrapper, which\ncopy-image needs.\n\nChange-Id: Id5251b77fe1671594912e0dcfef7296007dab9b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a3af0719975bd38cd6abdf17aa3a014adf3b46ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a3af0719975bd38cd6abdf17aa3a014adf3b46ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
