)]}'
{"id":"openstack%2Fglance~774830","triplet_id":"openstack%2Fglance~master~Ib3cacb4dd4d75de32e539f8a3b48bdaa762e6d8e","project":"openstack/glance","branch":"master","topic":"bump-image-api-version","hashtags":[],"change_id":"Ib3cacb4dd4d75de32e539f8a3b48bdaa762e6d8e","subject":"New API /v2/images/{id}/tasks","status":"MERGED","created":"2021-02-10 11:30:47.000000000","updated":"2021-02-26 02:49:31.000000000","submitted":"2021-02-26 00:27:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":28,"unresolved_comment_count":10,"has_review_started":true,"submission_id":"774830-bump-image-api-version","meta_rev_id":"7a55e6785f23aa7a22332e63df5bbd09d5d58b17","_number":774830,"virtual_id_number":774830,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-26 00:27:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-02-24 13:34:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-02-26 02:49:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2021-02-25 16:12:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-02-26 02:49:31.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-10 12:10:14.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-10 13:06:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-12 15:55:19.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-02-16 12:44:21.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2021-02-24 13:34:09.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-02-25 21:34:09.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"CC"}],"messages":[{"id":"102d0f6c88c84835bbe2e0442e38900b53d6b697","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-10 11:30:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ae6f172eff350402f99fcb8e8adc88ab8957095","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-10 12:10:14.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/774830/774830/1/check-not-vote/ext-glance-zuul/ee95bb1 : SUCCESS in 37m 49s","accounts_in_message":[],"_revision_number":1},{"id":"4ad8c85532366bec290ef1bb5410477ce9cba00a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 13:06:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/f09e48c87f4d45f9992687fc432bdae8 : FAILURE in 14m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d60bbdf079dc47da8c456f630cdc3cd7 : SUCCESS in 1h 29m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aed1e54856ac4c0991ac21418b1a6bbc : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0937ce99b5648d890df71aef27391cc : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df93782c50824a46a1bb2d7644ecfae2 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac801eea86f84867bb4cdbe1be827c02 : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/34ae354ad4304e0e8e0482497d5a97e1 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3822220b74b84d239e54902878a0f937 : SUCCESS in 6m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e1c48977603d46daa6770e19d5df89be : SUCCESS in 8m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6ee90ffebb87434cb8d768c50e78c710 : SUCCESS in 8m 09s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a8005589595741628c3a61b5f853c6a2 : SUCCESS in 6m 53s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3f8fe8e9c3824343a3e4f1bfd4eb1f66 : SUCCESS in 1h 16m 43s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/49f5dc180b524ddfbdbf24fcb0a3a3a5 : SUCCESS in 1h 15m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/903b445b1a264588a7d054ef5608a5b4 : SUCCESS in 1h 24m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1c15e52af02c4e3b88992b08c499f37b : SUCCESS in 1h 22m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4028022573d840e4a18f0b8a2c2651ec : SUCCESS in 50m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7911e7bc08604de281418387359f7032 : SUCCESS in 1h 14m 35s","accounts_in_message":[],"_revision_number":1},{"id":"265c3de329a753ba75dc2743e245f2dea7affd3d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-10 14:58:26.000000000","message":"Patch Set 1:\n\nExample curl;\n\ncurl -g -i -X GET http://\u003cip\u003e/image/v2/images/d972948c-eead-4fd8-8de0-28a248795534/tasks -H \"Content-Type: application/octet-stream\" -H \"User-Agent: python-glanceclient\" -H \"X-Auth-Token: \u003cauth-token\u003e\"","accounts_in_message":[],"_revision_number":1},{"id":"a1404a27eff82dfa7ac1a4c078c92584bf02d561","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-12 13:43:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bc234c59cf652c3ae43dd203506d911483ffc13d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 15:10:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/4d8b3f808cff4bada6e37fac3fae43fa : SUCCESS in 1h 05m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/675bdf0b0dbc4c5b8791e4053d765849 : SUCCESS in 1h 23m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf4ae082056a426894676a5c456dada7 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bb9fb64b5f84e7591b634543fa05890 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34217782cff0453e9f20b9647ded35ac : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56f516d481b444a1aa81dc93e7ef1f5b : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdfa223407e7458389eb7ca4d7b89175 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8bd497036054b61aabdc98610b479d6 : SUCCESS in 6m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/700788b7ca744acfae10c1de68f7e17b : SUCCESS in 7m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9947bc98b97c4cffbd05e5e96cd443c2 : SUCCESS in 7m 28s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a2000091793941fb9db2d71dcbb42454 : SUCCESS in 5m 54s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/538a94060a44411dafe28a441a119425 : SUCCESS in 1h 09m 04s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5d043240d32c4325a10fbe6ff12a2c40 : SUCCESS in 1h 04m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c4323ff23b0949ffafb06ede02f980fa : SUCCESS in 1h 06m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/415a7cf27b5d4d8c912c72b8f0f926ae : SUCCESS in 1h 10m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bed205b16cf4e17942d54c848c5c54b : SUCCESS in 1h 00m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f861b8ebeba44ee2b5f28447851d48f0 : SUCCESS in 59m 16s","accounts_in_message":[],"_revision_number":2},{"id":"0b1ada2d96a7c7ef6ad7e10a94daf3cd837caa42","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-12 15:55:19.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"26b7d1d394aa92c946b411e27ae2d0d4e37e662c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-15 10:49:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"07a4f403e14995db1a900d098b88ef1171cd26ce","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-15 10:51:20.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nThank you for review Dan!","accounts_in_message":[],"_revision_number":3},{"id":"7dda8e9cade9750811a93eba3d3cb1920e434ea9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 12:59:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/44a7ead12a0549d88f34aebed350bb9b : SUCCESS in 1h 25m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/89f6c89e83ee48c89fe6b5fa6949d7f8 : FAILURE in 1h 09m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8c4f270ef863431f92cf05fb23d2fe38 : SUCCESS in 8m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a203ecc2b3a34c8490529b5930ae9a04 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf379cffeb504d93acd30063193872cb : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/057cf823e27043158e168b2c1889a0bb : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00b399e8eeb44be5af59fddb175abd61 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/380d783a040549ee894c74aee838a45a : SUCCESS in 6m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ef4821a621b5427fb040128ec7672355 : SUCCESS in 8m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fb418aae4a174c6e833afee531d03972 : SUCCESS in 8m 09s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7569f5ab0a3e471b8a5d1be21161da1f : SUCCESS in 4m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f593413524f3419d849da02fb4180e09 : SUCCESS in 1h 12m 30s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6591a9ffe53d489995e1e1f1622387c9 : SUCCESS in 1h 05m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3f1c31aae359477e8ed78263402598c7 : SUCCESS in 1h 16m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/670a2a8c69f24a19934178017af7f258 : SUCCESS in 1h 27m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84d653c308dd4feba9823d0f3e1ab6ad : SUCCESS in 1h 11m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a517062fae364dbba47135f16d69ca29 : SUCCESS in 1h 25m 32s","accounts_in_message":[],"_revision_number":3},{"id":"d999cd3e5977a2c6037e198af49fca054019d822","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-15 14:00:26.000000000","message":"Patch Set 3: Review-Priority+1\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"70d802a58f6862138cb23d0f473a61d78335c8ad","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-15 15:15:15.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)\n\nThis is definitely getting close. Do you have a tempest test proposed somewhere? We likely will only be able to grab the task after import, but we should at least be able to validate it after the fact there.","accounts_in_message":[],"_revision_number":3},{"id":"9f3d9b3af0b8a2f519463dad9f7ea1c6bcc23f10","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-15 15:33:33.000000000","message":"Patch Set 3:\n\n(2 comments)\n\n\u003e Patch Set 3: Code-Review-1\n\u003e \n\u003e (4 comments)\n\u003e \n\u003e This is definitely getting close. Do you have a tempest test proposed somewhere? We likely will only be able to grab the task after import, but we should at least be able to validate it after the fact there.\n\nNo, didn\u0027t worked on tempest for this change, will try to add/modify existing one","accounts_in_message":[],"_revision_number":3},{"id":"2f3ec1fd1202f9624089cb71bdac574577f820cb","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-15 15:44:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"78600d92b5e877cf46f7ef1398984817636ffcbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 18:45:06.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/66703f8d91dc42ce9f0ac59930d50815 : SUCCESS in 55m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/87a4d22454da4ff3955b58c7c280fa1f : FAILURE in 1h 13m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1265d57929e64f92863dd92a30cf6d07 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3dac0e2b16c48399a6fd927428815cf : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/673d116f2aee4e22ac8007ccc214422f : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b70a777449d47f0a80a6390f9b5d59f : SUCCESS in 8m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf20b035a02043458da51940e1f62af5 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27dac8b4c1344ee6b8cb062d85e7d9c2 : SUCCESS in 7m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fde2cedd963f4203a3a2afb132961cca : SUCCESS in 8m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ce88ba59b56340a5973ef6090f98254c : SUCCESS in 7m 18s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c8f29094f9af41d5a97249a2c82322d3 : SUCCESS in 4m 18s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e911b5224a264e7ca272e8cd4fbe7352 : SUCCESS in 59m 21s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e10629bc420944dfbb38de436cb1e502 : SUCCESS in 1h 24m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1277ebe9164a41f4a497d94fa9ea2070 : SUCCESS in 1h 33m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b76d1bbbb526429dad14ecde66278e85 : SUCCESS in 1h 40m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25bfaab7ed3c4cd098fb214ca34d8d32 : SUCCESS in 1h 26m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6438051bafc640529ed5e0dd9f7c9267 : SUCCESS in 1h 20m 42s","accounts_in_message":[],"_revision_number":3},{"id":"91f85bdc2f88f2963f268ffecfdf634c1f67cc82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-16 07:04:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"104659efc5e519b240e25efec064fada7d208173","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-16 07:04:20.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2518cfc7deaea121763a37cc5eaceee7cf2a5287","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 08:34:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/691090af25d04df0a937573a4549903d : SUCCESS in 53m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2420f150c2f348c1bc9b1731928abbb4 : SUCCESS in 1h 06m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/351c747398a348acbdf82aef67332e63 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49cf373efdf04598a7c96ae5ea6e7c11 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e83ced8bd6b411eb7b1c8941abb1531 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/359233e022114c208edde84798f7931d : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8f5f17e43ac444eb43557749dec32e1 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e07b6fcdfb14e94af34e05056399e74 : SUCCESS in 6m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f424fcec46044ba5aead3c2d121a0067 : SUCCESS in 9m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9f369941e6c2447097a7d6e2d7719ade : SUCCESS in 7m 49s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6a184e196ec04ff78422edc6c78539e1 : SUCCESS in 4m 39s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/83c26e88422b4ed0a0d932e8a5a7515c : SUCCESS in 1h 01m 33s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3bb09ed7b2a84c62b8f0f121eb3f6945 : SUCCESS in 1h 25m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5d00c46e896e4969b36d7e4d88088872 : SUCCESS in 1h 13m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/dc15a841c1ac49c5a2d87f960190fe1e : SUCCESS in 1h 16m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b68c1b4b3623443bb0de10d6fcb64ab0 : SUCCESS in 56m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cec7a2c7435d45798bd7535e5f7354e3 : SUCCESS in 1h 13m 34s","accounts_in_message":[],"_revision_number":4},{"id":"883995dcfb337ceb0a7646dda2d33eb58b39c18d","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-02-16 12:44:21.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nComment inline about the filtering strategy.","accounts_in_message":[],"_revision_number":4},{"id":"1d2aee6c32af55b21a5f55a936a6da6555a5fa8b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-16 14:35:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9c096c753ae4cd60b433727cee401ae59696e027","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-16 15:34:27.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0f25444c8c35125055f9c6530b54ebb8fe3bbfc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 23:02:51.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/34bd08a36fd44a6485bc22f85f39bd0e : SUCCESS in 52m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f8941c11aa8a45ae936ec28ba050f064 : SUCCESS in 1h 14m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67fa96739c774cf58aae265950590681 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fda24afd76084fdd954a583fe1780b77 : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e07aeff5fd4f4d4394ef4be1b1f6e3de : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f62b87cdd45f405c9c2a47aa1ab75785 : SUCCESS in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eef681294fe44ec0aa9653a1c6a9e5d7 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/102ba699a69d4a49a55fc59af7f511db : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5e3f9679707c445abfcfeb61866af4b2 : SUCCESS in 9m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c7c32b8aab3a4da0b7cd59345cd2bd87 : SUCCESS in 7m 42s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9b87577d2a8942ab91d27f0e6d39cdae : SUCCESS in 5m 48s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e618729d6577462691945c16b94d847a : FAILURE in 1h 40m 03s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c62bd7e385b940f4b5cd6affb5dae712 : SUCCESS in 1h 08m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ce17f6d1a9af4887b73c830087087205 : SUCCESS in 1h 12m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f2a5e9d5a1cc43c08c53165f00ec2908 : SUCCESS in 1h 16m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83fb5e8a19964555832679231bbe46db : SUCCESS in 1h 11m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ca1d75de56a04b8d90e40b5428955242 : SUCCESS in 1h 06m 01s","accounts_in_message":[],"_revision_number":5},{"id":"9bd1bd88b8046afc8e35d8227fbdc915719edfde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-19 19:30:07.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"41eb15efa3c1d3d43e26bfc7dac82ddc6327097a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-19 19:31:36.000000000","message":"Patch Set 6: Code-Review+2\n\n(2 comments)\n\nI think this is good to go now. I just finished up a test assertion for Abhi.","accounts_in_message":[],"_revision_number":6},{"id":"3a510f38df63df1c8fe89f3d16f188fd55e2b08d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-19 19:33:45.000000000","message":"Patch Set 6: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"52f8b2faf93cfef55431dea725a17733b5bf015d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-19 20:00:17.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"881c872733b8d0d890a45718a2908b1d9f984cbb","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-19 20:00:54.000000000","message":"Patch Set 7:\n\n(2 comments)\n\nI added tests for the sqlalchemy function, but it left me with some questions, outlined here.","accounts_in_message":[],"_revision_number":7},{"id":"0f32c9fb432b18532f1b4ed704dbbd0419d7ce75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 22:07:58.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/0d849600fb4941f99d19d90daba595e4 : SUCCESS in 1h 00m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1b8f0893a8a94e8fa581610a4d5bf9a3 : SUCCESS in 1h 06m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb9622e2196e4031ab46c7a9f1a4a923 : SUCCESS in 9m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dc49d9f3d0e4c8d8405050598017924 : SUCCESS in 6m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8cbc6edf59a54819b0a67c3afb45dcc3 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fdc4678997cc4636906b3b96c43b8c49 : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c677ae7aad68497a98624c322303dd52 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf344ae4a58742449adc86e207a37ffa : SUCCESS in 6m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/34fb66de8b7a40078b2139a3e94f9c78 : SUCCESS in 8m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e53b6c230f054174ae93a06bb41a6924 : SUCCESS in 7m 45s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/bc1569f84d94427e8ed68fa98263593b : SUCCESS in 5m 04s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f042b893078b4248bdd46c9dd30b5f55 : SUCCESS in 1h 17m 01s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1efbbddf3e714c2b8c7418334da3fa28 : SUCCESS in 1h 08m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/cec48d78febe4d35a9696734552407a1 : SUCCESS in 1h 08m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0bd7be033a8c48cd94e07a3cb9fa665b : SUCCESS in 1h 21m 40s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2881a3e0a1114a749222844eb0ed45c7 : SUCCESS in 54m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a5bd2007cb784dfca093999cb061da0a : SUCCESS in 1h 10m 44s","accounts_in_message":[],"_revision_number":7},{"id":"e1e8c13dc9032896f15ecaff9b53a245c9f1254e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-19 23:22:15.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/774830/774830/7/check-not-vote/ext-glance-zuul/9d2b371 : SUCCESS in 1h 00m 53s","accounts_in_message":[],"_revision_number":7},{"id":"2678f04352e59330b0c9354d7eeab6fd15e0c19d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-22 05:02:03.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"f9d9d2b22905cb8dd972d9e1d77ef9a38ae84adb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-22 05:21:27.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b8065c99028ec9aaf57b14b8ddb41a798b877872","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-22 06:46:07.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"6beee025aef6f68104dc9dbd94068b66de584872","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-22 07:28:11.000000000","message":"Patch Set 8:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/774830/774830/8/check-not-vote/ext-glance-zuul/9626df8 : SUCCESS in 41m 08s","accounts_in_message":[],"_revision_number":8},{"id":"c9c22f042b38b381338e0f383d9e997b50f040a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 08:11:47.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/d96c421207834756a586cbb2cbd56c97 : SUCCESS in 59m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b7d19989d9e74762830897b79ccffb00 : SUCCESS in 1h 13m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ffebaa78804e4ce887c1b52aa4257c8e : SUCCESS in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/faf7fe87a58342c5a99866e4fc764bdb : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9cb891fd02ee4e60837df526f7a7c97b : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/69b593f294064573aa98f308186ddb87 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4423165a53fa439caad949c45d3285e0 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83df7a6bde41496db155684260a38d88 : SUCCESS in 5m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1735016a12dc4134b5bde109ce58877e : SUCCESS in 8m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a859782eecb744928e0e34fc8b18ee83 : SUCCESS in 9m 29s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/1215d523bd624fb08e677de32ab5744c : SUCCESS in 4m 45s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/292e82328c124f54ae33eea4a211692a : SUCCESS in 1h 06m 19s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fbe5aa35e41d40dbb3b2eb944ae61a53 : SUCCESS in 57m 25s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8e663aced2cb4e5390f200df19351405 : SUCCESS in 1h 11m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d53c5a753a4341e9a2c16db8a4ad6fc4 : SUCCESS in 1h 24m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f79e108f1d64e369ea68e1c8904de8c : SUCCESS in 54m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9a37976277b044ba9256a5b88060532c : SUCCESS in 1h 21m 09s","accounts_in_message":[],"_revision_number":8},{"id":"b9c6e089667a68708db41a2c5632c8c4bebfe177","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-23 15:02:19.000000000","message":"Patch Set 8: Code-Review+2\n\nI think this is ready to go, thanks Abhi!","accounts_in_message":[],"_revision_number":8},{"id":"1939bc0a5fb3ce8ecff4ea490bc5578551897816","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-24 06:00:03.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"5157795493dc428e4c173d57f91b4cec1713e832","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 08:30:23.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/f6ebf34a2b8d4bb9a864db2fe5c4ee95 : SUCCESS in 1h 12m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f4a5867fd96340f98d7f5c79d274bdf3 : SUCCESS in 1h 21m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/effa23bae5704e28a993f2ef56e946f4 : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c7bb4e7bf764bc598178c475624dfac : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd165c824487463fb40a72deb748b7a0 : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cdea5f3dade4fdd827afd2b83205eb2 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ceb4a700ee584e9d82ab91db4d263d8f : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e53d99c8a1c143268e7cc4dab2cb74c7 : SUCCESS in 9m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e95660fb27b343a5bf42b6640100cd73 : FAILURE in 12m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5f36fb3a22ed4d3eadde0399a9584f16 : SUCCESS in 10m 38s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6d26dd00f3a542f99995e4a4a1ef7c5e : SUCCESS in 5m 04s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f16b69962e864674803958deee7cf60d : SUCCESS in 1h 20m 52s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/97fd14435658450f868206d0d76a12c2 : SUCCESS in 1h 30m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/097cd40576664a87b2ab0a0b6569e1fb : SUCCESS in 1h 18m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/45c8f056cb1b4195b11525e385afa77e : SUCCESS in 1h 17m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/08130d5058f444ffa90d254157135e32 : SUCCESS in 1h 01m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0fb881118e2b4a7c8c7c06f82c5d4f5a : SUCCESS in 1h 09m 33s","accounts_in_message":[],"_revision_number":9},{"id":"a8a4ccd48988bde874f1f11c393fb0ed1660cd54","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-24 08:32:34.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"cf3e1a5432eb7df2587eb2e990f42e4fd1f0a5d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 10:03:28.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/b5b1302a3217499b99254c4da31fb5f9 : SUCCESS in 1h 16m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d7e9728b30444c8ca8bbd67b0532ff39 : SUCCESS in 1h 17m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3633e25e8fb4cc6bad0d8f82e842e49 : SUCCESS in 6m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fe3eca0ede5451f9a21d084327ea606 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3d03d4f693a743f689f47e6081b8ef90 : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d85cea88292a458eb35300eded9a0fff : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccda1c2a2e604d3480663fa43c712c64 : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9df0ee2e348345cf8ab9e1ae716aaf34 : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/99f9829362bc42c2b8f30f9c2d80a15a : SUCCESS in 8m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2636fae71b9d4dadb3a4ccc38193ca15 : SUCCESS in 8m 24s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/db51f54ad5b9415ab3300cfab34f8a2c : SUCCESS in 5m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8ac0c926830a465e9ea639db799d7a8a : SUCCESS in 1h 12m 21s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7643ea13e52949f7ab0d1561b9d1720b : SUCCESS in 1h 24m 00s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/eaaf450ad39345529f7d6d8e8c06d894 : SUCCESS in 1h 17m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b86161f39ec94f66bb4ccdc2dae81f7a : SUCCESS in 1h 25m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e81abd82b6f74deb96ec0ffbb8db664f : SUCCESS in 1h 11m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e8100db44bf540e29261525d890cc2d7 : SUCCESS in 1h 24m 24s","accounts_in_message":[],"_revision_number":9},{"id":"34b98f9841c0701ff6b0e159ef0d8db7a672b270","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-02-24 13:34:09.000000000","message":"Patch Set 9: Code-Review+2\n\nStill looks good.","accounts_in_message":[],"_revision_number":9},{"id":"804cca8cc9417dc3632b9ed0e706960f52d0a3e4","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-25 16:12:34.000000000","message":"Patch Set 9: Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"149ce3a476b4579987dce715e24b8342b6cf26e0","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-25 21:34:09.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"cf3757017561eb165ef1d9e0e1b27ae578a834e9","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-25 21:43:26.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"154036972c16c136771b32792c40f521db729228","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 22:06:40.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"14615b2b8e46244e4f265196680502f70f536ff8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 00:27:23.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/ad40eef73adf428bb20420583f15fd5b : SUCCESS in 58m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f424160eb76443439dbeea2bd93db234 : SUCCESS in 1h 19m 07s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0aa231c297f54cfeaa686152af33a4f7 : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a7f818186dc4a4985412a18875bc97c : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5dc0b23d92614895a4c6ea64d300b38c : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a9e59c642894d6da03d890f01bc12db : SUCCESS in 9m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf82c800810a450aa2b264bd1fece262 : SUCCESS in 6m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/69ab5eb695dd4319ae1f2932936053ae : SUCCESS in 8m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5cc5f7f1fedb4e07bc8c62f8acd1a249 : SUCCESS in 8m 07s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7dec3cbea6c844f88647b115774867a3 : SUCCESS in 1h 13m 38s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c4c68b7b5ebb41fe925b71fb05bd3197 : SUCCESS in 52m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5cbd00fb08145b1ada90fbb06e48022 : SUCCESS in 1h 37m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/abd360ee121f4d6a8bce693551fac51c : SUCCESS in 57m 10s","accounts_in_message":[],"_revision_number":9},{"id":"ba317df47fa8ffc5d1cd7b9d254895f38a650152","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 00:27:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"72b15a97e42e4be3dcf26add63e69d51e5dcc064","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-26 02:49:31.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"281fadc15c5e257e72514957701a5571a83d8c38","revisions":{"9e78e3c5c789b24b63a200fc8abe652c33db1fee":{"kind":"REWORK","_number":1,"created":"2021-02-10 11:30:47.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/30/774830/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/30/774830/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/774830/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/774830/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/774830/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/30/774830/1"}}},"commit":{"parents":[{"commit":"a463be66b3839feca384062348204afb210a36f2","subject":"WIP: Utilize newly added tasks database fields","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a463be66b3839feca384062348204afb210a36f2"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-10 11:20:47.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-10 11:20:47.000000000","tz":0},"subject":"WIP: New API /v2/images/{id}/tasks to show tasks associated with image","message":"WIP: New API /v2/images/{id}/tasks to show tasks associated with image\n\nToDo:\nUnit and functional tests\ndocumentation\n\nChange-Id: Ib3cacb4dd4d75de32e539f8a3b48bdaa762e6d8e\nImplements: blueprint messages-api\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9e78e3c5c789b24b63a200fc8abe652c33db1fee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9e78e3c5c789b24b63a200fc8abe652c33db1fee"}]},"branch":"refs/heads/master"},"463fa2e0af4cb7ebab3beae4fd71aa1550be8064":{"kind":"REWORK","_number":2,"created":"2021-02-12 13:43:22.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/30/774830/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/30/774830/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/774830/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/774830/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/774830/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/30/774830/2"}}},"commit":{"parents":[{"commit":"8bba37dc19f99c22b0028fba01b6a9141092149b","subject":"Utilize newly added tasks database fields","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8bba37dc19f99c22b0028fba01b6a9141092149b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-12 07:01:59.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-12 13:34:29.000000000","tz":0},"subject":"WIP: New API /v2/images/{id}/tasks to show tasks associated with image","message":"WIP: New API /v2/images/{id}/tasks to show tasks associated with image\n\nToDo:\nUnit and functional tests\ndocumentation\n\nChange-Id: Ib3cacb4dd4d75de32e539f8a3b48bdaa762e6d8e\nImplements: blueprint messages-api\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/463fa2e0af4cb7ebab3beae4fd71aa1550be8064"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/463fa2e0af4cb7ebab3beae4fd71aa1550be8064"}]},"branch":"refs/heads/master"},"9d959c7b4a41ddabf0e9c93748dba96531b7504a":{"kind":"REWORK","_number":3,"created":"2021-02-15 10:49:52.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/30/774830/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/30/774830/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/774830/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/774830/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/774830/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/30/774830/3"}}},"commit":{"parents":[{"commit":"758bb707c63224b8ff9a1d5eaf78e32f604ff122","subject":"Utilize newly added tasks database fields","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/758bb707c63224b8ff9a1d5eaf78e32f604ff122"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-15 09:04:17.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-15 09:04:17.000000000","tz":0},"subject":"New API /v2/images/{id}/tasks","message":"New API /v2/images/{id}/tasks\n\nAdded new API /v2/images/{id}/tasks to show tasks associated with\nimage. This API will return list of tasks associated for valid image\nelse returns 404 not found if image is not present. This API also\ninitiates task scrubbing before returning tasks to user.\n\nImplements: blueprint messages-api\nChange-Id: Ib3cacb4dd4d75de32e539f8a3b48bdaa762e6d8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9d959c7b4a41ddabf0e9c93748dba96531b7504a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9d959c7b4a41ddabf0e9c93748dba96531b7504a"}]},"branch":"refs/heads/master"},"9b1ba2b2d6bd6a1cf7ecd88224587ccb33ac3e7a":{"kind":"REWORK","_number":4,"created":"2021-02-16 07:04:06.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/30/774830/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/30/774830/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/774830/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/774830/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/774830/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/30/774830/4"}}},"commit":{"parents":[{"commit":"758bb707c63224b8ff9a1d5eaf78e32f604ff122","subject":"Utilize newly added tasks database fields","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/758bb707c63224b8ff9a1d5eaf78e32f604ff122"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-15 09:04:17.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-16 06:24:07.000000000","tz":0},"subject":"New API /v2/images/{id}/tasks","message":"New API /v2/images/{id}/tasks\n\nAdded new API /v2/images/{id}/tasks to show tasks associated with\nimage. This API will return list of tasks associated for valid image\nelse returns 404 not found if image is not present. This API also\ninitiates task scrubbing before returning tasks to user.\n\nImplements: blueprint messages-api\nChange-Id: Ib3cacb4dd4d75de32e539f8a3b48bdaa762e6d8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9b1ba2b2d6bd6a1cf7ecd88224587ccb33ac3e7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9b1ba2b2d6bd6a1cf7ecd88224587ccb33ac3e7a"}]},"branch":"refs/heads/master"},"6bbafa0a09f0cee9a58349eae23c35ec0bd4257b":{"kind":"REWORK","_number":5,"created":"2021-02-16 15:34:27.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/30/774830/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/30/774830/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/774830/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/774830/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/774830/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/30/774830/5"}}},"commit":{"parents":[{"commit":"758bb707c63224b8ff9a1d5eaf78e32f604ff122","subject":"Utilize newly added tasks database fields","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/758bb707c63224b8ff9a1d5eaf78e32f604ff122"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-15 09:04:17.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-16 15:12:51.000000000","tz":0},"subject":"New API /v2/images/{id}/tasks","message":"New API /v2/images/{id}/tasks\n\nAdded new API /v2/images/{id}/tasks to show tasks associated with\nimage. This API will return list of tasks associated for valid image\nelse returns 404 not found if image is not present. This API also\ninitiates task scrubbing before returning tasks to user.\n\nImplements: blueprint messages-api\nChange-Id: Ib3cacb4dd4d75de32e539f8a3b48bdaa762e6d8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6bbafa0a09f0cee9a58349eae23c35ec0bd4257b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6bbafa0a09f0cee9a58349eae23c35ec0bd4257b"}]},"branch":"refs/heads/master"},"d94e84e4e3df07b0258ec20524fb8c2f64e253a0":{"kind":"REWORK","_number":6,"created":"2021-02-19 19:30:07.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/30/774830/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/30/774830/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/774830/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/774830/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/774830/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/30/774830/6"}}},"commit":{"parents":[{"commit":"d798530b692bd9d54493fc452ebb9576e75f29cc","subject":"Utilize newly added tasks database fields","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d798530b692bd9d54493fc452ebb9576e75f29cc"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-15 09:04:17.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-19 20:28:17.000000000","tz":-480},"subject":"New API /v2/images/{id}/tasks","message":"New API /v2/images/{id}/tasks\n\nAdded new API /v2/images/{id}/tasks to show tasks associated with\nimage. This API will return list of tasks associated for valid image\nelse returns 404 not found if image is not present. This API also\ninitiates task scrubbing before returning tasks to user.\n\nImplements: blueprint messages-api\nChange-Id: Ib3cacb4dd4d75de32e539f8a3b48bdaa762e6d8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d94e84e4e3df07b0258ec20524fb8c2f64e253a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d94e84e4e3df07b0258ec20524fb8c2f64e253a0"}]},"branch":"refs/heads/master"},"1be3973e4ebe91e4f51597cd0cc764b18f3781d9":{"kind":"REWORK","_number":7,"created":"2021-02-19 20:00:17.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/30/774830/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/30/774830/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/774830/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/774830/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/774830/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/30/774830/7"}}},"commit":{"parents":[{"commit":"d798530b692bd9d54493fc452ebb9576e75f29cc","subject":"Utilize newly added tasks database fields","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d798530b692bd9d54493fc452ebb9576e75f29cc"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-15 09:04:17.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-19 20:58:54.000000000","tz":-480},"subject":"New API /v2/images/{id}/tasks","message":"New API /v2/images/{id}/tasks\n\nAdded new API /v2/images/{id}/tasks to show tasks associated with\nimage. This API will return list of tasks associated for valid image\nelse returns 404 not found if image is not present. This API also\ninitiates task scrubbing before returning tasks to user.\n\nImplements: blueprint messages-api\nChange-Id: Ib3cacb4dd4d75de32e539f8a3b48bdaa762e6d8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1be3973e4ebe91e4f51597cd0cc764b18f3781d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1be3973e4ebe91e4f51597cd0cc764b18f3781d9"}]},"branch":"refs/heads/master"},"8ca29d8379c8357f3e0d177397f7d551769c7af6":{"kind":"REWORK","_number":8,"created":"2021-02-22 06:46:07.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/30/774830/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/30/774830/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/774830/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/774830/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/774830/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/30/774830/8"}}},"commit":{"parents":[{"commit":"d798530b692bd9d54493fc452ebb9576e75f29cc","subject":"Utilize newly added tasks database fields","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d798530b692bd9d54493fc452ebb9576e75f29cc"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-15 09:04:17.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-22 06:30:26.000000000","tz":0},"subject":"New API /v2/images/{id}/tasks","message":"New API /v2/images/{id}/tasks\n\nAdded new API /v2/images/{id}/tasks to show tasks associated with\nimage. This API will return list of tasks associated for valid image\nelse returns 404 not found if image is not present. This API also\ninitiates task scrubbing before returning tasks to user.\n\nImplements: blueprint messages-api\nChange-Id: Ib3cacb4dd4d75de32e539f8a3b48bdaa762e6d8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8ca29d8379c8357f3e0d177397f7d551769c7af6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8ca29d8379c8357f3e0d177397f7d551769c7af6"}]},"branch":"refs/heads/master"},"281fadc15c5e257e72514957701a5571a83d8c38":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-02-24 06:00:03.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/30/774830/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/30/774830/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/774830/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/774830/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/774830/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/30/774830/9"}}},"commit":{"parents":[{"commit":"d54449af44212e01762285832b9c7c1ad4de1eb3","subject":"Utilize newly added tasks database fields","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d54449af44212e01762285832b9c7c1ad4de1eb3"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-15 09:04:17.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-24 05:19:43.000000000","tz":0},"subject":"New API /v2/images/{id}/tasks","message":"New API /v2/images/{id}/tasks\n\nAdded new API /v2/images/{id}/tasks to show tasks associated with\nimage. This API will return list of tasks associated for valid image\nelse returns 404 not found if image is not present. This API also\ninitiates task scrubbing before returning tasks to user.\n\nImplements: blueprint messages-api\nChange-Id: Ib3cacb4dd4d75de32e539f8a3b48bdaa762e6d8e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/281fadc15c5e257e72514957701a5571a83d8c38"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/281fadc15c5e257e72514957701a5571a83d8c38"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
