)]}'
{"id":"openstack%2Fglance~775246","triplet_id":"openstack%2Fglance~master~I04dd2d89793da975a79a4dd721071c19cd939d13","project":"openstack/glance","branch":"master","hashtags":[],"change_id":"I04dd2d89793da975a79a4dd721071c19cd939d13","subject":"DNM Test glance_store lazy volume type check","status":"ABANDONED","created":"2021-02-11 17:53:03.000000000","updated":"2021-02-24 16:25:44.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6bc408d07b19fc2ac0c5d6d33c62a6c83b67ffe5","_number":775246,"virtual_id_number":775246,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-12 21:20:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-11 18:31:55.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-11 22:45:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-12 17:46:50.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"}],"messages":[{"id":"5c27c5be8458d38fe6fd0d7a0dd10e0021550fab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-11 17:53:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92561b1be63038b57a51c5a51041fd6abccf73dd","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-11 18:31:55.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/775246/775246/1/check-not-vote/ext-glance-zuul/a1d7764 : SUCCESS in 38m 28s","accounts_in_message":[],"_revision_number":1},{"id":"5343df4fa7aed23c75d24b76de22e1ea1bd54154","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 22:45:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/9840ec51b4404bc18f4f8928c1dce281 : SUCCESS in 1h 22m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7c55582e87cb42e7ad3079e7f45e36b6 : SUCCESS in 1h 21m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20f5c0f5f695423c8ca051150d2f98f9 : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bd21b93abbd41c1a6b67ece7773cbfb : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e50bac5e21fa4f4380542cbb1016b42c : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c972f49c7d1494db91ed568c9d85555 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b5006dac27045cead0af0328afefc60 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e30206561fe744d6944daf84eeb5c247 : SUCCESS in 5m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c6ad10389c70419b8edd55038fefc426 : SUCCESS in 9m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/07134dd0811a4ea5a536770df0da3c70 : SUCCESS in 11m 24s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e21b83a933d945ccb595efe356a0700a : SUCCESS in 8m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ed64003af11d4561853a517c0c60f4ce : SUCCESS in 1h 27m 14s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f4d9ddd3240c485a8b7c20c1ffd06aa5 : SUCCESS in 1h 21m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/125d82fb9b1945a28cf1c79e7d550952 : SUCCESS in 1h 23m 08s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/593860463b984c32aeef9367f122ff1f : FAILURE in 1h 13m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/623b10372033419589383151007a466e : SUCCESS in 58m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4de1149ba0d2462898af2a43e0860e1b : SUCCESS in 1h 16m 09s","accounts_in_message":[],"_revision_number":1},{"id":"87b58d5dfbaeb9c4d382737fb4c9c79836b1aac7","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-12 17:46:00.000000000","message":"Patch Set 1:\n\nrecheck test latest version","accounts_in_message":[],"_revision_number":1},{"id":"4e41619144af15ed52749d856753dfc168a13762","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-12 17:46:50.000000000","message":"Patch Set 1:\n\nFirst run seemed to work, I think the volume fail was just one of the  many expected cinder fails we see (i.e. it was a VolumesTest not ImagesTest).","accounts_in_message":[],"_revision_number":1},{"id":"aba2b1a3dbed7f281c4b25f0a643130ca7a9a608","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-12 18:24:46.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/775246/775246/1/check-not-vote/ext-glance-zuul/62a8854 : SUCCESS in 38m 17s","accounts_in_message":[],"_revision_number":1},{"id":"1a2e1e838c510d92a8120b7229b8c2c430aed4f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 21:20:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/4cc25258ba824dad8b31eb77e75463c3 : SUCCESS in 1h 13m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c3e257ec961b4f1dad7291d1210be92e : SUCCESS in 1h 12m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/010b6c61f7214a198596ba608bb2dcdc : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66817a35a04b4c7f8d1c82c8ee4233e5 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8cd14679bf24193a0f39a4b298f5e4c : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a46169a930745dc8f0efb1e3a18b072 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16127340f5d04f44a77664cfdd2861b6 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a1e480a73554bd885eb1ceca513322b : SUCCESS in 6m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fe0884b420224833a176378c1907d28e : SUCCESS in 8m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/427e9f3c7c0244f7863f4f20317e4cf2 : FAILURE in 9m 28s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/155a42b84e284fd0a6efb2a848c4a68f : SUCCESS in 4m 52s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c896f9090d554d83a8f7d3f056a822ae : SUCCESS in 1h 28m 38s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/707b942fd01543fb98cb0bcb5a87b6e4 : SUCCESS in 1h 02m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/14c5471e2eb74f499bb1c995d860508e : SUCCESS in 1h 05m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6d924000745a41899cdfbc4c4198052c : SUCCESS in 1h 29m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a052c043526240eda9ae0fcccabf877d : SUCCESS in 1h 07m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c0aea6abee3a4c85b1e9b5b7aff633f0 : SUCCESS in 1h 19m 08s","accounts_in_message":[],"_revision_number":1},{"id":"6bc408d07b19fc2ac0c5d6d33c62a6c83b67ffe5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-24 16:25:44.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ff14eda4272ad533cab4ba031a277179720eae53","revisions":{"ff14eda4272ad533cab4ba031a277179720eae53":{"kind":"REWORK","_number":1,"created":"2021-02-11 17:53:03.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/46/775246/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/46/775246/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/775246/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/775246/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/775246/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/46/775246/1"}}},"commit":{"parents":[{"commit":"f2452863e74b563fdb4b1001af76ea492c63c484","subject":"Merge \"[goal] Deprecate the JSON formatted policy file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f2452863e74b563fdb4b1001af76ea492c63c484"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-11 18:51:08.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-11 18:51:08.000000000","tz":-480},"subject":"DNM Test glance_store lazy volume type check","message":"DNM Test glance_store lazy volume type check\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/774703\nChange-Id: I04dd2d89793da975a79a4dd721071c19cd939d13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ff14eda4272ad533cab4ba031a277179720eae53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ff14eda4272ad533cab4ba031a277179720eae53"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
