)]}'
{"id":"openstack%2Fglance~775276","triplet_id":"openstack%2Fglance~master~Ia6c556be679f9cd81f9920ff33fa089b79afde10","project":"openstack/glance","branch":"master","hashtags":[],"change_id":"Ia6c556be679f9cd81f9920ff33fa089b79afde10","subject":"trivial: Fix a typo in devstack plugin.sh","status":"MERGED","created":"2021-02-11 21:09:51.000000000","updated":"2021-02-12 16:17:42.000000000","submitted":"2021-02-12 16:04:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775276","meta_rev_id":"c7fc674eea5eeba81a28f9eced7dca0a1971d12e","_number":775276,"virtual_id_number":775276,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-12 16:03:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-02-11 21:11:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-02-12 13:48:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-02-12 13:48:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-11 21:10:37.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-02-11 21:10:37.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-02-11 21:11:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-02-11 21:52:04.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-12 00:41:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8ac44ebdb0b293a850c86362876e28839e71534f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-11 21:09:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"26142df3843e83a89f219c9b513d7ca38985b0f4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-02-11 21:11:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e039180a042f176f50b1b421a12026f59409a7fe","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-11 21:52:04.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/775276/775276/1/check-not-vote/ext-glance-zuul/2d19b2e : SUCCESS in 41m 41s","accounts_in_message":[],"_revision_number":1},{"id":"afd45aa9161efedbe9f303da6655d7769d61b6b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 00:41:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/9cf9b15b54f444b292927c0380c838bf : SUCCESS in 1h 07m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/874aa4cba8ac4dc8919e9ad287166cd2 : SUCCESS in 1h 17m 21s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/288796c3098c4afa9c51abdde2e41ed9 : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0736d9353eb745ec80f733c531956851 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d7e2cf52bda4399a1abdd8cb15d867c : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3cca67652fc47fbb26cd68e1557a407 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/884fa5669303465382a49997d145b80a : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfcf120953c5468b8d155286254dc214 : SUCCESS in 6m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eaea2217ef1d401597b045d238f147a2 : SUCCESS in 8m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fcfd00d0a10b47938033725ad31509fe : SUCCESS in 9m 07s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/66df6b9c7bf64605aaada42540824677 : SUCCESS in 5m 00s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9a85890e1d6a4adf9be56b0d08cf521a : SUCCESS in 1h 07m 10s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/215b7871efa74f6dba034719dc491d11 : SUCCESS in 1h 23m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6b5718004aa24d789d9d54008ed6e632 : SUCCESS in 1h 28m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/959fae68d09544f1a504447284c83b3a : SUCCESS in 1h 12m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ccd5ab5935224e428b48f044dfb3cfed : SUCCESS in 1h 03m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/297013ad85f54451b3dc31467b49f025 : SUCCESS in 1h 09m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b6e2c507660fd17d3e721a6f8f44033c73497968","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-12 13:48:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bb787b3221390c762f8889c8b2f42e9d40834590","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 13:48:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5770cf5922141551517eca49b95c35392ce67e84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 16:03:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/3e2cb66abe994fc8bedb1bf4888452e2 : SUCCESS in 57m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8495aa0b2db84504ae77b83cf14fde97 : SUCCESS in 1h 15m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/278265c8f96f4002b919f3f38555bc56 : SUCCESS in 8m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/514d61ab945446da873d007e47a405d9 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8590225b17044ea39c25bc8bc71783a3 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c793b986acd4469eac3b3346fb2b3fd0 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d950434c28154d4ba4640fd19524fa05 : SUCCESS in 7m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d2ffa4bd68f74c51bddc54faaf662fdf : SUCCESS in 7m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2b4975c8102d483a836f1b972cfaea06 : SUCCESS in 10m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7706618cff74471c9f6fce97f12de193 : SUCCESS in 1h 06m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ffeb91e3af6c4db9bebf70b42265f7cc : SUCCESS in 1h 21m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61c7cf241523469e987cb20cef4652f4 : SUCCESS in 57m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a0aa8187f6e847d784181367c2e662e3 : SUCCESS in 1h 12m 51s","accounts_in_message":[],"_revision_number":1},{"id":"0d22274c98e26e7765b12d2fdd48c3f35e90c775","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 16:04:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c7fc674eea5eeba81a28f9eced7dca0a1971d12e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 16:17:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1988814e2a9432790516e1c0ae61383 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"81f8349cde5f928517abd8be744a17e78925983c","revisions":{"81f8349cde5f928517abd8be744a17e78925983c":{"kind":"REWORK","_number":1,"created":"2021-02-11 21:09:51.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/76/775276/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/775276/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/775276/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/775276/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/775276/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/775276/1"}}},"commit":{"parents":[{"commit":"2c893fbd80d0241fad2515221b61266ced12f92d","subject":"Merge \"Add devstack plugin script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2c893fbd80d0241fad2515221b61266ced12f92d"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-11 21:09:14.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-11 21:09:14.000000000","tz":0},"subject":"trivial: Fix a typo in devstack plugin.sh","message":"trivial: Fix a typo in devstack plugin.sh\n\nThe enforce scope script should bounce the glance API after setting\nconfiguration values for scope enforcement, not keystone.\n\nChange-Id: Ia6c556be679f9cd81f9920ff33fa089b79afde10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/81f8349cde5f928517abd8be744a17e78925983c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/81f8349cde5f928517abd8be744a17e78925983c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
