)]}'
{"id":"openstack%2Fglance~775296","triplet_id":"openstack%2Fglance~master~Ib95b14da6fd70570a3a228bb1bfeac8792f30822","project":"openstack/glance","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"Ib95b14da6fd70570a3a228bb1bfeac8792f30822","subject":"WIP: Modify download_image policy to be read-only","status":"ABANDONED","created":"2021-02-11 23:29:36.000000000","updated":"2021-02-15 19:28:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"aae997a6c7a1fd9cd59bd86bddf41398ecf1e6ac","_number":775296,"virtual_id_number":775296,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-02-12 02:15:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-11 23:57:02.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-12 02:15:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"92a841dd22f8329dcce50c3d361c5409b6584d3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-11 23:29:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"930a000f3ce6d31df74018dce68acf7b2b2bdd45","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-11 23:57:00.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6dbda9606dc403083f94e05ba3906a3d6b3691da","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-11 23:57:02.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"d0f24bb5413960433c64d1a94e43c5408b524689","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-12 00:39:29.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/775296/775296/1/check-not-vote/ext-glance-zuul/4cfeb06 : SUCCESS in 39m 45s","accounts_in_message":[],"_revision_number":1},{"id":"4e6dbe0fb49dc3b7a77f8868098f84b6c14371c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 02:15:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/4e854a7873b54b88bcd07c4f10dfc5d5 : SUCCESS in 53m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b5fc27b0ee084fa0a92fa44d0bfe90bc : SUCCESS in 1h 03m 55s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8ea5d195e02b4f7b9a3df5bc7884ed8c : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a70f0f259d77409f856aa892cfe8ad63 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7852ce229c144295b97517efe13e14ef : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbaa85df6e5b4d1a88b0e8f795ba42e6 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1ef729b4aef7403f9b7b8e40fe6cbc58 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/636444f6efc841eda431241ae57f5952 : SUCCESS in 5m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c4e6c56b6f704e30be070f9a1959a690 : SUCCESS in 7m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2a7dec9c4290418d92c3a352f43c5597 : SUCCESS in 9m 56s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/255d0d8923dc4491b2e06889fa33be3d : SUCCESS in 6m 00s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/297eea284ac14ff3ad13cedede952d1b : SUCCESS in 1h 01m 49s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/20d03b990c604e65b841608ea9f7a74a : SUCCESS in 1h 23m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9bedf4a049e34edbb099cf6d6bdad744 : SUCCESS in 1h 03m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/64cfb7048ce448bc8ac665e70f59eb8b : SUCCESS in 1h 23m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36307f5bd8264dc18bdcbee173e702e7 : SUCCESS in 57m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/202398f909e6406fa1ed871a41b9d7c5 : SUCCESS in 1h 27m 10s","accounts_in_message":[],"_revision_number":1},{"id":"aae997a6c7a1fd9cd59bd86bddf41398ecf1e6ac","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-15 19:28:18.000000000","message":"Abandoned\n\nAbandoning this per Abhishek\u0027s comments in https://review.opendev.org/c/openstack/glance/+/764754/8","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f267896b3f3c4d37bd7c671aaf2cb8d81e5b9fdc","revisions":{"f267896b3f3c4d37bd7c671aaf2cb8d81e5b9fdc":{"kind":"REWORK","_number":1,"created":"2021-02-11 23:29:36.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/96/775296/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/96/775296/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/775296/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/775296/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/96/775296/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/96/775296/1"}}},"commit":{"parents":[{"commit":"1fd5c967257ed313a8c3f1eed77d1e57112fa273","subject":"Add logic to properly enforce secure RBAC when handling images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1fd5c967257ed313a8c3f1eed77d1e57112fa273"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-11 23:27:27.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-11 23:27:27.000000000","tz":0},"subject":"WIP: Modify download_image policy to be read-only","message":"WIP: Modify download_image policy to be read-only\n\nThis should probably be rolled into the previous commit, but\ndownload_image doesn\u0027t make any writable changes from what I can tell\nand I don\u0027t see a reason why it wouldn\u0027t be exposed to people with the\nreader role.\n\nChange-Id: Ib95b14da6fd70570a3a228bb1bfeac8792f30822\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f267896b3f3c4d37bd7c671aaf2cb8d81e5b9fdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f267896b3f3c4d37bd7c671aaf2cb8d81e5b9fdc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
