)]}'
{"id":"openstack%2Fglance~775466","triplet_id":"openstack%2Fglance~master~Ie0d9f8eee128205849b24cdad12aa9c2fbdf5278","project":"openstack/glance","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"Ie0d9f8eee128205849b24cdad12aa9c2fbdf5278","subject":"WIP: Refactor get_image policy to return 404 instead of 403","status":"ABANDONED","created":"2021-02-13 00:59:58.000000000","updated":"2021-02-15 19:25:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"efd6f908d27d49fdc7ea06c7dad71892fb87dff1","_number":775466,"virtual_id_number":775466,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-13 02:24:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-13 01:38:43.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-13 02:24:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"31593e0fcd5f4e105faa1c8cd9fd3124141d0860","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-13 00:59:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dec4e6f0a32fb2ab1eb5fb50c0653e90fb1f3d95","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-13 01:38:43.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/775466/775466/1/check-not-vote/ext-glance-zuul/4c9a581 : SUCCESS in 37m 26s","accounts_in_message":[],"_revision_number":1},{"id":"98e69ae7927dd606d3bc53a9f58c678fa6d29e13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-13 02:24:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/7af5bcff72c146d3a602284e3b2b501f : SUCCESS in 1h 17m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/94c9c1688c3042fcae8b0194dd12f4ed : SUCCESS in 1h 05m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/80dcfb530d0c412187af304f9de6f56f : FAILURE in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc3cb9d694b44512a871215092170dcf : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8ac92e2129c54fda9489a8b06f00ce93 : FAILURE in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b68c5bf10c74d0f8ad474b5974840fa : FAILURE in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71baf9ec185a4603b46645774fc203f8 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/446d21a31e244cbe918b12dd4fc1e8ba : SUCCESS in 6m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3830f392be67433fac67afc4c6a7829d : FAILURE in 9m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/eeaa932573474e33a6810baccc2f966f : FAILURE in 7m 34s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6b297a89d91949798c83599aaede78e9 : SUCCESS in 4m 06s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/27c464a5d7f8441298222bc1b8a5994c : SUCCESS in 1h 01m 11s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8771cf7d07104581a348f3258250b94d : SUCCESS in 1h 02m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6b09053476e244a09b0ddd3165ee196e : SUCCESS in 1h 02m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2383815909e745f585b6233736c18ce2 : SUCCESS in 1h 09m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2c9887099a44f559334933e8c4aacd1 : SUCCESS in 53m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0bfee7f1ae134621952a4895985561cd : SUCCESS in 1h 00m 16s","accounts_in_message":[],"_revision_number":1},{"id":"efd6f908d27d49fdc7ea06c7dad71892fb87dff1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-15 19:25:17.000000000","message":"Abandoned\n\nI rebased this on https://review.opendev.org/c/openstack/glance/+/775435 instead","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"170af081158282a30e2a87892f0fc688e531018e","revisions":{"170af081158282a30e2a87892f0fc688e531018e":{"kind":"REWORK","_number":1,"created":"2021-02-13 00:59:58.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/66/775466/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/775466/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/775466/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/775466/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/775466/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/775466/1"}}},"commit":{"parents":[{"commit":"f267896b3f3c4d37bd7c671aaf2cb8d81e5b9fdc","subject":"WIP: Modify download_image policy to be read-only","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f267896b3f3c4d37bd7c671aaf2cb8d81e5b9fdc"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-13 00:57:55.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-13 00:57:55.000000000","tz":0},"subject":"WIP: Refactor get_image policy to return 404 instead of 403","message":"WIP: Refactor get_image policy to return 404 instead of 403\n\nStealing this from https://review.opendev.org/c/openstack/glance/+/775435\nfor testing with the protection tests in glance_tempest_plugin.\n\nThis should be formally included in the chain.\n\nChange-Id: Ie0d9f8eee128205849b24cdad12aa9c2fbdf5278\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/170af081158282a30e2a87892f0fc688e531018e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/170af081158282a30e2a87892f0fc688e531018e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
