)]}'
{"id":"openstack%2Fglance~775597","triplet_id":"openstack%2Fglance~master~Ie69d4180e09fe4c670faaa7d3c31742d1fc581e2","project":"openstack/glance","branch":"master","topic":"bump-image-api-version","hashtags":[],"change_id":"Ie69d4180e09fe4c670faaa7d3c31742d1fc581e2","subject":"Update API docs for new /v2/images/{image_id}/tasks API","status":"MERGED","created":"2021-02-15 10:49:52.000000000","updated":"2021-02-26 00:34:08.000000000","submitted":"2021-02-26 00:27:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"775597-bump-image-api-version","meta_rev_id":"f26c08ea8b702794115c95196f0b338e0f3f1f93","_number":775597,"virtual_id_number":775597,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-26 00:27:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-02-24 13:35:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2021-02-24 13:53:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-02-24 13:53:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-15 11:38:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-19 19:30:07.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-02-23 22:58:34.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"b2b1d317617792f69ec126d746895541c94dea8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-15 10:49:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8064eb9762d27f91d2ffd2f89e55adae351ce7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 11:38:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/fc64390cb5394e3898b432f2ef38d19a : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8994329309f34e268c6f5900531958ca : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8e219f28188449a988017353b016b09 : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f9eeee74891444d8b90001d14bb11be : SUCCESS in 5m 56s","accounts_in_message":[],"_revision_number":1},{"id":"5072054182cfb1a2078e149531f6e60d3533ee50","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-16 07:04:06.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"565ebfda40377a8ff35d4ffc7a0715b39dd8e6ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 07:16:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/19a0be007ce646b9abb099bfabcc0b8b : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c62d90c631a5411596839a304592084a : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bad0d18c38284905b1ae03990f363cd3 : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff2a9dd518ea4e63a4e51abc89776be7 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":2},{"id":"35f7a026fe2f56720151feb17748be1651f1dece","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-16 15:34:27.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"2f2fa4cff4381de70ee54e34783b0f10bce2f322","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 16:56:48.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f3af5f6157d344609839bdfe587ebea7 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83a13ff632c84a87995fd5c2ca707ea4 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e9cc5e816fc4ea19571d46da0dd85a6 : SUCCESS in 7m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/978be1aa12f04305a2389f2f7cfc063d : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":3},{"id":"668929f6a9eb9e5191925f8a59b5580d5aa0f591","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-19 19:30:07.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"2969d244b863323b0a63dd365f79fdaa97d91671","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-19 20:00:17.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"37c327ac140afe4d0aada17811728247409ec321","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 20:53:48.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/04b79b45115a47b08fc241a1a4405fd3 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/859432ad7bb34280bdca1f53dbb258f7 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c40f8422b39a4664a81eb564860495fb : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c56f0e95b144721a3f241164e07d82d : SUCCESS in 6m 03s","accounts_in_message":[],"_revision_number":5},{"id":"6a2533f29e44701e3dc654e46a70f55fdd011116","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-22 06:46:07.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"29cceda40c0781154d50481edfd45c31d42048ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 06:56:57.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f6343a9c958743629c372a55ad91d7e6 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1f93f3acbfb4515a5998c78c661a981 : SUCCESS in 7m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f98fbdfe2d44da3b7945975b8423d6c : SUCCESS in 8m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dafe4e47d18f4da297e694ad56c62a13 : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":6},{"id":"2b38453cffd2049a870f3a08e05cbb442e5f5e07","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-02-23 22:58:34.000000000","message":"Patch Set 6: Code-Review-1\n\n(5 comments)\n\nA few suggestions inline, some nits but some improvements.","accounts_in_message":[],"_revision_number":6},{"id":"56fd3066501ec50ccf8cb625b9205b3e60c739b5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-24 05:43:00.000000000","message":"Patch Set 6:\n\n(5 comments)\n\nThank you for review Brian!!","accounts_in_message":[],"_revision_number":6},{"id":"a5b2dc2d045891447798abaf67a037342e51e389","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-24 06:00:03.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ebeb3ff2eef95924843fe52961307c9b9869dbef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 07:06:26.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0dc2ba066f914b79bd928b8a75431106 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07ca5e743b764a72a7991d037e5824a0 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a426a7da49444ef884118cfb412e002b : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d4103b979f5469dbaa45061a66c312d : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":7},{"id":"34588f2479ea51180aa6525edf445b05bdf2b0a0","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-02-24 13:35:52.000000000","message":"Patch Set 7: Code-Review+2\n\nMy concerns have been addressed.  LGTM.","accounts_in_message":[],"_revision_number":7},{"id":"91417342b3e29c6b780a71d77cb1b91ff5450a73","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-24 13:53:44.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)\n\nThis looks good modulo a nit, if you have to respin.","accounts_in_message":[],"_revision_number":7},{"id":"0bff73f7b0dc0167b63800af9a132c6d99a463bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 22:06:42.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"3defb0dc99d54524c5b2f732a1ac2c58ed9d79ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 00:27:29.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2544cb62f21b4f84a1a6d3e4271b394b : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4ae607cfd894773ad636850e6dda5da : SUCCESS in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/658a48eb6d6542e592dac70c236883ea : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f018f761f63a4e5991a1e2336107d41d : SUCCESS in 5m 37s","accounts_in_message":[],"_revision_number":7},{"id":"96f2c8f38f298c5c4b77cfe7d971cbe0d1f96155","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 00:27:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"f26c08ea8b702794115c95196f0b338e0f3f1f93","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 00:34:08.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/a0c4931d6f1e49d0bb2e4b290de7435e : SUCCESS in 1m 02s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/571e5c5007b849199d5038f4bb2f3d49 : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f601af7e2f7b4ff08ac656aeff995101 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"03696299d20debf8a5aa561bc80878775ad02695","revisions":{"17866a1b100a0e67ff5b62ed9cc80d8e4e212b36":{"kind":"REWORK","_number":1,"created":"2021-02-15 10:49:52.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/97/775597/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/97/775597/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/775597/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/775597/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/775597/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/97/775597/1"}}},"commit":{"parents":[{"commit":"9d959c7b4a41ddabf0e9c93748dba96531b7504a","subject":"New API /v2/images/{id}/tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9d959c7b4a41ddabf0e9c93748dba96531b7504a"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-15 10:04:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-15 10:04:41.000000000","tz":0},"subject":"Update API docs for new /v2/images/{image_id}/tasks API","message":"Update API docs for new /v2/images/{image_id}/tasks API\n\nUpdated API reference documentation for new API.\n\nChange-Id: Ie69d4180e09fe4c670faaa7d3c31742d1fc581e2\nRelated: blueprint messages-api\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/17866a1b100a0e67ff5b62ed9cc80d8e4e212b36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/17866a1b100a0e67ff5b62ed9cc80d8e4e212b36"}]},"branch":"refs/heads/master"},"f548847c15b3242c46da07262034ae3a9fce5fd8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-16 07:04:06.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/97/775597/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/97/775597/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/775597/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/775597/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/775597/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/97/775597/2"}}},"commit":{"parents":[{"commit":"9b1ba2b2d6bd6a1cf7ecd88224587ccb33ac3e7a","subject":"New API /v2/images/{id}/tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9b1ba2b2d6bd6a1cf7ecd88224587ccb33ac3e7a"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-15 10:04:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-16 06:27:22.000000000","tz":0},"subject":"Update API docs for new /v2/images/{image_id}/tasks API","message":"Update API docs for new /v2/images/{image_id}/tasks API\n\nUpdated API reference documentation for new API.\n\nChange-Id: Ie69d4180e09fe4c670faaa7d3c31742d1fc581e2\nRelated: blueprint messages-api\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f548847c15b3242c46da07262034ae3a9fce5fd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f548847c15b3242c46da07262034ae3a9fce5fd8"}]},"branch":"refs/heads/master"},"bc7598bcc2a1c2dd180caf6fe8bb40e74eb3737b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-02-16 15:34:27.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/97/775597/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/97/775597/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/775597/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/775597/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/775597/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/97/775597/3"}}},"commit":{"parents":[{"commit":"6bbafa0a09f0cee9a58349eae23c35ec0bd4257b","subject":"New API /v2/images/{id}/tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6bbafa0a09f0cee9a58349eae23c35ec0bd4257b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-15 10:04:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-16 15:14:52.000000000","tz":0},"subject":"Update API docs for new /v2/images/{image_id}/tasks API","message":"Update API docs for new /v2/images/{image_id}/tasks API\n\nUpdated API reference documentation for new API.\n\nChange-Id: Ie69d4180e09fe4c670faaa7d3c31742d1fc581e2\nRelated: blueprint messages-api\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc7598bcc2a1c2dd180caf6fe8bb40e74eb3737b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc7598bcc2a1c2dd180caf6fe8bb40e74eb3737b"}]},"branch":"refs/heads/master"},"832ef03ed571f26e2def07efaa2ca57a792678fa":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-02-19 19:30:07.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/97/775597/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/97/775597/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/775597/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/775597/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/775597/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/97/775597/4"}}},"commit":{"parents":[{"commit":"d94e84e4e3df07b0258ec20524fb8c2f64e253a0","subject":"New API /v2/images/{id}/tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d94e84e4e3df07b0258ec20524fb8c2f64e253a0"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-15 10:04:41.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-19 20:28:17.000000000","tz":-480},"subject":"Update API docs for new /v2/images/{image_id}/tasks API","message":"Update API docs for new /v2/images/{image_id}/tasks API\n\nUpdated API reference documentation for new API.\n\nChange-Id: Ie69d4180e09fe4c670faaa7d3c31742d1fc581e2\nRelated: blueprint messages-api\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/832ef03ed571f26e2def07efaa2ca57a792678fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/832ef03ed571f26e2def07efaa2ca57a792678fa"}]},"branch":"refs/heads/master"},"7d24da56395af70d24c45c67e1fdd4838bf305fd":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-02-19 20:00:17.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/97/775597/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/97/775597/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/775597/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/775597/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/775597/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/97/775597/5"}}},"commit":{"parents":[{"commit":"1be3973e4ebe91e4f51597cd0cc764b18f3781d9","subject":"New API /v2/images/{id}/tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1be3973e4ebe91e4f51597cd0cc764b18f3781d9"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-15 10:04:41.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-19 20:58:56.000000000","tz":-480},"subject":"Update API docs for new /v2/images/{image_id}/tasks API","message":"Update API docs for new /v2/images/{image_id}/tasks API\n\nUpdated API reference documentation for new API.\n\nChange-Id: Ie69d4180e09fe4c670faaa7d3c31742d1fc581e2\nRelated: blueprint messages-api\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7d24da56395af70d24c45c67e1fdd4838bf305fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7d24da56395af70d24c45c67e1fdd4838bf305fd"}]},"branch":"refs/heads/master"},"54dad4ba08832f3b99781885c65bd4c0e6c5dda3":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-02-22 06:46:07.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/97/775597/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/97/775597/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/775597/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/775597/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/775597/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/97/775597/6"}}},"commit":{"parents":[{"commit":"8ca29d8379c8357f3e0d177397f7d551769c7af6","subject":"New API /v2/images/{id}/tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8ca29d8379c8357f3e0d177397f7d551769c7af6"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-15 10:04:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-22 06:34:11.000000000","tz":0},"subject":"Update API docs for new /v2/images/{image_id}/tasks API","message":"Update API docs for new /v2/images/{image_id}/tasks API\n\nUpdated API reference documentation for new API.\n\nChange-Id: Ie69d4180e09fe4c670faaa7d3c31742d1fc581e2\nRelated: blueprint messages-api\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/54dad4ba08832f3b99781885c65bd4c0e6c5dda3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/54dad4ba08832f3b99781885c65bd4c0e6c5dda3"}]},"branch":"refs/heads/master"},"03696299d20debf8a5aa561bc80878775ad02695":{"kind":"REWORK","_number":7,"created":"2021-02-24 06:00:03.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/97/775597/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/97/775597/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/775597/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/775597/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/97/775597/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/97/775597/7"}}},"commit":{"parents":[{"commit":"281fadc15c5e257e72514957701a5571a83d8c38","subject":"New API /v2/images/{id}/tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/281fadc15c5e257e72514957701a5571a83d8c38"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-15 10:04:41.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-24 05:57:16.000000000","tz":0},"subject":"Update API docs for new /v2/images/{image_id}/tasks API","message":"Update API docs for new /v2/images/{image_id}/tasks API\n\nUpdated API reference documentation for new API.\n\nChange-Id: Ie69d4180e09fe4c670faaa7d3c31742d1fc581e2\nRelated: blueprint messages-api\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03696299d20debf8a5aa561bc80878775ad02695"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03696299d20debf8a5aa561bc80878775ad02695"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
