)]}'
{"id":"openstack%2Fglance~775860","triplet_id":"openstack%2Fglance~master~Icd56ce4f0fb0b70fdfc649474766b638a707fdd0","project":"openstack/glance","branch":"master","topic":"bump-image-api-version","hashtags":[],"change_id":"Icd56ce4f0fb0b70fdfc649474766b638a707fdd0","subject":"Bump Images API version to 2.12","status":"MERGED","created":"2021-02-16 15:34:27.000000000","updated":"2021-03-06 21:55:06.000000000","submitted":"2021-03-06 21:53:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"775860-bump-image-api-version","meta_rev_id":"95133b45e91838d56ce59196df4b28a5755e17c2","_number":775860,"virtual_id_number":775860,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-06 21:53:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":2,"date":"2021-03-02 17:26:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":1,"date":"2021-03-05 18:38:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-16 22:53:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-19 19:30:07.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-03-02 08:13:21.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-03-02 17:24:24.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"2fac8fde957d497d5107cb8ad454002659a384a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-16 15:34:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"831c9d77f06e02f8cb304801738d1d8e23de9096","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-16 15:38:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ab49dce424078d2ab2c832f26b5a14672694402","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 22:53:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/f283592d625d43c7b23d90dbc265fed1 : SUCCESS in 1h 00m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/216e7c0e4c214b48ab13c313a8deae81 : SUCCESS in 1h 19m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d25f2e0e9c64c0eba8436307563a88b : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e204621f0a0a47d2a91c9195474c3d96 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b0b8dc475bd4a3c836e0d378dfea460 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f69c0a7803a64f8b8267a83234a4f2bc : SUCCESS in 7m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9678f6cdced9419cb70e8683ff07ff15 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7554d7d2a374294a1eda86a073ff940 : SUCCESS in 6m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/490f7e57dd4948a3a23e04a9c191fd69 : SUCCESS in 9m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4cdc2ac6d91b437eb72ec5d2d390e889 : SUCCESS in 8m 17s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b385d091174c4f988cd668ceb2172970 : SUCCESS in 4m 29s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3031058318da4427819209b368aeb6ee : SUCCESS in 1h 21m 40s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/97fe2139b23c4886b3a3449d298af2d0 : SUCCESS in 1h 20m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9bd26b7621064ae09a31733906f7dae3 : SUCCESS in 1h 10m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c1acf98fc7e242f796ea1bdd670b4fa2 : SUCCESS in 1h 08m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa1dc24eb825451582ec24eabf3dc248 : SUCCESS in 1h 26m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3d3aac61b5eb4354b6243fa1939f563e : SUCCESS in 1h 30m 36s","accounts_in_message":[],"_revision_number":1},{"id":"d00922b397ed55a5fb3ecad330e7fb1de428ef9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-19 19:30:07.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"ed203b061d30f0a33a065194fcef4ba773561d10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-19 20:00:17.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"57deae28fa4d1f20ebf645c44c00c94bc749380f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 22:12:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/c7ccac131dd946fd92f490b706eb6dfd : SUCCESS in 55m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e7881a8010314095a832e0b1fd3df008 : SUCCESS in 1h 08m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/09aacde872be44e9ac784e4171978fb8 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b203664fde94fa9b3e6b5710670bb62 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/299d5d513bf240a181cdc2f32a12985b : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4184027fbc84af5afd2d7ecea017f29 : SUCCESS in 9m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5f43b03696b4cfbbf748a84d541fc7c : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46f3c0c93ee54e81832a50ff117bf6f6 : SUCCESS in 6m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d68d5e7f2acf45c499c10f4fca78efd0 : SUCCESS in 8m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c9d8b47a48ed41aead02c998842c4c87 : SUCCESS in 7m 32s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4c46b98e8ece45608047864f7a7094bd : SUCCESS in 6m 32s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fe52655cb8ac4f9b805c4c65258004a0 : SUCCESS in 1h 15m 40s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b00762db03ff4dd8b39288f032217d21 : SUCCESS in 1h 26m 25s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/041d2b684de749399f0966bec04f1a68 : SUCCESS in 1h 17m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d279b24e069d487aac3ae02b79080d78 : FAILURE in 1h 08m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff9c5e778fa64588ad362b1b6c312c6b : SUCCESS in 1h 00m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cf17c9ee56124a7aa9ff015284860eb8 : SUCCESS in 1h 06m 06s","accounts_in_message":[],"_revision_number":3},{"id":"8f91aaaa2be963d3e1ad47fb298d519df9095ca3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-22 06:46:07.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"555023dc63f9acde6fb3562e8cb4867f43331f73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 08:17:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/a184764e98444f1d9d0f56c1feb5003f : SUCCESS in 58m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8bfd821e552a4d6b8f7374edc63c0302 : SUCCESS in 1h 15m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee8b24677ed9401d827fa03073b4d7e4 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/688f43342a7b4f049e320fd4b608a9bf : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8dc1e2738b0a47e093271ad6809fb951 : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc52bc88d922499abe37242ba62713ac : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35036c765d70454bbf906ca2e65fa762 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74b8929a77ef47aa9f76b58d16d00dcc : SUCCESS in 6m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9e6527c8615a4c8791e51a05cf7b6305 : SUCCESS in 7m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/691f62ccdc794b62a891c49669495560 : SUCCESS in 8m 00s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/fa19b3968f5145cfbdb162e1b1e47877 : SUCCESS in 6m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/528c5d869282445fb984806d592d0729 : SUCCESS in 1h 14m 26s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7eac19bea2b44dd5aa8d0ce8e37b00a0 : FAILURE in 1h 04m 38s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f76272d9cfb64b34aafac8647a33a8e6 : SUCCESS in 1h 07m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9c4679c6453b4d28b628f63534154b94 : SUCCESS in 1h 25m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/43d8eacab0fe4e5696ae941a9c3a8643 : SUCCESS in 1h 04m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c045e2cfb09440368cfd119c7234ed86 : SUCCESS in 1h 20m 01s","accounts_in_message":[],"_revision_number":4},{"id":"d78fd599a283a5335eb99aae033d9037e09bb0f6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-22 08:28:10.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"9091e7806349c60cf6add94407900376e9b80bd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 09:56:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/b623b62a79174b589d13a86df11fef7b : SUCCESS in 1h 05m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/68764a1881ab49edb1e8037a7bf39bee : SUCCESS in 1h 05m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/510f4df9f7b54734922834c3cebc49ea : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96741c56a03540c6b41ff508fdf7666c : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31e625257bef4e3789baf770d1b4cc9d : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a33412cc76246ad9b988540dcd3e977 : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/246664cad14f4a7bb65c88172c597b12 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9df266829b749859b087181b9874bd3 : SUCCESS in 6m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/147e7a306d804da19fcc37a375194df4 : SUCCESS in 9m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/971b1e91f5724333803d05d9a97f0da9 : SUCCESS in 8m 01s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/151f018de3074e1799d83226882e3776 : SUCCESS in 6m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/72b197ed39b54eb6ac9e92868c6c4c4f : SUCCESS in 1h 19m 46s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7b10ffbf066e4c039c5fef2eef8065cb : SUCCESS in 1h 07m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a508aeb454654399ad867e7870f28766 : SUCCESS in 1h 18m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7f05dd9b8ea44d15af7faab54056d3fe : SUCCESS in 1h 17m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c13e118de9ba4044a02130384a5960c7 : SUCCESS in 57m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4e2f0e5d885b4013acfe16083f2c21d4 : SUCCESS in 1h 11m 38s","accounts_in_message":[],"_revision_number":4},{"id":"91d88682e96c781a3555d473f1afab6d7f4730c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-24 06:00:03.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"f21f00cb072f58eee33906b452e022107ebfaab6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 08:26:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/170f28f727a643798c24f0fe103b44c0 : SUCCESS in 54m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4c45e531fe0b43a49f8c51dfab0ec011 : SUCCESS in 1h 24m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea0e826480dd48b4bf7b032c5d385420 : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34d1df7083d7453ca5c612cdf39a2fb3 : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/10bb9ee3e724416c8688033d62c7083d : SUCCESS in 7m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0acce8a76f844de4844d71282f093d18 : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9266137d4e394c3e9380a019fe0a72e0 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e941acd1683442ad9c0bdccce88f6cd6 : SUCCESS in 7m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4c42483b59774613a16939c10adc4dde : SUCCESS in 9m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cd47a3ad7cf049d59aee8c9726939b40 : SUCCESS in 9m 51s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/167db127ba224bd08c2a1cc319a9ca34 : SUCCESS in 5m 13s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6fb4bb823c274c14ae40aa265d5cff52 : SUCCESS in 1h 13m 59s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/042b1408a6fe4833960370c06439eec4 : SUCCESS in 1h 22m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4919eedfe555449c83bea0601148ca56 : SUCCESS in 1h 07m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/540b2cee3b8945408f9f26c0128faa85 : SUCCESS in 1h 13m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0421b2b2f7a84110bc9359c1b3a65165 : SUCCESS in 57m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d173f0ca26e94f3db35f1f6f678d7dc1 : SUCCESS in 1h 05m 11s","accounts_in_message":[],"_revision_number":5},{"id":"efa18f76225bf65b6139dc5b464fa85a29848d8e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-01 19:35:35.000000000","message":"Patch Set 5: Code-Review+1\n\nJust to be clear, I think this is good. I don\u0027t know what Glance\u0027s API versioning scheme and policy is (like, do we do this once per release, or feature?), so I\u0027m hesitant to +2 without another reviewer. That said, I wrote the tempest test that depends on this, so I can confirm that it works. Consider my +1 a +2 alongside another :)","accounts_in_message":[],"_revision_number":5},{"id":"2f87d9e714e5d8518157e27ac1529095af604f04","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-02 08:14:21.000000000","message":"Patch Set 5:\n\n\u003e Patch Set 5: Code-Review+1\n\u003e \n\u003e Just to be clear, I think this is good. I don\u0027t know what Glance\u0027s API versioning scheme and policy is (like, do we do this once per release, or feature?), so I\u0027m hesitant to +2 without another reviewer. That said, I wrote the tempest test that depends on this, so I can confirm that it works. Consider my +1 a +2 alongside another :)\n\nOnce per feature if it is affecting API","accounts_in_message":[],"_revision_number":5},{"id":"2510d0ca92fedbe8ae751885287c35fc29232711","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-02 17:26:24.000000000","message":"Patch Set 5: Code-Review+2\n\n\u003e Once per feature if it is affecting API\n\nOkay then :)","accounts_in_message":[],"_revision_number":5},{"id":"1716ea8016abbc7902e63383f3518e28e9c30c98","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-05 18:38:25.000000000","message":"Patch Set 5: Workflow+1\n\nPer discussion about limited review bandwidth at crunch time, I\u0027m going to ninja approve this as it\u0027s nearly trivial, we have a test up to validate it, and it\u0027s needed for the client patch.","accounts_in_message":[],"_revision_number":5},{"id":"83831af6bcde0ae5e82601c3ae6e4375b8485b79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 02:09:39.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"2461990bbd40ab3070e58e9a1659a1629073fe69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 13:19:26.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/794fe42a55084697b059eb3e23da42ee : SUCCESS in 1h 04m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8080dbf5a8c643609205f9458d21fcc6 : SUCCESS in 1h 19m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/574a5a2bdfd54694987de82f46963744 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a66734c80cd747e4aaff0fe76edc01f9 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8aa07648dcc46ca9e959fbdd3402e66 : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98760ae0f266461486df2a70b995ff4d : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1745561d646f4452994300b34a98a7ef : SUCCESS in 7m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a7738b9a60514d40a092c81318c956b0 : SUCCESS in 9m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b5f0203fab9f4a54a0fd01ea4b29faf4 : SUCCESS in 11m 15s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e2316beba65d4ecc9d79771467720a2f : FAILURE in 1h 15m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/800bc3b7a96f4d66848c821e6e2d774e : SUCCESS in 1h 17m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c2fc77d8b5df476489265ac1cab04abc : SUCCESS in 1h 00m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/62f7969d28854b5d9a5624c66627f4f1 : SUCCESS in 1h 29m 44s","accounts_in_message":[],"_revision_number":5},{"id":"8ebb5f2f9a03a9b5b158e2cf1b735d01188a2a07","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-06 16:52:12.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"96a12643f2a4ee59f9d0198609e94b2d9dcffba9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 18:29:21.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/37ae8019282e409fb4545b4014c8b4b5 : SUCCESS in 1h 05m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/406d5076f1984694becdbc5095ad3ec2 : SUCCESS in 1h 14m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95452684e04e41a8bf6ce5b9d741e6b4 : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58cb58d144674e33ab2a0f811f215aa5 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a73f3e75d63a431f93bd6ef30d73c399 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b43699f57274257a07ee3f24a63299c : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83afd873caf24d9a8531da0fa4572ed3 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42a07993f0c14369bbfbfd2fe9c6f731 : SUCCESS in 6m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d7114f7778444a55a3805fba0a00e99b : SUCCESS in 9m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fe46a346aee64a1cba6a0b239075c88f : SUCCESS in 8m 50s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d90a71bf34dc460dbfc624aa9d883566 : SUCCESS in 5m 01s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/024e028c28694affbb6205cea93383f5 : SUCCESS in 1h 07m 46s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1c31dcc8946f435a8ce1680b7164c937 : SUCCESS in 1h 24m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/977e61063878493cafcdfe2ab7296cf8 : FAILURE in 53m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f9cc116ee542463aa17a1302bdafe5c3 : SUCCESS in 1h 30m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8974629e2ae54ffaa3258f224146aa96 : SUCCESS in 43m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/79fe4db08a4242539aaed22bfae9b743 : SUCCESS in 59m 27s","accounts_in_message":[],"_revision_number":5},{"id":"f48230c2bc9f2fc7869a053b57b64e97403aa318","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-06 18:33:11.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"40b1a878d517d8e468a02a3a1853174210b60004","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 20:10:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/91b466e704be4447b04dbb2b1ee70b5e : SUCCESS in 1h 24m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4dd9c91f807542738b130a6ccb5a93a4 : SUCCESS in 1h 13m 39s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21bda1868e8b40e58e4b31b45298cf0b : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12f90ece42984c9ba6494d0bae45e382 : SUCCESS in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/923b0c632b6442bf93e48b0b3ff6bf67 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8568e2e49a354b11ae01a754a3f982e7 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fabd15de5c734334b3d9fbdd02fc68bf : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/104943b5bcb540ef9cd9a3b49768ed8b : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d21d674ded25461b945ad739250b2412 : SUCCESS in 12m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0db04291e22b4ac6ae2b6f6c0e68f9e2 : SUCCESS in 13m 40s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3e122b6b33b143a1862769968fcc4952 : SUCCESS in 6m 52s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ff114b30be144a199ff1459eb60c8936 : SUCCESS in 1h 24m 18s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f1675dd8862442dbb80f8e1078e8255e : SUCCESS in 1h 14m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/72050de574614ec4a202bb7099898768 : SUCCESS in 1h 35m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/04b54ba6ac8d4f43bbd32f87d1afa888 : SUCCESS in 1h 10m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c140f3687af74d9fae5c098a32f35144 : SUCCESS in 1h 03m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3dfbe66c8fe444f7ad8cde63298cc531 : SUCCESS in 1h 09m 54s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e3bd9b2d823d412cbfdcd288743a2d69 : FAILURE in 30m 32s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/41e53baab67e40f8ad71814f15744b79 : SUCCESS in 29m 14s","accounts_in_message":[],"_revision_number":5},{"id":"10770304dc16a7e9bd8656a7567781d6094ceb65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 20:11:01.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"483d40efea89d112b4e951b625bca550172c2d60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 21:53:39.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/5227d7e555b9406398c3a9452f60c483 : SUCCESS in 58m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c89aae6aaadc46cea5cf554e61e849eb : SUCCESS in 1h 10m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d353621c7e01498aa1155866c1e7ec80 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05d84e7db5e14d55a986679588ed9666 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/37c533a5e5b841db8cff4176902744a0 : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f50accc673cd43fbbd5daeefafd6cf63 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38d79f4d5d4a4a2392efbe6a99ef6e54 : SUCCESS in 5m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/392ccc5381d440a6a75353d9915ef760 : SUCCESS in 10m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8a4af79ec3054e96a091b51097c325b8 : SUCCESS in 11m 46s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/572db228bfae45da9fae439e48c22584 : SUCCESS in 1h 11m 33s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7c21a03440ab4cb0bbe8568185a2f64c : SUCCESS in 1h 02m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be11da4613da4d58ab6e3e10ef3503d1 : SUCCESS in 53m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8dcd87842ee6445bbbf0bb2ba014ac74 : SUCCESS in 1h 14m 22s\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/66bab1d5574144d09eeeeb7c13d80c98 : SUCCESS in 33m 46s","accounts_in_message":[],"_revision_number":5},{"id":"f04cd5108dc2a658f7188bdbe762e8f0b938c594","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 21:53:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"95133b45e91838d56ce59196df4b28a5755e17c2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 21:55:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d205722272da4e28b303b38126fde686 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"af3ce1adc98cf652e06af0fdb87206c9897ace2f","revisions":{"bd6d0e4006d349278ef87f85a96641d50040aaf0":{"kind":"REWORK","_number":1,"created":"2021-02-16 15:34:27.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/60/775860/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/60/775860/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/775860/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/775860/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/775860/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/60/775860/1"}}},"commit":{"parents":[{"commit":"bc7598bcc2a1c2dd180caf6fe8bb40e74eb3737b","subject":"Update API docs for new /v2/images/{image_id}/tasks API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc7598bcc2a1c2dd180caf6fe8bb40e74eb3737b"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-16 15:24:33.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-16 15:24:33.000000000","tz":0},"subject":"Bump Images API version to 2.12","message":"Bump Images API version to 2.12\n\nChange-Id: Icd56ce4f0fb0b70fdfc649474766b638a707fdd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bd6d0e4006d349278ef87f85a96641d50040aaf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bd6d0e4006d349278ef87f85a96641d50040aaf0"}]},"branch":"refs/heads/master"},"98e8b883ec7e5f723d5524cae3456133d922d262":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-19 19:30:07.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/60/775860/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/60/775860/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/775860/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/775860/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/775860/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/60/775860/2"}}},"commit":{"parents":[{"commit":"832ef03ed571f26e2def07efaa2ca57a792678fa","subject":"Update API docs for new /v2/images/{image_id}/tasks API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/832ef03ed571f26e2def07efaa2ca57a792678fa"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-16 15:24:33.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-19 20:28:17.000000000","tz":-480},"subject":"Bump Images API version to 2.12","message":"Bump Images API version to 2.12\n\nChange-Id: Icd56ce4f0fb0b70fdfc649474766b638a707fdd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/98e8b883ec7e5f723d5524cae3456133d922d262"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/98e8b883ec7e5f723d5524cae3456133d922d262"}]},"branch":"refs/heads/master"},"b1969f54fb0e53bcda2bf1e17f6127cf8d888a2c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-02-19 20:00:17.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/60/775860/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/60/775860/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/775860/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/775860/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/775860/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/60/775860/3"}}},"commit":{"parents":[{"commit":"7d24da56395af70d24c45c67e1fdd4838bf305fd","subject":"Update API docs for new /v2/images/{image_id}/tasks API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7d24da56395af70d24c45c67e1fdd4838bf305fd"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-16 15:24:33.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-19 20:58:56.000000000","tz":-480},"subject":"Bump Images API version to 2.12","message":"Bump Images API version to 2.12\n\nChange-Id: Icd56ce4f0fb0b70fdfc649474766b638a707fdd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b1969f54fb0e53bcda2bf1e17f6127cf8d888a2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b1969f54fb0e53bcda2bf1e17f6127cf8d888a2c"}]},"branch":"refs/heads/master"},"37c8c75dbdf3c59d9e436878847af677f59102ea":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-02-22 06:46:07.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/60/775860/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/60/775860/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/775860/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/775860/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/775860/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/60/775860/4"}}},"commit":{"parents":[{"commit":"54dad4ba08832f3b99781885c65bd4c0e6c5dda3","subject":"Update API docs for new /v2/images/{image_id}/tasks API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/54dad4ba08832f3b99781885c65bd4c0e6c5dda3"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-16 15:24:33.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-22 06:35:54.000000000","tz":0},"subject":"Bump Images API version to 2.12","message":"Bump Images API version to 2.12\n\nChange-Id: Icd56ce4f0fb0b70fdfc649474766b638a707fdd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/37c8c75dbdf3c59d9e436878847af677f59102ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/37c8c75dbdf3c59d9e436878847af677f59102ea"}]},"branch":"refs/heads/master"},"af3ce1adc98cf652e06af0fdb87206c9897ace2f":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":5,"created":"2021-02-24 06:00:03.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/60/775860/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/60/775860/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/775860/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/775860/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/775860/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/60/775860/5"}}},"commit":{"parents":[{"commit":"03696299d20debf8a5aa561bc80878775ad02695","subject":"Update API docs for new /v2/images/{image_id}/tasks API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03696299d20debf8a5aa561bc80878775ad02695"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-16 15:24:33.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-24 05:58:14.000000000","tz":0},"subject":"Bump Images API version to 2.12","message":"Bump Images API version to 2.12\n\nRelated: blueprint messages-api\nChange-Id: Icd56ce4f0fb0b70fdfc649474766b638a707fdd0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/af3ce1adc98cf652e06af0fdb87206c9897ace2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/af3ce1adc98cf652e06af0fdb87206c9897ace2f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
