)]}'
{"id":"openstack%2Fglance~776406","triplet_id":"openstack%2Fglance~master~Ie860c63d60b800ecafc4ee655f17c53cc3381c51","project":"openstack/glance","branch":"master","topic":"bug/1916011","hashtags":[],"change_id":"Ie860c63d60b800ecafc4ee655f17c53cc3381c51","subject":"Skip test test_migrate_image_after_upgrade","status":"ABANDONED","created":"2021-02-18 08:46:38.000000000","updated":"2021-02-18 14:33:51.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"10bc83b453f4a7d4b701d0d1d29fdb5064ede0c7","_number":776406,"virtual_id_number":776406,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-02-18 09:24:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-02-18 08:48:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-18 08:48:01.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-02-18 09:24:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-18 09:25:55.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-18 14:21:14.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"}],"messages":[{"id":"8b3741b18e08ddca6dc28a3e54a469f746ee8d6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-02-18 08:46:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"67b7d927677d4f82c6e93b423ca8ee1b847be556","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-18 08:48:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e1c9801f5436b043d94c7a2dd924bcb90871db2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 09:24:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e17f4375d6cf4184bc7be1d5bdcb0b0d : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61145358e295443db4a9925f8bfaa08b : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e115138b736748ed857e2b97b4009748 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73bd95f85f42474e9575da2c502992ad : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5651d46069b40aea2f1cdb705ae002a : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1624a71f1d04ab68081fe23ee509930 : SUCCESS in 6m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/07691c422646497782f7eff0efc2a0b1 : SUCCESS in 10m 43s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ee2851d5054e49a5b83c8748eb728c35 : SUCCESS in 8m 20s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0c9ba034385948d692d490c30b76a031 : SUCCESS in 6m 42s","accounts_in_message":[],"_revision_number":1},{"id":"79f0c68d9b8e7434dbf07ff0d0b72c5247d32be2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-18 09:25:55.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/776406/776406/1/check-not-vote/ext-glance-zuul/d5c4824 : SUCCESS in 38m 49s","accounts_in_message":[],"_revision_number":1},{"id":"ee1c00272d17b8e46e67b271d1320f5aceb68971","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-18 14:21:14.000000000","message":"Patch Set 1:\n\nTalking with Rajat, I think we can fix this without the skip..release..fix..unskip dance. So let\u0027s hold off on this for a bit.","accounts_in_message":[],"_revision_number":1},{"id":"10bc83b453f4a7d4b701d0d1d29fdb5064ede0c7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-02-18 14:33:51.000000000","message":"Abandoned\n\nIn favor of approach followed here https://review.opendev.org/c/openstack/glance/+/776369","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5499a481d62b2f55dbe8def1e008ae4277bd64c1","revisions":{"5499a481d62b2f55dbe8def1e008ae4277bd64c1":{"kind":"REWORK","_number":1,"created":"2021-02-18 08:46:38.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/06/776406/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/06/776406/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/776406/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/776406/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/776406/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/06/776406/1"}}},"commit":{"parents":[{"commit":"7922092e4a8f8a76d2fb97c89a78b9ef88e2df73","subject":"Merge \"trivial: Fix a typo in devstack plugin.sh\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7922092e4a8f8a76d2fb97c89a78b9ef88e2df73"}]}],"author":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-02-18 08:42:16.000000000","tz":0},"committer":{"name":"whoami-rajat","email":"rajatdhasmana@gmail.com","date":"2021-02-18 08:45:08.000000000","tz":0},"subject":"Skip test test_migrate_image_after_upgrade","message":"Skip test test_migrate_image_after_upgrade\n\nglance_store change[1] causes failure of test_migrate_image_after_upgrade.\nA fix[2] has been proposed but would require glance_store to be released\nto pass, until then skip it.\n\n[1] https://review.opendev.org/c/openstack/glance_store/+/774703\n[2] https://review.opendev.org/c/openstack/glance/+/776369\n\nCloses-Bug: #1916011\nChange-Id: Ie860c63d60b800ecafc4ee655f17c53cc3381c51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5499a481d62b2f55dbe8def1e008ae4277bd64c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5499a481d62b2f55dbe8def1e008ae4277bd64c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
