)]}'
{"id":"openstack%2Fglance~776588","triplet_id":"openstack%2Fglance~master~I06293de08dd3fdfbd60b9a65501d1198f40ff434","project":"openstack/glance","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I06293de08dd3fdfbd60b9a65501d1198f40ff434","subject":"Fail to start if authorization and policy is misconfigured","status":"MERGED","created":"2021-02-19 08:34:51.000000000","updated":"2021-03-02 18:10:38.000000000","submitted":"2021-03-02 18:05:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":26,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"776588-secure-rbac","meta_rev_id":"7dd00dc7112ee9035e96dce745bcf6014758b3e9","_number":776588,"virtual_id_number":776588,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-02 18:05:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2021-03-02 15:08:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-03-02 15:19:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2021-03-02 15:19:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2021-03-02 15:08:32.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-19 11:16:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-19 14:07:39.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2021-02-19 14:44:56.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2021-02-22 19:34:14.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"CC"},{"updated":"2021-02-22 20:06:34.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-01 19:46:02.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"83cc9ad21c575a2cedd76148d5d003ef16bd5ea3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-19 08:34:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a8721d0fa628efaa99e894f5976b1535a298827e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 11:16:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/9fe2c0845d7b4bffa0a4389b6750647f : SUCCESS in 1h 00m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d4946a55191d40cdbabb5bf554d9e850 : FAILURE in 1h 12m 47s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e46302ddd8748048a7ec4ee6e0f1cd2 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7259a1f85249476a9b4acb393eef48bd : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b78abbf4c3246d9b832330b66c063f3 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66d751719f4941abae4c91f812199df1 : SUCCESS in 7m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4873ad18f8f948c999b6dda3084d1b63 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9812e1d7add9467f9166d43e45ad5118 : SUCCESS in 6m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b76655069668430f8beedbecee77158a : SUCCESS in 8m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1a3f9022f817497ebe7f80178f746e04 : SUCCESS in 8m 12s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/63b35ef5d8c04507a2fa61470c6a39f6 : SUCCESS in 5m 09s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2d69061f3ef340f1850ec459c44cd9bd : POST_FAILURE in 51m 33s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3d48ab1dd8254e42a70aa8dda3767dfb : FAILURE in 1h 11m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/71387c6de830402bb567aabbd02f1e57 : SUCCESS in 1h 20m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e705d53e2c724a37a46c9fed23253b02 : SUCCESS in 1h 22m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c7ced0b6a4b420da024567f672d2600 : SUCCESS in 1h 11m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f087c906b89e4826abb90c2a0f746c5b : SUCCESS in 1h 26m 28s","accounts_in_message":[],"_revision_number":1},{"id":"7cccada1bbf668438d2acf88c61c3eff13b978ab","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-19 13:37:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e158ca1c9933adb541a0d056d751bd58db367c64","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-02-19 14:07:39.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nLooks like good starting point, the actual usage logic missing, proposal of it inline.\n\nThe zuul error does not seem related.","accounts_in_message":[],"_revision_number":1},{"id":"0ec567054410fa487fb9750864982c8b12c0f82c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-19 14:19:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cfc55a848639adce34c4cab2946375d90b5f014e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-19 14:22:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"394f66393f384b69cce38de0edadaab2dfd8f50b","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-19 14:44:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0aea43988002582343808af94ea5b916d1c65ef8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 15:13:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/23cb9931ed7c4f7fb38eeb7483d0deb7 : SUCCESS in 51m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e446ba1786e446858119bc5b0e1743dd : FAILURE in 1h 08m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16c2a869e85f428d96ba883057050034 : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14ac63afb4604f5f8bddde8ee708f441 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6807a7b0ffc483696f0045330159c8f : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ec3c37f34d3451ab2be37efaf8c06b0 : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b754cd31d3154ed8bbd54a3aa3f285a4 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf933a432ca04cf1b3bd5ce51aeaa5a2 : SUCCESS in 7m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b0ecaa247ebb49afa29bf300f7ba9401 : SUCCESS in 9m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/89175c80c2534b94a67c97244d8bd626 : SUCCESS in 8m 16s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d656ce0e8e844bca94042f4d0bba5a31 : SUCCESS in 5m 07s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1987b048f6e548e68a3b369bde215891 : SUCCESS in 1h 09m 27s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8b7632e5b3cd4e6eb2636041da640df5 : SUCCESS in 1h 19m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7b5af9ff7a5744f4a95def4ae8f1c92b : SUCCESS in 1h 27m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/70ee46d59b8c4b17a17da0d2a5774613 : SUCCESS in 1h 11m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c7ccd6f0db84433eb907382db81df5e9 : SUCCESS in 59m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7a0b3ab248ad46fcaf35be2f587bea22 : SUCCESS in 1h 09m 57s","accounts_in_message":[],"_revision_number":1},{"id":"5f245781689e0d8b51e30ddfa55c6b5e89ffc9f4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-19 15:59:12.000000000","message":"Patch Set 1:\n\nshould I take push this patch upward somewhere to unblock others;\n\nlike on top of this; https://review.opendev.org/c/openstack/glance/+/767425/7","accounts_in_message":[],"_revision_number":1},{"id":"d3215551505e47bf89fd5fc762a39883286d6f2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-22 19:25:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1d75601a8bdcea77d1eab3cd5692c33a1f2241b0","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-22 19:34:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fdf51fc1fbd9d72b60fa4e43b9dc187782f2061d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-22 20:06:34.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/776588/776588/2/check-not-vote/ext-glance-zuul/e82dce3 : SUCCESS in 39m 40s","accounts_in_message":[],"_revision_number":2},{"id":"a9e684bc31df2c4a2a07f8e2cc6a58d8c9958366","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 21:06:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/957f4c412deb4b9d933c0ce39cee4def : SUCCESS in 1h 04m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bed9bb5d3fcf4ceb81f57324b3caac78 : SUCCESS in 1h 03m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5ebda9ef532c4ceaa84d6f5ec3ac184c : SUCCESS in 6m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cbac7a36c4447058413da4391d16cf3 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe723ee6b1c44056a5cad77b64923c21 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2681d2f9b0d452c8a814cef8594c365 : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4bce86d35be6484f98c848a540183515 : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75c251ec946b4c7fa3c8071ce3c6ad8f : SUCCESS in 9m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c6c8b823740d4701b2daf8dca2d8935b : SUCCESS in 7m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ecc574b616c540ea9266840cd0aec9f9 : SUCCESS in 11m 01s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0500357d9ec44657acf3eb97f50d2df8 : SUCCESS in 5m 28s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9a52c60b51334d5783c6eb3ed224ec3c : SUCCESS in 1h 17m 29s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f89750b748e24350b2f7515d396a71f1 : SUCCESS in 1h 12m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2a696478fda643e28074a6853e58d615 : SUCCESS in 1h 14m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/43c01d655fd146ef8f729f3753f3e6cb : FAILURE in 1h 04m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86459aa73f47494ba89427bc605aae27 : SUCCESS in 1h 03m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/638e3967832d4fd88c21c27ecc4e59b0 : SUCCESS in 1h 14m 41s","accounts_in_message":[],"_revision_number":2},{"id":"633e010cd9d487ec305672e14edb56b1032413fc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-23 10:37:37.000000000","message":"Patch Set 2: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"378c66833a5adfacfca8aec57b1e9e53ba999f5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-23 17:31:15.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"2c1d46a86a66ab32e54e7eec5ab530e62e3247c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-23 22:25:07.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"de6b45afc21fc4cfea1c1859dcfd511a60c01382","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 02:00:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/bb31ed05ca614767b5d2a6b46f4baeaa : SUCCESS in 1h 06m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5499617714f844e1b3331fd36b48fed4 : SUCCESS in 1h 21m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21a36c61b5034ca5a863d0d1ef8a67c1 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9089f11095db493eb41fa81239dbeebc : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a52e2e8304f453983a891e2cb0599f7 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b7e9db4bb994635b5e4d9406f878678 : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b9eb87c81a1499ca0524a2b2306e834 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eef0d75debd94abd9474c4b3797cc6f2 : SUCCESS in 6m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3fa4964d6edc453c9b0ab75135544320 : SUCCESS in 8m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/389d802e74964860a6f2cb9caa92f3e6 : SUCCESS in 9m 30s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/486c9fd688ea43b785fd3c693b7794ff : SUCCESS in 6m 26s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/434f20038553490e8a070b1d126e4b44 : SUCCESS in 1h 11m 04s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fe7d9d869fe14749a15e47e1a02fda82 : SUCCESS in 1h 11m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c99c6868b26a4bdda97a04018b3662f1 : SUCCESS in 1h 09m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f58559718bc34a17bf277a165c8d34ef : SUCCESS in 1h 20m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1489fd15ab240ef83911986b142a621 : SUCCESS in 49m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/501d645c0b244e4db690667f42ad7d08 : SUCCESS in 1h 10m 13s","accounts_in_message":[],"_revision_number":4},{"id":"476865142700e69ce01dcd1978d3301bdcdaced3","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-24 13:09:03.000000000","message":"Topic secure-rbac-handle-members removed","accounts_in_message":[],"_revision_number":4},{"id":"535b224dd3ae5248b9c143e7686bf86d39a66e12","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-24 13:09:41.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":4},{"id":"603b6aa82ba0f4fd7a87b562abd40ea374efa03e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-01 06:36:40.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e50cc5355d5afa9ff0be523519e40f09e1d158ae","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-01 17:29:56.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4f57a457519ee58b3f95b04a5bd55eb9bf279af0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-01 17:46:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6de767d87904e560f3c5d6b71ee3cd9234746bcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-01 19:36:07.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3a396c7819ce53a7ae2c58986ec7437deb26b373","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-01 19:46:02.000000000","message":"Patch Set 5: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"c22eb4bb3b269868beae3e932ffbb0d88cd4f808","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-01 19:50:28.000000000","message":"Patch Set 5:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/776588/776588/5/check-not-vote/ext-glance-zuul/dd80f51 : FAILURE in 13m 17s","accounts_in_message":[],"_revision_number":5},{"id":"81e24741b9d67a1e8aa2f72f360e93c52a704c63","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-01 20:02:22.000000000","message":"Patch Set 5:\n\n(5 comments)\n\nNew patch set on the way. Thanks for the quick review, Dan.","accounts_in_message":[],"_revision_number":5},{"id":"88bf749b03a6a613a4291a6dc233942674511ad0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-01 20:03:24.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"49a5c45cea70a9d69351378c56f74253b61a4e2e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-01 20:06:35.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"cc764af12cca4f7c2662c75c695dffcc884541c7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-01 20:19:59.000000000","message":"Patch Set 6:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/776588/776588/6/check-not-vote/ext-glance-zuul/a2777fc : FAILURE in 15m 51s","accounts_in_message":[],"_revision_number":6},{"id":"01a770aafb33c778e64bd12492584750f168e5c1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-01 21:53:22.000000000","message":"Topic bp/secure-rbac removed","accounts_in_message":[],"_revision_number":6},{"id":"4dc04b738cae8f1ec8bb12af405bbf8adeeb375a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-01 21:53:28.000000000","message":"Topic set to secure-rbac","accounts_in_message":[],"_revision_number":6},{"id":"0c0a09f7ef302fea367499b0f994feab7ef6048b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 22:51:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/aacd1c7f8e6d48698363b97265f03111 : SUCCESS in 57m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/83fbd6278ab9480ea9c8e103c1ed6b14 : SUCCESS in 1h 12m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8f978396d51c4835bd672238f91f14f8 : SUCCESS in 10m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8985295fbfdc43ceb632e4301d61c90f : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/201a134994604d0ab666a79e3869e0e9 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5be80536d474860b102aef300ca1516 : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/02e46c89ab6c413180e223f9783a95b1 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96ddb18061b343c784fe9812c42f3162 : SUCCESS in 6m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/79def1a9f5984de7b9ded2e9b14d8305 : SUCCESS in 12m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f5c36fea983b4706b4a0917c69b6e891 : SUCCESS in 8m 34s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/320cd5cc467c4c9bad54dac491a7727a : SUCCESS in 5m 13s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e68d05ce02574a5bab31e06cfcc63107 : SUCCESS in 1h 11m 53s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b9bd5f16a16f49efb3d600df44ce25a9 : SUCCESS in 1h 20m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4145f8cac008433090e6b7a6a7349b7f : SUCCESS in 1h 15m 38s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/df9cee5ef6c24112a1cfaa15a965e183 : SUCCESS in 1h 08m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60ba4b6220b844c18d5b3c3af3d48240 : SUCCESS in 1h 18m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b7149958bda14d25b1b18cf07a8eff15 : SUCCESS in 1h 09m 36s","accounts_in_message":[],"_revision_number":6},{"id":"f1272f43ee716181ffa50dc106d624314ede51b4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-02 06:51:59.000000000","message":"Patch Set 6: Review-Priority+1\n\n(2 comments)\n\nSome suggestions inline","accounts_in_message":[],"_revision_number":6},{"id":"08f42ff9eca7d7b6ccdb565635a301b4647b4e50","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-02 13:28:56.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b24f19699549a166a36ee80536b43f1df3fcf0ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-02 13:30:04.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3efc001c029cda73aa6af9c7746ac4bfb329edc9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-02 13:56:14.000000000","message":"Patch Set 7: Code-Review+2 Review-Priority+1\n\n(1 comment)\n\nWe can address the comment in Follow up if respin is not required.","accounts_in_message":[],"_revision_number":7},{"id":"6e92d204eed6964b7c70b7104f153ae922336729","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-02 14:12:19.000000000","message":"Patch Set 7:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/776588/776588/7/check-not-vote/ext-glance-zuul/adae62d : FAILURE in 16m 20s","accounts_in_message":[],"_revision_number":7},{"id":"f9697b9e61e5a27bc4017811296bad31fc0d473f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-02 14:54:25.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"73056d101c0af580e408063bf1b183f5d74e9591","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-02 14:54:32.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2f3b1229791861892e0b198f31c6acd2b41c93c4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-02 15:08:32.000000000","message":"Patch Set 8: Code-Review+2 Review-Priority+1\n\nThank you Lance;","accounts_in_message":[],"_revision_number":8},{"id":"3756dc59949db3167d914be00e29cf016f4f80d0","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-02 15:19:20.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"4fa001b9535d475c6d8e07651ca5faebda34d861","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 16:22:34.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/02b76c2a57d744e6b13241d48b0ecec3 : SUCCESS in 1h 19m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d471192bc56849e9b719fd197df9b597 : SUCCESS in 1h 20m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e22f4469eb54befb6d2e6a206a730b3 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/554ac57905084b54ba6dafbbab870771 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec0c1319b83345b6a2d1199412535e0d : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27b26976271547a6ba78e261b130440b : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53a78b6beb734d0eb365196d709808b2 : SUCCESS in 6m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dadb582bf9934008b2d6135be29a0f8a : SUCCESS in 7m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1e2d127e2e84455aba7af3939c00e1eb : SUCCESS in 11m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c789a17b3534492e8aae03ec9c04a818 : SUCCESS in 9m 38s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8e166f63df6b499faaa4fdbd48fb738d : SUCCESS in 6m 32s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c0e1399da614415ba285f1532b77e4dc : SUCCESS in 1h 17m 49s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/937623bcbf87469aac706969fbe3fb32 : SUCCESS in 1h 24m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7ca62733d0634bda827e0cfef655bf64 : SUCCESS in 1h 16m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/61a7539e06b94798a84182f5c3d6c283 : SUCCESS in 1h 21m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9627fa81d86d4bd692304d17e3a68805 : SUCCESS in 1h 00m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f90d1b7a36844f3180b4c1b6c214be0f : SUCCESS in 1h 13m 31s","accounts_in_message":[],"_revision_number":8},{"id":"85e26350fdf6b2f49e3bcd02fa8e58fdaaaa5715","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 16:27:05.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"496d9e4aa199e14a8fed110fa918163c8abd5b01","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-02 17:52:41.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/776588/776588/8/check-not-vote/ext-glance-zuul/356186a : FAILURE in 17m 37s","accounts_in_message":[],"_revision_number":8},{"id":"8fb549052491ad4e6d1c55d4f2cbfdc9c0eda71a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 18:05:06.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/cd2bd46fed4b4e8dad4ef3156d9831db : SUCCESS in 1h 00m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fc0918d359004387a72752b82fdcea8c : SUCCESS in 1h 16m 51s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d33fc92e9a824b3896f4362b467cdf13 : SUCCESS in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84dd36de4c5a492fb14e1ed19b8b14d6 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34235297b7304f1b89381cd9aa7a64c6 : SUCCESS in 7m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d44d387582274b3fba5b262d07bd7130 : SUCCESS in 10m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ba346a498ab4c27b0bab85a66402f81 : SUCCESS in 9m 12s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f7a7c8a0d8f24feba3d6ad855438c067 : SUCCESS in 13m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/72f8e688e5234db3805693cdf56f42c8 : SUCCESS in 10m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/743fab41f04744ffb7e774b7eeba90e7 : SUCCESS in 1h 15m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1c8b8dab03374e68ade24481b1f202e1 : SUCCESS in 1h 15m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/066be3d40e254a9d9284d1e6a04d886e : SUCCESS in 1h 03m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6747952131cd4343b105bd291414b024 : SUCCESS in 1h 12m 45s","accounts_in_message":[],"_revision_number":8},{"id":"fc6c16ad1f215fa4f1d01a90cab376cb264c86bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 18:05:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"7dd00dc7112ee9035e96dce745bcf6014758b3e9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 18:10:38.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/071c66915158469d93127302447c8b4c : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"bf838242ac4ccca4d7341e2d1ee8187e3101d425","revisions":{"dc15d66e7708280201740800139e1ce26489b623":{"kind":"REWORK","_number":1,"created":"2021-02-19 08:34:51.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/88/776588/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/776588/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/776588/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/776588/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/776588/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/776588/1"}}},"commit":{"parents":[{"commit":"ed930ec5123a618fea3a6e6e64bf2d6bfccf7312","subject":"Fix: cinder store test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ed930ec5123a618fea3a6e6e64bf2d6bfccf7312"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-19 05:22:25.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-19 05:22:25.000000000","tz":0},"subject":"New config option to mark Secure RBAC as Experimental","message":"New config option to mark Secure RBAC as Experimental\n\nAdded new config option `enforce_new_policies` default to False which\nwill ensure that Secure RBAC will be treated as EXPERIMENTAL. Once it\nis marked as stable we will swithc the default value of this option to\nTrue and remove this option in following cycle.\n\nRelated: blueprint secure-rbac\nChange-Id: I06293de08dd3fdfbd60b9a65501d1198f40ff434\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc15d66e7708280201740800139e1ce26489b623"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc15d66e7708280201740800139e1ce26489b623"}]},"branch":"refs/heads/master"},"f74f9f7f491072fac1913a7cd4f9f6ef294f422a":{"kind":"REWORK","_number":2,"created":"2021-02-22 19:25:48.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/88/776588/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/776588/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/776588/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/776588/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/776588/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/776588/2"}}},"commit":{"parents":[{"commit":"cf20d89829837027c84746c1d2ffcfde9f8c4b1d","subject":"Pass oslo.context RequestContext objects directly to policy enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cf20d89829837027c84746c1d2ffcfde9f8c4b1d"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-19 05:22:25.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-22 19:20:09.000000000","tz":0},"subject":"Log an experimental warning at startup for secure RBAC","message":"Log an experimental warning at startup for secure RBAC\n\nThis informs operators of glance\u0027s support status for secure RBAC as of\nthe Wallaby release. Eventually, this message will be removed when\nglance adopts more support for secure RBAC personas.\n\nRelated: blueprint secure-rbac\nChange-Id: I06293de08dd3fdfbd60b9a65501d1198f40ff434\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f74f9f7f491072fac1913a7cd4f9f6ef294f422a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f74f9f7f491072fac1913a7cd4f9f6ef294f422a"}]},"branch":"refs/heads/master"},"21091ae15faa9afe9fd68be292ca476b967ed3d0":{"kind":"NO_CHANGE","_number":3,"created":"2021-02-23 17:31:15.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/88/776588/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/776588/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/776588/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/776588/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/776588/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/776588/3"}}},"commit":{"parents":[{"commit":"c25e06ac336335589fa651d863d83960a5b624df","subject":"Pass oslo.context RequestContext objects directly to policy enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c25e06ac336335589fa651d863d83960a5b624df"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-19 05:22:25.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-23 17:29:29.000000000","tz":0},"subject":"Log an experimental warning at startup for secure RBAC","message":"Log an experimental warning at startup for secure RBAC\n\nThis informs operators of glance\u0027s support status for secure RBAC as of\nthe Wallaby release. Eventually, this message will be removed when\nglance adopts more support for secure RBAC personas.\n\nRelated: blueprint secure-rbac\nChange-Id: I06293de08dd3fdfbd60b9a65501d1198f40ff434\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/21091ae15faa9afe9fd68be292ca476b967ed3d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/21091ae15faa9afe9fd68be292ca476b967ed3d0"}]},"branch":"refs/heads/master"},"63ceecdbe702df47d5276d2ec09ea89aa6ca6fa3":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-02-23 22:25:07.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/88/776588/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/776588/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/776588/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/776588/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/776588/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/776588/4"}}},"commit":{"parents":[{"commit":"e6755ebab9eea071cb5c9e4fb09be8fe81911c0c","subject":"Pass oslo.context RequestContext objects directly to policy enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e6755ebab9eea071cb5c9e4fb09be8fe81911c0c"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-19 05:22:25.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-23 22:23:18.000000000","tz":0},"subject":"Log an experimental warning at startup for secure RBAC","message":"Log an experimental warning at startup for secure RBAC\n\nThis informs operators of glance\u0027s support status for secure RBAC as of\nthe Wallaby release. Eventually, this message will be removed when\nglance adopts more support for secure RBAC personas.\n\nRelated: blueprint secure-rbac\nChange-Id: I06293de08dd3fdfbd60b9a65501d1198f40ff434\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/63ceecdbe702df47d5276d2ec09ea89aa6ca6fa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/63ceecdbe702df47d5276d2ec09ea89aa6ca6fa3"}]},"branch":"refs/heads/master"},"09dd275b1a3c27c5143bc3332fcd2dc60d32c968":{"kind":"REWORK","_number":5,"created":"2021-03-01 19:36:07.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/88/776588/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/776588/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/776588/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/776588/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/776588/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/776588/5"}}},"commit":{"parents":[{"commit":"bdbad59dc9605afbb9178d79617ab54e78a9235d","subject":"Merge \"Pass oslo.context RequestContext objects directly to policy enforcement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bdbad59dc9605afbb9178d79617ab54e78a9235d"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-19 05:22:25.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-03-01 19:22:42.000000000","tz":0},"subject":"Fail to start if authorization and policy is misconfigured","message":"Fail to start if authorization and policy is misconfigured\n\nThis informs operators of glance\u0027s support status for secure RBAC as of\nthe Wallaby release. Eventually, this message will be removed when\nglance adopts more support for secure RBAC personas.\n\nThis also forces glance to fail if it\u0027s configured improperly. This is\ndone to explicitly prevent ambiguity with authoritative decisions.\n\nRelated: blueprint secure-rbac\nChange-Id: I06293de08dd3fdfbd60b9a65501d1198f40ff434\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/09dd275b1a3c27c5143bc3332fcd2dc60d32c968"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/09dd275b1a3c27c5143bc3332fcd2dc60d32c968"}]},"branch":"refs/heads/master"},"8ad6cd59bcbcf2ccec767adc6b4fa4f22233024d":{"kind":"REWORK","_number":6,"created":"2021-03-01 20:03:24.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/88/776588/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/776588/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/776588/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/776588/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/776588/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/776588/6"}}},"commit":{"parents":[{"commit":"bdbad59dc9605afbb9178d79617ab54e78a9235d","subject":"Merge \"Pass oslo.context RequestContext objects directly to policy enforcement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bdbad59dc9605afbb9178d79617ab54e78a9235d"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-19 05:22:25.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-03-01 20:00:42.000000000","tz":0},"subject":"Fail to start if authorization and policy is misconfigured","message":"Fail to start if authorization and policy is misconfigured\n\nThis informs operators of glance\u0027s support status for secure RBAC as of\nthe Wallaby release. Eventually, this message will be removed when\nglance adopts more support for secure RBAC personas.\n\nThis also forces glance to fail if it\u0027s configured improperly. This is\ndone to explicitly prevent ambiguity with authoritative decisions.\n\nRelated: blueprint secure-rbac\nChange-Id: I06293de08dd3fdfbd60b9a65501d1198f40ff434\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8ad6cd59bcbcf2ccec767adc6b4fa4f22233024d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8ad6cd59bcbcf2ccec767adc6b4fa4f22233024d"}]},"branch":"refs/heads/master"},"a4335022751066826d9aaf9bc6269ab1b86da747":{"kind":"REWORK","_number":7,"created":"2021-03-02 13:30:04.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/88/776588/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/776588/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/776588/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/776588/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/776588/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/776588/7"}}},"commit":{"parents":[{"commit":"bdbad59dc9605afbb9178d79617ab54e78a9235d","subject":"Merge \"Pass oslo.context RequestContext objects directly to policy enforcement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bdbad59dc9605afbb9178d79617ab54e78a9235d"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-19 05:22:25.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-03-02 13:27:27.000000000","tz":0},"subject":"Fail to start if authorization and policy is misconfigured","message":"Fail to start if authorization and policy is misconfigured\n\nThis informs operators of glance\u0027s support status for secure RBAC as of\nthe Wallaby release. Eventually, this message will be removed when\nglance adopts more support for secure RBAC personas.\n\nThis also forces glance to fail if it\u0027s configured improperly. This is\ndone to explicitly prevent ambiguity with authoritative decisions.\n\nRelated: blueprint secure-rbac\nChange-Id: I06293de08dd3fdfbd60b9a65501d1198f40ff434\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a4335022751066826d9aaf9bc6269ab1b86da747"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a4335022751066826d9aaf9bc6269ab1b86da747"}]},"branch":"refs/heads/master"},"bf838242ac4ccca4d7341e2d1ee8187e3101d425":{"kind":"REWORK","_number":8,"created":"2021-03-02 14:54:25.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/88/776588/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/776588/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/776588/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/776588/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/776588/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/776588/8"}}},"commit":{"parents":[{"commit":"bdbad59dc9605afbb9178d79617ab54e78a9235d","subject":"Merge \"Pass oslo.context RequestContext objects directly to policy enforcement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bdbad59dc9605afbb9178d79617ab54e78a9235d"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-02-19 05:22:25.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-03-02 14:52:21.000000000","tz":0},"subject":"Fail to start if authorization and policy is misconfigured","message":"Fail to start if authorization and policy is misconfigured\n\nThis informs operators of glance\u0027s support status for secure RBAC as of\nthe Wallaby release. Eventually, this message will be removed when\nglance adopts more support for secure RBAC personas.\n\nThis also forces glance to fail if it\u0027s configured improperly. This is\ndone to explicitly prevent ambiguity with authoritative decisions.\n\nRelated: blueprint secure-rbac\nChange-Id: I06293de08dd3fdfbd60b9a65501d1198f40ff434\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bf838242ac4ccca4d7341e2d1ee8187e3101d425"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bf838242ac4ccca4d7341e2d1ee8187e3101d425"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
