)]}'
{"id":"openstack%2Fglance~776693","triplet_id":"openstack%2Fglance~master~I4e87fa0fd58db1ddbb1876f4b3a58497c7906142","project":"openstack/glance","branch":"master","topic":"secure-rbac-handle-members","hashtags":[],"change_id":"I4e87fa0fd58db1ddbb1876f4b3a58497c7906142","subject":"Assert policy check strings are formatted as we expect","status":"ABANDONED","created":"2021-02-19 18:21:17.000000000","updated":"2021-02-22 17:45:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"434751b613ff532aafb49eafd7847167693015d9","_number":776693,"virtual_id_number":776693,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-02-19 21:04:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-19 20:28:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-19 23:01:43.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"32c6f363367a3149b32b4359e3e1279f9cea3c3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-19 18:21:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f338058dcd94f238e90252430ea35672d774c545","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 20:28:05.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b0ecd3a422d40d5957d6a861c14f21e : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2216c83999cd485b84d1abea5dcaebb0 : FAILURE in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a764473d98248c1a0940b55c09da795 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5818c48306cc4235b6d7abe23e37ef72 : SUCCESS in 7m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/648372a753e04ae99f1fa2f85ca71f32 : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0474aa8858fb48a7b0956644d27e7a82 : SUCCESS in 7m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/663c60264cd54a45b965825c32671ab8 : SUCCESS in 8m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/003613b2c2f140128279ae8a32438474 : SUCCESS in 8m 50s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9b42991e159b449bad96d2f51a39c638 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":1},{"id":"2b944d0e832d12c3109ea14f93a859226e8241d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-19 20:48:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5e7dddd8db1b94164e1e9e3a48d1a2030508ca84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 21:04:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a791b6ac803d4430ae4fe54fd142d7df : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe13099d0e1b4904ac9e71c49dcd9c8e : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70796784ad8e4819bab956962518fc4d : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a703d865062f403293ada7a34f0e18a7 : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a09bb824ef04c22a0f00a1a4ccc9247 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c93ec7b305a8487da179a959288ad35e : SUCCESS in 6m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a1de4d90bb5b43b99020d5cd467ac1ba : SUCCESS in 8m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ffcc7be30f7a4340b7cc16c70e4c8098 : SUCCESS in 8m 59s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c878a5881b484f9d848751cdf27397ba : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":2},{"id":"c48c8ddcbf72e7d3d49d9ffb0fd02c4062cb563c","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-19 23:01:43.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/776693/776693/2/check-not-vote/ext-glance-zuul/2d48aae : SUCCESS in 37m 24s","accounts_in_message":[],"_revision_number":2},{"id":"434751b613ff532aafb49eafd7847167693015d9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-22 17:45:05.000000000","message":"Abandoned\n\nRolled this into https://review.opendev.org/c/openstack/glance/+/764241/10","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3d1c4bb84226f2aa62b84eae2640e13b10cbff9a","revisions":{"1f6e77158279ec4f6dcf0c6ed4119cf9c3e984ea":{"kind":"REWORK","_number":1,"created":"2021-02-19 18:21:17.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/93/776693/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/776693/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/776693/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/776693/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/776693/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/776693/1"}}},"commit":{"parents":[{"commit":"d093dd7ffbe872fd3b752295325a087ac502a8ed","subject":"Add basic/common personas to base policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d093dd7ffbe872fd3b752295325a087ac502a8ed"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-19 18:18:57.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-19 18:18:57.000000000","tz":0},"subject":"Assert policy check strings are formatted as we expect","message":"Assert policy check strings are formatted as we expect\n\nThis commit adds a couple of tests to make sure the check strings we\u0027re\nusing to implement authorization are formatted in a way that makes\nsense.\n\nChange-Id: I4e87fa0fd58db1ddbb1876f4b3a58497c7906142\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1f6e77158279ec4f6dcf0c6ed4119cf9c3e984ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1f6e77158279ec4f6dcf0c6ed4119cf9c3e984ea"}]},"branch":"refs/heads/master"},"3d1c4bb84226f2aa62b84eae2640e13b10cbff9a":{"kind":"REWORK","_number":2,"created":"2021-02-19 20:48:48.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/93/776693/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/776693/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/776693/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/776693/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/776693/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/776693/2"}}},"commit":{"parents":[{"commit":"d093dd7ffbe872fd3b752295325a087ac502a8ed","subject":"Add basic/common personas to base policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d093dd7ffbe872fd3b752295325a087ac502a8ed"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-19 18:18:57.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-19 20:47:25.000000000","tz":0},"subject":"Assert policy check strings are formatted as we expect","message":"Assert policy check strings are formatted as we expect\n\nThis commit adds a couple of tests to make sure the check strings we\u0027re\nusing to implement authorization are formatted in a way that makes\nsense.\n\nChange-Id: I4e87fa0fd58db1ddbb1876f4b3a58497c7906142\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3d1c4bb84226f2aa62b84eae2640e13b10cbff9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3d1c4bb84226f2aa62b84eae2640e13b10cbff9a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
