)]}'
{"id":"openstack%2Fglance~776788","triplet_id":"openstack%2Fglance~master~Id3415057d60e1c8a7b81cd4d4003d578206e9082","project":"openstack/glance","branch":"master","topic":"secure-rbac-metadefs","hashtags":[],"change_id":"Id3415057d60e1c8a7b81cd4d4003d578206e9082","subject":"WIP: Implement scope checks for namespace APIs","status":"ABANDONED","created":"2021-02-20 05:09:00.000000000","updated":"2021-11-22 21:21:57.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"12ad4bb776608a6aa7ab3c0e4b33559c74403073","_number":776788,"virtual_id_number":776788,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-20 06:21:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-20 05:50:57.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-20 06:21:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ca759afa86780cbf198e03cc3c76a40c06797958","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-02-20 05:09:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f3ec1aef24b86be1ba42b20689522d2a9eb7cbf7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-20 05:50:57.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/776788/776788/1/check-not-vote/ext-glance-zuul/e40c786 : SUCCESS in 38m 17s","accounts_in_message":[],"_revision_number":1},{"id":"abe2feb14e30eeda42c73a76f513214672c7de3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-20 06:21:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/8a7fb1a9e9314b249bfee237f3eeee68 : SUCCESS in 50m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/63b534f728dd425b9d6811e25a72fe8d : SUCCESS in 59m 05s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b49737e660f4455bcc8c22b650c8700 : FAILURE in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b11e0ab4754f445a9b04f5b79d114ac1 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3716bdc7bf884f94949c1df7e0246d71 : FAILURE in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3be3e91c8294eb98b730055d7a88c67 : FAILURE in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6c85322ae1748a48bcbbf1903a4d87b : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cacfa81e6f04105b8c5c57ad8e61831 : SUCCESS in 6m 04s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8de382b6246843c6bfefb34434830c2f : SUCCESS in 8m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/74e6c864b91a4558848436ba546539f7 : SUCCESS in 9m 05s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e4b40898de0d4b87bd14b175296e4a47 : SUCCESS in 4m 41s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a0275248087a4a95b7dd132f55829f84 : SUCCESS in 1h 09m 10s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9d264252ac634950b8116648fff397a1 : SUCCESS in 1h 02m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0401a037aaca4503bbdb29b027acbf6e : SUCCESS in 1h 03m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/590af0d0cc284e5cbc8dc99557c6927a : SUCCESS in 1h 11m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aa200ecf26664faeb3f494932d95b5c5 : SUCCESS in 54m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6543a0e446ac47bdbaa3e2f831753048 : SUCCESS in 1h 10m 22s","accounts_in_message":[],"_revision_number":1},{"id":"12ad4bb776608a6aa7ab3c0e4b33559c74403073","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-11-22 21:21:57.000000000","message":"Abandoned\n\nI think this was superseded by https://review.opendev.org/q/topic:%2522secure-rbac%2522+(status:open+OR+status:merged)+project:openstack/glance","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8415d49924b3b60a75160c121d9550f5594907aa","revisions":{"8415d49924b3b60a75160c121d9550f5594907aa":{"kind":"REWORK","_number":1,"created":"2021-02-20 05:09:00.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/88/776788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/776788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/776788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/776788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/776788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/776788/1"}}},"commit":{"parents":[{"commit":"d764496192fd581d980285e40e82c4bfbb479b70","subject":"Implement secure RBAC for metadef APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d764496192fd581d980285e40e82c4bfbb479b70"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-20 05:07:19.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-02-20 05:07:19.000000000","tz":0},"subject":"WIP: Implement scope checks for namespace APIs","message":"WIP: Implement scope checks for namespace APIs\n\nChange-Id: Id3415057d60e1c8a7b81cd4d4003d578206e9082\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8415d49924b3b60a75160c121d9550f5594907aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8415d49924b3b60a75160c121d9550f5594907aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
