)]}'
{"id":"openstack%2Fglance~777012","triplet_id":"openstack%2Fglance~master~Ib80e9cfb58680f9e8ead5993dc206f4da882dd09","project":"openstack/glance","branch":"master","topic":"bug/1913625","hashtags":[],"change_id":"Ib80e9cfb58680f9e8ead5993dc206f4da882dd09","subject":"Add housekeeping module and staging cleaner","status":"MERGED","created":"2021-02-22 21:23:24.000000000","updated":"2021-03-09 01:16:04.000000000","submitted":"2021-03-09 01:14:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":34,"unresolved_comment_count":12,"has_review_started":true,"submission_id":"777012-bug/1913625","meta_rev_id":"ac05cee603aaacdbeab1ef0d0c910ec6f7cf13be","_number":777012,"virtual_id_number":777012,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-09 01:14:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":1,"date":"2021-03-04 12:57:40.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2021-03-04 05:43:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2021-03-08 07:43:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2021-03-04 05:43:15.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-22 22:00:02.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2021-02-22 22:00:02.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-02-22 22:00:02.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-02-22 22:58:21.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-22 23:22:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-03 01:13:05.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2021-03-04 16:38:44.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"c241a612516a38067b7fa7b47879ba014c5ddf0a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-22 21:23:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"751a16665e21366ae0f1cb5565ac4fac098c42ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-22 21:57:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dd8a34a26d2128d0d7b4a686ed6fdd361cb227bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-22 22:02:34.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"6c331086d68b0009e4f0126a0e4316966cc8c3ad","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-22 22:58:21.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/777012/777012/3/check-not-vote/ext-glance-zuul/87e32ad : SUCCESS in 55m 16s","accounts_in_message":[],"_revision_number":3},{"id":"74fdd9a1c816731486413b421d426fe709f1174e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 23:22:37.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/ed360afc669548fc858436afe5b9a868 : SUCCESS in 1h 10m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/762ccf7cb28c453e96a1df23d4eab2d9 : SUCCESS in 1h 09m 48s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a49b1adfd2db4ecea175131db27a262a : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27edcd05ea654d85b5d44990ddd0500a : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6767edd5733420e9d5ed322477746f3 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3feba38eb3ed4646b00bacc025263c9f : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9104844e67e49bea166ab3290107c6d : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6054ab49c8974a7faad17e29b0be38e8 : SUCCESS in 6m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6bf01c32db0f4031930dc9e8d8de9388 : SUCCESS in 8m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ceded8868c244443abd6fc21098c82a2 : SUCCESS in 7m 56s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/691a0d5348e14b0e9ad6498569069974 : SUCCESS in 5m 10s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1304f27681f64c3ea027c6fc6b754bf3 : SUCCESS in 1h 04m 27s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5847aabe4f304ccc833d5e18427b047a : SUCCESS in 1h 04m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4497e7bfb8174ee5bfaeef12e00a0c42 : SUCCESS in 1h 12m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/26bad32f716442c69df1dfafd7ca957e : SUCCESS in 1h 14m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/001cb91ceaa4412b86651831ea39c180 : SUCCESS in 1h 17m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9344b5ec86104fb1bfeee18a97e787da : SUCCESS in 1h 01m 30s","accounts_in_message":[],"_revision_number":3},{"id":"a2ff52cc49c819c8473387df474b6bf237bb7d9f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-23 11:57:40.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nApologies if missing something.","accounts_in_message":[],"_revision_number":3},{"id":"109c4180bf9fccfd83f321c96de158f6d418285d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-23 14:29:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ab05f16b61e9c6fa028a7b1674d570b907b28852","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-23 14:59:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ac5471adece12c24d2768d90261ea46d6da81ef2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-23 15:40:02.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/777012/777012/4/check-not-vote/ext-glance-zuul/ff6d0b8 : SUCCESS in 39m 57s","accounts_in_message":[],"_revision_number":4},{"id":"e72ac3ca632765081a5d712fed809f5eaa1d91ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-23 16:33:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"7f1b5ed847241b05a38467d17973f480b109d6f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-23 16:54:26.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"02378622982c835fa5dce6206719431abb25e1d7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-23 18:51:38.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/777012/777012/6/check-not-vote/ext-glance-zuul/0c45467 : SUCCESS in 37m 31s","accounts_in_message":[],"_revision_number":6},{"id":"bad9ea45eb978ab44875bb4fcc487eff30a48617","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 22:31:30.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/ae54e6d05a22446c8da94eba4b270847 : SUCCESS in 1h 06m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/76fb49cd1daf43f1a21bfb146f339572 : SUCCESS in 1h 13m 54s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7428ddefd228400b94d7c845a79eb4bd : SUCCESS in 9m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5852aa110474a32b7041fc4337fe829 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/435fa691fb3746a792449e831536835b : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73a609776dad4b97b25128dc71432f90 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2cb642d32584686ae9e3953c4a4f32d : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/791de99a225448c5acc2e6cfb28345a1 : SUCCESS in 7m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f25e81a440bf44c8ac2947f2ad29614e : FAILURE in 9m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/aa557457fad248368df79e1f9725c96e : FAILURE in 8m 01s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f6cba7d6d2924d83955582c47429e733 : SUCCESS in 7m 49s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/52166848b83b4147b25cdc884240cb70 : SUCCESS in 1h 21m 49s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5c67dc983da744e38e656482d268ffb2 : SUCCESS in 1h 15m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/80be60f671234240a5b30a6cee5b27a4 : SUCCESS in 1h 15m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7c4f041625874d9790fe84162575aae0 : FAILURE in 1h 10m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30afddf90dec4cd0b38abe2696d810c9 : SUCCESS in 58m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1a7beb7d078240379ee1d4f81c083a4f : SUCCESS in 1h 14m 45s","accounts_in_message":[],"_revision_number":6},{"id":"4b847df7804e3e603b9b0585d5649c2709badf44","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-23 23:23:24.000000000","message":"Patch Set 6:\n\nThis does not fail locally. I think the problem is that when single-store is in use, the test workers all share the same /tmp/staging area. If two tests that stage an image run concurrently, one will clean up the images that it doesn\u0027t know about from the other worker. AFAICT, the functional stuff doesn\u0027t let us override that conf option, so I\u0027ll have to extend it so that we can. The multistore tests use self.test_dir for staging, so they\u0027re isolated.","accounts_in_message":[],"_revision_number":6},{"id":"8bf82b2871051e8936aed453d42d58410f2d7230","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-23 23:30:10.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"514ade6710838dc67cbcd113f9ccea39c47d13b7","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-24 00:08:50.000000000","message":"Patch Set 7:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/777012/777012/7/check-not-vote/ext-glance-zuul/9757ba5 : SUCCESS in 37m 02s","accounts_in_message":[],"_revision_number":7},{"id":"63df882d3dcb9cf07d7d36130268d40682aaf9fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 02:22:33.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/bc37275e993f407090329754066155ac : SUCCESS in 1h 20m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3ec60417fcfa427a902e831acb0b3eae : SUCCESS in 1h 15m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed89c54e507b478380d9ed8ca3ccb830 : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da09dbe5738349bba935e78848bd9565 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf0976d1bb844510bb1bb4745e4f35c5 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db4be4b26a4b4f8bb2f3982dc8e4f30b : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0a1fd8d2f7c4427b1abba512967b6d5 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5ca0c47d4874867ad4aefa66be0eff8 : SUCCESS in 6m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d943740f4b3648d4be144a686f9662e5 : FAILURE in 7m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b590f5ca8e8b4d4da9d5a1461c66ca3e : FAILURE in 9m 04s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9febd510a4624b849ca8696bb1f3c6d6 : SUCCESS in 8m 05s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8dba68fa1ba94283bf6364221c22fc3c : SUCCESS in 1h 19m 02s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bd09bfdd4e40451b8d698039725bd53a : FAILURE in 1h 08m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b7bb83a1486a478f9a2fa1eaabc7a468 : SUCCESS in 1h 30m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f51245c4091743a6a790c2e54d3f9b05 : SUCCESS in 1h 27m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/effdb6cf4a6742a9a8a2554fb8de41a7 : SUCCESS in 58m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dcb165d80a1d481f9eb9925b19d51215 : SUCCESS in 1h 24m 08s","accounts_in_message":[],"_revision_number":7},{"id":"cbd3c58565cf2ebee36323af579d3338be51bc2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-24 03:06:11.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"0e1d3f09cb2f1ba2669b691621c508752b8d1ef6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 04:31:43.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/7a6131be50a84c59a186fbaebd66940b : SUCCESS in 52m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f7512afc9b1c4b66958a6a913089c7a5 : SUCCESS in 1h 22m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aed00dfd079f41c5b8ced45936c9ae9d : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f6ebdb99e9f4fdc9a00e02af1b2c50b : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f66b46072c8a47199c1748c63e092c6d : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab018ffa305c4f4281657802935abd2c : SUCCESS in 7m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2500c810e00445febb1042f79ee58b20 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efc6a2c15b2b484cb23576f2657cf484 : SUCCESS in 6m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6577253e13ea4bdc82144e7abfcf3370 : SUCCESS in 9m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/01406ede789f499daf78708519340246 : SUCCESS in 8m 26s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/30c662881b544f6a80120a89a34eb1be : SUCCESS in 5m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0f5ec669c440457eaec758704a9d6bcd : SUCCESS in 1h 19m 55s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/daa0e50327ee4ea7a1916dada9beedcc : SUCCESS in 1h 18m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f2c3e6a8c214435e90afad95beab0e81 : SUCCESS in 1h 18m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b918b699cacb4a7db089663b2deb1d14 : SUCCESS in 1h 20m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/573a36a685f74655bcf8899723a3ca0a : SUCCESS in 1h 04m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0e24dd23e282408aad764db49c1f4904 : SUCCESS in 1h 05m 59s","accounts_in_message":[],"_revision_number":8},{"id":"cd9d95898763499fcfb294e1b735dc7e31427e59","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-24 04:33:14.000000000","message":"Patch Set 8:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/777012/777012/8/check-not-vote/ext-glance-zuul/5cb2f72 : FAILURE in 16m 02s","accounts_in_message":[],"_revision_number":8},{"id":"5c9a2320f6175270df47596b63513109989daaf5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-24 16:01:59.000000000","message":"Patch Set 8: Code-Review+2 Review-Priority+1\n\nThank you Dan, looks good to me!!!","accounts_in_message":[],"_revision_number":8},{"id":"678de183dcb372e4c1ac67269757f425ac11901e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-24 17:32:51.000000000","message":"Uploaded patch set 9: New patch set was added with same tree, parent, and commit message as Patch Set 8.","accounts_in_message":[],"_revision_number":9},{"id":"6b3c5807da900788eb7a75962dcf8a198b63c06a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 01:11:35.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/483a68c4005147e3a8724808f906fe6c : SUCCESS in 1h 09m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f811f29c2259445593a774c76d30c9bf : FAILURE in 1h 15m 40s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8a55dd780cb4af49bf46062330e8dd9 : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/526f2bb515874bd4b50d70dff866c168 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ae1141a08f2947139f06efa12082beb0 : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a7baeb51d224e1c9d9a66fe4be78937 : SUCCESS in 7m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7e8a732d7214d12b6138f84bc8244b7 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b510e28824ef4bedb22b728f88118fd1 : SUCCESS in 7m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/32ce9db2fb4346a3a1f46267d55853cc : TIMED_OUT in 31m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f11a7202b8d04b26aa3e079a36578135 : SUCCESS in 9m 37s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/83155addbbd64b368eea8185a5332b31 : SUCCESS in 6m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b060428a335942a197b52e7d7c1d683b : SUCCESS in 1h 36m 43s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c48f4aca77e744379de29a6c4320bbd0 : SUCCESS in 1h 17m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4d6d43dba1a44974acf7da86ef78d27d : SUCCESS in 1h 10m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/aa1cc9ef139748f79bf6582d633c526e : SUCCESS in 1h 43m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/837a52ab8aa8417b93241fecd1fb61d9 : SUCCESS in 1h 02m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e42fad01dcb64d0e90cac248ab6a9cc2 : SUCCESS in 1h 16m 10s","accounts_in_message":[],"_revision_number":9},{"id":"9fdbdbf9feb153d53b8c9c39a3ea42d07c1b34c0","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-25 02:00:11.000000000","message":"Patch Set 9:\n\nrecheck cinder fail","accounts_in_message":[],"_revision_number":9},{"id":"6b9f2b0c02b1cfd7ed1dfd46cb10596fd296f9fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 04:19:24.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/e38f1274e37f4e1781aa8b2497c05d38 : SUCCESS in 1h 07m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bf5584d4bfd94d3c89548f78124d66ff : SUCCESS in 1h 13m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9befdb7e06a844af95cd046ef53ee001 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21af175b2a884c94b8265b517d29c2d7 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/69afb827691042668063127718b8aeef : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3ddd2944f83048ed95a3579efe0f7c23 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbb05a4ed1e848119cf05063762886e1 : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8648744de827462db5b9352730e29e57 : SUCCESS in 6m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9d5502a612a94c82b7136c92c8402522 : SUCCESS in 9m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/abd0acd1c61c4f0aa987fde036e7f57d : SUCCESS in 8m 23s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/5629069eb9f04c71afaea0fd89afe495 : SUCCESS in 4m 44s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f781d5ad52aa4d0eadf11db62928a39a : SUCCESS in 1h 22m 41s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/084d9e0af17248d9bdf6119a5918347e : SUCCESS in 1h 08m 24s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b93c7cd2042946f3ac9fe25a278694d9 : SUCCESS in 1h 12m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a7a68a95a01147338b40b7a06c311931 : SUCCESS in 1h 10m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/38bbe29aba50404db2f4b2ef66293a98 : SUCCESS in 1h 00m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f7950c47eb324e609091abb29fbd6a7f : SUCCESS in 1h 09m 38s","accounts_in_message":[],"_revision_number":9},{"id":"2702e0dde32917620c3dd42a30ffc4e414a12b0e","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-03-02 14:13:11.000000000","message":"Patch Set 9: Code-Review-1\n\n(7 comments)\n\nSome comments inline.","accounts_in_message":[],"_revision_number":9},{"id":"51a88b50b636e6dd465f7e1f7f31e47efbef5c0c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-02 19:49:39.000000000","message":"Patch Set 9:\n\n(7 comments)","accounts_in_message":[],"_revision_number":9},{"id":"182167342b1225cdfa5aa3ba474de759c5545ebd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-02 19:51:08.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"b4f45219368921826b6ab39eabec81df73dafb5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 21:31:27.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/e9c94ccb120f44c7a4877b77e4386653 : SUCCESS in 59m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/555181a5d5ee4c15b555b9324d10889a : SUCCESS in 1h 30m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27cfd7d6a5a540a5b4281729f8b63430 : SUCCESS in 8m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b5aa6f9f3dc48dba956e577cfd3adc5 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0822dfffba642dcb7041ca19c0b088f : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76cecb54dcea4e5088156c76d647f0ce : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d8086bb77b4445f947efc38af5d5171 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02970319bdb449ababda6685f251948f : SUCCESS in 8m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6cffb9b78eca498cadcd1ee90e658ec8 : SUCCESS in 9m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a0af95655c484c55959b48370cbb55d5 : SUCCESS in 8m 20s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9ef41ff073604939b4e3220a87ae5903 : SUCCESS in 5m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/810461fb0b154ecfaed074a75ce58831 : SUCCESS in 1h 14m 19s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/52258c6b06ae49ca981819f33ea97a0c : FAILURE in 1h 01m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/29edcd4a1da14545887a5e34e848e1b9 : SUCCESS in 1h 09m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6d4997b455ba46ffbe8166782788da8e : SUCCESS in 1h 16m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee4332ad195d4c5d9b7f57fd2080417a : SUCCESS in 1h 01m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/73df356779b54aac9630c0a6a8378bdc : SUCCESS in 1h 36m 44s","accounts_in_message":[],"_revision_number":10},{"id":"9a6ad0a0f89db0b6e286af020d23cb480489f5c9","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-02 21:47:51.000000000","message":"Patch Set 10:\n\nrecheck cinder dance","accounts_in_message":[],"_revision_number":10},{"id":"83c9b0465bcdb4843cf494319c2e528f4181da8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 23:10:12.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/6e3520ef99864bb1af5e0b56b5b661d0 : SUCCESS in 49m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0a018370d3514feea6cf037bebf71582 : SUCCESS in 1h 05m 03s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b53894dea7d43fba1e0b2579e1fcc58 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff766c1b1edb469cb86f9c2d4274851e : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff35fcb427c34ad19ac052528aa74613 : SUCCESS in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88bc9da164d742c9a3102b888bb9fc42 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/433edb17339a495bbab0a24eeaac665a : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c797583de77d460d8060c043b0ca87d4 : SUCCESS in 6m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0a2b47ed5158446495a7635a449d7fdb : SUCCESS in 8m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fc6c31775ca749359a83dcd53281758e : SUCCESS in 7m 38s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/74cb763674274963ad7062811679b757 : SUCCESS in 7m 42s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cbbf7b87cb934d48a068befbefea9cc4 : SUCCESS in 1h 03m 54s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e52f6f79c761462aa3057da84b284166 : SUCCESS in 1h 15m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/23b9012f7a5c4afca1b8f48ad1d348ed : SUCCESS in 1h 04m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/14dadc014c26452cb81356df6efb41d9 : SUCCESS in 1h 17m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58c8645a0c664785bbfd57d9c094e461 : SUCCESS in 52m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cf5ce677de3347c6a8a8f4e293d1130e : SUCCESS in 1h 18m 34s","accounts_in_message":[],"_revision_number":10},{"id":"9e644fdb2f139dc02b103344446a8c4906d40dab","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-03-03 01:13:05.000000000","message":"Patch Set 10:\n\n(2 comments)\n\nSome minor details inlined. I like the extensive testing and the clean design which will make this patch as easy to backport as possible, thanks a lot :)","accounts_in_message":[],"_revision_number":10},{"id":"3fe473da80b0cba5033e8860cb5ea1c1b659d423","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-03 01:30:59.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"a2b7fc807c4dbd21a7e3dd6aa3c2be567ea3bf04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-03 01:36:06.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"b38f656ef97d26da06f2d4740d6110e93de40a31","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-03 01:51:28.000000000","message":"Patch Set 11:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/777012/777012/11/check-not-vote/ext-glance-zuul/2c92d59 : FAILURE in 14m 54s","accounts_in_message":[],"_revision_number":11},{"id":"8eaac46c2edb30ee30e8aec0b022d274dd30468c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 03:03:56.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/d33b4ca453a543ae87affdaf82d46984 : SUCCESS in 1h 04m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/90d633d0a2064023bb59243d5413e430 : SUCCESS in 1h 21m 49s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da202a1237ee44c78581fca63ab9891a : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f409bc4bad5f484bbed70200ed796198 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0430c138f61a4e4b992a5a8e78f7b957 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab687c3925b243f9a7b524c8b4ee98f5 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/afa67c4dda804631a5378b8a066d774c : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c52159903ac4804b16a2687c08ed51e : SUCCESS in 6m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7227b66cdd504052971effb0c64e1c84 : SUCCESS in 9m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f82714a3ac65433c9c94dfac17e3cff9 : SUCCESS in 7m 53s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/4c57085c472c43da951ea338c8999270 : SUCCESS in 4m 40s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6436affd7a4e4af78bb82b484f17121e : SUCCESS in 1h 18m 20s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/97e232707f5b47928fde01054a5a0251 : SUCCESS in 1h 16m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6715d58bc30747f7be33f3a267bc6a59 : FAILURE in 1h 08m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/02b24b99d5414ba5a13fc9c046c8af7c : SUCCESS in 1h 21m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/564540b0d5894661be91233b4ff29a56 : SUCCESS in 1h 07m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f8fcfcd0683f48e89115fd63daaa1532 : SUCCESS in 1h 12m 12s","accounts_in_message":[],"_revision_number":11},{"id":"e8a0549a1590db21969503d712b04daeada00a06","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-03 03:08:15.000000000","message":"Patch Set 11:\n\nrecheck cinder dance","accounts_in_message":[],"_revision_number":11},{"id":"4f780d35f87d6d6565bbe417f65bbfdc6abf2b6e","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-03 03:23:02.000000000","message":"Patch Set 11:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/777012/777012/11/check-not-vote/ext-glance-zuul/5a42f8b : FAILURE in 14m 19s","accounts_in_message":[],"_revision_number":11},{"id":"2cd39ab731c86d50b22b21c01115e685b822e330","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 04:32:27.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/d7708d39a7594614875f2ff9c87cfdd1 : SUCCESS in 1h 10m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5ec961210bcd4eada7060ce48273d02a : SUCCESS in 1h 10m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2afcec00632345e6b718538fdc921031 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f581ab0fc72f4cb09525ac4faf2e70a1 : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43db896992bf4a5daf05e26cdc9e7bf8 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f6b8378f8fa4eb9be303d7e943f5817 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b78d6d0678f040908270c60b4559ee9e : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2777fb73d55643c49b655d9d41aecd3c : SUCCESS in 6m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0b2d9b86c9e24923846de5d934c07f02 : SUCCESS in 7m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6a9052fb8f014d989edd3a5398a16751 : SUCCESS in 13m 29s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/23b97656b6ba4d6a8ee05f2f6f5e7a12 : SUCCESS in 5m 10s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/36b0f7389c324b73815e26207a936289 : SUCCESS in 1h 17m 47s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9c0a9946fe014d72b70ce97be057d9c6 : SUCCESS in 1h 11m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/11ac9ab20fc943e0a7181e25279ed613 : SUCCESS in 1h 18m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/834c8bace9c74eb5ab6c758a57975fc0 : FAILURE in 1h 10m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b920be22e7b849e389671a0889da0468 : SUCCESS in 1h 01m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/776217a5300a4dec8f43d78f3e4ebd82 : SUCCESS in 1h 10m 31s","accounts_in_message":[],"_revision_number":11},{"id":"2eea7677fc184000bef04454e31f5ceb313485d1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-03 05:28:06.000000000","message":"Patch Set 11: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"8bcf7f15da6ac081f348d993a9edec6612387282","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-03-03 14:17:26.000000000","message":"Patch Set 11: Code-Review-1\n\n(5 comments)\n\n-1 carrying over from PS9, we\u0027re still ignoring the files created by conversion and decompression plugins.","accounts_in_message":[],"_revision_number":11},{"id":"f68c3317dde1f313c63336d3e7d124d18cdd7588","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-03 14:52:37.000000000","message":"Patch Set 11:\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"6bbea6d76dcae496f00c2ed91002dee42da0772c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-03 16:01:38.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"6688bdb9732c1ed60379a774f6ed4e7954e10b94","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-03 16:19:42.000000000","message":"Patch Set 12:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/777012/777012/12/check-not-vote/ext-glance-zuul/08256d9 : FAILURE in 16m 31s","accounts_in_message":[],"_revision_number":12},{"id":"9c3faa2501b9a76c13ec73f585c81700c260fe58","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-03 17:37:07.000000000","message":"Patch Set 12: Code-Review+2\n\n(1 comment)\n\nLooks good me, suggestion inline if sounds reasonable.","accounts_in_message":[],"_revision_number":12},{"id":"d24bb787a7ad9d322e785aedeb47de7557341faf","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-03 17:43:40.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"a496ae4b53f636cffc673f3bba7374011e18262b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-03 18:00:38.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"5600feff3c5d9b0866a3c3d9483f8adcd02f39a2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-03 18:12:23.000000000","message":"Patch Set 13: Code-Review+2\n\nPerfect, thank you!!!","accounts_in_message":[],"_revision_number":13},{"id":"a1071383f3b3de34f496cf35653077c3dc0814f3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-03 18:17:11.000000000","message":"Patch Set 13:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/777012/777012/13/check-not-vote/ext-glance-zuul/90c443e : FAILURE in 15m 22s","accounts_in_message":[],"_revision_number":13},{"id":"e430a117f4d5f2ef547ab3ec6dfc11821d045825","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-03 22:43:28.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"32e9795b219aa48998d0893a98a53b94b0ac67f9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-03 22:59:53.000000000","message":"Patch Set 14:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/777012/777012/14/check-not-vote/ext-glance-zuul/f928762 : FAILURE in 15m 23s","accounts_in_message":[],"_revision_number":14},{"id":"cea4dad2c4ab6b7dcde9a672c945d01822c76f4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 00:32:01.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/5f74a20aeacd42e5a6629b4c6d360c53 : SUCCESS in 1h 04m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eacd03c7d9264d53b4049b172610ed51 : SUCCESS in 1h 13m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e0254ce268b4e82803ec5e6c75a36d4 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3dc4b87140a4fdfa66ce234ffe616a5 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9856ff9950e42cfa8980d098674eb1f : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa53a610b6bc4a8c8095496a5f822942 : SUCCESS in 11m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/104cfcab413b4784893c7d93fbd002bc : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19c425290249473fa5b0ebc2797264a8 : SUCCESS in 8m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/864778ccee8b429ab9890701ef8fa624 : SUCCESS in 11m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/07ddd97f30e34d38829c81a6cb75b065 : SUCCESS in 12m 59s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/1271a7af816f48cca14cd2854fe5bf06 : SUCCESS in 6m 04s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3e3a90044199448eb94b34b08efc31bc : SUCCESS in 1h 24m 51s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5d98771c6fe546e9837111bf352fdd61 : SUCCESS in 1h 18m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b3ca96eb71d847bb9da5aacfb69b99e6 : SUCCESS in 1h 29m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a7a024d9f882489896d395423e551472 : SUCCESS in 1h 15m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac7933edff0f4549942d11c4284c1013 : SUCCESS in 1h 00m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ef01251114e84de49c820c26f06ebe1a : SUCCESS in 1h 16m 10s","accounts_in_message":[],"_revision_number":14},{"id":"578c55ff29eb9142ed0babe13ebd6c0754e38bf8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-04 05:43:15.000000000","message":"Patch Set 14: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":14},{"id":"267604e6ac1af3c9e436a01558fce49f76d10bca","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-03-04 12:57:40.000000000","message":"Patch Set 14: Code-Review+1\n\n(1 comment)\n\nOne nit if anything else needed. Otherwise looks to be fine","accounts_in_message":[],"_revision_number":14},{"id":"dbf9508048e6a1e970cb87dc7f1768d60b7db26a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-04 14:21:29.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"0c97a1f246e6982fa75f6f8560551b85202b2ae3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-08 07:43:30.000000000","message":"Patch Set 14: Workflow+1\n\nConsidering +1 from Erno as a go ahead for approval. Thank you Dan for working on this!!","accounts_in_message":[],"_revision_number":14},{"id":"c7fac6187a2738cb44dac3d0544452f912089f88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 07:43:50.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"693409430fede4f4b2ebaa65ef5f2798d5420ec7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 09:25:04.000000000","message":"Patch Set 14: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/663138298e0444ab96bc7e7994c690fe : SUCCESS in 56m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/87da94f5f85540c39a291fdd1016cb8f : SUCCESS in 1h 10m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fee26449ecf14337b22e3f515fc5903a : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f8ad04ef9b6411991d55f367700aa21 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c37e6ae8eea4c8f99ff4937eee330aa : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92eea90f1900431fa1944ca58f3c7c04 : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c526c9b305a94f148f1f463189c997bf : SUCCESS in 6m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/678c61ebf1cf47a7874f74c6a25811c6 : TIMED_OUT in 31m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6eb0d64818f744f8867a9d72688fe84c : FAILURE in 8m 51s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4683442026384879a2b596f0fb2ceff5 : SUCCESS in 1h 13m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1fe82db0247d4016a6d7c6d032c901b4 : SUCCESS in 57m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fbf7041700fc44278439cb03577264ea : SUCCESS in 57m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ea3c04f17fb94b35a6355b3ff258c9f4 : SUCCESS in 1h 21m 04s\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f3d88c817e514ceab2c1e823a1fa4e94 : SUCCESS in 31m 02s","accounts_in_message":[],"_revision_number":14},{"id":"1bc6ce0e314474ff1a5b0a5e8d41d6758a3f46c9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-08 09:27:53.000000000","message":"Patch Set 14:\n\nrecheck unrelated failure","accounts_in_message":[],"_revision_number":14},{"id":"1dc2dfe10a0cdd222ae0cba384d2c059e3da20f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 11:12:20.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/ab8bd7eeaebb4a2aae4399aa8a9f5390 : SUCCESS in 1h 00m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6aaa60ee2a1d40b4a9cd1f083651a782 : SUCCESS in 1h 10m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/17966f0547ae4d7ea92ca307c8c7f283 : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/006832d3b49143c899894ddeb3457346 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0197c6f953d3412492c029d2e918783e : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6a12c4fd8dd46feaa572339e5d5d602 : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/489be3e753b54e5c9d8e38ef3613a5d1 : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1454d3c333f498c85a6dff2457a27bd : SUCCESS in 9m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ac0a6a8c44aa4811a63618fe345819d9 : SUCCESS in 10m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7eac52b06ca347bca32604ad9c8b7ee1 : SUCCESS in 9m 42s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/5c8cee7e173d4825adf051cd2ab54eb7 : SUCCESS in 5m 39s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c7dc9f4393824f4fa53255d0a1a8e1a5 : FAILURE in 1h 13m 31s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ab9c38077c064e8e9c367798c5e6e426 : SUCCESS in 1h 14m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7167acdf0195454cb7134748f9a367ac : SUCCESS in 1h 15m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/472f12badeff478cb43b5fed293039cd : SUCCESS in 1h 27m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0aa6ca1831147b2845a1b26f950c1e7 : SUCCESS in 51m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/91437d5efa614d8984a96d2c8f2cd16c : SUCCESS in 1h 23m 00s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e67f6d7ab3f14526aa3431782a847164 : FAILURE in 40m 54s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/30ddbcc21da849eb9d7623c9d93722e3 : SUCCESS in 39m 43s","accounts_in_message":[],"_revision_number":14},{"id":"4243d3c0630de08d6c7982b900f30d0b2adf76fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 11:13:32.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"823c8f065df24287e6a20a94f686921cc1e536e6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-08 11:41:11.000000000","message":"Patch Set 14:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/777012/777012/14/check-not-vote/ext-glance-zuul/3177bef : SUCCESS in 39m 18s","accounts_in_message":[],"_revision_number":14},{"id":"446563b0588109fa0c2f49a251b6d07190148297","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 17:44:45.000000000","message":"Patch Set 14: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/4281b917dbce44eab295c0ba2132fda3 : SUCCESS in 1h 02m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/11028cf1efbe40c1ab769415bfb6aebc : SUCCESS in 1h 05m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f92b388c1b14c6a8b7a5a0c75c7dcc4 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b414002f583140f1b616ee680e200163 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b297637add53429186aecb9f3782ef6a : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62dd9ce56f7548b4a63d21e314165124 : SUCCESS in 6m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c183b5394d284cb8b09d6a9e7cca8fdb : SUCCESS in 7m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/021bab0ac9ee4f47b3c6db481e3f2527 : SUCCESS in 12m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/519150a55bed4fe0b3fdb46d88fdd0f2 : SUCCESS in 9m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5c9c806578064502ae49c29116136720 : FAILURE in 1h 04m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3c49c868b6ef4757aaabe7e6637d74a0 : SUCCESS in 1h 16m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69f82db4509c4d8b84f9c8bf07093a1e : SUCCESS in 53m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f0fb10ce131340a88758fb7a9852d00e : SUCCESS in 1h 23m 33s\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5d94026a0a78442e8bd093cc55691671 : SUCCESS in 37m 28s","accounts_in_message":[],"_revision_number":14},{"id":"940eafc312181b82b2cb3c35d4c42c3d5dac6106","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-08 17:54:07.000000000","message":"Patch Set 14:\n\nrecheck cinder related failure","accounts_in_message":[],"_revision_number":14},{"id":"edf23b9fe790abb8255b2f5f594320fe4a322304","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-08 19:29:50.000000000","message":"Patch Set 14:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/777012/777012/14/check-not-vote/ext-glance-zuul/836f9a0 : SUCCESS in 35m 35s","accounts_in_message":[],"_revision_number":14},{"id":"a4bfadcb3d33e1031c62f901493359b690eedb8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 19:59:46.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/94145dae4a794092a09889464949b44a : SUCCESS in 58m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c41ea78518424bb097154bd054b80ec2 : SUCCESS in 1h 03m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3884bbcedda54bc9a755f5552e520f00 : SUCCESS in 8m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85636bb1665246b8abbb0c265f25909d : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca760a16b35346cab33492c42f77905e : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22abd33987284debb5338f1719697d78 : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d87982c3b72447d5a0c4f7defa4a174b : SUCCESS in 8m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5ed133df4394f20833d2161bc947bc5 : SUCCESS in 7m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/92a56f873abb428fab5e23dff0aee069 : FAILURE in 12m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ff687d8c61ea400c86be7191f437b9b4 : SUCCESS in 9m 54s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/211169469b024aaa9a1ec258ba986e09 : SUCCESS in 8m 20s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fb88ee13acad40da8cf7d31da2cc5052 : FAILURE in 1h 05m 42s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9455b628fabe4285afd82136aa5bbb23 : SUCCESS in 1h 29m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2faaa20c615a4a589e81b742d2b409ed : SUCCESS in 1h 16m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/20a4381368d54b96b3751a56bd9709d4 : SUCCESS in 1h 00m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c917d415c650442c857a079003a32b45 : SUCCESS in 48m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/913cb20e572e47b1afc9f01d7539a77e : SUCCESS in 1h 12m 18s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cba2d102bc744c318522c2d84d9ef774 : FAILURE in 38m 06s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c8073ab7015b42a99b46c695b998c279 : SUCCESS in 47m 09s","accounts_in_message":[],"_revision_number":14},{"id":"26bfe958ebf91c482f9001667a4f656458e06504","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-08 20:16:57.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"f8e2381c41a5c5e94dd64c2be4b33c7579415d59","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-03-08 20:21:20.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"289d15c0f2247bd128aa02b587f36f23dec45332","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-08 21:07:43.000000000","message":"Patch Set 14:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/777012/777012/14/check-not-vote/ext-glance-zuul/271cb91 : SUCCESS in 35m 46s","accounts_in_message":[],"_revision_number":14},{"id":"898879ac4aef578c104274c9f89d8ad41e6c6d1c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-08 21:21:32.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"92ba505da5d30dee296ca73f15019848b90141ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 21:46:17.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/e9a4f9ebaa66499dacab23d64621e72e : SUCCESS in 56m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9402ae24a58e4eeea589fe7bb4fc70ef : SUCCESS in 1h 16m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b0ce82a611d74ba6aab58f5d352aba06 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaf7036f925c4ee7b4cafb7173f0a0e9 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a754be6c52e141858bafb1be39ed4d01 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21f7289e2acb4f90b6aaddce9738b63d : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f555c703f254533b86d1c1bd3a2d22e : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb0528f8024344ff9dcaa7c0f2eff959 : SUCCESS in 8m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/631527d751964f4e8fc8ad3957f1c5f2 : FAILURE in 12m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/45ad632fc7c74c0f81040a592bfb5489 : SUCCESS in 8m 08s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6fb89bf803184bd9aa77b6a938792d47 : SUCCESS in 5m 44s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/747b395d65a54651a533750807cb2e9a : POST_FAILURE in 1h 05m 00s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/829cdf6aead84867a7fc2b59f00371d0 : SUCCESS in 1h 11m 21s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5ace2ee16304453bacb368c4014b94dd : SUCCESS in 1h 12m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d383fc18d270445aa5ef8e46463b2b81 : SUCCESS in 1h 22m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f7fb04af6794cdebf8055ba9969d219 : SUCCESS in 1h 08m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cb233c3c8c46467a9025426c1eb55e76 : SUCCESS in 1h 13m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d96306572a0b4fc2b0326b67089086c8 : FAILURE in 40m 42s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/eb9cf4dcf83f4dedb4f96d9942b6bc89 : SUCCESS in 35m 29s","accounts_in_message":[],"_revision_number":14},{"id":"fc23c20a2190c2b340fe93efcd29a8e7710ada7f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-08 21:59:08.000000000","message":"Patch Set 14:\n\nrecheck functional spurious failure, debugging elsewhere","accounts_in_message":[],"_revision_number":14},{"id":"080b37d2b148b2a7cd9a53620e79e0ec240b5179","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 23:26:04.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/abebc4d957204d6aa95f5db9d871cc0b : SUCCESS in 1h 04m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f90f707c60c1465e8a35f070bfd3a308 : SUCCESS in 1h 10m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dfbe50844ac84b2f801d37e8247edc82 : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2b69da57acf4974b1b43deaae6cb7fb : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6ec168cc8164ae98f7418740099d335 : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6af348bfb1240beaad54d7b18fcbec4 : SUCCESS in 6m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28456eec35f148a78152890f2ebe7e2c : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c834dcacde174daa8dad2b632de75267 : SUCCESS in 5m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2641a499476346beaa0e1a4880d1a21c : SUCCESS in 9m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c57200c6dcc148049f52d4d1278a2700 : SUCCESS in 9m 35s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/732e02966d3a4f268ecab4a28944a3a9 : SUCCESS in 5m 43s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2a2b5e6e2f52409d9e95ca990ae54caa : POST_FAILURE in 1h 04m 02s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7d45f1c7583d49da92d083c26ad798e1 : SUCCESS in 1h 18m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9a6aff0ffc9246a4b24ed9c6763b912f : SUCCESS in 1h 22m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2e6bb0d53bce4d5ab7a5788c2a4862a2 : SUCCESS in 1h 24m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2c53640486f4096a5ee5aaf642ffcd7 : SUCCESS in 59m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c55e7e92d52b46e78893d0ec4cb44079 : SUCCESS in 1h 19m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bbc2bde3e170446cacdb5cc5046769e0 : FAILURE in 35m 22s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7e397b1b06b840b8b9a06051bcea3207 : SUCCESS in 36m 19s","accounts_in_message":[],"_revision_number":14},{"id":"0f50e452621e4e91a11518d94895aec9e69c754b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 23:26:35.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"222c8db85d3c8e1a5a6c2445ed77879a85c3af13","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-08 23:40:29.000000000","message":"Patch Set 14:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/777012/777012/14/check-not-vote/ext-glance-zuul/9fed8a6 : SUCCESS in 37m 23s","accounts_in_message":[],"_revision_number":14},{"id":"d34ede36d1e87efe414fc5e813e3a44bb858b162","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 01:14:07.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/17adc26079894cb29a04514097b075f2 : SUCCESS in 59m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9922476829bb4977b0443c6233965526 : SUCCESS in 1h 46m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d720f035fff4a5b92115ea0fa769e8e : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/601265b70c7b4e6aa309be51e46ef79c : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c175caa07644a19a3dbc90492c9af25 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d6363942a464f6f8fc5907b4df74547 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d33a6ef50994a8bb7eef469100f716a : SUCCESS in 6m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b5ded582ed214805b8491711861b7fa3 : SUCCESS in 8m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2a0881a6fb9b46179867871d0315d07f : SUCCESS in 8m 09s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c2e7526a300f4fa08b4c989b4e8a409e : SUCCESS in 1h 24m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7eca60431258459da6259b673d21300d : SUCCESS in 1h 19m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/413e9251c10c429d9ea9a9617992f15f : SUCCESS in 1h 08m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3f0b754e7d5243e386192db455e570f8 : SUCCESS in 1h 15m 53s\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a390f4f1f96642a4a13da4f187e3419a : SUCCESS in 45m 13s","accounts_in_message":[],"_revision_number":14},{"id":"90d6ee9e992ed6334e759d8093012e7e621316f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 01:14:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":14},{"id":"ac05cee603aaacdbeab1ef0d0c910ec6f7cf13be","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 01:16:04.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2090a0222f343a0b853be93c8a32f45 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"232177e68c547a815c31a2f30b6d1f95cdb3098d","revisions":{"165ebba32681f4a01540f457b2c2c30aa14f2297":{"kind":"REWORK","_number":1,"created":"2021-02-22 21:23:24.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/12/777012/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/777012/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/777012/1"}}},"commit":{"parents":[{"commit":"ed930ec5123a618fea3a6e6e64bf2d6bfccf7312","subject":"Fix: cinder store test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ed930ec5123a618fea3a6e6e64bf2d6bfccf7312"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 20:52:03.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 22:22:05.000000000","tz":-480},"subject":"Add housekeeping module and staging cleaner","message":"Add housekeeping module and staging cleaner\n\nAs noted in previous discussions, glance should clean its staging\ndirectory on startup. This is important for scenarios where we\nstarted an import operation, but failed in the middle. If, when we\nrecover, the image has already been deleted from the database, then\nwe will never remove the (potentially very large) residue from disk\nin our staging directory.\n\nThis is currently a problem with web-download, but will also occur\nwith glance-direct once we have the non-shared distributed import\nfunctionality merged.\n\nChange-Id: Ib80e9cfb58680f9e8ead5993dc206f4da882dd09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/165ebba32681f4a01540f457b2c2c30aa14f2297"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/165ebba32681f4a01540f457b2c2c30aa14f2297"}]},"branch":"refs/heads/master"},"2357e9a4b55def2229f25afdb46634a788ba3d7b":{"kind":"REWORK","_number":2,"created":"2021-02-22 21:57:21.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/12/777012/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/777012/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/777012/2"}}},"commit":{"parents":[{"commit":"ed930ec5123a618fea3a6e6e64bf2d6bfccf7312","subject":"Fix: cinder store test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ed930ec5123a618fea3a6e6e64bf2d6bfccf7312"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 20:52:03.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 22:56:03.000000000","tz":-480},"subject":"Add housekeeping module and staging cleaner","message":"Add housekeeping module and staging cleaner\n\nAs noted in previous discussions, glance should clean its staging\ndirectory on startup. This is important for scenarios where we\nstarted an import operation, but failed in the middle. If, when we\nrecover, the image has already been deleted from the database, then\nwe will never remove the (potentially very large) residue from disk\nin our staging directory.\n\nThis is currently a problem with web-download, but will also occur\nwith glance-direct once we have the non-shared distributed import\nfunctionality merged.\n\nChange-Id: Ib80e9cfb58680f9e8ead5993dc206f4da882dd09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2357e9a4b55def2229f25afdb46634a788ba3d7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2357e9a4b55def2229f25afdb46634a788ba3d7b"}]},"branch":"refs/heads/master"},"db2770439155572d29b65b1a3434539012dfb9a9":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-02-22 22:02:34.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/12/777012/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/777012/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/777012/3"}}},"commit":{"parents":[{"commit":"ed930ec5123a618fea3a6e6e64bf2d6bfccf7312","subject":"Fix: cinder store test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ed930ec5123a618fea3a6e6e64bf2d6bfccf7312"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 20:52:03.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 23:00:58.000000000","tz":-480},"subject":"Add housekeeping module and staging cleaner","message":"Add housekeeping module and staging cleaner\n\nAs noted in previous discussions, glance should clean its staging\ndirectory on startup. This is important for scenarios where we\nstarted an import operation, but failed in the middle. If, when we\nrecover, the image has already been deleted from the database, then\nwe will never remove the (potentially very large) residue from disk\nin our staging directory.\n\nThis is currently a problem with web-download, but will also occur\nwith glance-direct once we have the non-shared distributed import\nfunctionality merged.\n\nCloses-Bug: #1913625\nChange-Id: Ib80e9cfb58680f9e8ead5993dc206f4da882dd09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/db2770439155572d29b65b1a3434539012dfb9a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/db2770439155572d29b65b1a3434539012dfb9a9"}]},"branch":"refs/heads/master"},"c60222790448e7edde3f101d570c6b5af6065f53":{"kind":"REWORK","_number":4,"created":"2021-02-23 14:59:27.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/12/777012/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/777012/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/777012/4"}}},"commit":{"parents":[{"commit":"ed930ec5123a618fea3a6e6e64bf2d6bfccf7312","subject":"Fix: cinder store test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ed930ec5123a618fea3a6e6e64bf2d6bfccf7312"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 20:52:03.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-23 15:47:28.000000000","tz":-480},"subject":"Add housekeeping module and staging cleaner","message":"Add housekeeping module and staging cleaner\n\nAs noted in previous discussions, glance should clean its staging\ndirectory on startup. This is important for scenarios where we\nstarted an import operation, but failed in the middle. If, when we\nrecover, the image has already been deleted from the database, then\nwe will never remove the (potentially very large) residue from disk\nin our staging directory.\n\nThis is currently a problem with web-download, but will also occur\nwith glance-direct once we have the non-shared distributed import\nfunctionality merged.\n\nCloses-Bug: #1913625\nChange-Id: Ib80e9cfb58680f9e8ead5993dc206f4da882dd09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c60222790448e7edde3f101d570c6b5af6065f53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c60222790448e7edde3f101d570c6b5af6065f53"}]},"branch":"refs/heads/master"},"40a555070858c0573ea38f5ed6275815f87db823":{"kind":"REWORK","_number":5,"created":"2021-02-23 16:33:34.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/12/777012/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/777012/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/777012/5"}}},"commit":{"parents":[{"commit":"ed930ec5123a618fea3a6e6e64bf2d6bfccf7312","subject":"Fix: cinder store test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ed930ec5123a618fea3a6e6e64bf2d6bfccf7312"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 20:52:03.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-23 16:33:17.000000000","tz":-480},"subject":"Add housekeeping module and staging cleaner","message":"Add housekeeping module and staging cleaner\n\nAs noted in previous discussions, glance should clean its staging\ndirectory on startup. This is important for scenarios where we\nstarted an import operation, but failed in the middle. If, when we\nrecover, the image has already been deleted from the database, then\nwe will never remove the (potentially very large) residue from disk\nin our staging directory.\n\nThis is currently a problem with web-download, but will also occur\nwith glance-direct once we have the non-shared distributed import\nfunctionality merged.\n\nCloses-Bug: #1913625\nChange-Id: Ib80e9cfb58680f9e8ead5993dc206f4da882dd09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/40a555070858c0573ea38f5ed6275815f87db823"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/40a555070858c0573ea38f5ed6275815f87db823"}]},"branch":"refs/heads/master"},"03aec1396f6974a3e2211b4a25565d2992dd3d9c":{"kind":"REWORK","_number":6,"created":"2021-02-23 16:54:26.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/12/777012/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/777012/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/777012/6"}}},"commit":{"parents":[{"commit":"ed930ec5123a618fea3a6e6e64bf2d6bfccf7312","subject":"Fix: cinder store test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ed930ec5123a618fea3a6e6e64bf2d6bfccf7312"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 20:52:03.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-23 16:54:22.000000000","tz":-480},"subject":"Add housekeeping module and staging cleaner","message":"Add housekeeping module and staging cleaner\n\nAs noted in previous discussions, glance should clean its staging\ndirectory on startup. This is important for scenarios where we\nstarted an import operation, but failed in the middle. If, when we\nrecover, the image has already been deleted from the database, then\nwe will never remove the (potentially very large) residue from disk\nin our staging directory.\n\nThis is currently a problem with web-download, but will also occur\nwith glance-direct once we have the non-shared distributed import\nfunctionality merged.\n\nCloses-Bug: #1913625\nChange-Id: Ib80e9cfb58680f9e8ead5993dc206f4da882dd09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03aec1396f6974a3e2211b4a25565d2992dd3d9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03aec1396f6974a3e2211b4a25565d2992dd3d9c"}]},"branch":"refs/heads/master"},"18e6f93b2c3ae42067c260c59a5857dfda9a905e":{"kind":"REWORK","_number":7,"created":"2021-02-23 23:30:10.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/12/777012/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/777012/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/777012/7"}}},"commit":{"parents":[{"commit":"ed930ec5123a618fea3a6e6e64bf2d6bfccf7312","subject":"Fix: cinder store test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ed930ec5123a618fea3a6e6e64bf2d6bfccf7312"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 20:52:03.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-23 23:30:01.000000000","tz":-480},"subject":"Add housekeeping module and staging cleaner","message":"Add housekeeping module and staging cleaner\n\nAs noted in previous discussions, glance should clean its staging\ndirectory on startup. This is important for scenarios where we\nstarted an import operation, but failed in the middle. If, when we\nrecover, the image has already been deleted from the database, then\nwe will never remove the (potentially very large) residue from disk\nin our staging directory.\n\nThis is currently a problem with web-download, but will also occur\nwith glance-direct once we have the non-shared distributed import\nfunctionality merged.\n\nCloses-Bug: #1913625\nChange-Id: Ib80e9cfb58680f9e8ead5993dc206f4da882dd09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/18e6f93b2c3ae42067c260c59a5857dfda9a905e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/18e6f93b2c3ae42067c260c59a5857dfda9a905e"}]},"branch":"refs/heads/master"},"b0bd5ea02162903073ad6d75153cfc9eb74bba20":{"kind":"REWORK","_number":8,"created":"2021-02-24 03:06:11.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/12/777012/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/777012/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/777012/8"}}},"commit":{"parents":[{"commit":"344251b01dc9ddfe49d95ed0e5121fe54acec600","subject":"Make functional tests set node_staging_uri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/344251b01dc9ddfe49d95ed0e5121fe54acec600"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 20:52:03.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-24 03:02:04.000000000","tz":-480},"subject":"Add housekeeping module and staging cleaner","message":"Add housekeeping module and staging cleaner\n\nAs noted in previous discussions, glance should clean its staging\ndirectory on startup. This is important for scenarios where we\nstarted an import operation, but failed in the middle. If, when we\nrecover, the image has already been deleted from the database, then\nwe will never remove the (potentially very large) residue from disk\nin our staging directory.\n\nThis is currently a problem with web-download, but will also occur\nwith glance-direct once we have the non-shared distributed import\nfunctionality merged.\n\nCloses-Bug: #1913625\nChange-Id: Ib80e9cfb58680f9e8ead5993dc206f4da882dd09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b0bd5ea02162903073ad6d75153cfc9eb74bba20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b0bd5ea02162903073ad6d75153cfc9eb74bba20"}]},"branch":"refs/heads/master"},"928597a3ae8caa20f8fd56fb54f9cde800d61496":{"kind":"NO_CHANGE","_number":9,"created":"2021-02-24 17:32:51.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/12/777012/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/777012/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/777012/9"}}},"commit":{"parents":[{"commit":"5c1a781d30ec70f005aabc1092858f91dfee447c","subject":"Make functional tests set node_staging_uri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5c1a781d30ec70f005aabc1092858f91dfee447c"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 20:52:03.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-24 17:32:01.000000000","tz":-480},"subject":"Add housekeeping module and staging cleaner","message":"Add housekeeping module and staging cleaner\n\nAs noted in previous discussions, glance should clean its staging\ndirectory on startup. This is important for scenarios where we\nstarted an import operation, but failed in the middle. If, when we\nrecover, the image has already been deleted from the database, then\nwe will never remove the (potentially very large) residue from disk\nin our staging directory.\n\nThis is currently a problem with web-download, but will also occur\nwith glance-direct once we have the non-shared distributed import\nfunctionality merged.\n\nCloses-Bug: #1913625\nChange-Id: Ib80e9cfb58680f9e8ead5993dc206f4da882dd09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/928597a3ae8caa20f8fd56fb54f9cde800d61496"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/928597a3ae8caa20f8fd56fb54f9cde800d61496"}]},"branch":"refs/heads/master"},"fb2f59c180c61bf4dd0eb745755a6234cafd0d72":{"kind":"REWORK","_number":10,"created":"2021-03-02 19:51:08.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/12/777012/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/777012/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/777012/10"}}},"commit":{"parents":[{"commit":"e9852fb6251c655f0e384c6b0b8677bd556f17a1","subject":"Make functional tests set node_staging_uri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e9852fb6251c655f0e384c6b0b8677bd556f17a1"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 20:52:03.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-02 19:51:00.000000000","tz":-480},"subject":"Add housekeeping module and staging cleaner","message":"Add housekeeping module and staging cleaner\n\nAs noted in previous discussions, glance should clean its staging\ndirectory on startup. This is important for scenarios where we\nstarted an import operation, but failed in the middle. If, when we\nrecover, the image has already been deleted from the database, then\nwe will never remove the (potentially very large) residue from disk\nin our staging directory.\n\nThis is currently a problem with web-download, but will also occur\nwith glance-direct once we have the non-shared distributed import\nfunctionality merged.\n\nCloses-Bug: #1913625\nChange-Id: Ib80e9cfb58680f9e8ead5993dc206f4da882dd09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fb2f59c180c61bf4dd0eb745755a6234cafd0d72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fb2f59c180c61bf4dd0eb745755a6234cafd0d72"}]},"branch":"refs/heads/master"},"5b99ae1947655518edce2f8a059b54ee9d9ad433":{"kind":"REWORK","_number":11,"created":"2021-03-03 01:36:06.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/12/777012/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/777012/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/777012/11"}}},"commit":{"parents":[{"commit":"e9852fb6251c655f0e384c6b0b8677bd556f17a1","subject":"Make functional tests set node_staging_uri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e9852fb6251c655f0e384c6b0b8677bd556f17a1"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 20:52:03.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-03 01:35:32.000000000","tz":-480},"subject":"Add housekeeping module and staging cleaner","message":"Add housekeeping module and staging cleaner\n\nAs noted in previous discussions, glance should clean its staging\ndirectory on startup. This is important for scenarios where we\nstarted an import operation, but failed in the middle. If, when we\nrecover, the image has already been deleted from the database, then\nwe will never remove the (potentially very large) residue from disk\nin our staging directory.\n\nThis is currently a problem with web-download, but will also occur\nwith glance-direct once we have the non-shared distributed import\nfunctionality merged.\n\nCloses-Bug: #1913625\nChange-Id: Ib80e9cfb58680f9e8ead5993dc206f4da882dd09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5b99ae1947655518edce2f8a059b54ee9d9ad433"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5b99ae1947655518edce2f8a059b54ee9d9ad433"}]},"branch":"refs/heads/master"},"3934ab9dcc3e0049894e1086121558aa22384f9e":{"kind":"REWORK","_number":12,"created":"2021-03-03 16:01:38.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/12/777012/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/777012/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/777012/12"}}},"commit":{"parents":[{"commit":"e9852fb6251c655f0e384c6b0b8677bd556f17a1","subject":"Make functional tests set node_staging_uri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e9852fb6251c655f0e384c6b0b8677bd556f17a1"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 20:52:03.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-03 15:37:37.000000000","tz":-480},"subject":"Add housekeeping module and staging cleaner","message":"Add housekeeping module and staging cleaner\n\nAs noted in previous discussions, glance should clean its staging\ndirectory on startup. This is important for scenarios where we\nstarted an import operation, but failed in the middle. If, when we\nrecover, the image has already been deleted from the database, then\nwe will never remove the (potentially very large) residue from disk\nin our staging directory.\n\nThis is currently a problem with web-download, but will also occur\nwith glance-direct once we have the non-shared distributed import\nfunctionality merged.\n\nCloses-Bug: #1913625\nChange-Id: Ib80e9cfb58680f9e8ead5993dc206f4da882dd09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3934ab9dcc3e0049894e1086121558aa22384f9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3934ab9dcc3e0049894e1086121558aa22384f9e"}]},"branch":"refs/heads/master"},"b242f87bcebea26bc843564f3b5f25482bcbce01":{"kind":"REWORK","_number":13,"created":"2021-03-03 18:00:38.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/12/777012/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/777012/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/777012/13"}}},"commit":{"parents":[{"commit":"e9852fb6251c655f0e384c6b0b8677bd556f17a1","subject":"Make functional tests set node_staging_uri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e9852fb6251c655f0e384c6b0b8677bd556f17a1"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 20:52:03.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-03 18:00:25.000000000","tz":-480},"subject":"Add housekeeping module and staging cleaner","message":"Add housekeeping module and staging cleaner\n\nAs noted in previous discussions, glance should clean its staging\ndirectory on startup. This is important for scenarios where we\nstarted an import operation, but failed in the middle. If, when we\nrecover, the image has already been deleted from the database, then\nwe will never remove the (potentially very large) residue from disk\nin our staging directory.\n\nThis is currently a problem with web-download, but will also occur\nwith glance-direct once we have the non-shared distributed import\nfunctionality merged.\n\nCloses-Bug: #1913625\nChange-Id: Ib80e9cfb58680f9e8ead5993dc206f4da882dd09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b242f87bcebea26bc843564f3b5f25482bcbce01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b242f87bcebea26bc843564f3b5f25482bcbce01"}]},"branch":"refs/heads/master"},"232177e68c547a815c31a2f30b6d1f95cdb3098d":{"kind":"REWORK","_number":14,"created":"2021-03-03 22:43:28.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/12/777012/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/777012/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/777012/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/777012/14"}}},"commit":{"parents":[{"commit":"e9852fb6251c655f0e384c6b0b8677bd556f17a1","subject":"Make functional tests set node_staging_uri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e9852fb6251c655f0e384c6b0b8677bd556f17a1"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 20:52:03.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-03 22:36:46.000000000","tz":-480},"subject":"Add housekeeping module and staging cleaner","message":"Add housekeeping module and staging cleaner\n\nAs noted in previous discussions, glance should clean its staging\ndirectory on startup. This is important for scenarios where we\nstarted an import operation, but failed in the middle. If, when we\nrecover, the image has already been deleted from the database, then\nwe will never remove the (potentially very large) residue from disk\nin our staging directory.\n\nThis is currently a problem with web-download, but will also occur\nwith glance-direct once we have the non-shared distributed import\nfunctionality merged.\n\nCloses-Bug: #1913625\nChange-Id: Ib80e9cfb58680f9e8ead5993dc206f4da882dd09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/232177e68c547a815c31a2f30b6d1f95cdb3098d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/232177e68c547a815c31a2f30b6d1f95cdb3098d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
