)]}'
{"id":"openstack%2Fglance~777014","triplet_id":"openstack%2Fglance~master~I72a4372582adc142c965328fa82e8b2b388e5608","project":"openstack/glance","branch":"master","topic":"bug/1913625","hashtags":[],"change_id":"I72a4372582adc142c965328fa82e8b2b388e5608","subject":"MDNM: Test restarting glance for staging cleanup","status":"ABANDONED","created":"2021-02-22 21:52:43.000000000","updated":"2021-03-04 14:47:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"84d12a18a2335f6c7334f6651f9c1dae2ef0c57d","_number":777014,"virtual_id_number":777014,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-25 00:44:43.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-22 22:31:23.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-22 23:37:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0f88dcf0172270f831b51215f6379ad21f671bc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-22 21:52:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61eb28e03239b9f650d855f2a2b2226114ff07df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-22 21:57:21.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"5dce085ca37248e0af89b848eb061bb6457732b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-22 22:02:34.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"6018c34f27476c1e990eb48e051b495bd6bc03d2","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-22 22:31:23.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/777014/777014/1/check-not-vote/ext-glance-zuul/ac2d122 : SUCCESS in 38m 05s","accounts_in_message":[],"_revision_number":1},{"id":"29e9b0f624a8bbdcc3b0bb7073433327e9949f8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 23:37:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/270d5f02c4744b7abcd15bb53817f191 : SUCCESS in 1h 01m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e47dcbbb9f724bd791d9775e8f2424fe : SUCCESS in 1h 07m 43s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d6982a32da6b479ba0e5f291d3eacbeb : SUCCESS in 9m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c248dbc505114f669b2fdfe7d0553658 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf2099f2226f4aefba78da5fdc6ffc87 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/983fd6ba73c54a22a456525d02f1e17d : SUCCESS in 5m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2717f55efa9741ce9a72165a046d1da7 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf566a479c9b4aabaa69b7e3f432f2ee : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ec9d7db1255447408a83e42dd24dc23f : SUCCESS in 8m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1be7d8d4fdd34030a2e775bd6660fbd5 : SUCCESS in 8m 24s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/44b416cc4585483881b89f67c4e5201e : SUCCESS in 5m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9930366310844ee0ae76e2301e0693a7 : SUCCESS in 1h 13m 18s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2ba8b345d84b48e7848163363a4e3955 : SUCCESS in 1h 29m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3b8700135d584b34977898848fb23191 : SUCCESS in 1h 26m 08s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b2712f97ebc94a3c8550d3f90dd4647f : SUCCESS in 1h 15m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7422b1cf3534436f8c05dced135d49c7 : SUCCESS in 56m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2285c4d9187f43b7bc6bdc505a7f98e7 : SUCCESS in 1h 06m 29s","accounts_in_message":[],"_revision_number":3},{"id":"0f9b4798846e4b2d3d5323b4acc8aca91a6c31d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-23 01:45:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2aba4ea702281032d70ea768a39cfb7456061188","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-23 02:22:06.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/777014/777014/4/check-not-vote/ext-glance-zuul/b10ea45 : SUCCESS in 35m 53s","accounts_in_message":[],"_revision_number":4},{"id":"359100a5e0123532f615b465fc2f97baed71aa63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 03:06:17.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/aea0ae96e57043e0a505e30b0214b50d : SUCCESS in 59m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dddc5ce0e3344418a22937d2a7d6cf2c : SUCCESS in 1h 04m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/14466d84725241f7b73290a300b54895 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd4351e0216f4d8aaef28fe8109c9ae2 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2269c03dc81451b9c1fa927a0a69d04 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/258cb453eeac455fbfc5e4ee2b0c872d : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1bfac717c5641e18a79d0ab5e698e4d : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/863b1c0aae3b481bb9f3cb745f29f7f6 : SUCCESS in 6m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2b778428cf3049b183df7c9b3e1d5ebe : SUCCESS in 8m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/92e3e1088cef4a29869b2b47450cd075 : SUCCESS in 9m 05s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a61e42fec12144d0a604d5ee584362cc : SUCCESS in 5m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a4601a0026ce4af4a16665839a18aa01 : SUCCESS in 1h 15m 47s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b3a5350d09b44d6dbb1ed2f4d2805aaf : SUCCESS in 1h 13m 09s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4d2c105723954ed2ac334fc4ecca6335 : SUCCESS in 1h 11m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7bfdf20b78e140dda78e407072562726 : SUCCESS in 1h 08m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7365a6da37c4af6ac372b3b75d3461d : SUCCESS in 1h 02m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5266961d7344414f8495f921ff391f34 : SUCCESS in 1h 04m 31s","accounts_in_message":[],"_revision_number":4},{"id":"f959155d938b6695d95664e92a460baeac25aa25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-23 14:59:27.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a5fa53fbea406c4a082d1fb70b2b3a7bf18ef1c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 14:59:52.000000000","message":"Patch Set 5: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch master.  The error was:\n\n  while parsing a block collection\n  in \"openstack/glance/.zuul.yaml@master\", line 1, column 1\ndid not find expected \u0027-\u0027 indicator\n  in \"openstack/glance/.zuul.yaml@master\", line 281, column 2","accounts_in_message":[],"_revision_number":5},{"id":"a53b564bbeecc7666e716c9186dbe5cc7992e0f9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-23 15:13:24.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ba58fc87b58686e2f0d491be1ee31617763de4c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-23 16:33:34.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"af78180ce5514683d66501d9b533ac01b4849033","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-23 16:54:26.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"30012d9ae1b40f0f200f1f0c911bd8d8e9e03a99","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-23 17:36:28.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/777014/777014/6/check-not-vote/ext-glance-zuul/e8715fd : SUCCESS in 35m 49s","accounts_in_message":[],"_revision_number":6},{"id":"8a88f548b32830c7175211e3677ad0804d4e912e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 22:46:37.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/a77a37fcc0784980ab8f89dd049e5fdc : SUCCESS in 1h 17m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/69052505c77f402191c324b11cdce244 : SUCCESS in 1h 22m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fabf5d1d91a4c3296cea991da3cd12a : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/916da713a4614cb48143100c922eed89 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdb8459046164083aeb8985f328c1e11 : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22ebc443bafe4704bf05db9efd8cb1a7 : SUCCESS in 7m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c9f6baccaf54f78b3656ad0032f9b92 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bab2fc4423c4775857c15fa9a4e04f6 : SUCCESS in 7m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a823e9d1f3d2442b8803f7e147f29e51 : FAILURE in 8m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ca8e47900d4347b58bd0004b93511e7e : FAILURE in 7m 57s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e060ffc3937a49f3bb043b551e21a769 : SUCCESS in 5m 23s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/26774e4af2c14db98ed85f28ca4f0a31 : SUCCESS in 1h 30m 37s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c984cf2fec5441f483789a0576a78dae : FAILURE in 1h 02m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6aa2ed7a586f4881b796a8e0a4c70229 : SUCCESS in 1h 32m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ebfdaf909f424c75b066a1f3efac9e1f : SUCCESS in 1h 23m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20a4fc672c1244ac95e64881e87af94a : SUCCESS in 58m 49s\n- glance-nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d3bb421109c94eb2aa6e2989d4db316d : SUCCESS in 1h 11m 12s","accounts_in_message":[],"_revision_number":8},{"id":"44c309ca0a4556cb1e44c376456cc96435295e16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-23 23:30:10.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"0e34e6e15411a998cb294137fff6ff3f6d9b3001","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-23 23:35:45.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"d4f4c4aeaae0a43ea51405e695c462d257c030c6","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-24 00:15:49.000000000","message":"Patch Set 10:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/777014/777014/10/check-not-vote/ext-glance-zuul/6ad7cde : SUCCESS in 39m 25s","accounts_in_message":[],"_revision_number":10},{"id":"5fab02d496a58f8c6a9ae02aad524fea2453ba54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 02:26:40.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/d952398dfc444000834fffada770a835 : SUCCESS in 1h 05m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/59cc6c5074514faca7f8781595c087ff : SUCCESS in 1h 05m 37s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e458a8dec894eaaa61f64bf4aa44d5a : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fef95d46077e4e6bad77943b5f2534f3 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2e5709c3ae72476da2db047f8f9f4abd : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/860e99d5d5804eeabfa461fce9ce99c1 : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/49d782daa16d4858a91c515240360b19 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/154f237b00904a4ab29a1ad98e252a34 : SUCCESS in 7m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/024195f930e04283ab78b030081cf025 : FAILURE in 10m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6f5e67db22e04988a584fd6b74f68569 : FAILURE in 6m 27s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0fd9842b47864f32b63ef438beeff848 : SUCCESS in 6m 52s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e799dc59a0ea4719ad971b23923ba7ce : SUCCESS in 1h 18m 32s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/902b8c0a49914c818da2cd7349dae55c : SUCCESS in 1h 10m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1ef6cab3172d4fdcb60867f6d8d0057a : SUCCESS in 1h 09m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/87c28f56e69848db82c45cd62e13b6bb : SUCCESS in 1h 19m 21s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a61fb7d8bbb4127aed7b1a721b8d667 : SUCCESS in 1h 04m 22s\n- glance-nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/aafe41a445874267a39f3d4be76344f6 : SUCCESS in 1h 31m 52s","accounts_in_message":[],"_revision_number":10},{"id":"a2fb3c0f3397c9b5583fabd17a1c399e8f3278f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-24 03:06:11.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"b874e417c4bb411eeea383aed600cd414a01d53f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 04:49:30.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/8570dd5d77c947f48fc404ec454812f3 : SUCCESS in 1h 03m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0073a55f47a24a018d3c308326fd4d1c : SUCCESS in 1h 19m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4182d4ff0bb47fe94debc9e2e227237 : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9e70baad39e45c988a807d9a1148353 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f89f2c9b47814941959bf72078c366c2 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7baeb55ad0d643a281e3b97ac8d39312 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf4942c4e1fa4b20949993ddb950699f : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6594c6216394a27b8427321c145f93b : SUCCESS in 6m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4275f68f6f5b475da7fafc396fb8d42f : SUCCESS in 10m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cbc45c28a1b84556840c75f5e216b29b : SUCCESS in 10m 47s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ed4dced59d534ecfb145d2ea4ec503bb : SUCCESS in 4m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f8cdac4d9cf348608818f05e5ebcdf6b : SUCCESS in 1h 14m 33s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/87bf0b8d20b04b9faa2866f93ab4fad4 : SUCCESS in 1h 41m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b70ddaed30b145349b5fea50fab70b3c : SUCCESS in 1h 27m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d2ba45df9e7c4e5c84ae49a78c11d259 : FAILURE in 1h 29m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25057937f2c14924b8a12cf35b90eb8b : SUCCESS in 42m 13s\n- glance-nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/41efb5a593a24bf5a72ef67ce74b6b4e : SUCCESS in 1h 08m 43s","accounts_in_message":[],"_revision_number":11},{"id":"0dd473e8c61be779e3f47dab568ddfed3048ad2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-24 14:04:56.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"89e1f00e4fdb625a26304fe7563ab3276cbf20b3","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-24 16:47:01.000000000","message":"Patch Set 12:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/777014/777014/12/check-not-vote/ext-glance-zuul/4874231 : FAILURE in 17m 55s","accounts_in_message":[],"_revision_number":12},{"id":"0e64bc930019a31780efbbeac43b60125a1e614b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-24 17:32:51.000000000","message":"Uploaded patch set 13: Commit message was updated.","accounts_in_message":[],"_revision_number":13},{"id":"566d86a8e4936783f30b75e4883e30a1b711dd59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 00:44:43.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/880cbb98874a444294a6eccc23094bdc : SUCCESS in 1h 04m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3b025e90159c4bc192c57480fe385e79 : FAILURE in 1h 02m 42s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b6ac268f7ef45cdac8320741a4d3f7e : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f4e5718c69d4e0c9977a548ddaf8555 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0b56eb838d14848aaf398b5777b865c : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7111bb0a582423eb3568c919e0b58e3 : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/998646528945447aa28b48cbcb72df4f : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3aedb38d310447f68965a244051cbb7e : SUCCESS in 8m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/16e135c54aec483dba47233f4762b1cf : SUCCESS in 9m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/45d970f5a8cc445faf02ba00ca560cc4 : SUCCESS in 12m 33s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/aeb7405e5bed4e55a13b6d859dbf96e3 : SUCCESS in 6m 10s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d016c1d9537c4a09ba7e99ef1e20dbc8 : SUCCESS in 1h 20m 08s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/06e5ace597834513bc0a7b184df022fb : SUCCESS in 1h 13m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2493d4faf5fa4cf2b6a968a3cd254edc : SUCCESS in 1h 12m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/db7d42477ee148b38f6823dc3556a03b : SUCCESS in 1h 15m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/91af9e5ec2f843d69c65024ebb5234f7 : SUCCESS in 1h 03m 55s\n- glance-nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a96c9e7ebf8f48a1965b852a02bb21a5 : SUCCESS in 1h 16m 12s","accounts_in_message":[],"_revision_number":13},{"id":"84d12a18a2335f6c7334f6651f9c1dae2ef0c57d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-04 14:47:00.000000000","message":"Abandoned\n\nNo longer needed, as this is done in functional tests","accounts_in_message":[],"_revision_number":13}],"current_revision_number":13,"current_revision":"a267aacfc7cb37a7f9f00731a66b0dc64d0399a9","revisions":{"9b6cc3119a662665e6e40fe1e3c4fb35e40a3710":{"kind":"REWORK","_number":1,"created":"2021-02-22 21:52:43.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/777014/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/777014/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/777014/1"}}},"commit":{"parents":[{"commit":"165ebba32681f4a01540f457b2c2c30aa14f2297","subject":"Add housekeeping module and staging cleaner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/165ebba32681f4a01540f457b2c2c30aa14f2297"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 22:50:23.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 22:50:23.000000000","tz":-480},"subject":"Test restarting glance for staging cleanup","message":"Test restarting glance for staging cleanup\n\nTesting this in reality is quite hard, so this adds a hacky post-run\nscript that should create some cruft in the staging directory,\nas well as stage an image and then restart glance so we can examine\nthe log files.\n\nChange-Id: I72a4372582adc142c965328fa82e8b2b388e5608\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9b6cc3119a662665e6e40fe1e3c4fb35e40a3710"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9b6cc3119a662665e6e40fe1e3c4fb35e40a3710"}]},"branch":"refs/heads/master"},"edf45cc2c7ddfec0ae66f5953ec40ac594811907":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-02-22 21:57:21.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/777014/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/777014/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/777014/2"}}},"commit":{"parents":[{"commit":"2357e9a4b55def2229f25afdb46634a788ba3d7b","subject":"Add housekeeping module and staging cleaner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2357e9a4b55def2229f25afdb46634a788ba3d7b"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 22:50:23.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 22:56:04.000000000","tz":-480},"subject":"Test restarting glance for staging cleanup","message":"Test restarting glance for staging cleanup\n\nTesting this in reality is quite hard, so this adds a hacky post-run\nscript that should create some cruft in the staging directory,\nas well as stage an image and then restart glance so we can examine\nthe log files.\n\nChange-Id: I72a4372582adc142c965328fa82e8b2b388e5608\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/edf45cc2c7ddfec0ae66f5953ec40ac594811907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/edf45cc2c7ddfec0ae66f5953ec40ac594811907"}]},"branch":"refs/heads/master"},"18f1572973c00b916e5913bf0e9c3f8381fce93b":{"kind":"NO_CHANGE","_number":3,"created":"2021-02-22 22:02:34.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/777014/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/777014/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/777014/3"}}},"commit":{"parents":[{"commit":"db2770439155572d29b65b1a3434539012dfb9a9","subject":"Add housekeeping module and staging cleaner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/db2770439155572d29b65b1a3434539012dfb9a9"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 22:50:23.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 23:01:05.000000000","tz":-480},"subject":"Test restarting glance for staging cleanup","message":"Test restarting glance for staging cleanup\n\nTesting this in reality is quite hard, so this adds a hacky post-run\nscript that should create some cruft in the staging directory,\nas well as stage an image and then restart glance so we can examine\nthe log files.\n\nChange-Id: I72a4372582adc142c965328fa82e8b2b388e5608\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/18f1572973c00b916e5913bf0e9c3f8381fce93b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/18f1572973c00b916e5913bf0e9c3f8381fce93b"}]},"branch":"refs/heads/master"},"2dff6a69f58c6d944714da25c4b6ab9da3a05438":{"kind":"REWORK","_number":4,"created":"2021-02-23 01:45:08.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/777014/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/777014/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/777014/4"}}},"commit":{"parents":[{"commit":"db2770439155572d29b65b1a3434539012dfb9a9","subject":"Add housekeeping module and staging cleaner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/db2770439155572d29b65b1a3434539012dfb9a9"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 22:50:23.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-23 01:36:56.000000000","tz":-480},"subject":"Test restarting glance for staging cleanup","message":"Test restarting glance for staging cleanup\n\nTesting this in reality is quite hard, so this adds a hacky post-run\nscript that should create some cruft in the staging directory,\nas well as stage an image and then restart glance so we can examine\nthe log files.\n\nChange-Id: I72a4372582adc142c965328fa82e8b2b388e5608\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2dff6a69f58c6d944714da25c4b6ab9da3a05438"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2dff6a69f58c6d944714da25c4b6ab9da3a05438"}]},"branch":"refs/heads/master"},"f0375fab26bd7f02769382a36a323ae9e991da72":{"kind":"REWORK","_number":5,"created":"2021-02-23 14:59:27.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/777014/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/777014/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/777014/5"}}},"commit":{"parents":[{"commit":"c60222790448e7edde3f101d570c6b5af6065f53","subject":"Add housekeeping module and staging cleaner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c60222790448e7edde3f101d570c6b5af6065f53"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 22:50:23.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-23 15:57:47.000000000","tz":-480},"subject":"Test restarting glance for staging cleanup","message":"Test restarting glance for staging cleanup\n\nTesting this in reality is quite hard, so this adds a hacky post-run\nscript that should create some cruft in the staging directory,\nas well as stage an image and then restart glance so we can examine\nthe log files.\n\nChange-Id: I72a4372582adc142c965328fa82e8b2b388e5608\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f0375fab26bd7f02769382a36a323ae9e991da72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f0375fab26bd7f02769382a36a323ae9e991da72"}]},"branch":"refs/heads/master"},"a7aa3c809990dcb3c49ff8046245f93221b01caf":{"kind":"REWORK","_number":6,"created":"2021-02-23 15:13:24.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/777014/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/777014/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/777014/6"}}},"commit":{"parents":[{"commit":"c60222790448e7edde3f101d570c6b5af6065f53","subject":"Add housekeeping module and staging cleaner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c60222790448e7edde3f101d570c6b5af6065f53"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 22:50:23.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-23 16:12:00.000000000","tz":-480},"subject":"Test restarting glance for staging cleanup","message":"Test restarting glance for staging cleanup\n\nTesting this in reality is quite hard, so this adds a hacky post-run\nscript that should create some cruft in the staging directory,\nas well as stage an image and then restart glance so we can examine\nthe log files.\n\nChange-Id: I72a4372582adc142c965328fa82e8b2b388e5608\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a7aa3c809990dcb3c49ff8046245f93221b01caf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a7aa3c809990dcb3c49ff8046245f93221b01caf"}]},"branch":"refs/heads/master"},"96920c14525da79dc5f6ea4c407ec2395ae4db0b":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-02-23 16:33:34.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/777014/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/777014/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/777014/7"}}},"commit":{"parents":[{"commit":"40a555070858c0573ea38f5ed6275815f87db823","subject":"Add housekeeping module and staging cleaner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/40a555070858c0573ea38f5ed6275815f87db823"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 22:50:23.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-23 16:33:26.000000000","tz":-480},"subject":"Test restarting glance for staging cleanup","message":"Test restarting glance for staging cleanup\n\nTesting this in reality is quite hard, so this adds a hacky post-run\nscript that should create some cruft in the staging directory,\nas well as stage an image and then restart glance so we can examine\nthe log files.\n\nChange-Id: I72a4372582adc142c965328fa82e8b2b388e5608\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/96920c14525da79dc5f6ea4c407ec2395ae4db0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/96920c14525da79dc5f6ea4c407ec2395ae4db0b"}]},"branch":"refs/heads/master"},"74bce8d25bbbadf03dbb3a5e35fb7061babaf251":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-02-23 16:54:26.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/777014/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/777014/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/777014/8"}}},"commit":{"parents":[{"commit":"03aec1396f6974a3e2211b4a25565d2992dd3d9c","subject":"Add housekeeping module and staging cleaner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03aec1396f6974a3e2211b4a25565d2992dd3d9c"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 22:50:23.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-23 16:54:23.000000000","tz":-480},"subject":"Test restarting glance for staging cleanup","message":"Test restarting glance for staging cleanup\n\nTesting this in reality is quite hard, so this adds a hacky post-run\nscript that should create some cruft in the staging directory,\nas well as stage an image and then restart glance so we can examine\nthe log files.\n\nChange-Id: I72a4372582adc142c965328fa82e8b2b388e5608\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/74bce8d25bbbadf03dbb3a5e35fb7061babaf251"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/74bce8d25bbbadf03dbb3a5e35fb7061babaf251"}]},"branch":"refs/heads/master"},"e130a3510095e0ed32fe0ec17d26926712e73706":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-02-23 23:30:10.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/777014/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/777014/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/777014/9"}}},"commit":{"parents":[{"commit":"18e6f93b2c3ae42067c260c59a5857dfda9a905e","subject":"Add housekeeping module and staging cleaner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/18e6f93b2c3ae42067c260c59a5857dfda9a905e"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 22:50:23.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-23 23:30:03.000000000","tz":-480},"subject":"Test restarting glance for staging cleanup","message":"Test restarting glance for staging cleanup\n\nTesting this in reality is quite hard, so this adds a hacky post-run\nscript that should create some cruft in the staging directory,\nas well as stage an image and then restart glance so we can examine\nthe log files.\n\nChange-Id: I72a4372582adc142c965328fa82e8b2b388e5608\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e130a3510095e0ed32fe0ec17d26926712e73706"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e130a3510095e0ed32fe0ec17d26926712e73706"}]},"branch":"refs/heads/master"},"67099b278fce6c9a32693b581912ec65edfb45f9":{"kind":"REWORK","_number":10,"created":"2021-02-23 23:35:45.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/777014/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/777014/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/777014/10"}}},"commit":{"parents":[{"commit":"18e6f93b2c3ae42067c260c59a5857dfda9a905e","subject":"Add housekeeping module and staging cleaner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/18e6f93b2c3ae42067c260c59a5857dfda9a905e"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 22:50:23.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-23 23:35:41.000000000","tz":-480},"subject":"Test restarting glance for staging cleanup","message":"Test restarting glance for staging cleanup\n\nTesting this in reality is quite hard, so this adds a hacky post-run\nscript that should create some cruft in the staging directory,\nas well as stage an image and then restart glance so we can examine\nthe log files.\n\nChange-Id: I72a4372582adc142c965328fa82e8b2b388e5608\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/67099b278fce6c9a32693b581912ec65edfb45f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/67099b278fce6c9a32693b581912ec65edfb45f9"}]},"branch":"refs/heads/master"},"67fb9dd4b0b11dc176d7102e89633554c03a2910":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2021-02-24 03:06:11.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/777014/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/777014/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/777014/11"}}},"commit":{"parents":[{"commit":"b0bd5ea02162903073ad6d75153cfc9eb74bba20","subject":"Add housekeeping module and staging cleaner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b0bd5ea02162903073ad6d75153cfc9eb74bba20"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 22:50:23.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-24 03:02:04.000000000","tz":-480},"subject":"Test restarting glance for staging cleanup","message":"Test restarting glance for staging cleanup\n\nTesting this in reality is quite hard, so this adds a hacky post-run\nscript that should create some cruft in the staging directory,\nas well as stage an image and then restart glance so we can examine\nthe log files.\n\nChange-Id: I72a4372582adc142c965328fa82e8b2b388e5608\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/67fb9dd4b0b11dc176d7102e89633554c03a2910"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/67fb9dd4b0b11dc176d7102e89633554c03a2910"}]},"branch":"refs/heads/master"},"cf76204b1b46dfd4e01004dc7eb36f1c1f627c9e":{"kind":"REWORK","_number":12,"created":"2021-02-24 14:04:56.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/777014/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/777014/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/777014/12"}}},"commit":{"parents":[{"commit":"b0bd5ea02162903073ad6d75153cfc9eb74bba20","subject":"Add housekeeping module and staging cleaner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b0bd5ea02162903073ad6d75153cfc9eb74bba20"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 22:50:23.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-24 14:04:52.000000000","tz":-480},"subject":"Test restarting glance for staging cleanup","message":"Test restarting glance for staging cleanup\n\nTesting this in reality is quite hard, so this adds a hacky post-run\nscript that should create some cruft in the staging directory,\nas well as stage an image and then restart glance so we can examine\nthe log files.\n\nChange-Id: I72a4372582adc142c965328fa82e8b2b388e5608\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cf76204b1b46dfd4e01004dc7eb36f1c1f627c9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cf76204b1b46dfd4e01004dc7eb36f1c1f627c9e"}]},"branch":"refs/heads/master"},"a267aacfc7cb37a7f9f00731a66b0dc64d0399a9":{"kind":"NO_CODE_CHANGE","_number":13,"created":"2021-02-24 17:32:51.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/777014/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/777014/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/777014/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/777014/13"}}},"commit":{"parents":[{"commit":"928597a3ae8caa20f8fd56fb54f9cde800d61496","subject":"Add housekeeping module and staging cleaner","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/928597a3ae8caa20f8fd56fb54f9cde800d61496"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 22:50:23.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-24 17:32:09.000000000","tz":-480},"subject":"MDNM: Test restarting glance for staging cleanup","message":"MDNM: Test restarting glance for staging cleanup\n\nTesting this in reality is quite hard, so this adds a hacky post-run\nscript that should create some cruft in the staging directory,\nas well as stage an image and then restart glance so we can examine\nthe log files.\n\nSince I got the functional tests able to do this like I want, I\u0027m\nnot sure we need to merge this, but I still want to try to get it\nto show the logs I\u0027m looking for, just for demonstration purposes.\n\nChange-Id: I72a4372582adc142c965328fa82e8b2b388e5608\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a267aacfc7cb37a7f9f00731a66b0dc64d0399a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a267aacfc7cb37a7f9f00731a66b0dc64d0399a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
