)]}'
{"id":"openstack%2Fglance~777018","triplet_id":"openstack%2Fglance~master~I50f88d0eeaa54d9857e8a7e30326bc4013928be6","project":"openstack/glance","branch":"master","topic":"feature_dist_import_func","hashtags":[],"change_id":"I50f88d0eeaa54d9857e8a7e30326bc4013928be6","subject":"WIP: Staging refactor","status":"ABANDONED","created":"2021-02-22 22:41:12.000000000","updated":"2021-03-01 16:52:55.000000000","total_comment_count":68,"unresolved_comment_count":35,"has_review_started":true,"meta_rev_id":"fa4aa4bde2865f1829d4d4e93bce27209dfcc617","_number":777018,"virtual_id_number":777018,"owner":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-02-26 16:19:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":-1,"date":"2021-03-01 14:44:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-22 23:59:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-23 12:48:36.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2021-02-23 13:11:43.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-02-23 22:39:48.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"}],"messages":[{"id":"f23184c68e998ae847e94d3ad64ba7144a565202","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-02-22 22:41:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1bb1e6ab69495fb76c30d3213563829a1b1a3f66","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-02-22 23:25:41.000000000","message":"Patch Set 1:\n\n(12 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f83b9524750698399f86450509a08b3f74803d44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 23:59:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/66853e8732644d1b9dd15fb54c77ad17 : SUCCESS in 58m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9e89dd9407ed4d92aa57b352c059810e : SUCCESS in 1h 05m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c8e1129811da42c192e6bf3d1a7461fa : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0004e586842b40bca59dc844188d9797 : SUCCESS in 4m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e77ef87eeda94b27acd901a0f4b80b8a : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34c5edf612534ef2aa13bbce6fca7df5 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24b03194c9f6414e90b68f5b91b6103b : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b0d80b20a904fa2a202aec48616c82b : SUCCESS in 6m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8774c44e7a2249c8b028065f0d652c0d : FAILURE in 8m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bcc642c1a5ec4b02a4d3e704b44161da : FAILURE in 7m 56s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/fd693fd558e340709bc5d32331ea375b : SUCCESS in 4m 50s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/66c7bba0a98f4548ac1900720b9aa0c2 : SUCCESS in 1h 08m 32s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/519d15d12fef4d30b40f9cd2ec48e1ee : SUCCESS in 1h 08m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a95e9a4d093e479aa2514423c93506cf : SUCCESS in 1h 08m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3365da5163de4119bcf0311807854b39 : SUCCESS in 1h 02m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12e9114869934cfa9ee358d96edf6bf7 : SUCCESS in 53m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0647dc1f6469411ebc96685a77e515b9 : FAILURE in 1h 15m 55s","accounts_in_message":[],"_revision_number":1},{"id":"d2a2b6311e776e7017c8905c3a14b32f5b4539a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-02-23 12:26:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"90c9783c202686c3e9aec2bbd615bad6368d7f82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-02-23 12:30:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9e75d8bd062ecec6775b71f9fd24ba70063f1ef4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-23 12:48:36.000000000","message":"Patch Set 3:\n\n(7 comments)\n\nAlso somehow it has regression on image_conversion_plugin; (need to check on master as well)\n\nIn case of single as well as multiple store configuration and image_conversion plugin is enabled then after successful import process it is not deleting .raw file from staging location","accounts_in_message":[],"_revision_number":3},{"id":"213fec5a4afcdd67f279be98cfa4b8256b1d4ab1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-23 12:51:59.000000000","message":"Patch Set 3:\n\n\u003e Patch Set 3:\n\u003e \n\u003e (7 comments)\n\u003e \n\u003e Also somehow it has regression on image_conversion_plugin; (need to check on master as well)\n\u003e \n\u003e In case of single as well as multiple store configuration and image_conversion plugin is enabled then after successful import process it is not deleting .raw file from staging location\n\nConfirmed, this is not the case on current master","accounts_in_message":[],"_revision_number":3},{"id":"d5e830402102b9bfc8bdfb3591ad4c46fc8c5b93","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-23 13:11:43.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/777018/777018/3/check-not-vote/ext-glance-zuul/3675d74 : SUCCESS in 39m 59s","accounts_in_message":[],"_revision_number":3},{"id":"f0c266eab59d84a1d59497932dba29441cddf63a","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-02-23 13:35:09.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"39ede897fa0f2a2f6f780ceb1370043f2c97d2b1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-23 14:14:21.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"afe5a10f8e3ba820391ec1ab8d30eb62ef7520d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 16:08:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/f89dfbaa13dc45fa9064bc56bbb63b5b : SUCCESS in 55m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5a84130ef3d54fd88fa1f9a65b563b12 : SUCCESS in 56m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ab65e584e37d40038b46428562543017 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4ff29af26c44a02ac14ce6301bf593d : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a32fe2ef25b14013b8e5bad68d7fc352 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a82cc0fd48a54b6984166ecef3dbb9c4 : SUCCESS in 6m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c47d020dd23e493884bd513e6b389184 : SUCCESS in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/759cdda6625440218bab7f94edf0732c : SUCCESS in 6m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d1cb3ef986284a1ea0b451fde129b797 : FAILURE in 8m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e7239490e1ac4eda901874bfaa1d170d : FAILURE in 8m 30s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f834153be60c4820bb17d7d727453e2f : SUCCESS in 5m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6e7b9455fcfa4dc6a5fa376f61b3a9e8 : SUCCESS in 1h 33m 16s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1de79e493d9b49278e78658984085572 : SUCCESS in 1h 08m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e9f6ae2621a0443983cb6a8e8a96a462 : SUCCESS in 1h 11m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/514726742da0498093cea3eb67f41135 : SUCCESS in 1h 31m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40bcc7238dd140b68ef415a426915909 : SUCCESS in 59m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/299075bdfb504f69ae0b160b087148a0 : SUCCESS in 1h 15m 15s","accounts_in_message":[],"_revision_number":3},{"id":"cf2c3ae5c854b4caa2d589182df6fc1d9563a364","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-02-23 18:30:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"63bc24a85a58578314a8fe529be637d49f2cffef","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-02-23 19:45:41.000000000","message":"Patch Set 4:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/777018/777018/4/check-not-vote/ext-glance-zuul/17fe4b3 : SUCCESS in 41m 05s","accounts_in_message":[],"_revision_number":4},{"id":"6b941fb86331ff141693eaf34e79f07464859635","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-23 22:39:48.000000000","message":"Patch Set 4: Code-Review-1\n\n(24 comments)\n\nFair disclosure, I only really got through the code and to the tests. There are a lot of them that are skipped because they\u0027re not working, so I didn\u0027t spend much time there. I\u0027ve skimmed the test modules that would cover the new stuff that is here, and it doesn\u0027t seem like there is much coverage yet of the new stuff being added here, but maybe that\u0027s because it\u0027s still WIP.\n\nThere is a ton of stuff going on in this patch, and it makes it hard to tell what test fallout comes from which change, so I\u0027d definitely like to see this broken into pieces we can review each piece and make sure each is tested. As noted inline, this conflicts with this bug fix series, which does these refactors to make the flows use the action wrapper:\n\nhttps://review.opendev.org/q/topic:%2522bug/1914826%2522\n\nSo we should just merge that. There are a lot of missing tests for the flows and surrounding code, and that series tackles them one at a time, adding and fixing test coverage as it goes. This patch will be a little smaller after that lands.\n\nOtherwise my concerns with what is here are largely the same as when we saw this squashed into my set, and I\u0027ve tried to elaborate on those inline. I definitely think that the unprotected use of the location metadata and the potential clash with an API user seems like a big problem, and we should have a tempest test to validate whatever mitigation scheme comes out of the review here.\n\nMy next largest concern with this is how much change it requires across all the layers, along with the tight integration this ends up building between the API and the low-level location storage. I think the amount of special casing that needs to happen at several places in the stack should be a huge red flag that using a store driver to manage staging is maybe not the best plan in general, and adding the locations like this make that even more obvious.\n\nI also looked to see what the conflict space is between this and my current patches, and it\u0027s quite small:\n\nhttps://pastebin.com/54CxF2TG\n\nLiterally just that you and I added a new method in the same place. Otherwise there\u0027s not really any conflict. I think we all agreed to see this fully parallel because there was some assertion that this work was so inextricably tied to master that doing it after my series was infeasible. From what I\u0027ve seen here, I don\u0027t think that\u0027s really the case. Now that you\u0027ve posted this and it\u0027s clear that there\u0027s (effectively) no conflict, any chance we could revisit seeing it proposed as a delta on top?","accounts_in_message":[],"_revision_number":4},{"id":"0988d8c494e64820b21cb880bb84b78b13752d87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 23:44:19.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/27104669d47744aa8b01402a6bc797f6 : SUCCESS in 59m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e63f91f4a8c245f8826b928d8f9545af : FAILURE in 1h 57m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c06f169ac2d244ff93d8c145240d4c3d : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/679caaaad4c34a57aed3809c1b55138a : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b3e09aa78f54a70851e76475ed61a99 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c80dc3b344d436dac7614064fa7af7a : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/048f4dc6ffa8412db33d72264cdd9005 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62ac9fcf6e864d4fab3a0de68073a57a : SUCCESS in 6m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/77bd495551394241a0a31e362fee202b : SUCCESS in 8m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8a9c1eefbdb44857bc67614ee8171b3e : SUCCESS in 9m 03s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/351ea6b0f0d748d19a2569cddc41a0d4 : SUCCESS in 5m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fe9e4a244e2d4180b0de9badfec107d4 : SUCCESS in 1h 19m 44s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/97ab73bee5a545a58014de610dec1e47 : SUCCESS in 1h 15m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2cbf8206c1854b3c86d0090e5b61671f : SUCCESS in 1h 18m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/caec5a7ede9a4218b6d736e92f613950 : SUCCESS in 1h 16m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6a285a165ef4c86b8755957f7e780df : SUCCESS in 1h 03m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/540255bbfecf4951bf515bcea7cd3d5a : SUCCESS in 1h 08m 24s","accounts_in_message":[],"_revision_number":4},{"id":"59a2b0006e2ed47fe8dbec25a5054d5aa37b60a5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-02-24 08:43:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c20e890f1cdb7d2cb42d5cee589cd99b5c41b288","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-02-24 13:48:18.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"537f25638e5102e17904c589277c827868e52da4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-02-26 14:27:48.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"fc36cafd5556e43e2239d3c0a21a291eeb0b5a90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 16:19:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/5d43eb9116804db590212b796ebdbed7 : SUCCESS in 1h 33m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c870e1c44441462686694a28508a9421 : SUCCESS in 1h 19m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/456eb5090f6f40c5a171da010ee95382 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3e32b304ac241d791a649fb2135db2a : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5aa76f39d934a8c8882506c18dc38c8 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/effcd6c7e778407bb0e86c021a3b2589 : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/092607c398a04820a5a7db93e4ef8be3 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fcaf497c8f948f3ae442fb74e224078 : SUCCESS in 6m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c6d3d8ddfd8a4d318871133d069fb76c : SUCCESS in 10m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ee11cc15c33e41d3bac81d263197bc6d : SUCCESS in 8m 50s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ebeb8784051d4dfe96a087f98c2d0014 : SUCCESS in 5m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/be0259ba10cd4a019e7541762ca9127a : SUCCESS in 1h 49m 39s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f89eca9563e041e9ad3eb2fb430b54aa : SUCCESS in 1h 13m 26s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/79c5e73926fb47148d2294c9dba8cf6d : SUCCESS in 1h 11m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c6448240b6664f9fa9f0c737bc7ae28d : SUCCESS in 1h 31m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50d40f5171144a73ac7cf765f259b59d : SUCCESS in 54m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d0b9a923b1474148a6fc47318a62cc2f : SUCCESS in 1h 19m 19s","accounts_in_message":[],"_revision_number":5},{"id":"48c35e310f30b3f59b5eb51471c10567676b1631","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-01 14:44:26.000000000","message":"Patch Set 5: Code-Review-1\n\nWith the exception of the big change to expand the database schema to move the storage out of the location metadata, it doesn\u0027t look like the rest of my comments were addressed or answered. Since this is still WIP and has test skips, I\u0027m just copying my -1 up to this PS, with the assumption that the other comments will be addressed or answered going forward.","accounts_in_message":[],"_revision_number":5},{"id":"849a4ef7c92eeda285620ab64c325d7e4e90be36","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-03-01 16:10:11.000000000","message":"Patch Set 5:\n\n(14 comments)\n\nshould have all Dan\u0027s concerns addressed. Sorry, most of these had the previous response as Draft. Should all reflect to the current situation in PS5.","accounts_in_message":[],"_revision_number":5},{"id":"fa4aa4bde2865f1829d4d4e93bce27209dfcc617","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-03-01 16:52:55.000000000","message":"Abandoned\n\nout of time","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c7bd3835992fd32985bf0236f3e4c00357c1284c","revisions":{"4548a5a39227b24a4ed59d318693428e735f0592":{"kind":"REWORK","_number":1,"created":"2021-02-22 22:41:12.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/18/777018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/777018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/777018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/777018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/777018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/777018/1"}}},"commit":{"parents":[{"commit":"ac53e433e23419defc745bb42b2419daca87adc5","subject":"Add get_ksa_client() helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ac53e433e23419defc745bb42b2419daca87adc5"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-02-22 22:11:49.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-02-22 22:40:14.000000000","tz":0},"subject":"WIP: Staging refactor","message":"WIP: Staging refactor\n\nThis change is to address old technical debt by refactoring the\nstaging used related to Interoperable Image Import to utilize\nlocations mechanisms to track the user staging data through the\nprocess.\n\nHandling of \"staging_host\" location metadata as preparation for\nDistributed Image Import tht needs to track the host where user\nimage data is being staged to avoid the shared filesystem between\nservice nodes.\n\nHiding the \"staging_host\" from API user and never selecting\nstaging loction as direct_url provided by the API.\n\nTo be addressed beofre merging:\nRemoving or refactoring tests skipped as \"Irrelevant?\", likely\nso due to the code they were testing been removed from said\nmodules.\nFixing tests skipped as \"BROKEN, needs eyes\" that are broken\nrelying the code they were testing being in the module and mocking\nthe rest.\nAdditional API testing for the insecure deprecated locations API\naround how the locations are being presented.\n\nChange-Id: I50f88d0eeaa54d9857e8a7e30326bc4013928be6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4548a5a39227b24a4ed59d318693428e735f0592"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4548a5a39227b24a4ed59d318693428e735f0592"}]},"branch":"refs/heads/master"},"9c555fae018fc926f7ff8bb96ffccd2324e44737":{"kind":"REWORK","_number":2,"created":"2021-02-23 12:26:57.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/18/777018/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/777018/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/777018/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/777018/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/777018/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/777018/2"}}},"commit":{"parents":[{"commit":"ac53e433e23419defc745bb42b2419daca87adc5","subject":"Add get_ksa_client() helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ac53e433e23419defc745bb42b2419daca87adc5"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-02-22 22:11:49.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-02-22 23:31:23.000000000","tz":0},"subject":"WIP: Staging refactor","message":"WIP: Staging refactor\n\nThis change is to address old technical debt by refactoring the\nstaging used related to Interoperable Image Import to utilize\nlocations mechanisms to track the user staging data through the\nprocess.\n\nHandling of \"staging_host\" location metadata as preparation for\nDistributed Image Import that needs to track the host where user\nimage data is being staged to avoid the shared filesystem between\nservice nodes.\n\nHiding the \"staging_host\" from API user and never selecting\nstaging loction as direct_url provided by the API.\n\nTo be addressed before merging:\nRemoving or refactoring tests skipped as \"Irrelevant?\", likely\nso due to the code they were testing been removed from said\nmodules.\nFixing tests skipped as \"BROKEN, needs eyes\" that are broken\nrelying the code they were testing being in the module and mocking\nthe rest.\nAdditional API testing for the insecure deprecated locations API\naround how the locations are being presented.\n\nChange-Id: I50f88d0eeaa54d9857e8a7e30326bc4013928be6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9c555fae018fc926f7ff8bb96ffccd2324e44737"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9c555fae018fc926f7ff8bb96ffccd2324e44737"}]},"branch":"refs/heads/master"},"ae34cb91795a9dc92c98c3871d30f13383809208":{"kind":"REWORK","_number":3,"created":"2021-02-23 12:30:40.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/18/777018/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/777018/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/777018/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/777018/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/777018/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/777018/3"}}},"commit":{"parents":[{"commit":"ac53e433e23419defc745bb42b2419daca87adc5","subject":"Add get_ksa_client() helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ac53e433e23419defc745bb42b2419daca87adc5"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-02-22 22:11:49.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-02-23 12:30:23.000000000","tz":0},"subject":"WIP: Staging refactor","message":"WIP: Staging refactor\n\nThis change is to address old technical debt by refactoring the\nstaging used related to Interoperable Image Import to utilize\nlocations mechanisms to track the user staging data through the\nprocess.\n\nHandling of \"staging_host\" location metadata as preparation for\nDistributed Image Import that needs to track the host where user\nimage data is being staged to avoid the shared filesystem between\nservice nodes.\n\nHiding the \"staging_host\" from API user and never selecting\nstaging loction as direct_url provided by the API.\n\nTo be addressed before merging:\nRemoving or refactoring tests skipped as \"Irrelevant?\", likely\nso due to the code they were testing been removed from said\nmodules.\nFixing tests skipped as \"BROKEN, needs eyes\" that are broken\nrelying the code they were testing being in the module and mocking\nthe rest.\nAdditional API testing for the insecure deprecated locations API\naround how the locations are being presented.\n\nChange-Id: I50f88d0eeaa54d9857e8a7e30326bc4013928be6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ae34cb91795a9dc92c98c3871d30f13383809208"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ae34cb91795a9dc92c98c3871d30f13383809208"}]},"branch":"refs/heads/master"},"a1ddbf303c1c4195739a90d014058d4ea7c87967":{"kind":"REWORK","_number":4,"created":"2021-02-23 18:30:46.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/18/777018/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/777018/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/777018/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/777018/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/777018/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/777018/4"}}},"commit":{"parents":[{"commit":"ac53e433e23419defc745bb42b2419daca87adc5","subject":"Add get_ksa_client() helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ac53e433e23419defc745bb42b2419daca87adc5"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-02-22 22:11:49.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-02-23 18:28:06.000000000","tz":0},"subject":"WIP: Staging refactor","message":"WIP: Staging refactor\n\nThis change is to address old technical debt by refactoring the\nstaging used related to Interoperable Image Import to utilize\nlocations mechanisms to track the user staging data through the\nprocess.\n\nHandling of \"staging_host\" location metadata as preparation for\nDistributed Image Import that needs to track the host where user\nimage data is being staged to avoid the shared filesystem between\nservice nodes.\n\nHiding the \"staging_host\" from API user and never selecting\nstaging loction as direct_url provided by the API.\n\nHiding the staging_host details from the output is tested on the\nfolowing change that implements the actual value for it.\n\nTo be addressed before merging:\nRemoving or refactoring tests skipped as \"Irrelevant?\", likely\nso due to the code they were testing been removed from said\nmodules.\nFixing tests skipped as \"BROKEN, needs eyes\" that are broken\nrelying the code they were testing being in the module and mocking\nthe rest.\n\nChange-Id: I50f88d0eeaa54d9857e8a7e30326bc4013928be6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a1ddbf303c1c4195739a90d014058d4ea7c87967"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a1ddbf303c1c4195739a90d014058d4ea7c87967"}]},"branch":"refs/heads/master"},"c7bd3835992fd32985bf0236f3e4c00357c1284c":{"kind":"REWORK","_number":5,"created":"2021-02-26 14:27:48.000000000","uploader":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"ref":"refs/changes/18/777018/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/777018/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/777018/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/777018/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/777018/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/777018/5"}}},"commit":{"parents":[{"commit":"7454e969323bc0b0c0bc07942fcddb184e7a0c63","subject":"Add get_ksa_client() helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7454e969323bc0b0c0bc07942fcddb184e7a0c63"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-02-25 00:02:23.000000000","tz":0},"committer":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2021-02-26 14:27:07.000000000","tz":0},"subject":"WIP: Staging refactor","message":"WIP: Staging refactor\n\nThis change is to address old technical debt by refactoring the\nstaging used related to Interoperable Image Import to utilize\nlocations mechanisms to track the user staging data through the\nprocess.\n\nHandling of \"staging_host\" location metadata as preparation for\nDistributed Image Import that needs to track the host where user\nimage data is being staged to avoid the shared filesystem between\nservice nodes.\n\nHiding the \"staging_host\" from API user and never selecting\nstaging loction as direct_url provided by the API.\n\nHiding the staging_host details from the output is tested on the\nfolowing change that implements the actual value for it.\n\nTo be addressed before merging:\nRemoving or refactoring tests skipped as \"Irrelevant?\", likely\nso due to the code they were testing been removed from said\nmodules.\nFixing tests skipped as \"BROKEN, needs eyes\" that are broken\nrelying the code they were testing being in the module and mocking\nthe rest.\n\nChange-Id: I50f88d0eeaa54d9857e8a7e30326bc4013928be6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c7bd3835992fd32985bf0236f3e4c00357c1284c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c7bd3835992fd32985bf0236f3e4c00357c1284c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
