)]}'
{"id":"openstack%2Fglance~778072","triplet_id":"openstack%2Fglance~master~I726abe5d1104510e8da0e94f90f2b36d43b82cbe","project":"openstack/glance","branch":"master","topic":"bp/distributed-image-import","hashtags":[],"change_id":"I726abe5d1104510e8da0e94f90f2b36d43b82cbe","subject":"Add administrator docs for distributed-import","status":"MERGED","created":"2021-03-01 21:11:31.000000000","updated":"2021-03-04 00:49:14.000000000","submitted":"2021-03-04 00:46:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"778072-bp/distributed-image-import","meta_rev_id":"82fb5e979eeb5b2f2f950e46de57a9c70a59be98","_number":778072,"virtual_id_number":778072,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-04 00:46:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-03-03 17:12:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-03-03 17:21:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-03-03 17:21:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-03-04 00:49:14.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-01 21:12:08.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-03-01 21:23:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-01 21:28:55.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-03 17:12:03.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"5b5f4b40e3601965590398c6074e5a8fe73145cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-01 21:11:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"526c179f212863d47f158634cce0ba98f1f8942e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 21:23:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4252a71a4c44c459f54d1d0a32c793d : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c23cffd75cfd4ef2b53a2321854b985c : SUCCESS in 6m 21s","accounts_in_message":[],"_revision_number":1},{"id":"949b34cc39a613ca900fcf2e45578e3a51d09b7d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-01 21:28:55.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/778072/778072/1/check-not-vote/ext-glance-zuul/a1a90f0 : FAILURE in 16m 05s","accounts_in_message":[],"_revision_number":1},{"id":"cf78ef4d2194d7add619379828cdcd07fa6b6ba1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-02 04:35:23.000000000","message":"Patch Set 1: Review-Priority+1\n\n(1 comment)\n\nOne question inline, else looks perfect to me.","accounts_in_message":[],"_revision_number":1},{"id":"1056c45c607bc77deb1ecb1e17e22a5272615983","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-02 14:26:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"87bedd92411171762772feeafb545e507aa27b0e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-02 14:34:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0292be4c6c4367c469d01562b6b2279c6d16eba1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-02 14:45:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a400eb4a65e46a0d766acfab93af629317f2731a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 15:00:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/caac09f5293a4079a2dcbc7968509891 : SUCCESS in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25d30170ac7b44b3a2adbf74df351f11 : SUCCESS in 8m 16s","accounts_in_message":[],"_revision_number":2},{"id":"bd28c6c955b8b284a898388b024bff3f53a671b7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-02 17:05:01.000000000","message":"Patch Set 2: Code-Review+2\n\nHappy with the documentation, thank you dan!!","accounts_in_message":[],"_revision_number":2},{"id":"5d23c7ebff811536e057c1adb030e706ca55514f","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-02 17:05:55.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/778072/778072/2/check-not-vote/ext-glance-zuul/db99ca1 : FAILURE in 15m 52s","accounts_in_message":[],"_revision_number":2},{"id":"4e4f05f819ea71c4f3a44fb58d30d7cf5c69c9b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-02 19:53:18.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"117aa04c08fdf1df5b2cdb2ec773b1153c9c2a37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 20:07:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a496cd5c2324a82bfd4a4e00cb8a93f : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21f2bdffc8634f8b92151ce924a5aa0f : SUCCESS in 6m 51s","accounts_in_message":[],"_revision_number":3},{"id":"98814c32adb17d482f89c69a92d9038651143449","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-03 05:06:40.000000000","message":"Patch Set 3: Review-Priority+1\n\nI think you should rebase this patch with actual implementation patch and keep https://review.opendev.org/c/openstack/glance/+/770629/9 (Enable second glance worker for import testing) patch on top as it has dependency on devstack + tempest. So once actual functionality patch is approved this doc patch need not be wait until the test related patch is merged.\n\nAND if you decides to do what I suggested, could you please add release note in this doc patch?\n\nThank you","accounts_in_message":[],"_revision_number":3},{"id":"9bf7206d161bd87d043b3d343bb01b01897ba2d3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-03 14:25:02.000000000","message":"Patch Set 3:\n\n\u003e I think you should rebase this patch with actual implementation patch and keep https://review.opendev.org/c/openstack/glance/+/770629/9 (Enable second glance worker for import testing) patch on top as it has dependency on devstack + tempest. So once actual functionality patch is approved this doc patch need not be wait until the test related patch is merged.\n\u003e \n\u003e AND if you decides to do what I suggested, could you please add release note in this doc patch?\n\u003e \n\u003e Thank you\n\nAck, sure, will do that.","accounts_in_message":[],"_revision_number":3},{"id":"94bbfaa5524bfd1ddede106d816e198ba26982c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-03 14:37:56.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c64e8810605bb2916fd2ff70a17ca1936a857f6d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-03 14:53:21.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/778072/778072/4/check-not-vote/ext-glance-zuul/30ef99d : FAILURE in 14m 18s","accounts_in_message":[],"_revision_number":4},{"id":"b9c241a091318aa2da6f6ff44e9a40c14f1c1825","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-03-03 17:12:03.000000000","message":"Patch Set 4: Code-Review+2\n\nDocs look great.","accounts_in_message":[],"_revision_number":4},{"id":"2df1631bc01473d5237450188a8068454243d9e2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-03 17:21:42.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nThank you for reshuffling and release note!!","accounts_in_message":[],"_revision_number":4},{"id":"1cd706761bd3ad567681a38d73b3c6dca9f3aeed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 18:36:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a894ba9daa24854a4cbc871998c669b : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4335c47844db4982a1cbf2a8b120c19a : SUCCESS in 7m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19c1a1fe5b1442b3a0b450d268baa18a : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":4},{"id":"7f22b5cbc7d4619e70fca2f3a178e34428d88bfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 20:02:18.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"e10f30dadc5e39d2918c33525df2d76c21c47510","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 00:46:36.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad8a30f30dae4652a0a8d29137a76283 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34178aad536f4510be8bb852c1d1cb12 : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fb3ebabb52054d099d5d8d2889eea109 : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":4},{"id":"1df4beb58aa9676a6b1803b22ed9198f7be042bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 00:46:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"36519e09f2b0901efbe7e2402005ac30dea2b5c1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 00:49:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9cda96d577c48809c0f11b5894222f6 : SUCCESS in 1m 05s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4aaf133c4f840a386e34e5a41117343 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d8a6309893d249623a31ecec88bb373fa08dfd3f","revisions":{"bb25a065ed9504fb26b04c181b6c92f6e78e77d6":{"kind":"REWORK","_number":1,"created":"2021-03-01 21:11:31.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/72/778072/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/778072/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/778072/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/778072/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/778072/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/778072/1"}}},"commit":{"parents":[{"commit":"cd49b3f4418db8e873466aef41bf96bb83aec0d7","subject":"Enable second glance worker for import testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cd49b3f4418db8e873466aef41bf96bb83aec0d7"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-01 21:09:30.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-01 21:09:30.000000000","tz":-480},"subject":"Add administrator docs for distributed-import","message":"Add administrator docs for distributed-import\n\nThis adds some text to the documentation about configuring the import\nmechanism, including details about shared vs. local staging\ndirectories. It also clarifies that *all* import methods require the\nstaging directory to be configured, as well as cleans up some\nsingle-store-specific wording in this area.\n\nRelated to blueprint distributed-image-import\n\nChange-Id: I726abe5d1104510e8da0e94f90f2b36d43b82cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bb25a065ed9504fb26b04c181b6c92f6e78e77d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bb25a065ed9504fb26b04c181b6c92f6e78e77d6"}]},"branch":"refs/heads/master"},"6975eaf4198abda2e8ebf8bdd6d6300cfe51ed62":{"kind":"REWORK","_number":2,"created":"2021-03-02 14:45:19.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/72/778072/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/778072/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/778072/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/778072/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/778072/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/778072/2"}}},"commit":{"parents":[{"commit":"cd49b3f4418db8e873466aef41bf96bb83aec0d7","subject":"Enable second glance worker for import testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cd49b3f4418db8e873466aef41bf96bb83aec0d7"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-01 21:09:30.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-02 14:45:15.000000000","tz":-480},"subject":"Add administrator docs for distributed-import","message":"Add administrator docs for distributed-import\n\nThis adds some text to the documentation about configuring the import\nmechanism, including details about shared vs. local staging\ndirectories. It also clarifies that *all* import methods require the\nstaging directory to be configured, as well as cleans up some\nsingle-store-specific wording in this area.\n\nRelated to blueprint distributed-image-import\n\nChange-Id: I726abe5d1104510e8da0e94f90f2b36d43b82cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6975eaf4198abda2e8ebf8bdd6d6300cfe51ed62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6975eaf4198abda2e8ebf8bdd6d6300cfe51ed62"}]},"branch":"refs/heads/master"},"e1d78f8fbcc414899d7b4cdf871a58511e4914de":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-03-02 19:53:18.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/72/778072/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/778072/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/778072/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/778072/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/778072/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/778072/3"}}},"commit":{"parents":[{"commit":"ac9630c79fd667d0c7d1fef5dfe710c57940ef59","subject":"Enable second glance worker for import testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ac9630c79fd667d0c7d1fef5dfe710c57940ef59"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-01 21:09:30.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-02 19:52:13.000000000","tz":-480},"subject":"Add administrator docs for distributed-import","message":"Add administrator docs for distributed-import\n\nThis adds some text to the documentation about configuring the import\nmechanism, including details about shared vs. local staging\ndirectories. It also clarifies that *all* import methods require the\nstaging directory to be configured, as well as cleans up some\nsingle-store-specific wording in this area.\n\nRelated to blueprint distributed-image-import\n\nChange-Id: I726abe5d1104510e8da0e94f90f2b36d43b82cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e1d78f8fbcc414899d7b4cdf871a58511e4914de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e1d78f8fbcc414899d7b4cdf871a58511e4914de"}]},"branch":"refs/heads/master"},"d8a6309893d249623a31ecec88bb373fa08dfd3f":{"kind":"REWORK","_number":4,"created":"2021-03-03 14:37:56.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/72/778072/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/72/778072/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/778072/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/778072/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/72/778072/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/72/778072/4"}}},"commit":{"parents":[{"commit":"41e1cecbe63c778ce8e92519993c61588ea1f0cb","subject":"Distributed image import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/41e1cecbe63c778ce8e92519993c61588ea1f0cb"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-01 21:09:30.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-03 14:37:29.000000000","tz":-480},"subject":"Add administrator docs for distributed-import","message":"Add administrator docs for distributed-import\n\nThis adds some text to the documentation about configuring the import\nmechanism, including details about shared vs. local staging\ndirectories. It also clarifies that *all* import methods require the\nstaging directory to be configured, as well as cleans up some\nsingle-store-specific wording in this area.\n\nRelated to blueprint distributed-image-import\n\nChange-Id: I726abe5d1104510e8da0e94f90f2b36d43b82cbe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d8a6309893d249623a31ecec88bb373fa08dfd3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d8a6309893d249623a31ecec88bb373fa08dfd3f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
