)]}'
{"id":"openstack%2Fglance~778736","triplet_id":"openstack%2Fglance~master~Iaa4ec2a3331bdca276a6e8f622701c9ce860b001","project":"openstack/glance","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"Iaa4ec2a3331bdca276a6e8f622701c9ce860b001","subject":"DNM: Check some jobs with new policy defaults","status":"ABANDONED","created":"2021-03-04 15:49:09.000000000","updated":"2022-05-05 14:21:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b306ae4d3a2f06d7faf7a8d1ad731f33f021459b","_number":778736,"virtual_id_number":778736,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-03-07 17:47:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-04 15:50:32.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2021-03-04 17:20:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-04 17:31:24.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-03-05 08:52:36.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"3eead84d9d08779ac37b4f4392f77260a7fa04aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-04 15:49:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5dbcd5ea24f809427a7ee6df162b19334eccfe9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-04 16:12:45.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6897ff9dafabc1213f4d545d639be1c2b9ddd68a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 17:20:09.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"b711d2602d6916e60179d7cae08401cf0ec6a6c1","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-04 17:26:17.000000000","message":"Patch Set 2:\n\nDan, do you need to put this on top of https://review.opendev.org/c/openstack/glance/+/778258/4 ?","accounts_in_message":[],"_revision_number":2},{"id":"ddb3dbe00c7c32d5a9cb97b419197442830953b0","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-04 18:06:20.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e Dan, do you need to put this on top of https://review.opendev.org/c/openstack/glance/+/778258/4 ?\n\nIt already is?","accounts_in_message":[],"_revision_number":2},{"id":"bdacd73c8daa29d1dfaaeffea77e1b3142f016ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-04 18:15:31.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"a7fc5688e86668294d619088b54b7a66edcdade1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 20:33:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d8bbb18a86b46b4a3607a3010c59f62 : SUCCESS in 10m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3aaa4158fbcb46eebfcc4cdfd3f92d2d : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6943a90f67f4db5860a377cdb2a9193 : SUCCESS in 8m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b140547b132467787493a7dc5188605 : SUCCESS in 9m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/476c8b52d7d844aeb10abeaa833d72bd : SUCCESS in 8m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0b31ba4bff2403da754e7157fcbee22 : SUCCESS in 10m 51s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9ae7c912f2d94a24a2e73079014ccdf5 : SUCCESS in 10m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/616f45e02c57421e82c2824c69e76cd6 : SUCCESS in 13m 07s\n- nova-ceph-multistore-rbac https://zuul.opendev.org/t/openstack/build/5642b55feb7f44df84a91b3a1b2e2a22 : FAILURE in 58m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c8e6f80577dc44e6933d77003888213a : SUCCESS in 31m 26s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b74bbce4a3884392bbf3b2c5a8a45426 : SUCCESS in 38m 37s","accounts_in_message":[],"_revision_number":3},{"id":"47efa6172dec2b3bf3e18e525fc44ba623b52534","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-04 22:49:22.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"a63a692a7f176c4690f5263ffaff7259995c71d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 23:58:08.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0d2bc4cbaf394d4f86ae7c4a4d8ce075 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/817ff9de78ca4a18bf377880135845b0 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64018ef85bfa431b9c0e5cef675989ae : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/671245cd54fe47cbb214b2c8bc86e47b : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/062577dbdb964634a3ef9ff35e155ce3 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/857d284e9ec749469ffe136ae0389cfb : SUCCESS in 6m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1a12d4b0f8a7436d8e82638a574d47b8 : SUCCESS in 10m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c0d02488004d476b8347d536cb741c15 : SUCCESS in 10m 37s\n- nova-ceph-multistore-rbac https://zuul.opendev.org/t/openstack/build/b0eeefc0fb334cebbf1d7900f16fef06 : FAILURE in 1h 06m 35s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0478c688001a45619e528a98023ddeb0 : SUCCESS in 36m 45s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/92e37c6558ac4c869025ae29d554344f : SUCCESS in 25m 56s","accounts_in_message":[],"_revision_number":4},{"id":"a26f333fc3ccbd3be63273cfd159ae4399841db7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-04 23:58:27.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3a1bbca25ae12c3e1d53725ee8b38a63392e5e8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 01:17:48.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1426c600cb284cc58e6ccced027268c7 : FAILURE in 9m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e03d4829bbfc46fea6642b725745efe3 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a61612a44ca246ab8c414924373ddd3e : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/085f5fe68a104bd8b3a090fb43e7db6d : SUCCESS in 8m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65f08d0eece649d2b1c4298e6225b3f3 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7aa4077121cc431d84033e7e9a41b7be : SUCCESS in 5m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/223b28befcdb451ba3db180843b019a0 : SUCCESS in 8m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/28f3bd18927845aa8c7741c089bd64b4 : SUCCESS in 8m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/aa711ae6129a4a4783cfa3ca2422dd09 : FAILURE in 33m 03s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/78dbdd62ebe343d5a00187ac4abfcf23 : FAILURE in 1h 17m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3ae2021fadec41149e7d5ef86acdce02 : FAILURE in 1h 01m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b3e3c256cf28494ab57e8a6f5de9ca96 : FAILURE in 1h 13m 54s (non-voting)\n- nova-ceph-multistore-rbac https://zuul.opendev.org/t/openstack/build/2829259fe337401a9be7cebe75b4eef0 : FAILURE in 45m 26s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/81c1e87e5d2d4ca6bd461ce32e9f9436 : SUCCESS in 33m 14s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/64f9a67ba808438b9deb3fda8627eebf : SUCCESS in 27m 23s","accounts_in_message":[],"_revision_number":5},{"id":"fe14cbbc1f72704245e75a972fee56eb486adba1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-05 08:52:36.000000000","message":"Patch Set 5:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/778736/778736/5/check-not-vote/ext-glance-zuul/4a3dd67 : SUCCESS in 38m 06s","accounts_in_message":[],"_revision_number":5},{"id":"25eb8e25a85d2326c5e0d77f5462f522b7b32303","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-05 13:37:45.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"f88866fdefc290da8802f1f13aea99a44b9929ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-05 17:18:30.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"75fdfadd00e80be6f5c3f4a67e57183abeeccc9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-05 18:06:01.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"67f3234a0d378aa600e69053adb0f11565e49639","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-05 20:15:09.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"b7659bfcfea3a0337847760574abb7df871512bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-05 20:34:13.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"b96c260b941edded556f57cbb0422bc5c5e13b9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-05 20:40:24.000000000","message":"Uploaded patch set 11: New patch set was added with same tree, parent, and commit message as Patch Set 10.","accounts_in_message":[],"_revision_number":11},{"id":"a4ade3cb19ef7dcb4e755dfd23d6f312a2762c85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-05 20:51:55.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"105b95ce48f58d31e022faaadf78afb028a11c36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 02:05:36.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0412b2e096a142c28521309a14788061 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77f46841c92e440d99df7964c79bfb52 : FAILURE in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dccd7e34d0d346db87d576fe1db9acaa : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce1a1041b9114faf8900a4edd6bd50bd : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba91fa14caa04a718e47d5f2df803e2f : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0975533c33594c5ebb4c8bb4f9d46a72 : SUCCESS in 7m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c9057cfbfb084006b26a9a02c25414bb : SUCCESS in 8m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/63a49411fcf74426ab4f99b43dfa3eeb : SUCCESS in 8m 27s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/63869b7f487f4e858311daf2340168e9 : FAILURE in 37m 27s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/252bd5c589f04e7b95e13c3599abdafe : SUCCESS in 1h 07m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/315945354ab3469d81df2862cf1a0189 : SUCCESS in 1h 09m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/927a90c7f69b48a7afd0222c5efe5227 : SUCCESS in 1h 19m 38s (non-voting)\n- nova-ceph-multistore-rbac https://zuul.opendev.org/t/openstack/build/eae99a4bf29443eda2b2c50e82bd3cd9 : SUCCESS in 1h 43m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2ea438073c3145569a69b6cb03f8a08e : SUCCESS in 32m 43s\n\nWarning:\n  Comments left for invalid file glance/policies/tasks.py","accounts_in_message":[],"_revision_number":12},{"id":"35276a9cce78321d83a93fd04b44051a4d350606","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-06 04:03:13.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"420b6bc86cd1be192369defb64aa27703208134a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 05:24:08.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2799715ce4e546688833ab36a648f23e : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3bcb96aa38e4d37ac837113f9b1d080 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d7b4f10ef7874c6eb9c95132e12fe32e : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3894256a01eb440883b1e212a92a181d : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61845f45c59342d4b5848a68c95315fe : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61805ff7bc07406d8a6716a74b769619 : SUCCESS in 7m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/17b31339cacb49469db62174cf0d14e0 : SUCCESS in 9m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d2a84db2ce944806b5eb1c274b8379db : SUCCESS in 8m 39s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/488ab79787324785904317a8bde3d466 : FAILURE in 25m 21s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b02b3fde85a2462e8a677b2d1d1eedb4 : SUCCESS in 59m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4ff973c8f05e4062a6d44e17480e5fd4 : SUCCESS in 1h 02m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f0e97495992a4bb785bac046f953f8de : FAILURE in 50m 50s (non-voting)\n- nova-ceph-multistore-rbac https://zuul.opendev.org/t/openstack/build/dae376e2b4374b3e9a3c8b117346ecd4 : SUCCESS in 1h 17m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/315142d9fba74199901d9ce7615cdcd9 : SUCCESS in 40m 06s","accounts_in_message":[],"_revision_number":13},{"id":"5f9b11fa074dc375c94baba1025979b8b3398772","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-06 15:25:29.000000000","message":"Patch Set 13:\n\nNow that these are green, do we want to keep a variant that ensures we test this in some integrated fashion?\n\nOTOH, I want to remain mindful of gate resources.\n\nThoughts?","accounts_in_message":[],"_revision_number":13},{"id":"8be126ee9e2cac82b9c1776762ebdf640f5258fc","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-06 16:05:07.000000000","message":"Patch Set 13:\n\n\u003e Patch Set 13:\n\u003e \n\u003e Now that these are green, do we want to keep a variant that ensures we test this in some integrated fashion?\n\u003e \n\u003e OTOH, I want to remain mindful of gate resources.\n\u003e \n\u003e Thoughts?\n\nYes, I was thinking about this last night. We could maybe we use the nova-ceph-multistore job for that, since it will get a *lot* of the surface area in one go. The only problem with that is that we might not notice if we break something with the default config, which will will break nova and not us. Not sure how much of a concern that is once this gets settled though.\n\nOtherwise we could just pick one our -import jobs, which have a lot of import plugins configured. The -import-standalone maybe, since that is a leaf of the -import job and we can tweak its config without affecting anything else.","accounts_in_message":[],"_revision_number":13},{"id":"5c79788eca0628f7fdba3341d298f391c6e1fc05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-07 16:15:14.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"3c3f8a8f56ac51124bc67b158b65207b60ec82ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-07 16:32:13.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"d78078f8a1a0597465c51f42a2c9915bf450ecc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-07 17:47:15.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b7dd1d77a754a17bb83f3e635138598 : FAILURE in 9m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7da9cb19b3745a6aed2188bff96f82d : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/762de7f9848343028ea3d7e9f799f520 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5363e13234f04b60b58b329f0994fe9c : SUCCESS in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f90bb757e43b4d2fad2ec23f2ef5b4e1 : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa4c37f3e21b418488de6b5fa6657b0c : SUCCESS in 6m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/958e2c0b107e42e39f143e4469c37b98 : SUCCESS in 9m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/15c2c48bdd124b36b9e4327a686488c1 : SUCCESS in 9m 00s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/92f56c6ce52c4a1496a3b6e85c74ce2a : FAILURE in 41m 20s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3360c710fe89422c82c851c5f1f71b8e : SUCCESS in 1h 09m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/886ef6a8ecfe4c54bf68b83f47689192 : SUCCESS in 1h 04m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a2e723efe2674ba5916c93ae94d53401 : SUCCESS in 57m 03s (non-voting)\n- nova-ceph-multistore-rbac https://zuul.opendev.org/t/openstack/build/60a7e29d2dcf47d8a6f45b0d3491e98c : SUCCESS in 56m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/246203382f46449781587e74108d42bd : SUCCESS in 32m 46s","accounts_in_message":[],"_revision_number":15},{"id":"b306ae4d3a2f06d7faf7a8d1ad731f33f021459b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-05-05 14:21:17.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"6e7242a802fee7380a1b904deb9be475040361b2","revisions":{"ae7a1a62455d40206f2166300ac6409dca34d62c":{"kind":"REWORK","_number":1,"created":"2021-03-04 15:49:09.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/36/778736/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/778736/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/778736/1"}}},"commit":{"parents":[{"commit":"366b79b99287815f4d2dc1a3631900fca4dd173b","subject":"DNM: Make sure protection tests protect against policy regressions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/366b79b99287815f4d2dc1a3631900fca4dd173b"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-04 15:48:38.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-04 15:48:38.000000000","tz":-480},"subject":"DNM: Check nova-ceph-glance with new policy defs","message":"DNM: Check nova-ceph-glance with new policy defs\n\nChange-Id: Iaa4ec2a3331bdca276a6e8f622701c9ce860b001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ae7a1a62455d40206f2166300ac6409dca34d62c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ae7a1a62455d40206f2166300ac6409dca34d62c"}]},"branch":"refs/heads/master"},"45b3f4fb71813a73922b7774644eca3906921d2f":{"kind":"REWORK","_number":2,"created":"2021-03-04 16:12:45.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/36/778736/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/778736/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/778736/2"}}},"commit":{"parents":[{"commit":"366b79b99287815f4d2dc1a3631900fca4dd173b","subject":"DNM: Make sure protection tests protect against policy regressions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/366b79b99287815f4d2dc1a3631900fca4dd173b"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-04 15:48:38.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-04 16:12:38.000000000","tz":-480},"subject":"DNM: Check nova-ceph-glance with new policy defs","message":"DNM: Check nova-ceph-glance with new policy defs\n\nChange-Id: Iaa4ec2a3331bdca276a6e8f622701c9ce860b001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/45b3f4fb71813a73922b7774644eca3906921d2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/45b3f4fb71813a73922b7774644eca3906921d2f"}]},"branch":"refs/heads/master"},"1f570bc1e28a7fd6320565061690c700e2178952":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-03-04 18:15:31.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/36/778736/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/778736/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/778736/3"}}},"commit":{"parents":[{"commit":"8ce16aa30a11f0b27f8536e4283f62277e802ccf","subject":"Update default policies for task API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8ce16aa30a11f0b27f8536e4283f62277e802ccf"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-04 15:48:38.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-04 18:15:26.000000000","tz":-480},"subject":"DNM: Check nova-ceph-glance with new policy defs","message":"DNM: Check nova-ceph-glance with new policy defs\n\nChange-Id: Iaa4ec2a3331bdca276a6e8f622701c9ce860b001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1f570bc1e28a7fd6320565061690c700e2178952"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1f570bc1e28a7fd6320565061690c700e2178952"}]},"branch":"refs/heads/master"},"301dabfe599b6f9e86440b15d2d0f2f77973357a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-03-04 22:49:22.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/36/778736/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/778736/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/778736/4"}}},"commit":{"parents":[{"commit":"906b28376fb749ad2ab07537675a84f14fb0f157","subject":"Update default policies for task API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/906b28376fb749ad2ab07537675a84f14fb0f157"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-04 15:48:38.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-04 22:49:11.000000000","tz":-480},"subject":"DNM: Check nova-ceph-glance with new policy defs","message":"DNM: Check nova-ceph-glance with new policy defs\n\nChange-Id: Iaa4ec2a3331bdca276a6e8f622701c9ce860b001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/301dabfe599b6f9e86440b15d2d0f2f77973357a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/301dabfe599b6f9e86440b15d2d0f2f77973357a"}]},"branch":"refs/heads/master"},"8dbde7afda0a09d49df4c9cb8bc932aeba94535e":{"kind":"REWORK","_number":5,"created":"2021-03-04 23:58:27.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/36/778736/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/778736/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/778736/5"}}},"commit":{"parents":[{"commit":"906b28376fb749ad2ab07537675a84f14fb0f157","subject":"Update default policies for task API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/906b28376fb749ad2ab07537675a84f14fb0f157"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-04 15:48:38.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-04 23:36:35.000000000","tz":-480},"subject":"DNM: Check some jobs with new policy defaults","message":"DNM: Check some jobs with new policy defaults\n\nThis makes a bunch of our jobs use the new policy defaults so we can\nsee what breaks if an admin turns this on.\n\nChange-Id: Iaa4ec2a3331bdca276a6e8f622701c9ce860b001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8dbde7afda0a09d49df4c9cb8bc932aeba94535e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8dbde7afda0a09d49df4c9cb8bc932aeba94535e"}]},"branch":"refs/heads/master"},"849ffd9dae00be4501e8a0c1b70b8df34e2f9894":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-03-05 13:37:45.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/36/778736/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/778736/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/778736/6"}}},"commit":{"parents":[{"commit":"f9591501e16bc775cf21a9708f1acece54af0f00","subject":"Make secure RBAC protection job voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f9591501e16bc775cf21a9708f1acece54af0f00"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-04 15:48:38.000000000","tz":-480},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-03-05 13:35:08.000000000","tz":0},"subject":"DNM: Check some jobs with new policy defaults","message":"DNM: Check some jobs with new policy defaults\n\nThis makes a bunch of our jobs use the new policy defaults so we can\nsee what breaks if an admin turns this on.\n\nChange-Id: Iaa4ec2a3331bdca276a6e8f622701c9ce860b001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/849ffd9dae00be4501e8a0c1b70b8df34e2f9894"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/849ffd9dae00be4501e8a0c1b70b8df34e2f9894"}]},"branch":"refs/heads/master"},"91197838826b02ef1c0e36bca30715a268ce3fbd":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-03-05 17:18:30.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/36/778736/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/778736/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/778736/7"}}},"commit":{"parents":[{"commit":"1d2afb0546f9805f951fc2371ccb52526900d421","subject":"Make secure RBAC protection job voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1d2afb0546f9805f951fc2371ccb52526900d421"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-04 15:48:38.000000000","tz":-480},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-03-05 17:16:19.000000000","tz":0},"subject":"DNM: Check some jobs with new policy defaults","message":"DNM: Check some jobs with new policy defaults\n\nThis makes a bunch of our jobs use the new policy defaults so we can\nsee what breaks if an admin turns this on.\n\nChange-Id: Iaa4ec2a3331bdca276a6e8f622701c9ce860b001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/91197838826b02ef1c0e36bca30715a268ce3fbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/91197838826b02ef1c0e36bca30715a268ce3fbd"}]},"branch":"refs/heads/master"},"76b9ab4786acdf79e122f00ae572f7cee84e1890":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-03-05 18:06:01.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/36/778736/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/778736/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/778736/8"}}},"commit":{"parents":[{"commit":"148d327d174bdb58c6f042b90ea6c32e940d8c1f","subject":"Make secure RBAC protection job voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/148d327d174bdb58c6f042b90ea6c32e940d8c1f"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-04 15:48:38.000000000","tz":-480},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-03-05 18:03:36.000000000","tz":0},"subject":"DNM: Check some jobs with new policy defaults","message":"DNM: Check some jobs with new policy defaults\n\nThis makes a bunch of our jobs use the new policy defaults so we can\nsee what breaks if an admin turns this on.\n\nChange-Id: Iaa4ec2a3331bdca276a6e8f622701c9ce860b001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/76b9ab4786acdf79e122f00ae572f7cee84e1890"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/76b9ab4786acdf79e122f00ae572f7cee84e1890"}]},"branch":"refs/heads/master"},"2100cecf32073e5e1fe2d36f0bb29e5449b29e22":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-03-05 20:15:09.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/36/778736/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/778736/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/778736/9"}}},"commit":{"parents":[{"commit":"3fea15c7e457f4f8cf92f9c10c5167ee925133f8","subject":"Make secure RBAC protection job voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3fea15c7e457f4f8cf92f9c10c5167ee925133f8"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-04 15:48:38.000000000","tz":-480},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-03-05 20:13:02.000000000","tz":0},"subject":"DNM: Check some jobs with new policy defaults","message":"DNM: Check some jobs with new policy defaults\n\nThis makes a bunch of our jobs use the new policy defaults so we can\nsee what breaks if an admin turns this on.\n\nChange-Id: Iaa4ec2a3331bdca276a6e8f622701c9ce860b001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2100cecf32073e5e1fe2d36f0bb29e5449b29e22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2100cecf32073e5e1fe2d36f0bb29e5449b29e22"}]},"branch":"refs/heads/master"},"dd80e87739a10e8c327ed92ae28ac17b594ca59f":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2021-03-05 20:34:13.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/36/778736/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/778736/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/778736/10"}}},"commit":{"parents":[{"commit":"cd77e0d53433f5241cd28a8fd68b68f51de7535c","subject":"Make secure RBAC protection job voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cd77e0d53433f5241cd28a8fd68b68f51de7535c"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-04 15:48:38.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-05 20:34:12.000000000","tz":-480},"subject":"DNM: Check some jobs with new policy defaults","message":"DNM: Check some jobs with new policy defaults\n\nThis makes a bunch of our jobs use the new policy defaults so we can\nsee what breaks if an admin turns this on.\n\nChange-Id: Iaa4ec2a3331bdca276a6e8f622701c9ce860b001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dd80e87739a10e8c327ed92ae28ac17b594ca59f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dd80e87739a10e8c327ed92ae28ac17b594ca59f"}]},"branch":"refs/heads/master"},"9208a330048866e5b7187a581efec8c2d9ed3a06":{"kind":"NO_CHANGE","_number":11,"created":"2021-03-05 20:40:24.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/36/778736/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/778736/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/778736/11"}}},"commit":{"parents":[{"commit":"40946152c3c30b1380d5cd16412a3d15ccd06f72","subject":"Make secure RBAC protection job voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/40946152c3c30b1380d5cd16412a3d15ccd06f72"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-04 15:48:38.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-05 20:40:08.000000000","tz":-480},"subject":"DNM: Check some jobs with new policy defaults","message":"DNM: Check some jobs with new policy defaults\n\nThis makes a bunch of our jobs use the new policy defaults so we can\nsee what breaks if an admin turns this on.\n\nChange-Id: Iaa4ec2a3331bdca276a6e8f622701c9ce860b001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9208a330048866e5b7187a581efec8c2d9ed3a06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9208a330048866e5b7187a581efec8c2d9ed3a06"}]},"branch":"refs/heads/master"},"209a72371b656aff736ed588dc3f724e01a656d3":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2021-03-05 20:51:55.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/36/778736/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/778736/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/778736/12"}}},"commit":{"parents":[{"commit":"b94028a112e50fe3e18b7677e3e8ca77f7c6a6d6","subject":"Make secure RBAC protection job voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b94028a112e50fe3e18b7677e3e8ca77f7c6a6d6"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-04 15:48:38.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-05 20:51:54.000000000","tz":-480},"subject":"DNM: Check some jobs with new policy defaults","message":"DNM: Check some jobs with new policy defaults\n\nThis makes a bunch of our jobs use the new policy defaults so we can\nsee what breaks if an admin turns this on.\n\nChange-Id: Iaa4ec2a3331bdca276a6e8f622701c9ce860b001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/209a72371b656aff736ed588dc3f724e01a656d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/209a72371b656aff736ed588dc3f724e01a656d3"}]},"branch":"refs/heads/master"},"0b56d3c5a5c3000ca2927472fea1b2958848bbb4":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2021-03-06 04:03:13.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/36/778736/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/778736/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/778736/13"}}},"commit":{"parents":[{"commit":"d58b46629e2e67cf631d12eec097eb99af261c36","subject":"Make secure RBAC protection job voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d58b46629e2e67cf631d12eec097eb99af261c36"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-04 15:48:38.000000000","tz":-480},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-03-06 04:01:02.000000000","tz":0},"subject":"DNM: Check some jobs with new policy defaults","message":"DNM: Check some jobs with new policy defaults\n\nThis makes a bunch of our jobs use the new policy defaults so we can\nsee what breaks if an admin turns this on.\n\nChange-Id: Iaa4ec2a3331bdca276a6e8f622701c9ce860b001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0b56d3c5a5c3000ca2927472fea1b2958848bbb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0b56d3c5a5c3000ca2927472fea1b2958848bbb4"}]},"branch":"refs/heads/master"},"e8e483507f04a9768cc6d93e7544af8a8650b83e":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2021-03-07 16:15:14.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/36/778736/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/778736/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/778736/14"}}},"commit":{"parents":[{"commit":"ee3d8c6b37f7730c46ef2f97a4b330815a1d8f01","subject":"Make secure RBAC protection job voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ee3d8c6b37f7730c46ef2f97a4b330815a1d8f01"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-04 15:48:38.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-07 16:13:02.000000000","tz":-480},"subject":"DNM: Check some jobs with new policy defaults","message":"DNM: Check some jobs with new policy defaults\n\nThis makes a bunch of our jobs use the new policy defaults so we can\nsee what breaks if an admin turns this on.\n\nChange-Id: Iaa4ec2a3331bdca276a6e8f622701c9ce860b001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e8e483507f04a9768cc6d93e7544af8a8650b83e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e8e483507f04a9768cc6d93e7544af8a8650b83e"}]},"branch":"refs/heads/master"},"6e7242a802fee7380a1b904deb9be475040361b2":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2021-03-07 16:32:13.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/36/778736/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/778736/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/778736/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/778736/15"}}},"commit":{"parents":[{"commit":"f7431a6bbd29b19c635fb923b229842f1bb56b2e","subject":"Make secure RBAC protection job voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f7431a6bbd29b19c635fb923b229842f1bb56b2e"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-04 15:48:38.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-07 16:32:12.000000000","tz":-480},"subject":"DNM: Check some jobs with new policy defaults","message":"DNM: Check some jobs with new policy defaults\n\nThis makes a bunch of our jobs use the new policy defaults so we can\nsee what breaks if an admin turns this on.\n\nChange-Id: Iaa4ec2a3331bdca276a6e8f622701c9ce860b001\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6e7242a802fee7380a1b904deb9be475040361b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6e7242a802fee7380a1b904deb9be475040361b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
