)]}'
{"id":"openstack%2Fglance~779070","triplet_id":"openstack%2Fglance~master~I8f980cf7731d26b92b5392fdada21e5be0f541c4","project":"openstack/glance","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I8f980cf7731d26b92b5392fdada21e5be0f541c4","subject":"Add a release note for secure RBAC personas","status":"MERGED","created":"2021-03-06 04:36:09.000000000","updated":"2021-03-09 21:56:15.000000000","submitted":"2021-03-09 21:50:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"779070-secure-rbac","meta_rev_id":"82bb7d0629c7d0d3ea431d776685f73259508ab1","_number":779070,"virtual_id_number":779070,"owner":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-09 21:50:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-03-09 18:16:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-03-09 18:31:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2021-03-09 18:31:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-06 04:47:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-06 05:20:54.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-03-09 17:30:55.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-03-09 17:45:05.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"3cb931a07ce7f1f405243137b495b8a125743142","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-06 04:36:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c8b44e731db077bd32e355efc138932af2d09532","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-06 04:47:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce6200923cdd4a248d275b82753e3d43 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48bbd2ede7454c189bcdd2fa5965e162 : SUCCESS in 6m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7aadccc718254d47aa7ad530b0d497e0 : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":1},{"id":"89606fa35c59ad12a9c93d89adfff3eee43696f1","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-06 05:20:54.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/779070/779070/1/check-not-vote/ext-glance-zuul/69b8ea2 : SUCCESS in 43m 46s","accounts_in_message":[],"_revision_number":1},{"id":"d5154fb9f30e1e68d0421e2107d0a113fa5341af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-08 14:24:58.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"3ed08d6f629ca5b677e3d588c658192a583d0abb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-08 15:37:43.000000000","message":"Patch Set 2:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/779070/779070/2/check-not-vote/ext-glance-zuul/eb96dba : SUCCESS in 35m 04s","accounts_in_message":[],"_revision_number":2},{"id":"89ed6e63e81ea7f9d5876d1068e77328dc8cd84f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 15:57:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7f89599b55b467898a950fddcfa9d11 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18a609c28fcc48e59e081531174f8309 : SUCCESS in 7m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19fd6259c13a430a9b4547b63ce0d1ce : SUCCESS in 8m 34s","accounts_in_message":[],"_revision_number":2},{"id":"51ec5021927e7e8c04f9a37f7ffa03f2629d0d96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-09 17:30:55.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"e6e11cf70e520e42f7999d5aee5e8cff2e1c5382","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-09 17:34:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fc560c971da1ea628b183d85f764fe88ff9350c1","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-09 17:36:40.000000000","message":"Patch Set 4:\n\nAt Abhi\u0027s suggestion, I squashed the release note and job voting change together. There\u0027s no real reason for them to be separate, and you could make the argument that at the point at which we\u0027re claiming it to be implemented (the reno) we should also be testing and voting on it.\n\nHope that\u0027s okay with people.","accounts_in_message":[],"_revision_number":4},{"id":"376ef2e5b9f319f1dba087cae2ec6ff0f72ec2a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-09 17:42:25.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"6fc7f36f6e0c52d6a8326b001e3f393ae80f09e2","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-09 17:42:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"eccd8c8c77c3284cf30bffa3cb1406978760cd09","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-09 17:43:32.000000000","message":"Patch Set 5: Code-Review+2\n\nMy only fingerprints on this are squashing the job config patch into this, and updating the deprecation text to reflect the updated task policies. I\u0027ll leave it to Abhi to +W after proof-reading my changes.","accounts_in_message":[],"_revision_number":5},{"id":"cec5b72f8b12b664df106034f85bb93b7356f6bf","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-09 17:45:05.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)\n\nsome suggestions/comments inline.","accounts_in_message":[],"_revision_number":5},{"id":"d43a5e08fec795dee7e8345ddf170a088bb20e46","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-09 17:51:13.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nLooks good to me, squashing two commits in one sounds right approach at this moment.","accounts_in_message":[],"_revision_number":5},{"id":"3a5596a936269bcbc419f56f86b94434c3e364b2","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-09 17:51:18.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"f5ef65674dac86fe177a040680cf954656a04149","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-09 17:51:57.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"45ee8d4fc4d5ce44fbc6e184c6c5f0c7cea9d321","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-09 18:16:23.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"7f84654f2a7a3a7d3083f1e1a127a82b23c81c8e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-09 18:31:53.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"8acecadda3e0983430f8640660a6919130507b06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 18:42:05.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c1923892d89e4cb185f67f098f2c1991 : SUCCESS in 8m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2b705b2b45d44afbe70de0fa8586cc1 : SUCCESS in 7m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3758913f2ab4be68622dec18e14a671 : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6006c357010481c90d17a1f15bb2b1e : SUCCESS in 8m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9f0692536344bf58163e80003c9e1da : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38bbe6044c57485d9cae8f12f5775539 : SUCCESS in 8m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bab781136c1841b79e0324983fcdb725 : SUCCESS in 7m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c7f338717fc442aabd4b76ed3732c0e1 : SUCCESS in 10m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cc96fd3f959a47caa51d07edb850b20e : SUCCESS in 10m 10s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/983233e2090240bb97a2c2021bed1dff : SUCCESS in 36m 23s","accounts_in_message":[],"_revision_number":6},{"id":"3d5921a266225eea83adecd9cca2ae3437113406","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-09 19:28:50.000000000","message":"Patch Set 6:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/779070/779070/6/check-not-vote/ext-glance-zuul/00458a9 : SUCCESS in 39m 10s","accounts_in_message":[],"_revision_number":6},{"id":"42dbe459ebb062c89362d5b10025238a8e0ce6a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 20:03:36.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"a4129316d41d7d8d67839cec55e6de47aca0c131","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 21:50:49.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/758fce3b1da143198f868acf24ce983c : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fc56a6b0b45440bafa46096c14cf328 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a69c649c259744cb860683eab589b484 : SUCCESS in 7m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63af079bd14a4c05b3de8b9e851e13c2 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c5c71d5be8e471b99b7a4fa0eb49901 : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/063d9aee77254fb1b14ed0021e6ee398 : SUCCESS in 6m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c74187fe71524a6287b75db93ee8ae6f : SUCCESS in 9m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/25ef8d2b02e045efadd33911451e2882 : SUCCESS in 8m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8438b8db0ad046aa8f3749108cef1010 : SUCCESS in 29m 05s","accounts_in_message":[],"_revision_number":6},{"id":"8e8ab964a5f7d794c32c6f87d1ab43e8cf34d01d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 21:50:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"82bb7d0629c7d0d3ea431d776685f73259508ab1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 21:56:15.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67e40750c58c4d7b8a23656d1e58527d : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/632050c9dc4d493e966709c78d18ff52 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"aec2de7ffd21fb553114c9f27a260062cafbc3b8","revisions":{"5c865457e8bb7ca0f1cf1488ef68234cf626188c":{"kind":"REWORK","_number":1,"created":"2021-03-06 04:36:09.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/70/779070/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/779070/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/779070/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/779070/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/779070/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/779070/1"}}},"commit":{"parents":[{"commit":"d58b46629e2e67cf631d12eec097eb99af261c36","subject":"Make secure RBAC protection job voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d58b46629e2e67cf631d12eec097eb99af261c36"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-03-06 04:32:21.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-03-06 04:32:21.000000000","tz":0},"subject":"Add a release note for secure RBAC project personas","message":"Add a release note for secure RBAC project personas\n\nProvide some literature on what we introduced for operators in wallaby,\nhow they can configure it, and actions we recommend they take.\n\nImplements: blueprint secure-rbac\nChange-Id: I8f980cf7731d26b92b5392fdada21e5be0f541c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5c865457e8bb7ca0f1cf1488ef68234cf626188c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5c865457e8bb7ca0f1cf1488ef68234cf626188c"}]},"branch":"refs/heads/master"},"b65448d49c896860512e8f5243307cacf44d2350":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-03-08 14:24:58.000000000","uploader":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"ref":"refs/changes/70/779070/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/779070/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/779070/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/779070/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/779070/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/779070/2"}}},"commit":{"parents":[{"commit":"f7431a6bbd29b19c635fb923b229842f1bb56b2e","subject":"Make secure RBAC protection job voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f7431a6bbd29b19c635fb923b229842f1bb56b2e"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-03-06 04:32:21.000000000","tz":0},"committer":{"name":"Lance Bragstad","email":"lbragstad@redhat.com","date":"2021-03-08 14:24:58.000000000","tz":0},"subject":"Add a release note for secure RBAC project personas","message":"Add a release note for secure RBAC project personas\n\nProvide some literature on what we introduced for operators in wallaby,\nhow they can configure it, and actions we recommend they take.\n\nImplements: blueprint secure-rbac\nChange-Id: I8f980cf7731d26b92b5392fdada21e5be0f541c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b65448d49c896860512e8f5243307cacf44d2350"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b65448d49c896860512e8f5243307cacf44d2350"}]},"branch":"refs/heads/master","description":"Rebase"},"03fb80faeafb7ae0c5db160e46c6f9f1dfc2be03":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-03-09 17:30:55.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/70/779070/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/779070/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/779070/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/779070/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/779070/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/779070/3"}}},"commit":{"parents":[{"commit":"165cce6d6e11504c85a71449fdca9608c7fd1d6c","subject":"Update the task policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/165cce6d6e11504c85a71449fdca9608c7fd1d6c"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-03-06 04:32:21.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-09 17:29:57.000000000","tz":-480},"subject":"Add a release note for secure RBAC project personas","message":"Add a release note for secure RBAC project personas\n\nProvide some literature on what we introduced for operators in wallaby,\nhow they can configure it, and actions we recommend they take.\n\nImplements: blueprint secure-rbac\nChange-Id: I8f980cf7731d26b92b5392fdada21e5be0f541c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03fb80faeafb7ae0c5db160e46c6f9f1dfc2be03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03fb80faeafb7ae0c5db160e46c6f9f1dfc2be03"}]},"branch":"refs/heads/master"},"1ca0b098d4cadacc99f77d64efe5564b6098b5fb":{"kind":"REWORK","_number":4,"created":"2021-03-09 17:34:08.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/70/779070/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/779070/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/779070/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/779070/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/779070/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/779070/4"}}},"commit":{"parents":[{"commit":"165cce6d6e11504c85a71449fdca9608c7fd1d6c","subject":"Update the task policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/165cce6d6e11504c85a71449fdca9608c7fd1d6c"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-03-06 04:32:21.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-09 17:31:31.000000000","tz":-480},"subject":"Add a release note for secure RBAC personas","message":"Add a release note for secure RBAC personas\n\nProvide some literature on what we introduced for operators in wallaby,\nhow they can configure it, and actions we recommend they take. Since\nthis marks the point at which we consider the feature implemented,\nthis also removes the legacy-rbac job and makes the secure-rbac job\nvoting.\n\nImplements: blueprint secure-rbac\nChange-Id: I8f980cf7731d26b92b5392fdada21e5be0f541c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1ca0b098d4cadacc99f77d64efe5564b6098b5fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1ca0b098d4cadacc99f77d64efe5564b6098b5fb"}]},"branch":"refs/heads/master"},"dd96ff3342fc3b61e6962c3799dbedd7d74d2121":{"kind":"REWORK","_number":5,"created":"2021-03-09 17:42:25.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/70/779070/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/779070/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/779070/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/779070/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/779070/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/779070/5"}}},"commit":{"parents":[{"commit":"165cce6d6e11504c85a71449fdca9608c7fd1d6c","subject":"Update the task policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/165cce6d6e11504c85a71449fdca9608c7fd1d6c"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-03-06 04:32:21.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-09 17:42:22.000000000","tz":-480},"subject":"Add a release note for secure RBAC personas","message":"Add a release note for secure RBAC personas\n\nProvide some literature on what we introduced for operators in wallaby,\nhow they can configure it, and actions we recommend they take. Since\nthis marks the point at which we consider the feature implemented,\nthis also removes the legacy-rbac job and makes the secure-rbac job\nvoting.\n\nImplements: blueprint secure-rbac\nChange-Id: I8f980cf7731d26b92b5392fdada21e5be0f541c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dd96ff3342fc3b61e6962c3799dbedd7d74d2121"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dd96ff3342fc3b61e6962c3799dbedd7d74d2121"}]},"branch":"refs/heads/master"},"aec2de7ffd21fb553114c9f27a260062cafbc3b8":{"kind":"REWORK","_number":6,"created":"2021-03-09 17:51:57.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/70/779070/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/779070/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/779070/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/779070/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/779070/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/779070/6"}}},"commit":{"parents":[{"commit":"165cce6d6e11504c85a71449fdca9608c7fd1d6c","subject":"Update the task policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/165cce6d6e11504c85a71449fdca9608c7fd1d6c"}]}],"author":{"name":"Lance Bragstad","email":"lbragstad@gmail.com","date":"2021-03-06 04:32:21.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-09 17:51:47.000000000","tz":-480},"subject":"Add a release note for secure RBAC personas","message":"Add a release note for secure RBAC personas\n\nProvide some literature on what we introduced for operators in wallaby,\nhow they can configure it, and actions we recommend they take. Since\nthis marks the point at which we consider the feature implemented,\nthis also removes the legacy-rbac job and makes the secure-rbac job\nvoting.\n\nImplements: blueprint secure-rbac\nChange-Id: I8f980cf7731d26b92b5392fdada21e5be0f541c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aec2de7ffd21fb553114c9f27a260062cafbc3b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aec2de7ffd21fb553114c9f27a260062cafbc3b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
