)]}'
{"id":"openstack%2Fglance~779163","triplet_id":"openstack%2Fglance~master~Idc579cf15dc49f4c891aeabbb94c19fe010da7cd","project":"openstack/glance","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"Idc579cf15dc49f4c891aeabbb94c19fe010da7cd","subject":"Add API version 2.13","status":"ABANDONED","created":"2021-03-08 07:26:56.000000000","updated":"2021-03-09 13:42:35.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d1ba6e2c3d9b13565269f44c15ff73225b0c8b72","_number":779163,"virtual_id_number":779163,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-08 11:19:22.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":1,"date":"2021-03-08 08:08:06.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-08 08:08:06.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2021-03-08 08:08:06.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-03-08 09:04:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"64aba6254e3c2782438cdfab2e59ef2582ba71c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-08 07:26:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8b3f685ef49fd3ddf6346caac36a25c10de6d78","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-08 08:08:06.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"99cbc3d15487480d3f2a95c78b291975095e690d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 09:04:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/324606707394447593a872a924b84d54 : SUCCESS in 1h 13m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3f4cfe3faab24b869b91686e092150a0 : SUCCESS in 1h 29m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0477f0c9599d46c5a6ceb13fc503660a : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6409b59e8e9d40c78f2b96b68e297977 : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e57dafa05e944b8fa677e178fb1c9001 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7519a4a3b5f3470783b7feb7cdadd6f4 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3a49fe595ab4bf9b5b9240c782a0048 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f74190be33b24f358123ead87493ed22 : SUCCESS in 7m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/337431b556ac4981b980d58bed2c5bc4 : FAILURE in 8m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/28097ddefc5b4b4fb4f2abc08e1faf90 : SUCCESS in 7m 45s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/182add5542914f469e386357b6dd564e : SUCCESS in 4m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/01ba16972c3d4ce2b91e79b17e51fca9 : SUCCESS in 1h 22m 07s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3ad6f6131f0c4d038f2514212c4505a6 : SUCCESS in 57m 12s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/44b135b855204c6d81c6b3cfacab94dc : SUCCESS in 1h 32m 46s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7bbcd771630a4d428f6a10fb4450018c : SUCCESS in 1h 14m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e0f99d8ed1b413ba7216e2b04c6ec7e : SUCCESS in 51m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a29f8e0577da4ce38f0fa3dc7bb071dd : SUCCESS in 1h 06m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e1353eff9e5240f08169cd892ee9d6fc : SUCCESS in 38m 07s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6ab2b0744ad449baadf5f770817f7382 : SUCCESS in 28m 02s","accounts_in_message":[],"_revision_number":1},{"id":"b784ba34aa0a998d4aea146036fc52d63ce44e98","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-08 09:06:53.000000000","message":"Patch Set 1:\n\nrecheck unrelated failure","accounts_in_message":[],"_revision_number":1},{"id":"990a919222f3a84542f388e489720e0bcf225abe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 11:19:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/2be3f6604e9b493cbdcc0179b5895d88 : SUCCESS in 1h 05m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/de549f9f6ee44ad481a2f7b5608cec8a : SUCCESS in 1h 21m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53218e63c2424af09a182eb197353c8a : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9ad6df1889d49eca30c5909d0d6d85e : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6fd8fb8c177b43ef9ee9dfbfa2c21b3a : SUCCESS in 7m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44d93214b53d4dc8b9bd9292973ffc24 : SUCCESS in 7m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4638c88ee0d14eba8e3fc18164cc0af5 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9581140be48f482ead4ce290c4ad513c : SUCCESS in 7m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/10c12ff46f514a78a0f281e74b6ae139 : SUCCESS in 10m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d4ea6520180346518b30c690c4264cc9 : SUCCESS in 9m 23s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/3fbb1c225188431eb95df7048793074d : SUCCESS in 5m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a222127ce3cf427188870d42ed86d676 : SUCCESS in 1h 38m 44s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/da391872b2b44e40ad23d4550f67962a : SUCCESS in 1h 19m 09s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/491d41384e82476aa468207daee8cf0e : SUCCESS in 1h 13m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c0ca4997722e4b2ebf5f3cd96ec7e526 : SUCCESS in 1h 26m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b97abc492fac45a9a61250d8f6d93e1b : SUCCESS in 1h 01m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7e1b5bd3154b4eee847af9145bf0021a : SUCCESS in 1h 40m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c6abaddf6d8a408d837324fb0c191b0a : SUCCESS in 40m 35s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0ede9a788f424861bd80003a81424dfb : SUCCESS in 32m 36s","accounts_in_message":[],"_revision_number":1},{"id":"0340387ef0abe1dec7ee363b0078b8b46c746a48","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-08 14:02:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ee4ff4e4bf12f5afd96d4df364cce900e75d84a0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-08 14:07:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9fb329057063b270fd052ea3a166d6889ab441c1","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-08 14:23:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"be86569049bab0f1a6fd9d63c8d8b592b66b0f22","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-08 14:30:00.000000000","message":"Patch Set 1:\n\nCan we have a little discussion with Lance about this before we merge?\n\nMy understanding is that the only people that *might* do anything different via the API after secure-rbac are system admins, and only when they need to manage something restricted to system scope, or if they need to do something like delete images across projects. These are people that will know how glance is configured, and what actual code is enabled. Lance, is this right?\n\nObviously version numbers are cheap, so it doesn\u0027t matter that much, but unless a client library is going to do something different if this version is present or not, I don\u0027t see the point. Is there something that OSC or glanceclient would do (or not do) based on the presence of this?","accounts_in_message":[],"_revision_number":1},{"id":"a65eacd76772ec45fabb635d393a42eb3cbeadaf","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-08 14:51:00.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Can we have a little discussion with Lance about this before we merge?\n\u003e \n\u003e My understanding is that the only people that *might* do anything different via the API after secure-rbac are system admins, and only when they need to manage something restricted to system scope, or if they need to do something like delete images across projects. These are people that will know how glance is configured, and what actual code is enabled. Lance, is this right?\n\u003e \n\u003e Obviously version numbers are cheap, so it doesn\u0027t matter that much, but unless a client library is going to do something different if this version is present or not, I don\u0027t see the point. Is there something that OSC or glanceclient would do (or not do) based on the presence of this?\n\nI don\u0027t think there will be any difference in this case with OSC or glanceclient.","accounts_in_message":[],"_revision_number":1},{"id":"19ddce6c8ce1f2280b9d6473ffbcea2ed0e6b698","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-08 15:25:25.000000000","message":"Abandoned\n\nThis is not required now as we are only dealing with project personas in this cycle.","accounts_in_message":[],"_revision_number":1},{"id":"d1ba6e2c3d9b13565269f44c15ff73225b0c8b72","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-03-09 13:42:35.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Can we have a little discussion with Lance about this before we merge?\n\u003e \n\u003e My understanding is that the only people that *might* do anything different via the API after secure-rbac are system admins, and only when they need to manage something restricted to system scope, or if they need to do something like delete images across projects. These are people that will know how glance is configured, and what actual code is enabled. Lance, is this right?\n\nYes. Glance will continue to consume project-scoped tokens for all operations, admin or not. A version bump might be a more interesting conversation when glance supports system-scope.\n\n\u003e \n\u003e Obviously version numbers are cheap, so it doesn\u0027t matter that much, but unless a client library is going to do something different if this version is present or not, I don\u0027t see the point. Is there something that OSC or glanceclient would do (or not do) based on the presence of this?","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"22cc7c29a8692ebc07d18b4114f33f75a05bc983","revisions":{"22cc7c29a8692ebc07d18b4114f33f75a05bc983":{"kind":"REWORK","_number":1,"created":"2021-03-08 07:26:56.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/63/779163/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/63/779163/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/779163/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/779163/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/779163/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/63/779163/1"}}},"commit":{"parents":[{"commit":"605d653435481a84443035d439daebffc07006f0","subject":"Deprecate the external task API policies","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/605d653435481a84443035d439daebffc07006f0"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-03-08 07:23:36.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-03-08 07:23:36.000000000","tz":0},"subject":"Add API version 2.13","message":"Add API version 2.13\n\nThis change bumps the API to EXPERIMENTAL 2.13 when secure RBAC\nis enabled.\n\nChange-Id: Idc579cf15dc49f4c891aeabbb94c19fe010da7cd\nRelated: : blueprint secure-rbac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/22cc7c29a8692ebc07d18b4114f33f75a05bc983"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/22cc7c29a8692ebc07d18b4114f33f75a05bc983"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
