)]}'
{"id":"openstack%2Fglance~779181","triplet_id":"openstack%2Fglance~master~Ifc957de2bcf4d22c1b7cba31ce2b467df9d5aa85","project":"openstack/glance","branch":"master","topic":"glance-wallaby","hashtags":["glance-wallaby"],"change_id":"Ifc957de2bcf4d22c1b7cba31ce2b467df9d5aa85","subject":"Refresh Glance example configs for Wallaby milestone 3","status":"MERGED","created":"2021-03-08 09:02:06.000000000","updated":"2021-03-12 00:37:50.000000000","submitted":"2021-03-12 00:36:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779181-glance-wallaby","meta_rev_id":"e19f08c918cd3167aaaf2aaa4dbec341ea5903ce","_number":779181,"virtual_id_number":779181,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-12 00:36:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-03-11 14:11:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-03-08 14:58:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-03-11 14:11:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-03-08 09:03:02.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-08 09:03:02.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-03-08 09:03:02.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2021-03-08 09:03:02.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-03-08 11:01:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-08 11:13:03.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"a44d00dcddc1a42210741363f20106f1e0c90804","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-08 09:02:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25ab8cc91a1b9a133c531aa81d88004be41c282e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-08 09:03:02.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"8f20df898288801954b91a03b8a13ef685d7251c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 11:01:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/a740168340f544d0a2559bb3905a0ea7 : SUCCESS in 1h 05m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eac2af7ea6124118beff74c42fa44ba9 : SUCCESS in 1h 11m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4fc0fdee14584081973d5a5b72c26fe3 : SUCCESS in 10m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e5a85de624e4a03b8cf3f9b5041d6f5 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/43fbb4b0d93f42d7b62b78a886b90fc3 : SUCCESS in 7m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6cbd956280d4352b88f74b6f2b58473 : SUCCESS in 8m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/918029f585e4431799712f21456b8175 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1afb9c11ea1b4225a215a6041f7e9de5 : SUCCESS in 7m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/55daf04da3bb49e39437068dbf86483a : SUCCESS in 11m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/05e3b3906b2f47879b88c2e098d220ee : SUCCESS in 11m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6ee5da3e5d4743b985587db0d4fd6902 : SUCCESS in 1h 24m 26s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a14bb5b954fc465da62d6f4c0fb56a2e : SUCCESS in 1h 21m 18s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/dcc1e37e8ba944b0a0fd8ba86b1ef90e : SUCCESS in 1h 21m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e8f325879f61433e87708d41dabbfebc : SUCCESS in 1h 30m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7eaa42e40b3c419fbbbbf34c82883c12 : SUCCESS in 1h 01m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/57d7536a2ded4fb69ee1c99edad0ff82 : SUCCESS in 1h 09m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e0e63cae55cc4a6fb86a29acb385b3ce : FAILURE in 36m 23s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/def7ba5026c747a6937927af6a8297c0 : SUCCESS in 39m 44s","accounts_in_message":[],"_revision_number":1},{"id":"7b37fab3aa49ea6ef7832457a6213aab399cd023","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-08 11:13:03.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/779181/779181/1/check-not-vote/ext-glance-zuul/adfa1a2 : FAILURE in 26m 18s","accounts_in_message":[],"_revision_number":1},{"id":"2057e53c81d9e5a6ac23e1f3f23b79ac1a5f4102","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-08 14:58:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3956762c3f9f284f316edc6a84445db6eedd5444","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-10 15:02:38.000000000","message":"Topic wallaby-m3/config-refresh removed","accounts_in_message":[],"_revision_number":1},{"id":"c24504f05bced96d81eae0b66286024563559a9c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-10 15:04:31.000000000","message":"Topic set to glance-wallaby","accounts_in_message":[],"_revision_number":1},{"id":"f2ef9576e9085a38f05c5fbadfa65a0805d28f0e","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-10 15:10:27.000000000","message":"Hashtag added: glance-wallaby","accounts_in_message":[],"_revision_number":1},{"id":"335029ac1ad49cb8cbc57fb64bb9f120eb7c7641","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-03-11 14:11:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM.","accounts_in_message":[],"_revision_number":1},{"id":"77cd3f2ad95f626cd7419b45d4ea42bb3e84d713","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 14:11:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e2ac0540d8d18d2263be245324210ea4ffff36f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 18:26:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/7be3e22f7ab94400ac7cc7e743808f5f : SUCCESS in 1h 01m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/51788f850b73416fbeb1fbc64ce90613 : SUCCESS in 1h 31m 29s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fc7f1a888fc44f3b2bbee71646fe1a1 : FAILURE in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5272eca7c5514f0f9be1c2dd046ff070 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbc176e0fc0f427b8c28be1e0efce26d : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c92ece3e48a84deaace41116a4651132 : SUCCESS in 6m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7115df2845e43b5a2eb09f828af4bd7 : SUCCESS in 7m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8778eca189774aa5b3da8d47d2eeb8c3 : SUCCESS in 10m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7f2e42ed206e4e22a65a0690472fa81c : SUCCESS in 9m 34s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/34a67de5668f49a7b9b3f295dd84ccc9 : SUCCESS in 1h 17m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3ef9cae994684b099d150db2ab18e5a3 : SUCCESS in 1h 20m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0d4474b6f7e40159f733e0ecba6f0d9 : SUCCESS in 59m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8232bd96055a459eaa9b5b73571ca49c : SUCCESS in 1h 13m 53s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7ca5bcdd5e7b474e8733801dd6a6b2f3 : SUCCESS in 35m 16s","accounts_in_message":[],"_revision_number":1},{"id":"327c536a53906210c799c00eb732a5e2c7a879fd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-11 18:27:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4aea5685f020282aa610ec403af4d15565d3eedb","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-11 19:06:40.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/779181/779181/1/check-not-vote/ext-glance-zuul/0d3e12b : SUCCESS in 37m 55s","accounts_in_message":[],"_revision_number":1},{"id":"89c17e7a0241a89fc816fd9b871f2629aa1687ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 21:56:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/36db4674ef1e481ca4efb447a7a7d0dc : SUCCESS in 1h 04m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/73b417115ea24e72ab8e60f96a59ea73 : SUCCESS in 1h 26m 22s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30b6c897f96e4a18be320c8fce8daced : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe612f75700b4738bdc524dcb4b8526e : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2f964a667ea414cb87e0f147aad26cd : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7edd6e84c6dc47e49f1199f260c87056 : SUCCESS in 7m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fad08eae88c4435da05d16eaf687c9cf : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85fa313db2f84fffa868dbdd9f301c3b : SUCCESS in 6m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/34dfcf21b2174c6e814bc2d04bf25d77 : SUCCESS in 9m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c8a9423cd5504620848cb49d5f0067a3 : SUCCESS in 9m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4fc6d738c726468a865a725a2acdf179 : SUCCESS in 1h 14m 12s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/af56858d2fe841e09a05437361a140e4 : SUCCESS in 1h 11m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/16f61115d5c9443fb6d15e4bbbdb0c41 : SUCCESS in 1h 18m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c9cf06b1d40349e38bcdbefcdd26caca : SUCCESS in 1h 15m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5471a8a775c4429bd6f9a8b7993b168 : SUCCESS in 52m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1c747eb09353457d854336f7467ce236 : SUCCESS in 1h 05m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3cdf978792ba4e638640bb17018eda97 : SUCCESS in 37m 20s","accounts_in_message":[],"_revision_number":1},{"id":"0ac2f16e4469ac386ae8349dad2d01f13eb20713","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 21:56:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4b951065fb69dd0ef98e10b0b063d7d795cfd3f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 00:36:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/5f0644f466ac41419181aeeffb044bfb : SUCCESS in 59m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e7d8b89d69514eafb22ff0961732c447 : SUCCESS in 1h 13m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/df89fddbbc1d49b196e36f64baf3f320 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6fe9972a59f46499b44120e0094588c : SUCCESS in 7m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18a49aa20e514d61824594e6bed96ecf : SUCCESS in 7m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ce4aebeec7e43b39aa4abd4dc1a8c27 : SUCCESS in 9m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b76c9945fee64a449fb2ebf724ec17ed : SUCCESS in 8m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/34246c1ca0f445a8a28a1c6f4ce57362 : SUCCESS in 9m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0c923f46351247329e68e38e38413df6 : SUCCESS in 10m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/55544320e9834fdfba3c2ffa38d627df : SUCCESS in 1h 09m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/114052d17db647c0870abc8501b0ca14 : SUCCESS in 1h 07m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99ae750347564cf58d9ae0214fcd81f5 : SUCCESS in 57m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/85e8ecb8fddf4091b2bda74d7201913e : SUCCESS in 1h 11m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/78631bc69753452d9f0d37fd71f37621 : SUCCESS in 43m 59s","accounts_in_message":[],"_revision_number":1},{"id":"6f6ab02924eba4db2560394e7c833356eefbc7c4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 00:36:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"e19f08c918cd3167aaaf2aaa4dbec341ea5903ce","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 00:37:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37034a586ead44389938e9feb65f3f24 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dbab6640398013dd2dfa2cad2bede365309d4fb0","revisions":{"dbab6640398013dd2dfa2cad2bede365309d4fb0":{"kind":"REWORK","_number":1,"created":"2021-03-08 09:02:06.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/81/779181/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/81/779181/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/779181/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/779181/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/779181/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/81/779181/1"}}},"commit":{"parents":[{"commit":"23ed884c4bf8aaa9076c76f4aeeae8cfba7984d3","subject":"Merge \"Fix test_cache_middleware ImageStub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/23ed884c4bf8aaa9076c76f4aeeae8cfba7984d3"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-03-08 08:59:50.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-03-08 08:59:50.000000000","tz":0},"subject":"Refresh Glance example configs for Wallaby milestone 3","message":"Refresh Glance example configs for Wallaby milestone 3\n\nChange-Id: Ifc957de2bcf4d22c1b7cba31ce2b467df9d5aa85\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dbab6640398013dd2dfa2cad2bede365309d4fb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dbab6640398013dd2dfa2cad2bede365309d4fb0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
