)]}'
{"id":"openstack%2Fglance~779369","triplet_id":"openstack%2Fglance~master~I8d9ba50f46e22b6740fdbdec6f8ef7c61dddbcf1","project":"openstack/glance","branch":"master","hashtags":[],"change_id":"I8d9ba50f46e22b6740fdbdec6f8ef7c61dddbcf1","subject":"Fix erroneous exit from copy wait loop","status":"MERGED","created":"2021-03-08 21:31:19.000000000","updated":"2021-03-09 12:57:51.000000000","submitted":"2021-03-09 12:49:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"779369","meta_rev_id":"7eda820fcbb8c90303ff8c9b93c3fa7cb7e7b4ba","_number":779369,"virtual_id_number":779369,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-09 12:49:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-03-09 10:54:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-03-09 10:54:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-08 21:33:23.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-03-08 22:09:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-08 23:06:24.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"31518902999ed5b9cec7efeefd77759396b8e898","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-08 21:31:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0095a1e9960ce3540df7822f08e9fb495facca78","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-08 21:33:23.000000000","message":"Patch Set 1:\n\nI found this while trying to debug the copy_image_lifecycle() failure. I don\u0027t think it\u0027s related because we assert both stores after the wait, and because that doesn\u0027t explain the missing task. However, I couldn\u0027t get it to fail for further debugging after I added this. So I\u0027m pulling it out because I think we should merge it on its own, and to see if I can get more debug info in the other patch without it.","accounts_in_message":[],"_revision_number":1},{"id":"0fc18c9bc35ad0971aa8c10ae010777707caf498","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-08 21:34:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"150b6bf2fe3fcfc5feaa33591b181dfd97cb7642","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-08 22:09:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fbdaf09c2264ca2b94ca806a418a968 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e83dbe3f8de41f2aac26b475c3d07f4 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/053efb2d858c4649b1a2652647ad9e0d : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0de920015a94ef5aacf7636c18dccc3 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a47d51ea1914751b6ba95d8c7fbdd29 : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8eebf7c5acb40c280e6b3ee1ad2e11c : SUCCESS in 8m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/87da590c19164d2da605d57e803af28f : SUCCESS in 7m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8c5cd53429174ffba0c48ab00963f435 : SUCCESS in 9m 41s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9eaafa5201ec4022aceb3fc7e2151817 : SUCCESS in 7m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/582cda3f1a3b4cbab75e33a0fe2dbc37 : FAILURE in 34m 30s (non-voting)\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8ae60c225bf64bfc90fd54a8d7cfd2ad : SUCCESS in 35m 07s","accounts_in_message":[],"_revision_number":1},{"id":"09df04698665e8125429e76e06898fa5627e02dc","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-03-08 23:06:24.000000000","message":"Patch Set 1:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/779369/779369/1/check-not-vote/ext-glance-zuul/5500611 : SUCCESS in 36m 56s","accounts_in_message":[],"_revision_number":1},{"id":"2284753c88bd2ca1f9b359b0ec1e56318f36cde6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-09 10:54:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLet\u0027s find out if this could rule out the failure or decrease the possibility of failure.","accounts_in_message":[],"_revision_number":1},{"id":"fac462675cc782fdbb1d7c0da7827a4fe6be2d9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 10:54:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6ba33b2ba69ae21628f5c47fa8adedfcf4ca5421","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 12:49:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a53dcb552c34465996d72cdac431ba8e : SUCCESS in 10m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e32b13a370c4c8990661d9a2a64ac5e : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3e532cdcacc4a47934cb4f221c7eb13 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c701a6e1419b48fbb5d81f73161f9e6b : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fb1961323864483b0409455200eac39 : SUCCESS in 6m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2a6177bfa64e4e208b7752be404059d8 : SUCCESS in 11m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b18d08b0113442f0b3bf1da9048bf985 : SUCCESS in 8m 43s\n- glance-legacy-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d95e07bc2d1f4643a21bce669067a374 : SUCCESS in 36m 28s","accounts_in_message":[],"_revision_number":1},{"id":"ba077f20526941841677716d6b956e376f9116af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 12:49:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7eda820fcbb8c90303ff8c9b93c3fa7cb7e7b4ba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 12:57:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab17555ad72d42da9df6a58fd8ddf154 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"80b84d4e970704bda6cb17ec74925c94637531d7","revisions":{"80b84d4e970704bda6cb17ec74925c94637531d7":{"kind":"REWORK","_number":1,"created":"2021-03-08 21:31:19.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/69/779369/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/69/779369/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/779369/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/779369/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/779369/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/69/779369/1"}}},"commit":{"parents":[{"commit":"23ed884c4bf8aaa9076c76f4aeeae8cfba7984d3","subject":"Merge \"Fix test_cache_middleware ImageStub\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/23ed884c4bf8aaa9076c76f4aeeae8cfba7984d3"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-08 21:28:58.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-08 21:30:34.000000000","tz":-480},"subject":"Fix erroneous exit from copy wait loop","message":"Fix erroneous exit from copy wait loop\n\nThe wait_for_copying() helper will exit the loop if the *last*\nstore is found in the list, instead of *all* of them. This technically\nworks if the stores are processed in the same order we are checking,\nbut it\u0027s fragile and likely to fail in confusing ways.\n\nThis makes us only exit if all of them are present.\n\nChange-Id: I8d9ba50f46e22b6740fdbdec6f8ef7c61dddbcf1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/80b84d4e970704bda6cb17ec74925c94637531d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/80b84d4e970704bda6cb17ec74925c94637531d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
