)]}'
{"id":"openstack%2Fglance~779636","triplet_id":"openstack%2Fglance~stable%2Fvictoria~Ib80e9cfb58680f9e8ead5993dc206f4da882dd09","project":"openstack/glance","branch":"stable/victoria","topic":"victoria/bug/1913625","hashtags":[],"change_id":"Ib80e9cfb58680f9e8ead5993dc206f4da882dd09","subject":"Add housekeeping module and staging cleaner","status":"MERGED","created":"2021-03-10 01:22:42.000000000","updated":"2021-04-01 13:42:57.000000000","submitted":"2021-04-01 13:39:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"779636-victoria/bug/1913625","meta_rev_id":"3eb41bb58ce547139bc8479850451012d1b4c4ef","_number":779636,"virtual_id_number":779636,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-01 13:38:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-03-31 07:52:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2021-03-30 14:34:26.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-04-01 07:32:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-10 01:22:42.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-03-10 02:51:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-10 05:08:05.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2021-03-31 07:52:36.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"a1016f222f584e4563d58f64ef42e568b5d6c41f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-03-10 01:22:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"137005a0133310581cd6291e6f766be28d2c75fd","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-10 02:36:13.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nLooks okay to me. I would tend to almost call this a feature more than a bug. It\u0027s clearly bad behavior, but it\u0027s not a regression in W and has never been a thing that was broken. But, I\u0027ll defer to the glance-stable-maint people on that :)","accounts_in_message":[],"_revision_number":1},{"id":"00da4d0f3373937ade49ce4c2984bcdab7792c70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 02:51:20.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/5159f462a77840e586877aa88510f8e4 : SUCCESS in 1h 03m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9c4cc61b02ac4be1bbebc41052dc01c0 : SUCCESS in 1h 13m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a81ca2f3799b487a95e5489ef5d6ae9c : FAILURE in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a0f1a03b78c4431965c109f7b28724d : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49762b8f2cc14c67a26b7361e4743727 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f4494b2da5c847dc8d0af30dd7b70234 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a43e3449ff10420ba3b94e14dc234a6b : SUCCESS in 7m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a0c591ae502a4cc9a2c3dcf2167f24e9 : FAILURE in 8m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/17ff372aa6e242cd9b7ebb37a1961d24 : FAILURE in 10m 19s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a9a750ed058b46a4a7f74562f1b986a3 : SUCCESS in 5m 33s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4d42643f97fa429683858d62917727c2 : SUCCESS in 1h 25m 33s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/591b32534a194d8bbd4f2200c6a65b24 : SUCCESS in 1h 19m 02s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/7dc17bd38ff142e8a6b50aebf0446d49 : SUCCESS in 1h 22m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/364c0badde6b469da6153159b54cbc98 : SUCCESS in 1h 02m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0755d7f98a964d4eb1f92fd51257f15b : SUCCESS in 1h 19m 36s","accounts_in_message":[],"_revision_number":1},{"id":"3d2ace05fe9ccd07fec419d860096d1398cf6eec","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-10 05:08:05.000000000","message":"Patch Set 1:\n\nwe need this patch to backport as well;\nhttps://review.opendev.org/c/openstack/glance/+/777277/3","accounts_in_message":[],"_revision_number":1},{"id":"9f4f0d299219c6a487f00b79156b209016e931d6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-18 13:58:00.000000000","message":"Patch Set 1:\n\nEither add dependency on this patch https://review.opendev.org/c/openstack/glance/+/781426 or add recheck once this patch is merged.","accounts_in_message":[],"_revision_number":1},{"id":"9c56a55fe3e36dad50d793d9fe72349a7e346e90","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-24 05:48:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"45d1f786a0cb786f13f5eefbeb1d631acea07bf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 07:28:53.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/509ff9d930454d7c9230bb1b6e9879ad : SUCCESS in 1h 07m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d4ebf4526c064e60b4550a1396020fd5 : SUCCESS in 1h 20m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56099a2a3f244492b5698c9867f69573 : SUCCESS in 8m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c009c676ee1b48ab9442959f3d99d602 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e432d927daa34992a4ce6c9880c008e6 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c206771b13c34585a42242d13ef1a228 : SUCCESS in 7m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6f1c3d8f6874184a75be1b8edc45cf3 : SUCCESS in 5m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7c26ca97dee843508c88e6ef8edb2791 : FAILURE in 7m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dd23ebdd2e0f47cf9c7ba2168deaf014 : FAILURE in 9m 39s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f9c6c6c0be0b4466becb12eb9c8a5ab8 : SUCCESS in 4m 39s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5c498a6a5f944f2194f12f8f2fa08d90 : SUCCESS in 1h 18m 03s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/1a025ae9e9ed4e6db616af243471ff43 : SUCCESS in 1h 18m 59s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/dd7966421da94ca88d4ac7c97b2b7020 : SUCCESS in 1h 17m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb3209ba988c44188a402353d05035bc : SUCCESS in 1h 04m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9f5194d91eea4bf58026e09d2cb918f3 : SUCCESS in 1h 37m 00s","accounts_in_message":[],"_revision_number":1},{"id":"eb553324a2515d551a41d071f7fd965940f32161","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-24 07:33:30.000000000","message":"Patch Set 1:\n\nThis should be depends on https://review.opendev.org/c/openstack/glance/+/782431","accounts_in_message":[],"_revision_number":1},{"id":"9bc0b09df295a72a639558cf2fa7361c0c0e47dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-24 14:16:35.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"28fa502c47475a74be6724a1d1d51a3778a06c19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 16:15:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/1821b357079f4aafa1d232021b724f54 : SUCCESS in 57m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8495be6c8e694960a5d9bc03f0653f5e : SUCCESS in 1h 13m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72edb9737aae415594ea58f7dd2a305b : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94d32a1b3f9f44fea261ca6deded6fe2 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/622646296829412382af99649c3b753c : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f10660eaab64016a6ee5d9bddef5da8 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfde67d3e33f40afa1096b25fb94591d : SUCCESS in 7m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/99356dca886e45d5be8632c83f665f14 : SUCCESS in 9m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/264f1ee547d643b0beb34271a931906f : SUCCESS in 8m 20s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/24e8099e295944719040a618122268e3 : SUCCESS in 5m 10s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2932d4f10ab742689971ded897186350 : SUCCESS in 1h 20m 42s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/1c85a45368004c17930569a38e3518fa : SUCCESS in 1h 09m 20s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/a683e37b48384fdba89dc9c3afa6be8d : SUCCESS in 1h 33m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4da5923c381441c900f78e3ae0f26c7 : SUCCESS in 1h 05m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b70979d641c84692ad7db55a8e5fc502 : SUCCESS in 1h 35m 03s","accounts_in_message":[],"_revision_number":2},{"id":"3170d8beb3ea645e56098e296943392396aa5229","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-30 14:34:26.000000000","message":"Patch Set 2: Code-Review+1\n\nThe fact that the functional tests were failing without the prefixing patch below proves that this is doing its job :)","accounts_in_message":[],"_revision_number":2},{"id":"dd5e3e6bfe7bcb0d1ddc2db580e71fa4b4b3f27e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-31 07:52:36.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nClean and important backport","accounts_in_message":[],"_revision_number":2},{"id":"cab0bb3e9351da4c83dba3f30150af9ae404041e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 07:53:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e9b595176dd71fdcc7aa0657381769c8b9a6ca71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 12:51:30.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/25ebf8c0f7e34db39650785d0a67d0df : SUCCESS in 1h 09m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/df5b78b1c5624be6ba7178fdae2f8095 : FAILURE in 1h 15m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8336c3bf6b64eceb402c0d9cc908083 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c0a9d9d9b6d41ef8d41cdefd19f28d9 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41b2c5f6b4904b16a6d391043e0fc1aa : SUCCESS in 7m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/790f12b8b0684c258dc57fe8885ac125 : SUCCESS in 7m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1bc4cf1eb8074a48a016ee245e096011 : SUCCESS in 17m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6353be683dd94b3dad884c46e17164d0 : SUCCESS in 9m 59s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/c2d431d29ceb4bfebbb467a5fb1d9ffa : SUCCESS in 1h 09m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be965ec810bf43828bf37a1f3b7ca3d8 : SUCCESS in 1h 21m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9014653c06444d598b1e0b9492318650 : FAILURE in 1h 59m 56s","accounts_in_message":[],"_revision_number":2},{"id":"6dfda883e5bc663be213ded7ff871339dd7dc0ef","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-31 15:02:18.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3487c65ed3c288cec94ace9a50a12fdf0d4b6087","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 18:15:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/f8e56347d4454695899e74835a3f6e60 : SUCCESS in 1h 00m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9396f53985814ddfb9c15fb93c19d8b2 : SUCCESS in 1h 12m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/130d1509adb047e2b5ce6c6ae44378f3 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8698227e64cd454598211119673915fa : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d76c142f7c5948ed8b54fe72324a434b : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2aecb072fbd94baf9cec448d9f6a178f : SUCCESS in 10m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5f8705a074f94146b34ad0950afc9543 : SUCCESS in 10m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e095bf6bbf944cadb28fcbccb5e6572e : SUCCESS in 11m 17s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6b52651e20d049ec952beaeb7e5a09ac : SUCCESS in 6m 12s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/05c8488ffff94effa340b51f0800ff52 : FAILURE in 1h 59m 02s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/7156d5615b72485aa610336d9cb9cd65 : SUCCESS in 1h 39m 14s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/5eb4d27067e840d7bd68568a884cf916 : SUCCESS in 1h 07m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e32e81841ea74261afad23a512ef5219 : SUCCESS in 1h 00m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a6ebb12d521340b6bfcbab6c486a048c : SUCCESS in 1h 20m 33s","accounts_in_message":[],"_revision_number":2},{"id":"cce20e1e7e0cfc3e795049b74a24c19118776b58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 18:44:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aec0d1c8cd8f2f96ceba32cc3afe3197d910c502","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-04-01 07:32:15.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"7e234bfaab39cc4701060055c8f748cc26322061","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-04-01 07:32:20.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d8090b067fc3f00bf03d90b9fde04e2d8f095391","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-04-01 08:21:46.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6a5fb8d57a3aaaced704ffb3ddacc4de38dcf168","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 10:16:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/3b4700d8b4d34f6ebc3da9ed64bb159f : SUCCESS in 54m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/afa436470ea841c4bef17c9d58301990 : SUCCESS in 1h 19m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/defbe7a100364325b3a81580610f7e5c : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/daee1be7d4b34a00b7ecd1d7fcceb603 : SUCCESS in 8m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd18bf9562eb451cb5ea5850b4c60494 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9dc9b817ba9490dba3a296ddcce5636 : SUCCESS in 6m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fcf5cea2586e49f38ceaaaa36a010026 : SUCCESS in 8m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/19b177decb6342c09c14a75374cafdf7 : SUCCESS in 9m 58s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0a16903830454e119ac8edad7e74b382 : SUCCESS in 5m 12s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/683ea52090ad4c2ca097456c6183227f : SUCCESS in 1h 49m 48s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/11ceb33fdfde4306b05a5620a8232f1f : SUCCESS in 1h 12m 06s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/714fcbbe854c470e9c4df254a53201fb : SUCCESS in 1h 21m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d630ff11e95436eac251671b62ec4e5 : SUCCESS in 1h 04m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dfb86b61b1ea48629c72cb020b5c2d51 : SUCCESS in 1h 07m 34s","accounts_in_message":[],"_revision_number":2},{"id":"8d1d9909543bd45f8be723872d1536b4f8a25647","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 10:17:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5c80df2f4079865bb2da0a4b848807aef6a8eeb4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 13:38:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/dbba6e296db04152960a5e44bc229ca3 : SUCCESS in 1h 26m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f921024b3dc64fdc811002777aa9f679 : SUCCESS in 1h 59m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cd21f28bf4e4fb4842083b3e1036670 : SUCCESS in 11m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3fba8abb5a3d49a29dcaa968b47c0acd : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92f488e53ba442ed93eb7da7b36feb6f : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10bd84967d2b4a969303b9629d852e7e : SUCCESS in 12m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/869d0eec932947dfb669d63b7d7d4d8c : SUCCESS in 8m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dabaf924869c486788528406ea70a949 : SUCCESS in 11m 46s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/f3e1efe8eb21457f8caf3f594e3f6177 : SUCCESS in 1h 31m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4795c1f4d64401eb316234851b0fe81 : SUCCESS in 58m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8db02199a07f42b894e86a327aadfce2 : SUCCESS in 1h 29m 04s","accounts_in_message":[],"_revision_number":2},{"id":"1ce725eb3cbcf5707c19622b046f7dbb61766ea9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 13:39:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3eb41bb58ce547139bc8479850451012d1b4c4ef","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 13:42:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/206bd50cbcc442d0b3cabda2246ae050 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"94679941368d37520975e0a1dc81d27424167191","revisions":{"057d32c3839eff8df6b26dc91252d38d2cb2d6cf":{"kind":"REWORK","_number":1,"created":"2021-03-10 01:22:42.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/36/779636/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/779636/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/779636/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/779636/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/779636/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/779636/1"}}},"commit":{"parents":[{"commit":"794c0fa19883c245ced7a886f9f5d11bd50228c4","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/victoria\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/794c0fa19883c245ced7a886f9f5d11bd50228c4"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 20:52:03.000000000","tz":-480},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2021-03-10 01:14:43.000000000","tz":60},"subject":"Add housekeeping module and staging cleaner","message":"Add housekeeping module and staging cleaner\n\nAs noted in previous discussions, glance should clean its staging\ndirectory on startup. This is important for scenarios where we\nstarted an import operation, but failed in the middle. If, when we\nrecover, the image has already been deleted from the database, then\nwe will never remove the (potentially very large) residue from disk\nin our staging directory.\n\nThis is currently a problem with web-download, but will also occur\nwith glance-direct once we have the non-shared distributed import\nfunctionality merged.\n\nConflicts:\n\tglance/tests/unit/common/test_wsgi_app.py: Minor differences\n\nCloses-Bug: #1913625\nChange-Id: Ib80e9cfb58680f9e8ead5993dc206f4da882dd09\n(cherry picked from commit 232177e68c547a815c31a2f30b6d1f95cdb3098d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/057d32c3839eff8df6b26dc91252d38d2cb2d6cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/057d32c3839eff8df6b26dc91252d38d2cb2d6cf"}]},"branch":"refs/heads/stable/victoria"},"94679941368d37520975e0a1dc81d27424167191":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-03-24 14:16:35.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/36/779636/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/779636/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/779636/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/779636/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/779636/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/779636/2"}}},"commit":{"parents":[{"commit":"8e216b4913e8b842cab983c100165949bed7def2","subject":"Make functional tests set node_staging_uri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8e216b4913e8b842cab983c100165949bed7def2"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 20:52:03.000000000","tz":-480},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-24 14:16:46.000000000","tz":-420},"subject":"Add housekeeping module and staging cleaner","message":"Add housekeeping module and staging cleaner\n\nAs noted in previous discussions, glance should clean its staging\ndirectory on startup. This is important for scenarios where we\nstarted an import operation, but failed in the middle. If, when we\nrecover, the image has already been deleted from the database, then\nwe will never remove the (potentially very large) residue from disk\nin our staging directory.\n\nThis is currently a problem with web-download, but will also occur\nwith glance-direct once we have the non-shared distributed import\nfunctionality merged.\n\nConflicts:\n\tglance/tests/unit/common/test_wsgi_app.py: Minor differences\n\nCloses-Bug: #1913625\nChange-Id: Ib80e9cfb58680f9e8ead5993dc206f4da882dd09\n(cherry picked from commit 232177e68c547a815c31a2f30b6d1f95cdb3098d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/94679941368d37520975e0a1dc81d27424167191"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/94679941368d37520975e0a1dc81d27424167191"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
