)]}'
{"id":"openstack%2Fglance~779842","triplet_id":"openstack%2Fglance~stable%2Fvictoria~I5df000eaeeaa4831eb5a021adeb6153cec0734b5","project":"openstack/glance","branch":"stable/victoria","topic":"victoria/lower-constraints","hashtags":[],"change_id":"I5df000eaeeaa4831eb5a021adeb6153cec0734b5","subject":"Remove openstack-tox-lower-constraints voting job","status":"ABANDONED","created":"2021-03-10 17:55:58.000000000","updated":"2021-04-02 00:09:22.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7a2e247b7fce7dded2f78e7c62bc654f9371ae3b","_number":779842,"virtual_id_number":779842,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-11 01:41:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":-1,"date":"2021-03-18 13:56:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}],"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-11 01:41:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-11 20:14:17.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"CC"},{"updated":"2021-03-11 20:14:17.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2021-03-18 13:56:46.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"f248eca3d73aa9fc0e2f716afca7c515a70a7cc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-03-10 17:55:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18efe5c97eb77a7f4b5dd0b5452c609454a1aeb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 01:41:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14adabe49dce4a8e9364bd67c41a2d6c : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7cb32fe6d06c4686a29bff0796202976 : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/920222e5269743b58f76d46037302700 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65926918a5be475bbec43e15d8932905 : SUCCESS in 7m 08s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/678d935bd94b40988d306bc076855cb0 : SUCCESS in 7m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d51d589286054fa8ad6c3657daa2b95a : SUCCESS in 11m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7030bb35cd1b4b67a306ec82e8c09cdc : SUCCESS in 1h 31m 23s","accounts_in_message":[],"_revision_number":1},{"id":"aefb95566c6a7b983ce16389e626447284203b75","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-12 05:30:25.000000000","message":"Patch Set 1:\n\nI think we should discuss this in our upcoming weekly meeting and then decide the action.","accounts_in_message":[],"_revision_number":1},{"id":"1daa14425c504eeee59400056684178d9e933329","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-18 13:56:46.000000000","message":"Patch Set 1: Code-Review-1\n\nthis backport will fix the lower-constraint failure, so no need to remove it now.\nhttps://review.opendev.org/c/openstack/glance/+/781426\n\nYou can abandon this patch.","accounts_in_message":[],"_revision_number":1},{"id":"7a2e247b7fce7dded2f78e7c62bc654f9371ae3b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-04-02 00:09:22.000000000","message":"Abandoned\n\nIn favor of https://review.opendev.org/c/openstack/glance/+/782769","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4ac08e639d311caddbcfa6fcd07309e5ae16c8ae","revisions":{"4ac08e639d311caddbcfa6fcd07309e5ae16c8ae":{"kind":"REWORK","_number":1,"created":"2021-03-10 17:55:58.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/42/779842/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/42/779842/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/779842/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/779842/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/42/779842/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/42/779842/1"}}},"commit":{"parents":[{"commit":"794c0fa19883c245ced7a886f9f5d11bd50228c4","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/victoria\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/794c0fa19883c245ced7a886f9f5d11bd50228c4"}]}],"author":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2021-03-10 17:42:23.000000000","tz":60},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2021-03-10 17:42:23.000000000","tz":60},"subject":"Remove openstack-tox-lower-constraints voting job","message":"Remove openstack-tox-lower-constraints voting job\n\nChange-Id: I5df000eaeeaa4831eb5a021adeb6153cec0734b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4ac08e639d311caddbcfa6fcd07309e5ae16c8ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4ac08e639d311caddbcfa6fcd07309e5ae16c8ae"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
