)]}'
{"id":"openstack%2Fglance~781235","triplet_id":"openstack%2Fglance~stable%2Fvictoria~I6db86b3e17a6a2f78745b40381b9419fb4404a4e","project":"openstack/glance","branch":"stable/victoria","topic":"bug/1914826","hashtags":[],"change_id":"I6db86b3e17a6a2f78745b40381b9419fb4404a4e","subject":"Pass ImageActionWrapper to internal plugins","status":"MERGED","created":"2021-03-18 09:48:27.000000000","updated":"2021-03-18 22:31:27.000000000","submitted":"2021-03-18 22:25:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781235-bug/1914826","meta_rev_id":"1b4766f2e1ba0a77751849976301d1763dac342f","_number":781235,"virtual_id_number":781235,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-18 22:25:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-03-18 22:31:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-03-18 22:31:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2021-03-18 22:31:27.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-03-18 22:31:27.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-18 09:48:27.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-03-18 11:03:53.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-03-18 13:50:58.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-03-18 14:36:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ca8a18017517c9a64292e019c581001f3f925509","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-03-18 09:48:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7151e47fdab09dade40dbb948b79bae1624e1d6","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-18 10:30:47.000000000","message":"Topic set to bug/1914826","accounts_in_message":[],"_revision_number":1},{"id":"956289cd348f822730ef9f1c69285f47d5976884","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-18 11:03:53.000000000","message":"Patch Set 1: Code-Review+1\n\nFrom stable core perspective this looks valid to backport. Also:\n* merged on master\n* clean cherry pick from there\n* unblocks nova\u0027s stable/victoria gate with the patch on top of this (see patch: [1])\nLGTM. (Giving +1 until all tests pass)\n\n[1] https://review.opendev.org/c/openstack/nova/+/781237","accounts_in_message":[],"_revision_number":1},{"id":"23919292a2a670d94934dcf8c70bbcbc005a5945","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-03-18 11:43:00.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"0719ab0d460c9941d2e1e7a08fce35d29b71663e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-18 13:27:13.000000000","message":"Patch Set 2:\n\nNeeded for the actual bugfix. This is just an internal refactor, so should be fine for stable.","accounts_in_message":[],"_revision_number":2},{"id":"da7540f8932def83abbe5f012b968a14b6ccbdbf","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-18 13:27:34.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"64a43abd5324881d6ba4eb7467d49c39ac4b5af5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-18 13:50:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nClean backport and is required to backport which fixes actual bug.\n\nNinja approving as we are short on cores as well as this is a blocker for nova gate.\n\nThank you.","accounts_in_message":[],"_revision_number":2},{"id":"9fb2aa423952f2f68d439db4628d2119e6c11727","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-18 14:01:57.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"038ddcaf842e743ae8683f602d24ee01ec49fc4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-03-18 14:34:18.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"ece259d163fb681d7e84650146bb43b9ee18a8f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 14:36:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/20bfa2dfa98b4411a6812f4c50091463 : SUCCESS in 1h 01m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e8de88424c7d4893a21daf9e6330e29d : SUCCESS in 1h 17m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8aa82936445640719d277419c491a155 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c2f2f40ed2649d49f0155c9838a935c : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55da0851f55e4908bc22f8d7113270a9 : SUCCESS in 7m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1706f41df30b45c3b1659ebe4a9c33c3 : SUCCESS in 8m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5a810b798cc42919749152b982cd57e : SUCCESS in 9m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dd2a6cf2c2084f45823373c90ec03cee : SUCCESS in 9m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2cf01d4bdc4d41a2adc341f882c8645b : SUCCESS in 9m 33s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d5fba6084e08481cbd81ef1f915364fe : SUCCESS in 8m 25s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2d34f4aa3c63407087b9eb0871078880 : SUCCESS in 1h 29m 44s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/6c0691dc3b0d464b83b9c8126bd96f04 : SUCCESS in 1h 16m 14s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/035479c918cc4074bae196c3bd098b2d : SUCCESS in 1h 14m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bcead8d164ef4821b60bacb265ad8ea4 : SUCCESS in 1h 01m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/69a73e008a594235bc92931b06bb5ef4 : FAILURE in 1h 19m 29s","accounts_in_message":[],"_revision_number":2},{"id":"7744baa3deba4899ab2e01a659a12c1bac518852","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-03-18 14:46:12.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent, and commit message as Patch Set 3.","accounts_in_message":[],"_revision_number":4},{"id":"d3753f6d9cce0ea259d3a24d5a4cd66e0d5d3c5f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-18 14:49:51.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"abbce364920cc69ecd2092584deb053b042b3fe1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-03-18 15:03:51.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent, and commit message as Patch Set 4.","accounts_in_message":[],"_revision_number":5},{"id":"291e45f90afab9408bca7c1f6b658c099000a07d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 17:41:18.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/292288cf66e6401bbe5793327125170d : SUCCESS in 1h 08m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f977d30f828f4d73b4899d4d27467351 : SUCCESS in 1h 21m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7a9de5622dca4e239eed57685b561d99 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c57823895814cd9bd9c995fe9123daf : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d218290351da4b9aad047609e988162f : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5cc698ee6ec84546916ff1d99a506d32 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9d13c816618474e90d5dd230c72c0ea : SUCCESS in 7m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/243a1c4b517e4dc0ba0f034b4f5ec9b2 : SUCCESS in 9m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c5d5ec2de789437f8f740e354e70a828 : FAILURE in 8m 45s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/892a9032f3d340beb7db37aba635a154 : SUCCESS in 5m 49s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/00b418260e6b45a7ab9a15f4de48851d : SUCCESS in 1h 35m 44s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/4d3c9bb4a50e4732a4846d069cefdc98 : SUCCESS in 1h 12m 55s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/8b67d41089d1460fba8a48d0039c615d : SUCCESS in 1h 19m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6538c08854324e24a7f6dc72609ca5a2 : SUCCESS in 52m 51s","accounts_in_message":[],"_revision_number":5},{"id":"98bdb051bf55e70b5593579304cf27ad13c8646c","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-03-18 17:52:44.000000000","message":"Patch Set 5:\n\nrecheck intermittent func failure","accounts_in_message":[],"_revision_number":5},{"id":"ee8d7c222ff3b7885335a53ffc309bd9f73fbf0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 19:32:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/8c684b015e9a44d798a3ae20edbe1c32 : SUCCESS in 1h 03m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/89e152d2d9d94385882cf949d0852283 : SUCCESS in 1h 14m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fbb5da064e44200adbf8dd541df9b88 : SUCCESS in 9m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9abd98d0c39433f98874b6235095fbe : SUCCESS in 7m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/006ed5d12492400dbe626b4e7fcb0566 : SUCCESS in 8m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54385152294247f79c9ad4d2175712ed : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08f67ea93c534331b19584c479373a24 : SUCCESS in 9m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/49c933779dcd4b9cb35ad4b342db4e94 : SUCCESS in 10m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/33ca166da41d4e3199193fa3b7dd8afe : SUCCESS in 11m 04s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/376ed8509a744acc8d3ada78ae32aec7 : SUCCESS in 7m 09s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e70a1c58d67349df87002b3c387094f5 : SUCCESS in 1h 25m 22s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/06e72b436f934724b2c84393917d5ff9 : SUCCESS in 1h 24m 28s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/7424eefe35a3439a98d92a5ac3e85113 : SUCCESS in 1h 17m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb0e3e32a812477cb6450e59d06938e7 : SUCCESS in 1h 06m 26s","accounts_in_message":[],"_revision_number":5},{"id":"900f34bf0e6dae84723f47e5e4c26ea86b529644","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 19:33:23.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b4f9b8f862d96de586a5ac29bd18432d734145ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 22:25:16.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/09255e96884a47a8a1ee45ca555fad3f : SUCCESS in 1h 00m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/82a6fb55d7144516a47388fd67f58afb : SUCCESS in 1h 10m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fedf9529a2af486f8d2de34b2818a5a2 : SUCCESS in 8m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc42f252036b4977b917ce2e813e3be9 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1da2a3ef4d7a42e5af4b620bfc98a6df : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2deee821116e406e9f2a65e47597674b : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/689641ec4c37454c96067564934eec73 : SUCCESS in 9m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6b17a388cce24fbaaedce0631aac087c : SUCCESS in 8m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b57f649625e44795a1e68e85a587943d : SUCCESS in 9m 21s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/74ea0e06e738412d8f73e6ddd0e2063a : SUCCESS in 1h 13m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8918c3ffe5b8408c999a1ac7ab79969a : SUCCESS in 53m 49s","accounts_in_message":[],"_revision_number":5},{"id":"a5ecd174df29a5e36a809693482c7eec935079de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 22:25:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"73be4c1588c3a382fe77dd7f7721b8bdc588818d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 22:31:27.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34af1a3c1c294e7baf29a60324c831d9 : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"43d41159787f10e51d154a08877bade78f3698f2","revisions":{"9d8b63c653bf170656825e01285d9b762099f75b":{"kind":"REWORK","_number":1,"created":"2021-03-18 09:48:27.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/35/781235/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/781235/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/781235/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/781235/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/781235/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/781235/1"}}},"commit":{"parents":[{"commit":"794c0fa19883c245ced7a886f9f5d11bd50228c4","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/victoria\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/794c0fa19883c245ced7a886f9f5d11bd50228c4"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-05 22:09:34.000000000","tz":-480},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-03-18 09:47:16.000000000","tz":0},"subject":"Pass ImageActionWrapper to internal plugins","message":"Pass ImageActionWrapper to internal plugins\n\nThe _internal_plugins/* tasks are not getting passed the\nImportActionWrapper, and thus are doing things against the image\nthat do not respect the task lock, nor apply multiple updates\natomically.\n\nThis fixes that and brings them in line with the base tasks, using the\nwrapper for things like getting the image_id. The copy_image task\nlooks at the image for a few other things that are not exposed out\nof the wrapper, so I left the image_repo.get() in that task until I\ncan extend it. They are read-only though, so not a big deal.\n\nThe web-download test also was not passing the right options to\nthat task (confusing task_repo and image_repo), which this cleans\nup as well. It was also missing validation of the code that reverts\nthe state to \u0027queued\u0027 when web-download fails, so that is added here\nas well.\n\nChange-Id: I6db86b3e17a6a2f78745b40381b9419fb4404a4e\nRelated-Bug: #1914826\n(cherry picked from commit 68646db35d4a389237b9c1db75c074ba90fa1ec6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9d8b63c653bf170656825e01285d9b762099f75b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9d8b63c653bf170656825e01285d9b762099f75b"}]},"branch":"refs/heads/stable/victoria"},"83973f2dcee53bf96af314763a35979518f83637":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-03-18 11:43:00.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/35/781235/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/781235/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/781235/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/781235/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/781235/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/781235/2"}}},"commit":{"parents":[{"commit":"2486837a505c2a269ed00bc679b25df63281c332","subject":"Bump lower_constraints and requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2486837a505c2a269ed00bc679b25df63281c332"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-05 22:09:34.000000000","tz":-480},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-03-18 11:42:15.000000000","tz":0},"subject":"Pass ImageActionWrapper to internal plugins","message":"Pass ImageActionWrapper to internal plugins\n\nThe _internal_plugins/* tasks are not getting passed the\nImportActionWrapper, and thus are doing things against the image\nthat do not respect the task lock, nor apply multiple updates\natomically.\n\nThis fixes that and brings them in line with the base tasks, using the\nwrapper for things like getting the image_id. The copy_image task\nlooks at the image for a few other things that are not exposed out\nof the wrapper, so I left the image_repo.get() in that task until I\ncan extend it. They are read-only though, so not a big deal.\n\nThe web-download test also was not passing the right options to\nthat task (confusing task_repo and image_repo), which this cleans\nup as well. It was also missing validation of the code that reverts\nthe state to \u0027queued\u0027 when web-download fails, so that is added here\nas well.\n\nChange-Id: I6db86b3e17a6a2f78745b40381b9419fb4404a4e\nRelated-Bug: #1914826\n(cherry picked from commit 68646db35d4a389237b9c1db75c074ba90fa1ec6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/83973f2dcee53bf96af314763a35979518f83637"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/83973f2dcee53bf96af314763a35979518f83637"}]},"branch":"refs/heads/stable/victoria"},"5dba7a5ae88c67843472667700343cfbc4f86b4d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-03-18 14:34:18.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/35/781235/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/781235/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/781235/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/781235/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/781235/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/781235/3"}}},"commit":{"parents":[{"commit":"29f03047d2b6c830a4bbfe2e18380bb41832d6c4","subject":"[stable-only] Remove nova-ceph-multistore to avoid blocking fix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/29f03047d2b6c830a4bbfe2e18380bb41832d6c4"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-05 22:09:34.000000000","tz":-480},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-03-18 14:30:52.000000000","tz":0},"subject":"Pass ImageActionWrapper to internal plugins","message":"Pass ImageActionWrapper to internal plugins\n\nThe _internal_plugins/* tasks are not getting passed the\nImportActionWrapper, and thus are doing things against the image\nthat do not respect the task lock, nor apply multiple updates\natomically.\n\nThis fixes that and brings them in line with the base tasks, using the\nwrapper for things like getting the image_id. The copy_image task\nlooks at the image for a few other things that are not exposed out\nof the wrapper, so I left the image_repo.get() in that task until I\ncan extend it. They are read-only though, so not a big deal.\n\nThe web-download test also was not passing the right options to\nthat task (confusing task_repo and image_repo), which this cleans\nup as well. It was also missing validation of the code that reverts\nthe state to \u0027queued\u0027 when web-download fails, so that is added here\nas well.\n\nChange-Id: I6db86b3e17a6a2f78745b40381b9419fb4404a4e\nRelated-Bug: #1914826\n(cherry picked from commit 68646db35d4a389237b9c1db75c074ba90fa1ec6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5dba7a5ae88c67843472667700343cfbc4f86b4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5dba7a5ae88c67843472667700343cfbc4f86b4d"}]},"branch":"refs/heads/stable/victoria"},"0a4dbc2e1bfc6920101053250084ade733758d86":{"kind":"NO_CHANGE","_number":4,"created":"2021-03-18 14:46:12.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/35/781235/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/781235/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/781235/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/781235/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/781235/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/781235/4"}}},"commit":{"parents":[{"commit":"a94a2e8abaa898670217a0cf9ed11cdf0ee63bef","subject":"Bump lower_constraints and requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a94a2e8abaa898670217a0cf9ed11cdf0ee63bef"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-05 22:09:34.000000000","tz":-480},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-03-18 14:45:33.000000000","tz":0},"subject":"Pass ImageActionWrapper to internal plugins","message":"Pass ImageActionWrapper to internal plugins\n\nThe _internal_plugins/* tasks are not getting passed the\nImportActionWrapper, and thus are doing things against the image\nthat do not respect the task lock, nor apply multiple updates\natomically.\n\nThis fixes that and brings them in line with the base tasks, using the\nwrapper for things like getting the image_id. The copy_image task\nlooks at the image for a few other things that are not exposed out\nof the wrapper, so I left the image_repo.get() in that task until I\ncan extend it. They are read-only though, so not a big deal.\n\nThe web-download test also was not passing the right options to\nthat task (confusing task_repo and image_repo), which this cleans\nup as well. It was also missing validation of the code that reverts\nthe state to \u0027queued\u0027 when web-download fails, so that is added here\nas well.\n\nChange-Id: I6db86b3e17a6a2f78745b40381b9419fb4404a4e\nRelated-Bug: #1914826\n(cherry picked from commit 68646db35d4a389237b9c1db75c074ba90fa1ec6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0a4dbc2e1bfc6920101053250084ade733758d86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0a4dbc2e1bfc6920101053250084ade733758d86"}]},"branch":"refs/heads/stable/victoria"},"43d41159787f10e51d154a08877bade78f3698f2":{"kind":"NO_CHANGE","_number":5,"created":"2021-03-18 15:03:51.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/35/781235/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/35/781235/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/781235/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/781235/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/35/781235/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/35/781235/5"}}},"commit":{"parents":[{"commit":"508800512d7716736425f88102a0fb09730fdfe8","subject":"Bump lower_constraints and requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/508800512d7716736425f88102a0fb09730fdfe8"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-05 22:09:34.000000000","tz":-480},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-03-18 15:02:17.000000000","tz":0},"subject":"Pass ImageActionWrapper to internal plugins","message":"Pass ImageActionWrapper to internal plugins\n\nThe _internal_plugins/* tasks are not getting passed the\nImportActionWrapper, and thus are doing things against the image\nthat do not respect the task lock, nor apply multiple updates\natomically.\n\nThis fixes that and brings them in line with the base tasks, using the\nwrapper for things like getting the image_id. The copy_image task\nlooks at the image for a few other things that are not exposed out\nof the wrapper, so I left the image_repo.get() in that task until I\ncan extend it. They are read-only though, so not a big deal.\n\nThe web-download test also was not passing the right options to\nthat task (confusing task_repo and image_repo), which this cleans\nup as well. It was also missing validation of the code that reverts\nthe state to \u0027queued\u0027 when web-download fails, so that is added here\nas well.\n\nChange-Id: I6db86b3e17a6a2f78745b40381b9419fb4404a4e\nRelated-Bug: #1914826\n(cherry picked from commit 68646db35d4a389237b9c1db75c074ba90fa1ec6)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/43d41159787f10e51d154a08877bade78f3698f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/43d41159787f10e51d154a08877bade78f3698f2"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
