)]}'
{"id":"openstack%2Fglance~781236","triplet_id":"openstack%2Fglance~stable%2Fvictoria~Iebbb2dcb767ecf3c965f34f1ca04af20a2039be1","project":"openstack/glance","branch":"stable/victoria","topic":"bug/1914826","hashtags":[],"change_id":"Iebbb2dcb767ecf3c965f34f1ca04af20a2039be1","subject":"Make web-download revert all stores on fail","status":"MERGED","created":"2021-03-18 09:48:27.000000000","updated":"2021-03-18 23:01:24.000000000","submitted":"2021-03-18 22:56:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781236-bug/1914826","meta_rev_id":"ac5cb89a07b495fb3cbe16ab810087e3e10e7d1d","_number":781236,"virtual_id_number":781236,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-18 22:56:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-03-18 15:17:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-03-18 15:27:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-03-18 15:27:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-18 09:48:27.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-03-18 11:04:41.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2021-03-18 13:53:42.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-03-18 14:22:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d5902ea413e7719a751da0f6070a88a173d8ce4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-03-18 09:48:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"31612b460785d3f738594b71611725a82e0c21ac","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-18 10:30:59.000000000","message":"Topic set to bug/1914826","accounts_in_message":[],"_revision_number":1},{"id":"62d913d35e057e9528f70fef8ccd29121c354420","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-18 11:04:41.000000000","message":"Patch Set 1: Code-Review+1\n\nFrom stable core perspective this looks valid to backport. Also:\n* merged on master\n* clean cherry pick from there\n* unblocks nova\u0027s stable/victoria gate with the parent patch (see test patch: [1])\nLGTM. (Giving +1 until all tests pass)\n\n[1] https://review.opendev.org/c/openstack/nova/+/781237","accounts_in_message":[],"_revision_number":1},{"id":"0a3e4696b74945be0e9113ee31cdf06f785dea8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-03-18 11:43:00.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"ef722aeab9e245e32c12b6e6b1563fd33eee09f0","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-18 13:30:55.000000000","message":"Patch Set 2: Code-Review+1\n\nThis is a bug fix which affects the correctness of the only API we have in victoria to determine that a web-download has failed, so this seems like a good addition to me.","accounts_in_message":[],"_revision_number":2},{"id":"7da0b6092be33742123d8f1eefa3815c5c5bc8b6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-18 13:53:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nClean backport, good to have this fix in victoria as well.\n\nNinja approving as we are short on cores as well as this is a blocker for nova gate.\n\nThank you.","accounts_in_message":[],"_revision_number":2},{"id":"f5836daf94ff3a5b80a033d4d617e9eda284e813","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-18 14:02:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9bc1e41e431a3200ae0ad3ad395665464b136dd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 14:22:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/bd3f30165d2849aba63b6b46f12cc27b : SUCCESS in 1h 00m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6e8ffab728ca41dea6d51dea2d8093b3 : SUCCESS in 1h 14m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/69db180de2454de880a240ad34bdd83a : SUCCESS in 7m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d754dd4b55ec46c8adec1d879b98ae31 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d8b9d7628554b4a9fe4572a35bcfd99 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c5af2ff7ba7d454ab99b71248860266d : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81f7511959d54eb68369fa7f2aa813b7 : SUCCESS in 6m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/089242726b3c44219e0ae2bcfd3b66c8 : SUCCESS in 9m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4b1580032e6749bbbf1414fb5c6ac706 : SUCCESS in 9m 47s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/5efd1876fe7940379393d696362490d5 : SUCCESS in 5m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/523245e206a64ebe8d18ce98833cea44 : SUCCESS in 1h 17m 38s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/f698b54b758d4fb1a63ac32b13defe4f : SUCCESS in 1h 12m 46s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/685ac92681e14992940bb096829df075 : SUCCESS in 1h 14m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75e680e48aa4497f985c3ffaceecceba : SUCCESS in 58m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fa15b734f316496eb633c6baea6ea904 : SUCCESS in 1h 15m 16s","accounts_in_message":[],"_revision_number":2},{"id":"38a4edcb8bda4bbc522ab2ab0c867a04df42fa0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-03-18 14:34:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6cd8bd78a8045bbb315d78c6052768db00138127","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-03-18 14:41:56.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"6524129d591261723a4e5eddf316f9f44b3eb07c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-03-18 14:46:12.000000000","message":"Uploaded patch set 5: New patch set was added with same tree, parent, and commit message as Patch Set 4.","accounts_in_message":[],"_revision_number":5},{"id":"627cc02ae25bf16f7240e2298940f33e10c56fe9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-18 14:51:53.000000000","message":"Patch Set 5: Code-Review+2\n\nThis adds the dropped nova-ceph-multistore job again with the actual fix.\nLGTM!","accounts_in_message":[],"_revision_number":5},{"id":"158b3edb7c36635dced294dd64aeb9d4a9994a67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-03-18 15:03:51.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"2471da9de2114d1f9dbaf7a723ac32211d218d4c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-18 15:17:37.000000000","message":"Patch Set 6: Code-Review+2\n\nThe original clean cherry-pick is extended with the restoration of nova-ceph-multnode job, as noted. LGTM.","accounts_in_message":[],"_revision_number":6},{"id":"9dc1a15a1a7d13c92cd58289eb1b6c93e2a58b32","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-18 15:27:31.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nLooks good to me, this restores the dropped nova-ceph job along with actual fix.","accounts_in_message":[],"_revision_number":6},{"id":"8e8371ffb1d674b75f78ebf1147b3685dc1d6a90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 17:40:42.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/87cf1128867745b5b32f43a45c19c120 : SUCCESS in 56m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e6df6ac8bb824e87a7e5ec68a4fc45f6 : FAILURE in 1h 02m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3e7df737a3144ed7bbda3e26700f162d : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dbe5d279287442eb982424f8d46f1be : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f8c39e59de6a4607ac9d4a50c930891b : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3af7691fdcf46c98063d1ddf65a9084 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c750426fb67498fa3501d8b3d1921d0 : SUCCESS in 6m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/04ea1c6825d84dbe9258b8baadd8057c : SUCCESS in 8m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e79ef907cc2c4e53ba42f5b1a0b76846 : SUCCESS in 9m 37s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d7ae015db77c4d8ab23087e8345c3866 : SUCCESS in 7m 42s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/10333725eedc4057a94aa8b55b37c8e8 : SUCCESS in 1h 31m 52s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/1c62f64573434d4894911095e539b39c : SUCCESS in 1h 17m 55s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/7d0e52c246b149079852389ddcf12847 : SUCCESS in 1h 35m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6bc65816eeeb464b8b535583e00b09f1 : SUCCESS in 1h 09m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ee837fe9ade24a239b44a7396eabe7ea : SUCCESS in 1h 14m 41s","accounts_in_message":[],"_revision_number":6},{"id":"ec357c2186f7aaf3359c03da1f1821586f932d27","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-03-18 17:53:20.000000000","message":"Patch Set 6:\n\nrecheck bug #1882521","accounts_in_message":[],"_revision_number":6},{"id":"a15472d647042bbf238dd721a82292fd2c3e0610","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 19:38:50.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/ffcd62d8dbca4621896b9208c66cc70d : SUCCESS in 1h 00m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f4534372ef6c4a4193c0dbad8cbe69dc : SUCCESS in 1h 28m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3112b3936d248a881c6b92ab2ae99b2 : SUCCESS in 9m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d078ee3d43b4b6f9485a13616d95c0b : SUCCESS in 8m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6fa9b417b9940dcafada958716ecc87 : SUCCESS in 7m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acff811f560d430c8b3249d2ba05b1e9 : SUCCESS in 9m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d53e9f18853456b975f0fbf4985ca7a : SUCCESS in 10m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/087972422d8e445fb05e2964ba374499 : SUCCESS in 11m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/318a858799fc4b8789594d43f820f214 : SUCCESS in 10m 21s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6dd279ff12134b35bb148d67cf5c634d : SUCCESS in 7m 23s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/70bd2db6815c40668631c4115be9952a : SUCCESS in 1h 19m 25s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/dcc9d2399fe84e139e51cfd918b95c82 : SUCCESS in 1h 16m 44s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/be6ee7b57c97469f83fb3509bdb4ff27 : FAILURE in 1h 06m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96ee0ef03d8c418ea11072a1449ca518 : SUCCESS in 58m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c3dfa3b3b53f4b6fac78ec871006e42f : SUCCESS in 1h 19m 34s","accounts_in_message":[],"_revision_number":6},{"id":"c9581cf43d2487491bb036c198ff7dacfb105fed","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-03-18 19:59:16.000000000","message":"Patch Set 6:\n\nrecheck c-bak","accounts_in_message":[],"_revision_number":6},{"id":"bf70a330c09e668a3545b1a7c26f89ea244dcb83","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-03-18 19:59:35.000000000","message":"Patch Set 6:\n\nI\u0027m not entirely sure why this change is running c-bak tests tbh.","accounts_in_message":[],"_revision_number":6},{"id":"10647aa79a450a3b623eb90ae5b5bc4eb2bde2f4","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-18 20:05:12.000000000","message":"Patch Set 6:\n\n\u003e Patch Set 6:\n\u003e \n\u003e I\u0027m not entirely sure why this change is running c-bak tests tbh.\n\nProbably because it\u0027s just a default, but some of the cinder stuff uses glance and there\u0027s maybe some overlap between glance, cinder, backups, and snapshots?","accounts_in_message":[],"_revision_number":6},{"id":"664a861766662b8627b33731b25487940f95cb24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 21:40:48.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/b02eb3f6576f45d9b1470867c83f92f0 : SUCCESS in 59m 50s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/62cd2cadb25e4a67b97145946236c47c : SUCCESS in 1h 14m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cc8e4214662242c0a47def07fd8d2abc : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8131b855df414a01bfce0e49a8760feb : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0a5359d2d7e4cc88c1f3ebc2899e4c4 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4715fbb159c4da2afbf07a4d31b6e7a : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9604d7d57dfe4caa86b766e09f03181d : SUCCESS in 6m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ed9bba16b58a4cf2a222e8b2699b7145 : SUCCESS in 10m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d3eb1d1e2ef94b8daea70af247222859 : SUCCESS in 7m 49s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/98339fe96c9d4ec88a670384159fe151 : SUCCESS in 9m 06s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/362a72771fa849cbbb35d56b166a019f : SUCCESS in 1h 33m 30s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/0c82337dc27e4ff29f2fec4ed4e7110f : SUCCESS in 1h 15m 39s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/7f7dd37cf8b04c568c9fed853c37202b : SUCCESS in 1h 17m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37854f02e71c420a9a89e7c90bde3298 : SUCCESS in 1h 03m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/03e7a61885674cc3baa6e408caf48712 : SUCCESS in 1h 13m 33s","accounts_in_message":[],"_revision_number":6},{"id":"732a7e1f870107365dd5dfccc39655bf88a05c0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 21:41:14.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"965191f3a336198ce941091020c207e12e194fd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 22:56:39.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/0b06382de2c34a178941c7182cb637e8 : SUCCESS in 1h 08m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3437ede7f7a440fab9be29cf6a5b0767 : SUCCESS in 1h 14m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/87cfc9d46e8b43ba9b23e557fcbbf916 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fd8f76022bc4713bb314870ca9a3a5b : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2ea51a57c4347ffb70c5247f609ce39 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce3d58a082284901a3fb2e1ffd8e3350 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/003401f4964c4b6a8f48d224f4bc5802 : SUCCESS in 6m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fe921d5dce4244f7947bdd9fd6217072 : SUCCESS in 7m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9e8de83182c14716958f177a9f1561e4 : SUCCESS in 8m 07s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/4c60141f2efb4273aaf91e9184028475 : SUCCESS in 1h 07m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed4ee8c6018849338c4c9ff12489113f : SUCCESS in 54m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/14ae0b2ecdad4a7b9d8138f692e9b738 : SUCCESS in 1h 10m 40s","accounts_in_message":[],"_revision_number":6},{"id":"5f9c5319e2a0ed55ea5a60dbc93846fb6882a96f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 22:56:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"ac5cb89a07b495fb3cbe16ab810087e3e10e7d1d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 23:01:24.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c3e4da6a068484ba98f38efb08e82c4 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"9cb3240b94dc4ca7fa288fc66eba1499fb99295f","revisions":{"d84354e81e2a6436163e6c925d26a6bc78e42f67":{"kind":"REWORK","_number":1,"created":"2021-03-18 09:48:27.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/36/781236/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/781236/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/781236/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/781236/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/781236/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/781236/1"}}},"commit":{"parents":[{"commit":"9d8b63c653bf170656825e01285d9b762099f75b","subject":"Pass ImageActionWrapper to internal plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9d8b63c653bf170656825e01285d9b762099f75b"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-05 22:35:51.000000000","tz":-480},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-03-18 09:47:22.000000000","tz":0},"subject":"Make web-download revert all stores on fail","message":"Make web-download revert all stores on fail\n\nIf web-download fails to stage the image, no subsequent import to\nstore tasks could have run. Thus, we should remove all of them from\n\"importing\" and add them to \"failed\" so that it is obvious from the\noutside that we failed.\n\nThis is another good reason for having the $image/tasks API, but we\nalso should not continue advertising \"yes we\u0027re importing to $store\"\nwhen we are not.\n\nChange-Id: Iebbb2dcb767ecf3c965f34f1ca04af20a2039be1\nCloses-Bug: #1914826\n(cherry picked from commit cbb2af6e329c1bf6c1c39cb59f95aaad2c0d7a07)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d84354e81e2a6436163e6c925d26a6bc78e42f67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d84354e81e2a6436163e6c925d26a6bc78e42f67"}]},"branch":"refs/heads/stable/victoria"},"065600decc6bc17a027374964066729f5f329962":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-03-18 11:43:00.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/36/781236/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/781236/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/781236/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/781236/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/781236/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/781236/2"}}},"commit":{"parents":[{"commit":"83973f2dcee53bf96af314763a35979518f83637","subject":"Pass ImageActionWrapper to internal plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/83973f2dcee53bf96af314763a35979518f83637"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-05 22:35:51.000000000","tz":-480},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-03-18 11:42:15.000000000","tz":0},"subject":"Make web-download revert all stores on fail","message":"Make web-download revert all stores on fail\n\nIf web-download fails to stage the image, no subsequent import to\nstore tasks could have run. Thus, we should remove all of them from\n\"importing\" and add them to \"failed\" so that it is obvious from the\noutside that we failed.\n\nThis is another good reason for having the $image/tasks API, but we\nalso should not continue advertising \"yes we\u0027re importing to $store\"\nwhen we are not.\n\nChange-Id: Iebbb2dcb767ecf3c965f34f1ca04af20a2039be1\nCloses-Bug: #1914826\n(cherry picked from commit cbb2af6e329c1bf6c1c39cb59f95aaad2c0d7a07)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/065600decc6bc17a027374964066729f5f329962"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/065600decc6bc17a027374964066729f5f329962"}]},"branch":"refs/heads/stable/victoria"},"ea32b37378d89ee29aab93ff2a7e875b8782e909":{"kind":"REWORK","_number":3,"created":"2021-03-18 14:34:18.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/36/781236/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/781236/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/781236/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/781236/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/781236/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/781236/3"}}},"commit":{"parents":[{"commit":"5dba7a5ae88c67843472667700343cfbc4f86b4d","subject":"Pass ImageActionWrapper to internal plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5dba7a5ae88c67843472667700343cfbc4f86b4d"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-05 22:35:51.000000000","tz":-480},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-03-18 14:33:28.000000000","tz":0},"subject":"Make web-download revert all stores on fail","message":"Make web-download revert all stores on fail\n\nIf web-download fails to stage the image, no subsequent import to\nstore tasks could have run. Thus, we should remove all of them from\n\"importing\" and add them to \"failed\" so that it is obvious from the\noutside that we failed.\n\nThis is another good reason for having the $image/tasks API, but we\nalso should not continue advertising \"yes we\u0027re importing to $store\"\nwhen we are not.\n\nNOTE(lyarwood): This includes a squashed revert of\nI274d62f2c55fea0af4adea383c2dad04f031b3d2 that removed\nnova-ceph-multistore from the gate to ensure this change resolves bug\n\nChange-Id: Iebbb2dcb767ecf3c965f34f1ca04af20a2039be1\nCloses-Bug: #1914826\n(cherry picked from commit cbb2af6e329c1bf6c1c39cb59f95aaad2c0d7a07)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ea32b37378d89ee29aab93ff2a7e875b8782e909"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ea32b37378d89ee29aab93ff2a7e875b8782e909"}]},"branch":"refs/heads/stable/victoria"},"9aa01e1e84826d9b292952a7d0671b50cf52875b":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-03-18 14:41:56.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/36/781236/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/781236/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/781236/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/781236/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/781236/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/781236/4"}}},"commit":{"parents":[{"commit":"5dba7a5ae88c67843472667700343cfbc4f86b4d","subject":"Pass ImageActionWrapper to internal plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5dba7a5ae88c67843472667700343cfbc4f86b4d"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-05 22:35:51.000000000","tz":-480},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-03-18 14:39:04.000000000","tz":0},"subject":"Make web-download revert all stores on fail","message":"Make web-download revert all stores on fail\n\nIf web-download fails to stage the image, no subsequent import to\nstore tasks could have run. Thus, we should remove all of them from\n\"importing\" and add them to \"failed\" so that it is obvious from the\noutside that we failed.\n\nThis is another good reason for having the $image/tasks API, but we\nalso should not continue advertising \"yes we\u0027re importing to $store\"\nwhen we are not.\n\nNOTE(lyarwood): This includes a squashed revert of\nI274d62f2c55fea0af4adea383c2dad04f031b3d2 that removed\nnova-ceph-multistore. This is reintroduced here to ensure this change\nresolves the stable/victoria specifci bug #1919993 and the original\nbug #1914826.\n\nChange-Id: Iebbb2dcb767ecf3c965f34f1ca04af20a2039be1\nCloses-Bug: #1914826\nCloses-Bug: #1919993\n(cherry picked from commit cbb2af6e329c1bf6c1c39cb59f95aaad2c0d7a07)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9aa01e1e84826d9b292952a7d0671b50cf52875b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9aa01e1e84826d9b292952a7d0671b50cf52875b"}]},"branch":"refs/heads/stable/victoria"},"7d5b8c6c1082676fbb7bb100e0b949b4f931b61b":{"kind":"NO_CHANGE","_number":5,"created":"2021-03-18 14:46:12.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/36/781236/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/781236/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/781236/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/781236/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/781236/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/781236/5"}}},"commit":{"parents":[{"commit":"0a4dbc2e1bfc6920101053250084ade733758d86","subject":"Pass ImageActionWrapper to internal plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0a4dbc2e1bfc6920101053250084ade733758d86"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-05 22:35:51.000000000","tz":-480},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-03-18 14:45:33.000000000","tz":0},"subject":"Make web-download revert all stores on fail","message":"Make web-download revert all stores on fail\n\nIf web-download fails to stage the image, no subsequent import to\nstore tasks could have run. Thus, we should remove all of them from\n\"importing\" and add them to \"failed\" so that it is obvious from the\noutside that we failed.\n\nThis is another good reason for having the $image/tasks API, but we\nalso should not continue advertising \"yes we\u0027re importing to $store\"\nwhen we are not.\n\nNOTE(lyarwood): This includes a squashed revert of\nI274d62f2c55fea0af4adea383c2dad04f031b3d2 that removed\nnova-ceph-multistore. This is reintroduced here to ensure this change\nresolves the stable/victoria specifci bug #1919993 and the original\nbug #1914826.\n\nChange-Id: Iebbb2dcb767ecf3c965f34f1ca04af20a2039be1\nCloses-Bug: #1914826\nCloses-Bug: #1919993\n(cherry picked from commit cbb2af6e329c1bf6c1c39cb59f95aaad2c0d7a07)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7d5b8c6c1082676fbb7bb100e0b949b4f931b61b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7d5b8c6c1082676fbb7bb100e0b949b4f931b61b"}]},"branch":"refs/heads/stable/victoria"},"9cb3240b94dc4ca7fa288fc66eba1499fb99295f":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2021-03-18 15:03:51.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/36/781236/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/781236/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/781236/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/781236/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/781236/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/781236/6"}}},"commit":{"parents":[{"commit":"43d41159787f10e51d154a08877bade78f3698f2","subject":"Pass ImageActionWrapper to internal plugins","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/43d41159787f10e51d154a08877bade78f3698f2"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-05 22:35:51.000000000","tz":-480},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-03-18 15:02:29.000000000","tz":0},"subject":"Make web-download revert all stores on fail","message":"Make web-download revert all stores on fail\n\nIf web-download fails to stage the image, no subsequent import to\nstore tasks could have run. Thus, we should remove all of them from\n\"importing\" and add them to \"failed\" so that it is obvious from the\noutside that we failed.\n\nThis is another good reason for having the $image/tasks API, but we\nalso should not continue advertising \"yes we\u0027re importing to $store\"\nwhen we are not.\n\nNOTE(lyarwood): This includes a partial squashed revert of the\n.zuul.yaml changes in Ifd55c44fef4e2187052d77084dc9c0fa9c9a0d16 that\nremoved nova-ceph-multistore. This is reintroduced here to ensure this\nchange resolves the stable/victoria specific bug #1919993 and the\noriginal bug #1914826.\n\nChange-Id: Iebbb2dcb767ecf3c965f34f1ca04af20a2039be1\nCloses-Bug: #1914826\nCloses-Bug: #1919993\n(cherry picked from commit cbb2af6e329c1bf6c1c39cb59f95aaad2c0d7a07)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9cb3240b94dc4ca7fa288fc66eba1499fb99295f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9cb3240b94dc4ca7fa288fc66eba1499fb99295f"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
