)]}'
{"id":"openstack%2Fglance~781426","triplet_id":"openstack%2Fglance~stable%2Fvictoria~Ifd55c44fef4e2187052d77084dc9c0fa9c9a0d16","project":"openstack/glance","branch":"stable/victoria","topic":"bug/1914826","hashtags":[],"change_id":"Ifd55c44fef4e2187052d77084dc9c0fa9c9a0d16","subject":"Bump lower_constraints and requirements","status":"MERGED","created":"2021-03-18 11:43:00.000000000","updated":"2021-03-18 18:05:30.000000000","submitted":"2021-03-18 17:48:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781426-bug/1914826","meta_rev_id":"7dd1ac898fb6371aa609cb3434ebe058533f6d51","_number":781426,"virtual_id_number":781426,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-18 17:48:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2021-03-18 15:19:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2021-03-18 15:24:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2021-03-18 15:24:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-18 11:43:00.000000000","updated_by":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2021-03-18 13:48:10.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-03-18 14:33:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-18 15:19:27.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"657c5353e6b603df2b7b250f9ea13fceb358d406","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-03-18 11:43:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1cbfbb74ee26f913b364a95f8680c2c4dfc91d44","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-03-18 13:34:53.000000000","message":"Topic set to bug/1914826","accounts_in_message":[],"_revision_number":1},{"id":"13a11eac3cea7367cbfb449f314c88415ba8363b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-18 13:48:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThis is fixing the broken lower constraint job in stable/victoria branch.\nClean backport and is required to backport which fixes actual bug.\n\nNinja approving as we are short on cores as well as this is a blocker for nova gate.\n\nThank you.","accounts_in_message":[],"_revision_number":1},{"id":"f9a29b18865df634490bd3e934523a2ea2bda01e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 14:33:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a21ab273b2d2423fa221ba32f6c13967 : SUCCESS in 3m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/35dd9140aa13404295465be7b8c5a5ab : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a4bdc9227524fbc94cbbc5a53458f7c : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3020bc4ab67540f1992abd5bbd72c449 : SUCCESS in 7m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d959bef9145d487e95c777976cadece6 : SUCCESS in 9m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c22087c29cc54e7ba10189ec1dd0db90 : SUCCESS in 8m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ccd9f3ca9e6148429c77f60198020de8 : SUCCESS in 10m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2f3a08b3ac1248029341720593162204 : SUCCESS in 10m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/758ecce27a0840bdbe6b28079123e45f : FAILURE in 1h 28m 17s","accounts_in_message":[],"_revision_number":1},{"id":"79afa1470bcb1145bf29894b3c79baab42523e30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-03-18 14:46:12.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"4507fd87c81ca0eb882419218b1122fc8f7620dc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-18 14:56:59.000000000","message":"Patch Set 2: Workflow+1\n\nCarrying out Elod\u0027s previous +2 for approval as this is just a rebase on top of other patch.","accounts_in_message":[],"_revision_number":2},{"id":"f3aaa1b7fecad82f46f61f8d014149af9c83bd7c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-03-18 15:03:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d8f58645336003e19b5dce33bd1093ffd1428749","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-03-18 15:19:27.000000000","message":"Patch Set 3: Code-Review+2\n\nThe squash looks good, let\u0027s just wait the tests to pass.","accounts_in_message":[],"_revision_number":3},{"id":"5062f0bacb8d1491c333d316923f820a134f8929","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-18 15:24:38.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nFinally, this looks good now.\nThe job dropped in this patch will be added back in the actual fix.\n\nThank you!","accounts_in_message":[],"_revision_number":3},{"id":"ea7aac1ec47858e2bf525a8858934dfd01d108d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 16:14:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b4969775a74d46579850c9ae1f6e17e4 : SUCCESS in 3m 53s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7906173b3ded45de902972aadaba967a : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f9c9157483842318a41b4d8fdeed634 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1584bc85db84468a163f12655e6752d : SUCCESS in 6m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81da638e0b8547ec9538fb155ab0b65b : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28c0b13891da4bb5837f4e91e79c7ba3 : SUCCESS in 6m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2e29de7629354f09be7aadd4bc56b2f2 : SUCCESS in 7m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/792c11bb4b284d45a611add555e20a29 : SUCCESS in 11m 19s","accounts_in_message":[],"_revision_number":3},{"id":"297354b71890c903ca9cbc49bccc05168f0a208f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 16:28:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0c1c6fde91b81a0717311ebcb65a94bc20dcd00e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 17:48:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/07926da6e694430c8fece8875a6ea30f : SUCCESS in 6m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c05230e094a845f9b6278c432a309250 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad9cd2d9c0d644dfbae3df3a80f51213 : SUCCESS in 8m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5daada664f44829a06456b7b86c1852 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04cece85670a4b69b41037da1427179b : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86f17878397d403f83bb5f94f4825d19 : SUCCESS in 7m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/88934a6a931f4787b008031bb8af6abe : SUCCESS in 12m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6728cc1306a04c8db1b38eb0c0953700 : SUCCESS in 10m 53s","accounts_in_message":[],"_revision_number":3},{"id":"28869c71c1f18bb0725f611e9ba625d53152decf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 17:48:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"7dd1ac898fb6371aa609cb3434ebe058533f6d51","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 18:05:30.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecdcd5ed6dcf4b25aa45e0ef844447e3 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"508800512d7716736425f88102a0fb09730fdfe8","revisions":{"2486837a505c2a269ed00bc679b25df63281c332":{"kind":"REWORK","_number":1,"created":"2021-03-18 11:43:00.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/26/781426/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/781426/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/781426/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/781426/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/781426/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/781426/1"}}},"commit":{"parents":[{"commit":"794c0fa19883c245ced7a886f9f5d11bd50228c4","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/victoria\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/794c0fa19883c245ced7a886f9f5d11bd50228c4"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-12-11 18:20:33.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-03-18 11:42:15.000000000","tz":0},"subject":"Bump lower_constraints and requirements","message":"Bump lower_constraints and requirements\n\nThis is to unblock the stuck gate due to the lower_constraints\njob failing.\n\nChange-Id: Ifd55c44fef4e2187052d77084dc9c0fa9c9a0d16\n(cherry picked from commit 795dad763315884223b79698f3611e1ac1b50a17)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2486837a505c2a269ed00bc679b25df63281c332"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2486837a505c2a269ed00bc679b25df63281c332"}]},"branch":"refs/heads/stable/victoria"},"a94a2e8abaa898670217a0cf9ed11cdf0ee63bef":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-03-18 14:46:12.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/26/781426/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/781426/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/781426/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/781426/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/781426/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/781426/2"}}},"commit":{"parents":[{"commit":"2e5d836f8553b7b8ae6bff44e4eb23c254e585fa","subject":"[stable-only] Remove nova-ceph-multistore to avoid blocking fix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2e5d836f8553b7b8ae6bff44e4eb23c254e585fa"}]}],"author":{"name":"Erno Kuvaja","email":"jokke@usr.fi","date":"2020-12-11 18:20:33.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-03-18 14:45:33.000000000","tz":0},"subject":"Bump lower_constraints and requirements","message":"Bump lower_constraints and requirements\n\nThis is to unblock the stuck gate due to the lower_constraints\njob failing.\n\nChange-Id: Ifd55c44fef4e2187052d77084dc9c0fa9c9a0d16\n(cherry picked from commit 795dad763315884223b79698f3611e1ac1b50a17)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a94a2e8abaa898670217a0cf9ed11cdf0ee63bef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a94a2e8abaa898670217a0cf9ed11cdf0ee63bef"}]},"branch":"refs/heads/stable/victoria"},"508800512d7716736425f88102a0fb09730fdfe8":{"kind":"REWORK","_number":3,"created":"2021-03-18 15:03:51.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/26/781426/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/781426/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/781426/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/781426/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/781426/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/781426/3"}}},"commit":{"parents":[{"commit":"794c0fa19883c245ced7a886f9f5d11bd50228c4","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/victoria\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/794c0fa19883c245ced7a886f9f5d11bd50228c4"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-03-18 14:28:53.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-03-18 15:00:35.000000000","tz":0},"subject":"Bump lower_constraints and requirements","message":"Bump lower_constraints and requirements\n\nThis is to unblock the stuck gate due to the lower_constraints\njob failing.\n\nNOTE(lyarwood): This change also drops the nova-ceph-multistore job\nwhile we work on fixing it via Iebbb2dcb767ecf3c965f34f1ca04af20a2039be1\n\nChange-Id: Ifd55c44fef4e2187052d77084dc9c0fa9c9a0d16\n(cherry picked from commit 795dad763315884223b79698f3611e1ac1b50a17)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/508800512d7716736425f88102a0fb09730fdfe8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/508800512d7716736425f88102a0fb09730fdfe8"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
