)]}'
{"id":"openstack%2Fglance~782431","triplet_id":"openstack%2Fglance~stable%2Fvictoria~Ic2ac2a528206c50c38e948a096daf9eb8e5eb715","project":"openstack/glance","branch":"stable/victoria","topic":"bug/1913625-stable/victoria","hashtags":[],"change_id":"Ic2ac2a528206c50c38e948a096daf9eb8e5eb715","subject":"Make functional tests set node_staging_uri","status":"MERGED","created":"2021-03-24 07:32:26.000000000","updated":"2021-03-31 07:33:19.000000000","submitted":"2021-03-31 07:22:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"782431-bug/1913625-stable/victoria","cherry_pick_of_change":777277,"cherry_pick_of_patch_set":3,"meta_rev_id":"02140a8a489eccf1b6f777a5e639193ea75ba2a4","_number":782431,"virtual_id_number":782431,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-31 07:22:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-03-30 00:46:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-03-30 14:21:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-24 07:32:26.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-03-24 08:47:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-30 00:46:44.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"d84c575b94972b8fd42e3865c80095f4381a8698","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-24 07:32:26.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b11bae2a78bdef5c31c2ad103b43545f7bffed1f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-24 07:33:04.000000000","message":"Patch Set 1:\n\nThis should be depend on https://review.opendev.org/c/openstack/glance/+/782431","accounts_in_message":[],"_revision_number":1},{"id":"f096c45ab8e4fee0190ab50dd332f33e529ef332","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-24 07:33:51.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e This should be depend on https://review.opendev.org/c/openstack/glance/+/782431\n\nIgnore this","accounts_in_message":[],"_revision_number":1},{"id":"11444fedc027ebc1c92bfc8f81a9d34aa4cf2862","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 08:47:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d339b0f022da413db728d85a7a5b9e46 : FAILURE in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/476a8a3833db40dcb74b6ffc8524a0f1 : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bef9ee5ecd5144e9b29e4db69170d840 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fe6aa23576a40e48d6558152b91fa74 : SUCCESS in 10m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36180658beb54ada8904c7e21d84d8a5 : SUCCESS in 6m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5f2e09f80df04846bd582a1e9a848a31 : SUCCESS in 8m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/db4db6b3f8704926951b3fa32a5cbdcb : SUCCESS in 8m 58s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/38bbe98b5b274355b4e2cc05c160a301 : SUCCESS in 4m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6b3acb2557b34188b7a946b227ec5306 : SUCCESS in 1h 13m 18s","accounts_in_message":[],"_revision_number":1},{"id":"72ff1024a70612ceca66c2e5fd7dc5aafd5de612","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-26 04:51:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dde0b565e3836ba36a13f8c4eb9906d8c97e03a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 06:16:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83db084142fa44b5ae00352eb393fc7a : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/616c9164c5234c84a9cbb574d2dfe07e : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb53e3afc3244c2fa4d55d19138860c1 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55641c190abe4662b760326aa4b8942e : SUCCESS in 6m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e5ed05a7560d4e0297ec6862d7e43d69 : SUCCESS in 9m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b130374a23e1469e8798d9eb1df5cc4a : SUCCESS in 9m 45s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d78fa662b71d4ec4900af1330e038f85 : SUCCESS in 5m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2e17402735ad4ef1a056ae259a73ae27 : SUCCESS in 1h 22m 40s","accounts_in_message":[],"_revision_number":1},{"id":"a6da8a2d3d5246fa74df6a1926503e3798916ba3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-03-30 00:46:44.000000000","message":"Patch Set 1: Code-Review+2\n\nLooks good, merged to master.","accounts_in_message":[],"_revision_number":1},{"id":"ee21f679918fbaddeee521d285960b60e6372e73","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-30 14:21:51.000000000","message":"Patch Set 1: Workflow+1\n\nTest related backport which needs for actual backport which is fixing a bug.","accounts_in_message":[],"_revision_number":1},{"id":"4d287dbe3ef860ee769205cfbf8f14184a0716a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 14:22:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"10961f96c3c3f5b1d320c447fadf62611caeb0a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 20:21:50.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8987555c2fcd7b1595328e05e6767fb6b0cf3f84","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 07:22:21.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1213ae88d1f64d6f97bda7d583dd47b0 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b21ba0251dc4d07bfa23d1918e14fa4 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b006f3580d7464a89d3deb8d0ab711c : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b87719a3476749adadb817f3d30ca546 : SUCCESS in 7m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/09fc927d468044f39910af58f3be728b : SUCCESS in 8m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/84d393394194463ebebe11f31ab0fe3c : SUCCESS in 12m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f7fdb8ea29f04048a11567b76409d144 : SUCCESS in 1h 08m 58s","accounts_in_message":[],"_revision_number":1},{"id":"6b41f5b60e93b1ca0e0a5e3c8aa142aac72ea632","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 07:22:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"02140a8a489eccf1b6f777a5e639193ea75ba2a4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 07:33:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95cd00e592d74ea19e4044fba90bf000 : SUCCESS in 1m 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8e216b4913e8b842cab983c100165949bed7def2","revisions":{"8e216b4913e8b842cab983c100165949bed7def2":{"kind":"REWORK","_number":1,"created":"2021-03-24 07:32:26.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/31/782431/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/31/782431/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/782431/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/782431/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/31/782431/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/31/782431/1"}}},"commit":{"parents":[{"commit":"9cb3240b94dc4ca7fa288fc66eba1499fb99295f","subject":"Make web-download revert all stores on fail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9cb3240b94dc4ca7fa288fc66eba1499fb99295f"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-24 02:59:07.000000000","tz":-480},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-03-24 07:32:26.000000000","tz":0},"subject":"Make functional tests set node_staging_uri","message":"Make functional tests set node_staging_uri\n\nCurrently it is not possible to configure the staging directory\nURI of the functional workers. We need to be able to do that in order\nto enable the stage cleaning behavior. Right now, they\u0027re all sharing\n/tmp/staging, which will cause workers to conflict while running in\nparallel. This causes them to use their private test directory,\nwhich may also help some other spurious failures due to interaction.\n\nRelated-Bug: #1913625\nChange-Id: Ic2ac2a528206c50c38e948a096daf9eb8e5eb715\n(cherry picked from commit e9852fb6251c655f0e384c6b0b8677bd556f17a1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8e216b4913e8b842cab983c100165949bed7def2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8e216b4913e8b842cab983c100165949bed7def2"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
