)]}'
{"id":"openstack%2Fglance~782768","triplet_id":"openstack%2Fglance~master~Id030b56b721cb62255936d56f0409c4be9ce21d4","project":"openstack/glance","branch":"master","topic":"bp/glance-unified-quotas","hashtags":[],"change_id":"Id030b56b721cb62255936d56f0409c4be9ce21d4","subject":"Drop lower-constraints jobs","status":"MERGED","created":"2021-03-24 14:49:01.000000000","updated":"2021-06-28 19:00:01.000000000","submitted":"2021-06-28 18:56:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"782768-bp/glance-unified-quotas","meta_rev_id":"e0e75aa9a05c0781dfa492042491f0ca3f5e286a","_number":782768,"virtual_id_number":782768,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-28 18:56:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-06-24 19:21:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"date":"2021-06-28 17:01:40.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":2,"date":"2021-06-28 19:00:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-06-28 15:40:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-28 19:00:01.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-06-28 17:01:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":1,"date":"2021-06-28 17:02:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-24 16:16:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-24 16:47:23.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-03-25 04:26:35.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-03-29 12:35:14.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"},{"updated":"2021-05-01 09:24:53.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-01 14:41:45.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-06-28 15:40:48.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"3d1c9099f060a8d1f9ffc0b19ca083d0858698dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-24 14:49:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77fcc70b53d33767f4c3c418afc42c162e408fde","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 16:16:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3994301399374d6aa249a3f73803afdb : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c86161f455d45aea982bfe79f365587 : SUCCESS in 9m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c771fe08fa914343b1e80801f059aeca : SUCCESS in 5m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74e052a7c8cd4a0695c35b4abdced89d : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bce937909574d7c9b288e1a5a973227 : SUCCESS in 6m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/69856460a6bc4949969bd87cb14179e0 : SUCCESS in 8m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1481c18cb37044c5a5ceb8fb5032e6ee : SUCCESS in 8m 28s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5837d4a3cbbc491795168536109595fb : SUCCESS in 43m 51s","accounts_in_message":[],"_revision_number":1},{"id":"4f729565fa0458082a70e6cc9770ab41442cf436","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-03-24 16:52:55.000000000","message":"Patch Set 1: Code-Review+1\n\nwe can drop tox env as well as the lower-constraints.xt file too.","accounts_in_message":[],"_revision_number":1},{"id":"17a3d2267eb04332239836dbc3d7ae109fb263a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-24 17:04:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9305d3e363e9e97b6c061138f09c555b2369e3eb","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-03-24 17:23:58.000000000","message":"Patch Set 2: Code-Review+1\n\nthanks","accounts_in_message":[],"_revision_number":2},{"id":"553e98dea01af714ecc6f18f15f96a4142c73a36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 19:29:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/51f57fb80b85415ca3cd14e568af2864 : SUCCESS in 5m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f5374b211e51475584dc6c78084bb6f2 : SUCCESS in 5m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/229c0a3a9d04444db2ead01131595c3f : SUCCESS in 1h 05m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3596026a8e4d467e829e5d162773405a : SUCCESS in 1h 11m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/318dccafbf6247dbb037518eb41cb40c : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f0b82291fee4da781c443eed88e1256 : SUCCESS in 8m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70fd39f5bd5444ebb1db6418660ceb79 : SUCCESS in 8m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/219d1b8a3b944aa397b634ee520ded00 : SUCCESS in 12m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2947371a03bd428b91a8f8e15e925ce9 : SUCCESS in 8m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96899e5856ce482ea046fdac212d1d75 : SUCCESS in 7m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a43402c846cf4cbd8c6c234e611e1cc5 : SUCCESS in 12m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/18832f527a694eaa8f1287a0e962391e : SUCCESS in 12m 59s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/b34b3d35c01b4efea9cf59df06a02d48 : SUCCESS in 5m 42s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e6d006e66ec24371b2521ad091c15fba : SUCCESS in 1h 18m 46s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7624b30e72574151838ee5547d7981e6 : SUCCESS in 1h 19m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f28f33ff50894da8baafdedd3e945f3d : SUCCESS in 1h 13m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ff2dc35d88da45b5a7d33d7738606f95 : FAILURE in 1h 16m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff4950f788064be6832e6c1934f710ec : SUCCESS in 56m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/493906a5a79d4e53b5244cf0f3d05b2a : SUCCESS in 1h 09m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c5ccc4b0c8074af3a651b34b6d2cf75e : SUCCESS in 28m 24s","accounts_in_message":[],"_revision_number":2},{"id":"528c12041205172e8c5b1a75f8fef64122606242","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-25 04:26:35.000000000","message":"Patch Set 2: Code-Review+2\n\nWe have 1 patch for stable/victoria, I guess we need 1 for stable/wallaby as well.","accounts_in_message":[],"_revision_number":2},{"id":"76660e32ee2e74377246ab5b1e61144a19d906ec","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-03-29 12:35:14.000000000","message":"Patch Set 2: Code-Review-1\n\nI don\u0027t think we should do this on Master as the job is actually working as intended finally. Dropping these jobs from stable branches I\u0027m all up to.","accounts_in_message":[],"_revision_number":2},{"id":"a031a9cca7495cb4d946f42daa27d099fc437458","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-04-01 20:29:17.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2: Code-Review-1\n\u003e \n\u003e I don\u0027t think we should do this on Master as the job is actually working as intended finally. Dropping these jobs from stable branches I\u0027m all up to.\n\nIts up to glance team. It might be working now but it can be broken anytime and if glance team is ok to fix and maintain it on master then fine. In my experience working on last few community-wide goals like upgrading CI CD from Bionic to Focal was 50-60% work was to update the l-c which was not worth as per the need of these l-c by users.\n\nIn TC meeting and on ML also we checked and found except Debian(which use these as ref) no body use these lower bounds.\n\n- http://lists.openstack.org/pipermail/openstack-discuss/2021-January/019877.html\n\nmany projects including Ironic, Oslo removed/removing these from master branch also.","accounts_in_message":[],"_revision_number":2},{"id":"33a5ce105f660abd5bae11403387aac95e254b61","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-01 21:00:41.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e Patch Set 2: Code-Review-1\n\u003e \u003e \n\u003e \u003e I don\u0027t think we should do this on Master as the job is actually working as intended finally. Dropping these jobs from stable branches I\u0027m all up to.\n\u003e \n\u003e Its up to glance team. It might be working now but it can be broken anytime and if glance team is ok to fix and maintain it on master then fine. In my experience working on last few community-wide goals like upgrading CI CD from Bionic to Focal was 50-60% work was to update the l-c which was not worth as per the need of these l-c by users.\n\u003e \n\u003e In TC meeting and on ML also we checked and found except Debian(which use these as ref) no body use these lower bounds.\n\u003e \n\u003e - http://lists.openstack.org/pipermail/openstack-discuss/2021-January/019877.html\n\u003e \n\u003e many projects including Ironic, Oslo removed/removing these from master branch also.\n\nYeah, nobody is using these as a hard line, and Debian is just advisory. I think it makes no sense to remove them from stable (which we have already done) but keep them on master. The distros pull from stable and those are the more likely to matter in terms of saying \"this was the lower bound at the time of the release.\" Even if Debian uses them as a reference, if we delete them from stable from which they pull, these are of no use to them at that point. Of course, we have to keep testing them there as well to make sure we don\u0027t violate the lower constraints with a backport. Further, once current master becomes stable/xena, we\u0027d have to re-delete them from there if they\u0027re in master, and keep doing that for each release in the future.\n\nSo I think think only two outcomes make sense: (1) delete from master too since these are not used and can\u0027t be used from stable anyway or (2) revert the patch that has already removed these from stable and drop the proposed patches to do so on the other branches.\n\nI\u0027m very very much in favor of #1.","accounts_in_message":[],"_revision_number":2},{"id":"31ad58bafbe7ebdb5ae7a3c6952397fb11890157","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-04-01 22:10:34.000000000","message":"Patch Set 2:\n\n\u003e Patch Set 2:\n\u003e \n\u003e \u003e Patch Set 2:\n\u003e \u003e \n\u003e \u003e \u003e Patch Set 2: Code-Review-1\n\u003e \u003e \u003e \n\u003e \u003e \u003e I don\u0027t think we should do this on Master as the job is actually working as intended finally. Dropping these jobs from stable branches I\u0027m all up to.\n\u003e \u003e \n\u003e \u003e Its up to glance team. It might be working now but it can be broken anytime and if glance team is ok to fix and maintain it on master then fine. In my experience working on last few community-wide goals like upgrading CI CD from Bionic to Focal was 50-60% work was to update the l-c which was not worth as per the need of these l-c by users.\n\u003e \u003e \n\u003e \u003e In TC meeting and on ML also we checked and found except Debian(which use these as ref) no body use these lower bounds.\n\u003e \u003e \n\u003e \u003e - http://lists.openstack.org/pipermail/openstack-discuss/2021-January/019877.html\n\u003e \u003e \n\u003e \u003e many projects including Ironic, Oslo removed/removing these from master branch also.\n\u003e \n\u003e Yeah, nobody is using these as a hard line, and Debian is just advisory. I think it makes no sense to remove them from stable (which we have already done) but keep them on master. The distros pull from stable and those are the more likely to matter in terms of saying \"this was the lower bound at the time of the release.\" Even if Debian uses them as a reference, if we delete them from stable from which they pull, these are of no use to them at that point. Of course, we have to keep testing them there as well to make sure we don\u0027t violate the lower constraints with a backport. Further, once current master becomes stable/xena, we\u0027d have to re-delete them from there if they\u0027re in master, and keep doing that for each release in the future.\n\u003e \n\u003e So I think think only two outcomes make sense: (1) delete from master too since these are not used and can\u0027t be used from stable anyway or (2) revert the patch that has already removed these from stable and drop the proposed patches to do so on the other branches.\n\u003e \n\u003e I\u0027m very very much in favor of #1.\n\nYes, I agree and this is good point. If any team want to maintain then maintaining in stable as well as master make more sense otherwise just remove it. That is what will match with its usage.\n\nAs discussed overall on ML, I too agree on #1.","accounts_in_message":[],"_revision_number":2},{"id":"9d5d2dade1daddfee15b2fd512e6246504f93d4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-27 13:15:09.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"44c6b869037485fbf354b1ca7b07531f03b2f78f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 14:50:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a00855eec54849c08cd4f5612b0c1bf6 : SUCCESS in 5m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/92794db4a1bd4d57892c43f1ce45d02c : SUCCESS in 5m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/d7587ed17338431a96e6912fb30a5c15 : SUCCESS in 1h 03m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/248b6b8c9e3e4937b0c1e9b4a1509593 : SUCCESS in 1h 06m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b80e4855ee7145fda72ef10cdb26c2a4 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c603b16899ef4917afa73ca8a92edd23 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c68487a6e8894643902c4ad67e90ae6a : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7df0badf19b4936b40aa89eaf41b0e4 : SUCCESS in 12m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e45fdc66e53143f1b7f824e7c02edb45 : SUCCESS in 7m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4b81d39a3054888bb0b7d18c166520f : SUCCESS in 6m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/52c25ed1e42d446299b20c22f6fae817 : SUCCESS in 17m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f5ee90ab9ca74b16bdb2b30b787ff44c : SUCCESS in 10m 51s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/935f1858c9b643c0b99d53eb2f6e57f0 : SUCCESS in 6m 32s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/bc72418925b44076be82dd993344ed4b : SUCCESS in 1h 04m 53s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6efb0d955bcc4f62b81777edb3bc0b15 : SUCCESS in 1h 29m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/dd144bca0ee941f7a5e771946a55adc2 : SUCCESS in 1h 04m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9f66b3d5076f40abb3a4ae0d14263259 : SUCCESS in 1h 09m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/656ea3b020ea4cda9c8d196761cc9b44 : SUCCESS in 49m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b269e36f110246ee8463aa5b14fd18c7 : SUCCESS in 1h 05m 24s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3e0f9311dfee4386b7c0bd8fa070d4de : SUCCESS in 29m 23s","accounts_in_message":[],"_revision_number":3},{"id":"31b6d841ae49b5952f25295aae07e2bd73798803","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-05-01 09:24:53.000000000","message":"Patch Set 3:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/782768/782768/3/check-not-vote/ext-glance-zuul/e68d0cd : SUCCESS in 40m 42s","accounts_in_message":[],"_revision_number":3},{"id":"949ba0667318faaa5780c0f3015b00ae264c4222","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-05-07 16:55:21.000000000","message":"Patch Set 3:\n\nI\u0027ve based the quotas patches on this patch. More context here:\n\nhttp://eavesdrop.openstack.org/irclogs/%23openstack-qa/%23openstack-qa.2021-05-07.log.html#t2021-05-07T16:09:33\n\nBut the short story is that adding oslo.limit to our requirements means we have to add it to l-c. There is a dependency between glance-\u003eglance_store-\u003ekeystoneauth-\u003eoslo.limit which causes the pip resolver to go into an infinite loop (for me at least) making it not possible to add that dependency without some hacks, re-releasing some oslo libraries to tweak dependencies, etc.\n\nOslo itself has already dropped l-c testing on master. Problems like this are exactly why I think it makes no sense to continue doing this, and with oslo drifting over time, this is an inevitable future.","accounts_in_message":[],"_revision_number":3},{"id":"17e275d3fbdc7261bb7c9a235d6546db19f0e076","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-06-01 14:41:45.000000000","message":"Patch Set 3:\n\nI am still in favor of removing this from master branch.","accounts_in_message":[],"_revision_number":3},{"id":"d9bfe63eba9e55494414043927752ca5744d54d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-02 15:08:00.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"986b4d015dbd20b4084e8a76019191417d35a048","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-02 17:48:20.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"51f35527568d7d5f7b3ef1b83897f72c4f4bc03d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 20:15:53.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b308debbf9814eb8bdb9903aa071f441 : SUCCESS in 6m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6098790e8c40427aac99b49c86806e8b : SUCCESS in 5m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/436d6118c7554bd39c29d14ce24cf0a2 : SUCCESS in 1h 02m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a138adba8fce49e99eb66266856fd105 : SUCCESS in 1h 19m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a998f7f0d0f470f860414a010411b5d : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/169f8db455ad457fa23e69092bd16f69 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e11040487174cc9aad3c65a152c5ce7 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa186caff7284c759154118bcbbbc569 : SUCCESS in 8m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acf4578ea5b941238f6774e4b09e093b : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfe472b5983247bfbd384a573383dfec : SUCCESS in 7m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9c5cc155039d45e6910e34db2654cb28 : SUCCESS in 10m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/73de6694f65842e395152729279e3ce3 : SUCCESS in 9m 09s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/752377cd12604363a56f74ac4b45a226 : SUCCESS in 6m 33s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/33f230a3701a42e79208b88c289fc8ec : SUCCESS in 1h 15m 48s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ad15cbbfd13a42f384d159ffbd7f8f88 : SUCCESS in 1h 13m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/880ce329d30143a1a969d9560c7476c6 : SUCCESS in 1h 23m 18s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7cb1f8a487654faf9c5cf84e00dbb6a1 : SUCCESS in 1h 41m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed4cdefc11514420839bba8456c00ea9 : SUCCESS in 53m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/351cca754c6a4c5ba186cfb3ee858d47 : SUCCESS in 1h 23m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b015827263fc466eb2c1e2e3b905a05c : FAILURE in 29m 54s","accounts_in_message":[],"_revision_number":5},{"id":"a73049dab6afbbbd95ac796e8d47776c05eba696","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-02 20:43:17.000000000","message":"Patch Set 5:\n\nrecheck infra issue resolved","accounts_in_message":[],"_revision_number":5},{"id":"31090b33cd958eaa3ce1b58b87c9875b4f6d2b96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 22:18:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/50aef2b90edc41ffaf538c932b5112a3 : SUCCESS in 5m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/fe168a77236e4d9a94c3efb9438e5b2d : SUCCESS in 3m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/d49b377098524d778f97e6316a0b293a : SUCCESS in 1h 09m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/add9f30cf57f4e72b625552ae81a00a0 : SUCCESS in 1h 02m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/838c549a07a64b17a3d3be8bed786bb9 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db307d1719a145f78962e99aa832e75b : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11bff94ce8794f3e94d95d1a514c8f88 : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc521c6ab23f4242a346777d6f968a54 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8342aa4ee86f46f4a1ee49258a2348b6 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07d1e104e0494202aa661a9230916578 : SUCCESS in 6m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/172a5e5c9db64d83b6943c7c6fcb9c68 : SUCCESS in 9m 12s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/da36b2cce71745f68f52419b03155f44 : SUCCESS in 8m 36s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f41f821ded754d91bc0467b07d73f7e2 : SUCCESS in 4m 31s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7767f6b798274582837530bda7f8b96f : SUCCESS in 1h 26m 14s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f92f17f377604c4897a5a280c424bf3b : SUCCESS in 1h 24m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d66d24c62b5d46ad8608cfeec9629ec9 : SUCCESS in 1h 29m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/34c78afdce4a467faa1f4a23ba48cf35 : SUCCESS in 1h 12m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e49f2e00cad54afbb10a5f4b666f2fba : SUCCESS in 59m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1339ec1519d04b3f8ba7f9ef405c83d6 : SUCCESS in 1h 20m 33s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/98491b1ade634911a91dd24062f71ed7 : SUCCESS in 22m 02s","accounts_in_message":[],"_revision_number":5},{"id":"2abeda291b604596028438c2854acc226a5bf8a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-04 14:09:35.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"85fe4614f436dc7895951149c85ccd7b1a8f48df","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-04 14:29:15.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/782768/782768/4/check-not-vote/ext-glance-zuul/dd1fa48 : FAILURE in 19m 36s","accounts_in_message":[],"_revision_number":4},{"id":"ffc270563e92d9b97798b60e9f4eff765cf45c0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 16:19:30.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1c091b29b2a844d4aa3fa68ed51debfe : SUCCESS in 6m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/07124fa5d7144bd49da2f490cb0831d5 : SUCCESS in 4m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/9ce0da3ee400416e85b264b19fa01a95 : SUCCESS in 1h 03m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f9a65c93f6d1416699b0dd3c6d5a30ab : SUCCESS in 1h 23m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7240f8108dbd45538251c908edc4aa63 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/189c98158894449bbba245161504f1ab : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8020bd951f224cb584a8100f70a96be1 : SUCCESS in 14m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b75a0e53fa8643be9ce0e302219b1494 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21a0a6d8fec640d784992a2a8d76cd78 : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eec849ee66f04e569ecc12ea9b3fb5a6 : SUCCESS in 6m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/475f88932a9c470a9661b9e5848a046d : SUCCESS in 9m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d1d2cab3ae10466688789f81e52b8b26 : SUCCESS in 10m 16s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7f9ce334d5d84d298df1504749974f5e : SUCCESS in 6m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f3c26d41a29f44b199e9a701fefeb4bf : SUCCESS in 1h 26m 06s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c7dad9cbee5e4c4d9015baacc34dc5d5 : SUCCESS in 1h 20m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a7929d084be4410ab8737062fe7b7ef9 : SUCCESS in 1h 27m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2e32d01de29e45d8b4c30c694c2670b1 : SUCCESS in 1h 23m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/174e9c4b95d049358bb6ebcf14c5c625 : SUCCESS in 1h 04m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cb511a8be34448cebae2f83a8de74fcb : SUCCESS in 1h 19m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/882b1182831a4c27b0c74c007b91185e : SUCCESS in 32m 31s","accounts_in_message":[],"_revision_number":6},{"id":"b1c34a6c4f34489ecdbd10d53379dc42e85f3f36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-04 17:08:07.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"24642fef92e24f1c74c79000ab02ba4ec414948a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-04 18:34:23.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"1e632148fa772691c030521976b96992b30e49c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 19:46:51.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/065ad52cbd8d4fcc9740028739bb055b : SUCCESS in 5m 07s\n- requirements-check https://zuul.opendev.org/t/openstack/build/90cc74f49d224bee8f660f0842f351d5 : SUCCESS in 3m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/1e20eb82c5d74818b45d32e0032b3d50 : SUCCESS in 1h 04m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac161fff82ba4f8f98276405b85f5c8e : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7e56d1163ec4721b46abc56b77a5da0 : SUCCESS in 9m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd246dccf44d4739bfcc717a755f5432 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87c47c74c0984694a1293969d10aeb7d : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3caa8ecb8b554e48a0604e0fa618bd3e : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c46c4a4548884b28831554237ae2c46a : SUCCESS in 7m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/675896fc5291470c898766a165940cf1 : SUCCESS in 8m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/05a861156da94e1c834924e4b4a7d042 : POST_FAILURE in 18m 49s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a592187efc3c4a579901333509ff9181 : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":8},{"id":"47007c86759516840f3c09389463c68235efc23e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-04 20:09:44.000000000","message":"Patch Set 8:\n\nrecheck infra fail","accounts_in_message":[],"_revision_number":8},{"id":"877a462d021d3bb0c1ea446ce358e42b7103a4bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 21:26:04.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/67adec27c04f43279f64f1b59eb5749c : SUCCESS in 4m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/eff5a26937e941418f6b3517a03d1499 : SUCCESS in 3m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/6a6e47cdc5044bbbb4a234d756e670f5 : SUCCESS in 1h 10m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/735669cab7a44f2aba141a9cf82af371 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bc75026239045b0a136442df001da16 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ef8429eda8d4a6ab3499758950a72a5 : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17d2e23ae0c842aaaa50a443da806833 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6662938c7d04535b1b9bc5ded4571d4 : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9da6f8db0aba4d62a1182d901af53f6c : SUCCESS in 6m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/55dfa19c25054f0788cbef9ce39005f1 : SUCCESS in 9m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0559598e03e74d13bfa8e99eba17e5f0 : SUCCESS in 8m 13s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0c969f6225c74cd2b634d528ed65474a : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":8},{"id":"58e2c7577a85fd208d5f7368eb8b6c3e4b968deb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-14 14:56:36.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"2325663b7b35e1c10097f9cd68fa00234158a335","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 16:42:29.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/b2f9859251dd4c76a60cf6de6f474767 : SUCCESS in 5m 54s\n- requirements-check https://zuul.opendev.org/t/openstack/build/94834f9b9d2247a1899e99e3cbf2ce58 : SUCCESS in 4m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/9780bb8968b34b4a999dd9a14dd62c8e : SUCCESS in 1h 03m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d160394aed3c40398b5a4aa2f7534ccd : SUCCESS in 1h 09m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00644cc0716f4f4da736f69ab1760375 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6dd63dad7aa34462a4c3f872d795cad0 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab89dda60efe4ef49630486adc7e07f3 : SUCCESS in 7m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3a53290ccf143849a4f1fb81b9a587f : SUCCESS in 7m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94c17936cc01407ba39cbd7bbcbfa453 : SUCCESS in 9m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a978ff797cea4e109455115b3663b64a : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/56bb3c6450f74550bb4140960f537c73 : SUCCESS in 11m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e53f7a7a51ac486086d8116e0aa1b39a : SUCCESS in 13m 02s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/23360d4bbeff4f6d9b6e4417f715d987 : SUCCESS in 6m 10s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/64efb6bce42a441886d1bdefa2e3eb95 : SUCCESS in 1h 18m 31s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/59962d72ed54436497469077c6ad50a0 : SUCCESS in 1h 38m 13s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/50cf3187c3c846bd9d250a045d41674d : SUCCESS in 1h 11m 38s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/36f111b02a1541bcae5d4393b1fdcbb2 : SUCCESS in 1h 02m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b26572e22ef49798cd466fab820668a : SUCCESS in 49m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e2b55e50a2a44492804d664e4e88e0a0 : FAILURE in 39m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/132d5fd51e374d2a97a44cb45e4b0b20 : SUCCESS in 28m 56s","accounts_in_message":[],"_revision_number":9},{"id":"e798ad14fd2c49d245202e424a3174dd7638b665","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-21 15:46:41.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"a22efae21e3435d95e7fe41b45a9b53bdf2ba7da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 17:07:44.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/fb1afd799b5244f18dbef11c2b4c56b4 : SUCCESS in 5m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/0d400cae11cd49fca114313fcc3f666a : SUCCESS in 5m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/3c91dd24d6f346beac0db35cd0ec715c : SUCCESS in 56m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/477b3d20d49241d68075720e2239890d : SUCCESS in 1h 15m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f39e38aee3ee4bebb5ef3c77854da582 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d75d6b5b582d463098259baf92002940 : SUCCESS in 7m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef8a53887ae9480bbee15e3ab415ec3c : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82684f062594402e9c2c1b61e583ee8d : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8da1a4351754a96927f1ea613a8d403 : SUCCESS in 7m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6739cfbf5d5c4756ab52898f86f0c0af : SUCCESS in 8m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/75f0bbf5b8fd4e6298e7896fd87f96ec : SUCCESS in 8m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1a4344bcf33d430abf8a48a9e6c637a3 : SUCCESS in 9m 40s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c9ebb62ce0304d478ab6ae8d62989584 : SUCCESS in 5m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6d7c44fbe55d4003a1764fe23a9c8df1 : SUCCESS in 1h 07m 06s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e3c60617e1454c81b4c8b50be18413df : SUCCESS in 1h 05m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c646a1c5610141198154d71098c07bdb : SUCCESS in 1h 06m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/dffaff25796a42d8b2c211d5bc999124 : SUCCESS in 1h 14m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49fed77e32ed4295911bf0167692ab04 : SUCCESS in 51m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8148281895ce44319d4f0aa4ff862c69 : SUCCESS in 1h 12m 23s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4b3faec933ff48a5ba083e65c1dda8ba : SUCCESS in 30m 42s","accounts_in_message":[],"_revision_number":10},{"id":"8f498bf1be9ed2cdd08f60a0acf4f3cd7613e300","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-24 19:21:44.000000000","message":"Patch Set 10: Code-Review+2\n\nI\u0027ve been thinking about this.  Given that nobody seems to really use the lower-constraints.txt, the utility of having the job is to make sure that the versions of dependencies we\u0027re using haven\u0027t drifted too far from the minima in requirements.txt (because that\u0027s what is in l-c.txt for direct dependencies).  That might still be worth checking occasionally--or you can aggressively update the minima in requirements each cycle to whatever pip freeze gives you at Milestone-3 or so.  I\u0027m inclined to go with the M-3 update strategy on the theory that the minima in requirements.txt are supposed to mean something, and if we have somelib\u003e\u003d1.0 but are running unit, functional, and tempest tests with somelib\u003d\u003d3.3, we really have no idea if you can really run our software with somelib\u003d\u003d1.0 (even if UTs pass with somelib\u003d\u003d1.0).  One downside to aggressive minima updates is that not all platforms may have packages for the exact patch version of some requirement that was updated to the latest version available on pip at M-3.  The other downside is that if any consuming project is maintaining lower-constraints, they\u0027ll have to update their l-c to match our minima.\n\nIn short, the only reason I can see for keeping l-c is if we want to run a sanity check on what we\u0027re claiming are the minimum acceptable versions in requirements.txt--and it\u0027s just a sanity check because it only runs UTs.  And if it fails, my inclination would be to update requirements.txt to the version that\u0027s currently being used in passing jobs, which would basically put us at the aggressive update strategy, just strung out over a bunch of cycles as stuff breaks, instead of all at once at each M-3.\n\nSo I am pro-aggressive-update and anti-lower-constraints job.","accounts_in_message":[],"_revision_number":10},{"id":"d2ec082d4a654019138913dcd586c4a87ca2341f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-24 19:23:51.000000000","message":"Patch Set 10:\n\nI didn\u0027t approve in case Erno can spot a flaw in my logic.","accounts_in_message":[],"_revision_number":10},{"id":"63fedbf0e9363941b37a3b20191490693ad2f7f5","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-28 14:20:57.000000000","message":"Patch Set 10:\n\n\u003e I didn\u0027t approve in case Erno can spot a flaw in my logic.\n\nNote the earlier explanation that I provided about pip going into an infinite loop when adding the oslo.limit dependency. Infra and QA people\u0027s suggestion to solve it was \"why are you running the l-c tests anyway?\"\n\nSo if we don\u0027t approve this, someone needs to figure out how to reshuffle the reqs so that we can add that limit and not send pip off into a loop.","accounts_in_message":[],"_revision_number":10},{"id":"4b36cb00c3ffd0eb9d79d317ff527a2754aaf9ea","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-06-28 15:40:48.000000000","message":"Patch Set 10: Code-Review+2\n\nThanks Dan, your latest comment convinced me to +2 this :)","accounts_in_message":[],"_revision_number":10},{"id":"23c7f2532daff04f7ac8cee117c44106acfe073f","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-06-28 16:59:07.000000000","message":"Removed Code-Review-1 by Erno Kuvaja \u003cjokke@usr.fi\u003e\n","accounts_in_message":[],"_revision_number":10},{"id":"2abffbecef83962e10b9614ac431cc28bacaa362","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-06-28 17:01:40.000000000","message":"Patch Set 10: Workflow+1\n\nNow better understanding the job mechanics as well, lets just stop wasting the infra resources on this. Shame to see we couldn\u0027t figure out how to do this with reasonable maintenance and actual meaningful tests.","accounts_in_message":[],"_revision_number":10},{"id":"4e017bcd89e0032a7ae8116af617da376926ec35","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-06-28 17:02:13.000000000","message":"Patch Set 10: Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"c9701fbc74a3918d1c0d0e6b7bb7a71b5490672f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 17:03:09.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"2bd9b69b9ee000927effbe74131885923f11b6f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 18:56:49.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d0d25e6d55bd4d608f5d392b49ca899e : SUCCESS in 6m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/608938b3b9ee432faed343a151bbd507 : SUCCESS in 5m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/60fe363c579b4ca78e4ef7ae146de0db : SUCCESS in 1h 12m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b7982a6d5a244709ab95878bcda58e58 : SUCCESS in 1h 12m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d0a92f2d19f42d5ac59ceee8e16d9cf : SUCCESS in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d33c873d51a4a888014cf8ec0bb9d75 : SUCCESS in 7m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ac83302d0874daf9ed07e7a86563cd5 : SUCCESS in 8m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52a32ed13f234c04a47e6e1cf717f66d : SUCCESS in 8m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a134d739900943e0858d6ae808c03ed0 : SUCCESS in 8m 03s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/beb5a2dad1cf4986ad73724f761c2fff : SUCCESS in 10m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9bdd3f17e9dd480cae7d702661bbac8e : SUCCESS in 10m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/07a4d4a633b146abbf57ffc164d2c328 : SUCCESS in 1h 11m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f8c268fb52024089baddbdca415dc804 : SUCCESS in 1h 35m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/773e693aaf2444fbbeb5a054a127c202 : SUCCESS in 57m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ede805ce133345ffb23b36aa30d1e473 : SUCCESS in 1h 24m 53s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/abb7ac6784fc419aa6b1ab91e2f9776c : SUCCESS in 36m 33s","accounts_in_message":[],"_revision_number":10},{"id":"89ebedcc38444a43558f98ef19422f2447b62e35","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 18:56:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"60722dbb22c6bd577cef20698ecabe9b08f1cabd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 19:00:01.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/bad7a4a85c3e41edaececa1012c4151e : SUCCESS in 1m 01s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a7d431334bb4084ab01bf966907e029 : SUCCESS in 1m 09s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/149f11cc70b643408ac4fc1cda92e339 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"fa1bc2f904d647735365937488cddb27bd830b1b","revisions":{"354bc74cf5b84d82525e93ee9aa196709d2ada6a":{"kind":"REWORK","_number":1,"created":"2021-03-24 14:49:01.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/68/782768/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/782768/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/782768/1"}}},"commit":{"parents":[{"commit":"8fc9b653936de840371a1cec914d298c131f1575","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8fc9b653936de840371a1cec914d298c131f1575"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-24 14:49:21.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-24 14:49:21.000000000","tz":-420},"subject":"Drop lower-constraints jobs","message":"Drop lower-constraints jobs\n\nChange-Id: Id030b56b721cb62255936d56f0409c4be9ce21d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/354bc74cf5b84d82525e93ee9aa196709d2ada6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/354bc74cf5b84d82525e93ee9aa196709d2ada6a"}]},"branch":"refs/heads/master"},"f7aa7a165fd56224564309b9221462897cef79f1":{"kind":"REWORK","_number":2,"created":"2021-03-24 17:04:13.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/68/782768/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/782768/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/782768/2"}}},"commit":{"parents":[{"commit":"8fc9b653936de840371a1cec914d298c131f1575","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8fc9b653936de840371a1cec914d298c131f1575"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-24 14:49:21.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-24 17:04:41.000000000","tz":-420},"subject":"Drop lower-constraints jobs","message":"Drop lower-constraints jobs\n\nChange-Id: Id030b56b721cb62255936d56f0409c4be9ce21d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f7aa7a165fd56224564309b9221462897cef79f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f7aa7a165fd56224564309b9221462897cef79f1"}]},"branch":"refs/heads/master"},"667280c6f6c5d80f87186634cf43b33ae142d605":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-04-27 13:15:09.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/68/782768/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/782768/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/782768/3"}}},"commit":{"parents":[{"commit":"1157c19a9c1408faad5b845cc9b1873f82fc8f31","subject":"Merge \"Guidelines for core reviewers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1157c19a9c1408faad5b845cc9b1873f82fc8f31"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-24 14:49:21.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 13:16:26.000000000","tz":-420},"subject":"Drop lower-constraints jobs","message":"Drop lower-constraints jobs\n\nChange-Id: Id030b56b721cb62255936d56f0409c4be9ce21d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/667280c6f6c5d80f87186634cf43b33ae142d605"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/667280c6f6c5d80f87186634cf43b33ae142d605"}]},"branch":"refs/heads/master"},"e269451cb1ca66e685901aeebb09d94d3ff84ffc":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-06-02 15:08:00.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/68/782768/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/782768/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/782768/4"}}},"commit":{"parents":[{"commit":"c5c7a2a6342fc3de0b9cc2ec3f0ff7b907cc1fc7","subject":"[Doc] Update Freenode to OFTC as our IRC server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c5c7a2a6342fc3de0b9cc2ec3f0ff7b907cc1fc7"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-24 14:49:21.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-02 15:09:50.000000000","tz":-420},"subject":"Drop lower-constraints jobs","message":"Drop lower-constraints jobs\n\nChange-Id: Id030b56b721cb62255936d56f0409c4be9ce21d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e269451cb1ca66e685901aeebb09d94d3ff84ffc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e269451cb1ca66e685901aeebb09d94d3ff84ffc"}]},"branch":"refs/heads/master"},"7d8734e2472ecd89a3c85465b068dd83041200d4":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-06-02 17:48:20.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/68/782768/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/782768/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/782768/5"}}},"commit":{"parents":[{"commit":"e69d621346a074169c6666ce0ae285d92bfb5fb9","subject":"Fix missing context args to get_flow()","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e69d621346a074169c6666ce0ae285d92bfb5fb9"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-24 14:49:21.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-02 17:50:03.000000000","tz":-420},"subject":"Drop lower-constraints jobs","message":"Drop lower-constraints jobs\n\nChange-Id: Id030b56b721cb62255936d56f0409c4be9ce21d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7d8734e2472ecd89a3c85465b068dd83041200d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7d8734e2472ecd89a3c85465b068dd83041200d4"}]},"branch":"refs/heads/master"},"2f1698c05dc2e3a931474cdf08d675de660f93df":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-06-04 14:09:35.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/68/782768/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/782768/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/782768/6"}}},"commit":{"parents":[{"commit":"cbce06b80735281c8ad5faee8bd8448404057703","subject":"Make image stage set image.size","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cbce06b80735281c8ad5faee8bd8448404057703"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-24 14:49:21.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-04 14:11:14.000000000","tz":-420},"subject":"Drop lower-constraints jobs","message":"Drop lower-constraints jobs\n\nChange-Id: Id030b56b721cb62255936d56f0409c4be9ce21d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f1698c05dc2e3a931474cdf08d675de660f93df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f1698c05dc2e3a931474cdf08d675de660f93df"}]},"branch":"refs/heads/master"},"0bf91123e3e21357079e2cf9996760c4ddfcfbaf":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-06-04 17:08:07.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/68/782768/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/782768/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/782768/7"}}},"commit":{"parents":[{"commit":"201bb3b73a9099a4f2915ece1ca4fd6526c5888d","subject":"Make image stage set image.size","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/201bb3b73a9099a4f2915ece1ca4fd6526c5888d"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-24 14:49:21.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-04 16:31:01.000000000","tz":-420},"subject":"Drop lower-constraints jobs","message":"Drop lower-constraints jobs\n\nChange-Id: Id030b56b721cb62255936d56f0409c4be9ce21d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bf91123e3e21357079e2cf9996760c4ddfcfbaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0bf91123e3e21357079e2cf9996760c4ddfcfbaf"}]},"branch":"refs/heads/master"},"ddbbf7bd7748e2ef025c32804a8dc4f8a85ae38f":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-06-04 18:34:23.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/68/782768/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/782768/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/782768/8"}}},"commit":{"parents":[{"commit":"a91d3496cbbb8608d9ba7d4f102d4cff406a8d5d","subject":"Make image stage set image.size","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a91d3496cbbb8608d9ba7d4f102d4cff406a8d5d"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-24 14:49:21.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-04 18:21:01.000000000","tz":-420},"subject":"Drop lower-constraints jobs","message":"Drop lower-constraints jobs\n\nChange-Id: Id030b56b721cb62255936d56f0409c4be9ce21d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ddbbf7bd7748e2ef025c32804a8dc4f8a85ae38f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ddbbf7bd7748e2ef025c32804a8dc4f8a85ae38f"}]},"branch":"refs/heads/master"},"30f0118d9098bc580c47d3897d3432861b673566":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-06-14 14:56:36.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/68/782768/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/782768/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/782768/9"}}},"commit":{"parents":[{"commit":"663043904b2b3c3d0f0aafdcac2f7622284bd977","subject":"Make image stage set image.size","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/663043904b2b3c3d0f0aafdcac2f7622284bd977"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-24 14:49:21.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-14 14:58:35.000000000","tz":-420},"subject":"Drop lower-constraints jobs","message":"Drop lower-constraints jobs\n\nChange-Id: Id030b56b721cb62255936d56f0409c4be9ce21d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/30f0118d9098bc580c47d3897d3432861b673566"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/30f0118d9098bc580c47d3897d3432861b673566"}]},"branch":"refs/heads/master"},"fa1bc2f904d647735365937488cddb27bd830b1b":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2021-06-21 15:46:41.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/68/782768/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/782768/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/782768/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/782768/10"}}},"commit":{"parents":[{"commit":"c290c4af5a06957bdc16158a3dff6971d1d3ad65","subject":"Make image stage set image.size","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c290c4af5a06957bdc16158a3dff6971d1d3ad65"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-24 14:49:21.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-21 14:19:01.000000000","tz":-420},"subject":"Drop lower-constraints jobs","message":"Drop lower-constraints jobs\n\nChange-Id: Id030b56b721cb62255936d56f0409c4be9ce21d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fa1bc2f904d647735365937488cddb27bd830b1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fa1bc2f904d647735365937488cddb27bd830b1b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
