)]}'
{"id":"openstack%2Fglance~782769","triplet_id":"openstack%2Fglance~stable%2Fvictoria~I774569d4788c7c67273fbe72757af07813c62f1d","project":"openstack/glance","branch":"stable/victoria","topic":"drop-lc-victoria","hashtags":[],"change_id":"I774569d4788c7c67273fbe72757af07813c62f1d","subject":"Drop lower-constraints jobs","status":"MERGED","created":"2021-03-24 14:50:21.000000000","updated":"2021-03-25 17:53:27.000000000","submitted":"2021-03-25 17:47:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"782769-drop-lc-victoria","meta_rev_id":"0c916f8576e507b506a92145f83242798406c23b","_number":782769,"virtual_id_number":782769,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-25 17:46:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-03-24 18:00:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2021-03-24 17:24:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-03-24 18:00:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-24 16:47:55.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-03-24 18:00:46.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-03-24 19:45:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"47c6ceaaadb4b4f3e066ad4ce5d145b94e89e5be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-24 14:50:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92ca8dcf7e50d689251540c507d89099b6cebb4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-24 17:03:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5c78afd8d75b7e8ab470f553150290121426c6b9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-03-24 17:24:03.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"5e6d4db0810f8b73e480f440dc891e636b02c925","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-24 18:00:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"56864fe3e23b652db36750886e71b676ca15fa5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 19:45:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e9a8521461404de3bee4a2f33e2e8909 : SUCCESS in 4m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/493c7fc09690445e910119750ded1485 : SUCCESS in 1h 01m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/704096e658a54358b1265884770f4108 : SUCCESS in 1h 27m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ad270222ad3436d8ac7eabcaddc3eb6 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e07c705ace546648f6c9ad0c169794a : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a16062f48174bc58506f6c2f111c519 : SUCCESS in 11m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e1dda7dc57748e8a8b6e21be6377f53 : SUCCESS in 11m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17627761392441d38b38aa5a19d8bfd2 : SUCCESS in 7m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4277e379109c4459921f145184279c44 : SUCCESS in 8m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/97293fd1d5024b4f8392941ddf0a9913 : SUCCESS in 11m 12s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f0779b36762b4bd7a31ef770236df1bc : SUCCESS in 7m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9cb1354a102b4f3684464079015e3574 : SUCCESS in 1h 17m 46s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/e075053243384e97a7ac667c7b8698e5 : SUCCESS in 1h 23m 00s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/d001c0ac81f241b88d8d38bc62c0bb1a : SUCCESS in 1h 28m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67a3a16324be486c97e8e305e742884e : SUCCESS in 58m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7e545a6786364cc28ab02a2d0f3fde31 : SUCCESS in 1h 16m 38s","accounts_in_message":[],"_revision_number":2},{"id":"98c9a0c8085f6e3e909c3be81d30339c7a7f9f79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 19:46:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"86d0ddc525570769e15ebd37b1749883ce9d150c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 00:22:54.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3d5790b40f044873bfcc6a98f6e487af : SUCCESS in 3m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/53ede89eb8354545b2245da717cd28f8 : SUCCESS in 47m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6548079d5ba744938708103dde67ae48 : SUCCESS in 1h 02m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0931eeda791f49638270ba3917bf6893 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/faa71da42b7d44b8aaeecb0813476c01 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e396d0ae00154b2bb378b1c97dbe8dc7 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1688893778fe4977bff81ca266824031 : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c0c7d2ad42f4191b8a5da3653e116d4 : SUCCESS in 5m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8730898ea8cc477c87ef84e9ef81f2da : TIMED_OUT in 30m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8b7c96c89b0540b4984f9e323a29175a : SUCCESS in 7m 08s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/91d0f83753024d83b86dbf9f470a6dad : SUCCESS in 1h 10m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ecb6237690644198519d4b0a6d0b6e1 : SUCCESS in 54m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/354986b018a7479dbf45b6cbea13c5b7 : SUCCESS in 1h 08m 39s","accounts_in_message":[],"_revision_number":2},{"id":"f7b0adc9f2514f47c82a14408b878ffba98b0608","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-25 04:24:30.000000000","message":"Patch Set 2:\n\nrecheck timeout","accounts_in_message":[],"_revision_number":2},{"id":"c8d6d0aa2a4ac508d75dc7735291bb962985bf82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 05:51:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/bf9823be84e54f7c95966589e17f5533 : SUCCESS in 3m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/b91b53a2c56d426f8f15fbbe4b27a23c : SUCCESS in 1h 03m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/00fa0c52798e4e9a92a978ee52d998aa : SUCCESS in 1h 01m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10073f36ce984c98b3d36046e228f31f : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f6d70e7ea342478181388ff72bdf5ffa : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7b3e0da37304faeaf2893628518e1f5 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf6ddf3130b44e95bae700411484d753 : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2bf892f90a2491cbba29d4e6b30323e : SUCCESS in 6m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e2faaf3b0c9c4089843bdc52baa96e67 : SUCCESS in 9m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ec94558c34784b53bf65bb825724c5a3 : SUCCESS in 11m 36s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0283d30a403742ad8af83fd9efaf083f : SUCCESS in 4m 37s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8bd3940dc0cc422eb6b0bedf72dd314c : SUCCESS in 1h 15m 33s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/2fdf32d2474a4e43b850dca1fa43c78e : SUCCESS in 1h 13m 35s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/13d08289074a44d3ab0641068b268f15 : SUCCESS in 1h 15m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb84c4736df646b581c5ecdb8992f97c : SUCCESS in 1h 04m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/37a100254391468f806b4f879248f06e : SUCCESS in 1h 11m 56s","accounts_in_message":[],"_revision_number":2},{"id":"dc4dd5c32b99f4119f7bd9686a4927f2640e9541","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 05:52:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7699fd0e768f5c1dba2e8c7eb64a09785ff0c618","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 07:19:25.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/6020ee8b078d4f9283b378f4588c09e7 : SUCCESS in 3m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/5f67f3fd97574275834a8ad58b95bae6 : FAILURE in 27m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d8afd67e0dd444e1a260eef00fd95b0d : SUCCESS in 1h 22m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71aa8cd78a8244ee99d653a1b566ef06 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3acbbd0e630495c9e7759a2b1446fe9 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1738a88611194a5198ac3d260894d66a : SUCCESS in 7m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5a795367e624334ad782b4531c4f023 : SUCCESS in 7m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f9969aac3dd420886ff7fcb4bec2ad9 : SUCCESS in 6m 56s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/dd35af8b658a4e9c8e1b56efdaabe456 : SUCCESS in 8m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/886bb28de98e4153aeb3d739da91d68d : SUCCESS in 11m 25s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/d03913d244a24070984e95f8e90ad40c : SUCCESS in 1h 18m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16a3da568b5942bcadf3befe87e272ca : SUCCESS in 59m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/50d2cb5abde44bdeb8eda318b75329b7 : SUCCESS in 1h 23m 33s","accounts_in_message":[],"_revision_number":2},{"id":"064727781349deb34cb15146dbf81512db0838ff","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-25 08:59:20.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8bca94f759cc3ee3784d3e15a9ec3423c9557a83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 10:25:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/19ed2a51e43546d29a72e89cabb0539d : SUCCESS in 3m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/084075aabd3d4e75b347859589a7f656 : SUCCESS in 57m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/255f65f346654b0a97e814476758264b : SUCCESS in 1h 19m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3faa7b635e7c48f4a4f2c067a08697ba : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f8eb8df5810497aacb01e53de847788 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ac6d3215b744217b6f5aed528cb9571 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fad7fb46a7843cfb7dfb3971c45c32e : SUCCESS in 10m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07b55d607c2a43008a272980c4832a10 : SUCCESS in 6m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/84d1f7b50cde44b49166d40250da84f7 : SUCCESS in 11m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/222541a64ff04f46979cd427c1f7d2f6 : SUCCESS in 8m 23s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6e282055909a4bfa960d98ae99e936d8 : SUCCESS in 4m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/68f50b2ea32242d2bcf42b940df7e591 : SUCCESS in 1h 20m 35s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/d894ed44901d457ea88ee6b1b81d12d8 : FAILURE in 1h 17m 14s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/4102c5a65f0641e2910eaa357dc700e6 : SUCCESS in 1h 13m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa1790bdc53a47e088f3941e422e73b7 : SUCCESS in 50m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/40a08db42187483c802d3622831b225e : SUCCESS in 1h 13m 38s","accounts_in_message":[],"_revision_number":2},{"id":"d47816ea57b9471fa58d345cf3acd6f8781711e6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-25 13:19:56.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"86a00722b068695a441a077d3c4d76aca003e22f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 15:19:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ce7c7a0f0d874030873a381efc7392c3 : SUCCESS in 3m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/6979186006474436b92341fccec93dcb : SUCCESS in 1h 00m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cd850ef2b0fa483ebb9d19ec54959d13 : SUCCESS in 1h 43m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a105d5e560e04dd5a7fbbbfa358515df : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97d03038ee0b4d47968d8eb21ec48210 : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/78c80ffebac94a40a526c92860a81ea7 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20e5460eb36543cda01cdde70fb0cff8 : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8dcd1976629f4a2a91869329a1a25747 : SUCCESS in 6m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/495ed6a2d9ce47878e137625728b155b : SUCCESS in 10m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/77254d2d1ea94f73bc76ef928eaede35 : SUCCESS in 8m 55s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/92ed6f26b36e441d886a0b9975473340 : SUCCESS in 6m 34s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/01b0f92e4e0c4ec1995593b462814dd1 : SUCCESS in 1h 22m 30s (non-voting)\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/c1655cd8ac0c4ae0aa746cabc9027f49 : SUCCESS in 1h 26m 23s\n- tempest-integrated-storage-wsgi-import https://zuul.opendev.org/t/openstack/build/5934d1508a3548b4856bb6bc7344bac9 : SUCCESS in 1h 13m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24594d16c4184552bdc719d4b02b0d3b : SUCCESS in 54m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d6df8b50fa804515b6b2c7aa8ba46575 : SUCCESS in 1h 10m 00s","accounts_in_message":[],"_revision_number":2},{"id":"43974e90419c88b0cea2b0da50ed002481d1f218","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 15:19:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8682e246b0150f3ff6a755c9c3d380c15bd223b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 17:46:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1ddb207dabc04e288546eba361563f40 : SUCCESS in 3m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/f3e225c893444177ad61d78c103e651a : SUCCESS in 1h 07m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e248329a28144595b45ad369a4db89f4 : SUCCESS in 1h 19m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8572b4cd15d415cb097c18ac62e5be7 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28b8a33fd7f744618ee8d7436486a8cc : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/176300916c4a45d893a04d8d0136da0d : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba5ffe18c7e94c17a6cb37ada6895a74 : SUCCESS in 6m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afdac73c343747da8dfd0fb12b5343cf : SUCCESS in 6m 37s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f867d31f7f8c482f9168d21f15b59de0 : SUCCESS in 8m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b252dbe83de94e33b42c4a5c05e8e1fd : SUCCESS in 8m 06s\n- tempest-integrated-storage-import-workflow https://zuul.opendev.org/t/openstack/build/ca25d16a47144b2682fae744b516813a : SUCCESS in 1h 07m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be3c2890452841b18c96c69976c7476d : SUCCESS in 56m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/88a960bc418e4e70b04ffcd9c639db05 : SUCCESS in 1h 13m 12s","accounts_in_message":[],"_revision_number":2},{"id":"ac69794e0601d5a796c4408407ddc79155d6b7cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 17:47:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"0c916f8576e507b506a92145f83242798406c23b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 17:53:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90e23f5ec236445c8bebc2da1443cf36 : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/564b6e3bd70e4c24ad9096a8f852344a : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0f6f5166f0e97abcc73dbc03d94998f253cd210c","revisions":{"7ce31d084734cce0f510366fb96941c8eb231d32":{"kind":"REWORK","_number":1,"created":"2021-03-24 14:50:21.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/69/782769/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/69/782769/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/782769/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/782769/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/782769/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/69/782769/1"}}},"commit":{"parents":[{"commit":"9cb3240b94dc4ca7fa288fc66eba1499fb99295f","subject":"Make web-download revert all stores on fail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9cb3240b94dc4ca7fa288fc66eba1499fb99295f"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-24 14:50:39.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-24 14:50:39.000000000","tz":-420},"subject":"Drop lower-constraints jobs","message":"Drop lower-constraints jobs\n\nChange-Id: I774569d4788c7c67273fbe72757af07813c62f1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7ce31d084734cce0f510366fb96941c8eb231d32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7ce31d084734cce0f510366fb96941c8eb231d32"}]},"branch":"refs/heads/stable/victoria"},"0f6f5166f0e97abcc73dbc03d94998f253cd210c":{"kind":"REWORK","_number":2,"created":"2021-03-24 17:03:49.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/69/782769/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/69/782769/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/782769/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/782769/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/782769/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/69/782769/2"}}},"commit":{"parents":[{"commit":"9cb3240b94dc4ca7fa288fc66eba1499fb99295f","subject":"Make web-download revert all stores on fail","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9cb3240b94dc4ca7fa288fc66eba1499fb99295f"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-24 14:50:39.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-24 17:04:10.000000000","tz":-420},"subject":"Drop lower-constraints jobs","message":"Drop lower-constraints jobs\n\nChange-Id: I774569d4788c7c67273fbe72757af07813c62f1d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0f6f5166f0e97abcc73dbc03d94998f253cd210c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0f6f5166f0e97abcc73dbc03d94998f253cd210c"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
