)]}'
{"id":"openstack%2Fglance~783668","triplet_id":"openstack%2Fglance~master~Ida5bc659b7481fa0a7f3926a54520628a7a4c406","project":"openstack/glance","branch":"master","topic":"bug/1895173","hashtags":[],"change_id":"Ida5bc659b7481fa0a7f3926a54520628a7a4c406","subject":"Ignore stale image property removal","status":"MERGED","created":"2021-03-29 15:10:04.000000000","updated":"2021-07-22 03:43:50.000000000","submitted":"2021-07-22 03:42:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"783668-bug/1895173","meta_rev_id":"ca03ff1b7c445ae2d4c058fd4424395dd6a3c8b0","_number":783668,"virtual_id_number":783668,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-22 03:42:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-07-22 02:08:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2021-04-09 13:50:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":1,"date":"2021-07-22 02:08:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-04-09 13:50:19.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-29 15:59:39.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2021-03-29 20:02:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-30 09:39:40.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-03-30 13:49:30.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-04-01 16:08:47.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2021-04-16 04:17:06.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-22 02:08:54.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"358d496989bf736cb36d5efa6e4b47437497e349","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-29 15:10:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e0a58d5f2b6e8bbf18d38a7cab084bc39b783513","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-29 15:42:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3ba896a1cecbaa986ac69888df0c525dff1b8a73","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-29 15:59:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5af71ff7d1a5aed50a236165a6a71de2d19fa364","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-29 16:02:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"19ac846015552e459247c705d66508384f1b6be4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-29 18:21:43.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"1d4011f9482097adbe4f9721615ddddd638040b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 20:02:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/89b70702962f4663b0ddeffe8a97b909 : SUCCESS in 1h 08m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f07763d4fa434c29ac3f715203417652 : SUCCESS in 1h 10m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e677e4bb1274d6e8f00607b7bcd1366 : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6e8559f01c64dc99642c3bd632d6e57 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/50dc445484bf4de5957677fcc461a347 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/716062e1c20f4f9295fd555fad247c68 : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f385650bd6234c14a94abba87482d196 : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/000991aad0c24dcf8f7b549a5387fffb : SUCCESS in 8m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7464b6289152475ea11adc18ffd6c377 : SUCCESS in 11m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d84ce4e975c14898bf353582dfe0d696 : SUCCESS in 9m 36s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0ac3629b83cd4a5cb820346c4fad9167 : SUCCESS in 5m 58s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9deb6c91c4054dfda29d6d89f337a0cf : SUCCESS in 1h 26m 26s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/57e1c9f133d443ea893cfa76c3c32c44 : FAILURE in 1h 33m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b12de0e744fa44379ac35d87b2ef5fc0 : SUCCESS in 1h 14m 57s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5c780d8e097c4920908e13fe04b139ce : SUCCESS in 1h 24m 41s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6fdff25b60b24e40afcf504be1f013a0 : SUCCESS in 1h 13m 24s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/35d7eba2b09342a9971b6deaab7974e6 : SUCCESS in 1h 14m 55s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/916d2b2789ee426090fff0f666e12e66 : SUCCESS in 39m 58s","accounts_in_message":[],"_revision_number":3},{"id":"9db6e874a360df6bb45cbd02d18e9eaebd3d60cf","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-29 21:12:11.000000000","message":"Patch Set 3:\n\nrecheck cinder dance","accounts_in_message":[],"_revision_number":3},{"id":"19693b0b96f36e640d866b4de0556788aa19556d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 22:50:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/d04188adb8194e4f93622280c3053c3a : SUCCESS in 1h 01m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/185f84cb8ae6456ea89a07e7d7ab15c1 : SUCCESS in 1h 10m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0f54bc929e548cfa67a96b262c851e5 : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83330f04272441eb8af7caf29c750a51 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/159fed6ceef34cbd96c893fe240513b4 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8c516e2b14c43a7ab6a484eddfb2be2 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/784301f437bf4f72bc9ac4a3360718c8 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f22634bdb56453d95d76b00935f582b : SUCCESS in 5m 54s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/92acf24849274329b1433f96c9af72d5 : SUCCESS in 9m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4f58b1613bf0411ab50d24047a483c8e : SUCCESS in 7m 51s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/1e8669e864ab4241b5409915f9fe8a03 : SUCCESS in 5m 12s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1e1cf37938f04900906220f617200eed : SUCCESS in 1h 11m 53s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/704b68256cba43ad93f8bc8dd563325d : SUCCESS in 1h 15m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4bbf8caf0f6a4d659c4c852d3446d7db : SUCCESS in 1h 12m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cb9faeedce6c48fea1f78585fcd39abd : SUCCESS in 1h 21m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9e0dfaa276af4d3e9bba826cf85fa74a : SUCCESS in 48m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8c6267ebc3ec4d43b6753980c120a9ec : SUCCESS in 1h 23m 35s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f93689750fd846e4a3c93ca901256a81 : SUCCESS in 23m 12s","accounts_in_message":[],"_revision_number":3},{"id":"979ace0105e9ea64c62e2888483ff818473fd1ee","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-30 09:39:40.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"bcb0287a7afdc85e57fb39f098b10f75081ef025","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-04-01 16:08:47.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nQuestion inline.","accounts_in_message":[],"_revision_number":3},{"id":"ff604b04af2a1950c3f3d2929ee11e83dc41398d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-01 17:04:35.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nI\u0027ll wait for your reply on the second thing before I revise this for the first.","accounts_in_message":[],"_revision_number":3},{"id":"cf5ff94cfa44be3477bfacbdc3124d706889b93e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-04-08 14:37:42.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nReply inline.","accounts_in_message":[],"_revision_number":3},{"id":"6b54562b881b5345f05b53f6eb6f284b073ab70f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-08 14:38:57.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c4a08281610d1a39a9f8e97e75c355cb48674ab5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-08 14:42:22.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"34005762bf70435a622e02b797c178fd67065a28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 17:28:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/577ab2aefda54bbfb801ee0e4d13bed7 : SUCCESS in 43m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d19fb457de1943acaad5662b3fde413e : SUCCESS in 1h 21m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3318587da4014160a657c532c6d75f6e : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72373a397f75483ea1052a2d5fadb37d : SUCCESS in 8m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/60e2fb73f03643de8c193e33235dceea : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19d03fff524a46a2938dac0c9501c9cd : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11937489541940268dda2a9170cc28c0 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3427caec17bd4e0995b42134df9f4566 : SUCCESS in 6m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7b6617b05c8b4d63aae547aa4b7506a6 : SUCCESS in 8m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/19efe3af483144f7bc282f70ca47c7d2 : SUCCESS in 8m 12s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d52f60a763d04b7e9677c5509d757958 : SUCCESS in 5m 49s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5a8c2273ebe944b891b83c126ecb567b : FAILURE in 1h 09m 16s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e891cecb4abe4252b38e7ced51259e66 : SUCCESS in 1h 13m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d1a4d93b6c5f4a04bc961fc532885241 : SUCCESS in 1h 28m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b6652a2e46dd4bbfb5f7e83b4cb93da9 : SUCCESS in 1h 17m 08s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/346da7f7ec4e456db37244e9c6c7ca74 : SUCCESS in 1h 03m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/befd5f3eee924e41a994a1793a7ebf27 : SUCCESS in 1h 17m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5b993781236b4eaa853bdbbbff73f4cc : SUCCESS in 35m 17s","accounts_in_message":[],"_revision_number":4},{"id":"2f79c171dc6614ae2b71d36dd50130461cddee11","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-04-09 13:50:19.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"6e3feea9b56512edd57e604fd204a505f8f0435d","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-04-16 04:17:06.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/783668/783668/4/check-not-vote/ext-glance-zuul/0247de5 : FAILURE in 16m 39s","accounts_in_message":[],"_revision_number":4},{"id":"5d784ec0d91b0e02f5701c43026a27962917f5ff","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-07-22 02:08:54.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nSorry it took me so long to get back to this.  The changes look fine.","accounts_in_message":[],"_revision_number":4},{"id":"ee017fff20bc6b067562423aac83d7a0b2ec3a85","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 02:10:03.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"eaf43aeed21821114a785033dec52725eb18f9d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 03:42:37.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/30f94470555142e99be5151bda989d98 : SUCCESS in 1h 08m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/131de7f1faf04d83b7f0c177abaacdff : SUCCESS in 1h 07m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21371e8aca214828b04a1324841ec3c1 : SUCCESS in 8m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c4e992be86e4f89ad8f3542cc143995 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd4118448fab42df9e6776aff5ad74d1 : SUCCESS in 10m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4025a9aa9e64cdab637c7f1560cb1b5 : SUCCESS in 6m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fd38f4c38ec24e55ae55f6cd88d4d360 : SUCCESS in 7m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6543e025691646c7a383f7f947570d4b : SUCCESS in 11m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6ceb5345ee57494dbc5586127845e723 : SUCCESS in 55m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/301d2fc2f5e54ad5982d4a1e82100268 : SUCCESS in 1h 07m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a65a18fa8ea4919b3f0d38f6f2c9944 : SUCCESS in 39m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f14f46e42a2041618bd8a3a06cb142d7 : SUCCESS in 1h 16m 25s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b60887292c0a49a9b067b66f5c051f9f : SUCCESS in 33m 01s","accounts_in_message":[],"_revision_number":4},{"id":"4739caef231a2749fb4e5bcf7a3630c0e6f0a0e6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 03:42:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"ca03ff1b7c445ae2d4c058fd4424395dd6a3c8b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-22 03:43:50.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35c7e62c609d44c09a7bb8d956fbd3b6 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7779c20198c4ee09ee2c31ba0f5752151331abaf","revisions":{"62ace347d8ae72aa2c06381c096bad857f6cf745":{"kind":"REWORK","_number":1,"created":"2021-03-29 15:10:04.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/68/783668/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/783668/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/783668/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/783668/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/783668/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/783668/1"}}},"commit":{"parents":[{"commit":"8fc9b653936de840371a1cec914d298c131f1575","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8fc9b653936de840371a1cec914d298c131f1575"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-29 15:07:41.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-29 15:07:41.000000000","tz":-420},"subject":"Debug stale image property removal","message":"Debug stale image property removal\n\nOccasionally we fail to delete a property from an image and get a\nStaleDataError from SQLAlchemy. It looks like maybe we are just racing\nto delete the same thing and can ignore this error, but we need more\ninformation. This catches that error and logs information about\nthe property we are trying to delete before re-raising.\n\nChange-Id: Ida5bc659b7481fa0a7f3926a54520628a7a4c406\nRelated-Bug: #1895173\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/62ace347d8ae72aa2c06381c096bad857f6cf745"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/62ace347d8ae72aa2c06381c096bad857f6cf745"}]},"branch":"refs/heads/master"},"17a500a01a5a280ef2187686b88db1520df1fb5e":{"kind":"REWORK","_number":2,"created":"2021-03-29 15:42:53.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/68/783668/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/783668/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/783668/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/783668/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/783668/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/783668/2"}}},"commit":{"parents":[{"commit":"8fc9b653936de840371a1cec914d298c131f1575","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8fc9b653936de840371a1cec914d298c131f1575"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-29 15:07:41.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-29 15:42:01.000000000","tz":-420},"subject":"Ignore stale image property removal","message":"Ignore stale image property removal\n\nOccasionally we fail to delete a property from an image and get a\nStaleDataError from SQLAlchemy. It looks like maybe we are just racing\nwith another operation to delete a single property (i.e. a race\nbetween image delete and the end of an import cleanup).\n\nThis ignores that \"can\u0027t delete because it is already deleted\" error\nand logs to debug about the situation for later auditing and\nconfirmation.\n\nChange-Id: Ida5bc659b7481fa0a7f3926a54520628a7a4c406\nCloses-Bug: #1895173\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/17a500a01a5a280ef2187686b88db1520df1fb5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/17a500a01a5a280ef2187686b88db1520df1fb5e"}]},"branch":"refs/heads/master"},"1dbb003dc9751fb29bfad1c086f8332b12c84bd4":{"kind":"REWORK","_number":3,"created":"2021-03-29 18:21:43.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/68/783668/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/783668/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/783668/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/783668/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/783668/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/783668/3"}}},"commit":{"parents":[{"commit":"8fc9b653936de840371a1cec914d298c131f1575","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8fc9b653936de840371a1cec914d298c131f1575"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-29 15:07:41.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-29 18:22:18.000000000","tz":-420},"subject":"Ignore stale image property removal","message":"Ignore stale image property removal\n\nOccasionally we fail to delete a property from an image and get a\nStaleDataError from SQLAlchemy. It looks like maybe we are just racing\nwith another operation to delete a single property (i.e. a race\nbetween image delete and the end of an import cleanup).\n\nThis ignores that \"can\u0027t delete because it is already deleted\" error\nand logs to debug about the situation for later auditing and\nconfirmation.\n\nChange-Id: Ida5bc659b7481fa0a7f3926a54520628a7a4c406\nCloses-Bug: #1895173\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1dbb003dc9751fb29bfad1c086f8332b12c84bd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1dbb003dc9751fb29bfad1c086f8332b12c84bd4"}]},"branch":"refs/heads/master"},"7779c20198c4ee09ee2c31ba0f5752151331abaf":{"kind":"REWORK","_number":4,"created":"2021-04-08 14:42:22.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/68/783668/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/783668/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/783668/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/783668/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/783668/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/783668/4"}}},"commit":{"parents":[{"commit":"8fc9b653936de840371a1cec914d298c131f1575","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8fc9b653936de840371a1cec914d298c131f1575"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-03-29 15:07:41.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-08 14:43:09.000000000","tz":-420},"subject":"Ignore stale image property removal","message":"Ignore stale image property removal\n\nOccasionally we fail to delete a property from an image and get a\nStaleDataError from SQLAlchemy. It looks like maybe we are just racing\nwith another operation to delete a single property (i.e. a race\nbetween image delete and the end of an import cleanup).\n\nThis ignores that \"can\u0027t delete because it is already deleted\" error\nand logs to debug about the situation for later auditing and\nconfirmation.\n\nChange-Id: Ida5bc659b7481fa0a7f3926a54520628a7a4c406\nCloses-Bug: #1895173\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7779c20198c4ee09ee2c31ba0f5752151331abaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7779c20198c4ee09ee2c31ba0f5752151331abaf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
