)]}'
{"id":"openstack%2Fglance~783893","triplet_id":"openstack%2Fglance~master~Ia8ff60f8184c533b9fc2a709d33cbdfc02793c41","project":"openstack/glance","branch":"master","topic":"core-reviewer-guidelines","hashtags":[],"change_id":"Ia8ff60f8184c533b9fc2a709d33cbdfc02793c41","subject":"Guidelines for core reviewers","status":"MERGED","created":"2021-03-30 15:08:55.000000000","updated":"2021-04-15 20:36:58.000000000","submitted":"2021-04-09 17:00:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":16,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783893-core-reviewer-guidelines","meta_rev_id":"7c903a92e4bcbc50b51b476c1cf1f52786693eda","_number":783893,"virtual_id_number":783893,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-09 17:00:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2021-04-08 14:40:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-04-05 13:30:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2021-04-09 15:07:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-30 15:40:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-30 18:50:13.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2021-03-31 13:22:10.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-04-01 19:00:56.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2021-04-01 19:35:12.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-04-15 20:36:58.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"b1bee064cb10cb75cb7e85b048202b3a99b488dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-30 15:08:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aa230f5a93d014d2edb3831a70b31c1c52dbfde5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 15:40:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5fa9a82e1a54a2992c965846622f6a1 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d97668c823a4219a106d13002f35b2e : SUCCESS in 8m 04s","accounts_in_message":[],"_revision_number":1},{"id":"70ee24037874d7fb12c3db7956a64e24d941c82c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-30 16:25:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2c720b2f229927811faa841e672c8889e5963d25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 17:00:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f800a928f9154ab1bd080a00362be80c : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cfe9b72b28646db8dd8f8445fbd2fb5 : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":2},{"id":"44c25f8c8fae25e70cadbcf80821b4dbe0136e1c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-30 18:50:13.000000000","message":"Patch Set 2:\n\n(5 comments)\n\nNice!\n\nJust a small number of grammar nits. Otherwise, I\u0027m hugely in favor of this, and thanks a lot for taking the time to put this together. This fits with my feeling on the expectations of core reviewers, and I think it matches the rest of the projects\u0027 intentions as well.","accounts_in_message":[],"_revision_number":2},{"id":"bdb74b079ab2fbff5e755d52420090bf1667d4be","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-31 05:41:24.000000000","message":"Patch Set 2:\n\n(5 comments)\n\nThank you for the review Dan!","accounts_in_message":[],"_revision_number":2},{"id":"5fa4b518073266fd29345edd5ce6fab0f8a473a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-03-31 05:59:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4e5904dd2c7e4f13b19bfc3cccf716502017555d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 06:10:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a134d3b7c2154a95b8f4e6321a871e26 : SUCCESS in 7m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b377c7dc1c62404083c0f0eeb9062a7a : SUCCESS in 8m 15s","accounts_in_message":[],"_revision_number":3},{"id":"92e7eae6ba3f21a20e39484c7cfa3e773da78ecc","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-03-31 13:22:10.000000000","message":"Patch Set 3: Code-Review+2\n\nThanks for the edits, looks great!","accounts_in_message":[],"_revision_number":3},{"id":"ea14822db83ca62e5c2bcf097cad057acdac4cdb","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-04-01 19:00:56.000000000","message":"Patch Set 3:\n\n(1 comment)\n\nJust one typo there :) I like the idea of being more explicit regarding our review process, thanks for writing this!","accounts_in_message":[],"_revision_number":3},{"id":"85142dc7e61cf66511dbf21f64528db5162ad239","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-04-01 19:35:12.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)\n\nOne nit inline, but otherwise this is great.  Thanks, Abhishek!  (Not approving in case anyone else wants to express an opinion before merging.)","accounts_in_message":[],"_revision_number":3},{"id":"c05bf3ac2b509b836b551bfe2d4b26e11de45e40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-04-05 05:41:52.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fd569012b287b6b43f838fc0c68c73f52ff1ab06","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-04-05 05:42:08.000000000","message":"Patch Set 3:\n\n(3 comments)\n\nThank you rosmaita and cyril for reviews, made changes as per your suggestions.","accounts_in_message":[],"_revision_number":3},{"id":"02128aa4e4bf9b643a804138b57b3efb51b3b8b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 05:57:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d02b30b0f5f4423982c9a08bc80adf88 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f942bf2927464ea5b9980ed615236e19 : SUCCESS in 7m 37s","accounts_in_message":[],"_revision_number":4},{"id":"a156d35a7d6dbebb8677c41cdeb8565659137789","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-05 13:30:24.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"c4d0547042787691e2a906341de2cb3be0cfaed8","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-04-08 14:40:31.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM!","accounts_in_message":[],"_revision_number":4},{"id":"e3549eb22cbdae345c23de2cfb4ad66629161dfb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-04-09 15:07:34.000000000","message":"Patch Set 4: Workflow+1\n\nApproving as per discussed in last weekly meeting. \nThank you all for the inputs!!","accounts_in_message":[],"_revision_number":4},{"id":"a9923cb3605801674d8fafd3ebac196519bf982e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 15:09:33.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"cccb088d71a7da1528a79f18b185e5effa6abaa6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 15:26:24.000000000","message":"Patch Set 4:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"4026b6cc90655ed76c8cdcde32c9635d6249c729","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 17:00:16.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92900ffdea514391aa288111c4024cf6 : SUCCESS in 7m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c190315778b442a8e138cdafbf6ed82 : SUCCESS in 10m 10s","accounts_in_message":[],"_revision_number":4},{"id":"28c8af3b4371257abce67a65defcff052f77bc1f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 17:00:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"0d50c618eaae8e7262df7299bea051f978e9e483","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 17:02:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51a7440b3da744419fe2b5ae58071f29 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":4},{"id":"7c903a92e4bcbc50b51b476c1cf1f52786693eda","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-04-15 20:36:58.000000000","message":"Patch Set 4:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/783893/783893/4/check-not-vote/ext-glance-zuul/003278b : FAILURE in 11m 44s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"120cb146ab48a61967ed4ded06d752fcb0839eaa","revisions":{"4673e76ed270a40341c452d4fb2418cc3150002d":{"kind":"REWORK","_number":1,"created":"2021-03-30 15:08:55.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/93/783893/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/783893/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/783893/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/783893/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/783893/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/783893/1"}}},"commit":{"parents":[{"commit":"8fc9b653936de840371a1cec914d298c131f1575","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8fc9b653936de840371a1cec914d298c131f1575"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-03-30 15:03:52.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-03-30 15:03:52.000000000","tz":0},"subject":"Guidelines for core reviewers","message":"Guidelines for core reviewers\n\nChange-Id: Ia8ff60f8184c533b9fc2a709d33cbdfc02793c41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4673e76ed270a40341c452d4fb2418cc3150002d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4673e76ed270a40341c452d4fb2418cc3150002d"}]},"branch":"refs/heads/master"},"3ecc0a296d34b5bb4957161918221641bfe1c4d5":{"kind":"REWORK","_number":2,"created":"2021-03-30 16:25:54.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/93/783893/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/783893/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/783893/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/783893/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/783893/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/783893/2"}}},"commit":{"parents":[{"commit":"8fc9b653936de840371a1cec914d298c131f1575","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8fc9b653936de840371a1cec914d298c131f1575"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-03-30 15:03:52.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-03-30 16:21:46.000000000","tz":0},"subject":"Guidelines for core reviewers","message":"Guidelines for core reviewers\n\nChange-Id: Ia8ff60f8184c533b9fc2a709d33cbdfc02793c41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3ecc0a296d34b5bb4957161918221641bfe1c4d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3ecc0a296d34b5bb4957161918221641bfe1c4d5"}]},"branch":"refs/heads/master"},"4862aa6e541f31e8c5b26dc78f4dae8f090eefe2":{"kind":"REWORK","_number":3,"created":"2021-03-31 05:59:18.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/93/783893/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/783893/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/783893/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/783893/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/783893/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/783893/3"}}},"commit":{"parents":[{"commit":"8fc9b653936de840371a1cec914d298c131f1575","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8fc9b653936de840371a1cec914d298c131f1575"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-03-30 15:03:52.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-03-31 05:55:05.000000000","tz":0},"subject":"Guidelines for core reviewers","message":"Guidelines for core reviewers\n\nChange-Id: Ia8ff60f8184c533b9fc2a709d33cbdfc02793c41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4862aa6e541f31e8c5b26dc78f4dae8f090eefe2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4862aa6e541f31e8c5b26dc78f4dae8f090eefe2"}]},"branch":"refs/heads/master"},"120cb146ab48a61967ed4ded06d752fcb0839eaa":{"kind":"REWORK","_number":4,"created":"2021-04-05 05:41:52.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/93/783893/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/93/783893/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/783893/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/783893/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/93/783893/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/93/783893/4"}}},"commit":{"parents":[{"commit":"8fc9b653936de840371a1cec914d298c131f1575","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8fc9b653936de840371a1cec914d298c131f1575"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-03-30 15:03:52.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-04-05 05:33:01.000000000","tz":0},"subject":"Guidelines for core reviewers","message":"Guidelines for core reviewers\n\nChange-Id: Ia8ff60f8184c533b9fc2a709d33cbdfc02793c41\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/120cb146ab48a61967ed4ded06d752fcb0839eaa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/120cb146ab48a61967ed4ded06d752fcb0839eaa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
