)]}'
{"id":"openstack%2Fglance~785218","triplet_id":"openstack%2Fglance~master~I1b6971888673b64ef60bed8fbcc97bbcbcf5c2ac","project":"openstack/glance","branch":"master","topic":"bug/1922928","hashtags":[],"change_id":"I1b6971888673b64ef60bed8fbcc97bbcbcf5c2ac","subject":"Fix image/tasks API for in-progress tasks","status":"MERGED","created":"2021-04-07 15:09:02.000000000","updated":"2021-04-16 03:10:13.000000000","submitted":"2021-04-08 07:11:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785218-bug/1922928","meta_rev_id":"1d9e3f4c855960c7973f005a89f44c5367493edc","_number":785218,"virtual_id_number":785218,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-08 07:11:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-04-08 04:21:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-04-08 04:21:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-07 16:31:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-08 04:21:19.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-04-16 03:10:13.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"206b581e535f206ff2640148291c7bb2300f285e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-07 15:09:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c750ecfb7ae9fe4e67026bdb006577cff4aad23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-07 15:09:30.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"53d02d7c7a4028e4f1917569225aac4376f7bac1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 16:31:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/ef24f45d300e4b0da478ea8c6c8165fb : SUCCESS in 1h 04m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f89024a9fc1f435f9c296122647be672 : SUCCESS in 1h 11m 02s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d4b61245e104dfc90b083469aae8fb2 : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b80ea76bf7849d8be21067c2f4cbd46 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93993f0a8b5f44cf81ea2fecd44a6ad8 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c4a8b93e46b487d99de1892db8b1260 : SUCCESS in 7m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1383db5e061f4a929fcbfe607b7e4ee7 : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09caebb61b6f4a8fb7886e6f04da85de : SUCCESS in 6m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/51b14a783587492abb55a25f7ca082b2 : SUCCESS in 17m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/62e376e761e445ad805da811e25e7780 : SUCCESS in 9m 42s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/31afd22eab4a4833b700a64f9f9eeb4d : SUCCESS in 5m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d097770d6644489780d0cfb6ae2a066f : SUCCESS in 1h 11m 03s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c69553c61d3e4aec87ec51e16fc6cfb5 : FAILURE in 1h 02m 21s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/27364c585365448db1b7ab46f1353c94 : SUCCESS in 1h 20m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a15c27bcae84406aa1970a91069d7204 : SUCCESS in 1h 19m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30c904181abb4266a8cb457a3ac9edc3 : SUCCESS in 41m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a73adf60ba3840a198cd5644ae443e7a : SUCCESS in 1h 11m 03s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/74b5adbab85a4b70872aaa504c80f625 : SUCCESS in 28m 04s","accounts_in_message":[],"_revision_number":2},{"id":"1e3dfc73fea425480b6327f6d3cd6444df56cb91","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-07 16:45:35.000000000","message":"Patch Set 2:\n\nrecheck cinder dance","accounts_in_message":[],"_revision_number":2},{"id":"a9fbd3ec0fea513d30833661ae20077e2339f1f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 18:48:45.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/8ea253f397c24d95a26573bbf5a1bf03 : SUCCESS in 1h 04m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ad17440d106c43eeabfccfc5c96d88fe : FAILURE in 1h 25m 17s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b658c13dc2647bfa5bf1fac5ec9ded3 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f64f25a8727f49aeadb363182c0d2c9f : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d51ae8b2063642f8a99c1e549185d4e9 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2146f502f8fd4f61a30734e2a7e6ced3 : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1f131149ab2476ebba11a23e7d97211 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6990d14ed2847288130c5204abc608d : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4cb6351a96a74a2dbfb6ffd2b641099e : SUCCESS in 9m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cb5e4cab336b4755810a84e6d8b0b205 : SUCCESS in 8m 20s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d703a63854954704af1d10296614918d : SUCCESS in 5m 28s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/992eca5a75864c55b56ec94abcb46a4a : SUCCESS in 1h 23m 50s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b682936e33de4da4a8e40f0186669c54 : SUCCESS in 1h 13m 35s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/31626baf0170431e849187b437a77a8b : SUCCESS in 1h 10m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f09af2b9026f4d48b53aa81ff869a873 : SUCCESS in 1h 12m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ef83165682845208a1b19ea7725b245 : SUCCESS in 1h 06m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e832524d822e4a40bb795fefce29888e : SUCCESS in 1h 27m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f9d3c4ec08494dc9b98a45771484e887 : SUCCESS in 28m 32s","accounts_in_message":[],"_revision_number":2},{"id":"f198ac76c6405c5f0a45b150b02e90e2ce8cf5c5","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-07 19:04:58.000000000","message":"Patch Set 2:\n\nrecheck cinder dance again","accounts_in_message":[],"_revision_number":2},{"id":"df6f9ce7bac07fb4a048ccb554b830b184d85df5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 20:34:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/afbd45609e8d4779966693f52387849c : SUCCESS in 1h 13m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1aa605c1edff4c71b0055efe739b97c9 : SUCCESS in 1h 18m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7384eec8e50b4f8e8e56cf215df0cfe3 : SUCCESS in 7m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d17317390d44386ba72f18c84d22610 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56e2e9b256f645f4b65b9caa5e3b47f0 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/451337fe313f4fcbb8653703c738f151 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47adbe3b6b4a4219a1fad6b1c7f1c8a1 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1912dcbb72a74fdfa26b2d951301e2b6 : SUCCESS in 7m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/58a3b300a34f4756ad43c42b1d811eac : SUCCESS in 8m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/31b4d258ae4f4148a6e26808c54d5ec8 : SUCCESS in 8m 16s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/43f9005570be410996a9b36ab5a1e0ff : SUCCESS in 5m 10s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2a238d5150c44fbe97849c4c34b09ca0 : SUCCESS in 1h 15m 28s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d392f9e1444a40808bf8ee66cbc45739 : SUCCESS in 1h 23m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f832fb6c116a49a68370a3168eb2db1e : SUCCESS in 1h 07m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7b43ce1ed20641db85643dfa44987142 : FAILURE in 1h 13m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9272b90635ff4ff7a3375feb0024eb62 : SUCCESS in 1h 00m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3a9838bc91bb46d7ab1eabff19119686 : SUCCESS in 1h 17m 00s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0555f7688b9a40d5a7386130803835e2 : SUCCESS in 40m 53s","accounts_in_message":[],"_revision_number":2},{"id":"b47813932b2bd0a428f59a2bf48dbe7f2894505c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-04-08 04:21:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nNinja approving as straightforward and easy change, need to backport as well.\nThank you Dan!!","accounts_in_message":[],"_revision_number":2},{"id":"c6de8501a2abfa60410468e1e9b5b8f15a7b2605","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 04:21:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5f5c7c3bb253b2987f6796e58399cd74a59ef632","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 07:11:08.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/d9daac4bf42d4f3c84ccd005f2ec863b : SUCCESS in 45m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/736212e0785c466c8bd6069d9f0ea974 : SUCCESS in 1h 07m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/498c80e2447a4d47a2d66dbced753edc : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/592cd97982194549852bee8a3782bf8a : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/21b5ec3fcd054f7195c63ef187856d56 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54f14fa2c92d490795212af93e615604 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90351452d7eb4174a2f7b8d3ebf70a7b : SUCCESS in 6m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/aa0bd5ee02834a1581761c8ed27a25c6 : SUCCESS in 9m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6ed86fcd414c4d7584b5d1f1c01b9382 : SUCCESS in 7m 21s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/aae83dcaafe349dc89b68e5b03aaed29 : SUCCESS in 1h 13m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/52103f2d3562411eace26fabfcab25ad : SUCCESS in 1h 10m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a99a578eb55448bf892669e01750b042 : SUCCESS in 45m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d42a363648b14dc094db3205658e3ae4 : SUCCESS in 1h 07m 00s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6cfb926acc674480943160be3a5afe45 : SUCCESS in 30m 33s","accounts_in_message":[],"_revision_number":2},{"id":"bf6390f595526af5d1992e254d5e234f98a5eafe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 07:11:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e657b4e3964c094c34e8c822139c2d4e87f49eed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 07:17:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efb09e4c5b4d449d9370ed4fa2ed8ca8 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":2},{"id":"1d9e3f4c855960c7973f005a89f44c5367493edc","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-04-16 03:10:13.000000000","message":"Patch Set 2:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/785218/785218/2/check-not-vote/ext-glance-zuul/0710a3b : FAILURE in 17m 30s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2a0d2303c3346b3e771a40c0fe4d4dceaf16f992","revisions":{"3194b2302c2061e9db5b3e3878c83eb8d8794c80":{"kind":"REWORK","_number":1,"created":"2021-04-07 15:09:02.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/18/785218/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/785218/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/785218/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/785218/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/785218/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/785218/1"}}},"commit":{"parents":[{"commit":"8fc9b653936de840371a1cec914d298c131f1575","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8fc9b653936de840371a1cec914d298c131f1575"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-07 15:07:05.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-07 15:07:05.000000000","tz":-420},"subject":"Fix image/tasks API for in-progress tasks","message":"Fix image/tasks API for in-progress tasks\n\nA slight error in the tasks_get_by_image() DB API method resulted in\nour excluding in-progress tasks from the returned list. This is\nbecause those tasks have expires_at\u003dNULL, and we were comparing\nthe expires_at\u003e\u003d$NOW to find unexpired tasks. This makes us check for\n\"NULL or not expired\" instead.\n\nWe did have a test asserting the wrong behavior, but it was done to\nincrease coverage and thus was asserting the behavior of the code and\nnot the *desired* behavior. This fixes that as well.\n\nChange-Id: I1b6971888673b64ef60bed8fbcc97bbcbcf5c2ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3194b2302c2061e9db5b3e3878c83eb8d8794c80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3194b2302c2061e9db5b3e3878c83eb8d8794c80"}]},"branch":"refs/heads/master"},"2a0d2303c3346b3e771a40c0fe4d4dceaf16f992":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-04-07 15:09:30.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/18/785218/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/18/785218/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/785218/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/785218/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/18/785218/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/18/785218/2"}}},"commit":{"parents":[{"commit":"8fc9b653936de840371a1cec914d298c131f1575","subject":"Merge \"Imported Translations from Zanata\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8fc9b653936de840371a1cec914d298c131f1575"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-07 15:07:05.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-07 15:10:03.000000000","tz":-420},"subject":"Fix image/tasks API for in-progress tasks","message":"Fix image/tasks API for in-progress tasks\n\nA slight error in the tasks_get_by_image() DB API method resulted in\nour excluding in-progress tasks from the returned list. This is\nbecause those tasks have expires_at\u003dNULL, and we were comparing\nthe expires_at\u003e\u003d$NOW to find unexpired tasks. This makes us check for\n\"NULL or not expired\" instead.\n\nWe did have a test asserting the wrong behavior, but it was done to\nincrease coverage and thus was asserting the behavior of the code and\nnot the *desired* behavior. This fixes that as well.\n\nCloses-Bug: #1922928\nChange-Id: I1b6971888673b64ef60bed8fbcc97bbcbcf5c2ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2a0d2303c3346b3e771a40c0fe4d4dceaf16f992"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2a0d2303c3346b3e771a40c0fe4d4dceaf16f992"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
