)]}'
{"id":"openstack%2Fglance~785420","triplet_id":"openstack%2Fglance~stable%2Fwallaby~I1b6971888673b64ef60bed8fbcc97bbcbcf5c2ac","project":"openstack/glance","branch":"stable/wallaby","hashtags":[],"change_id":"I1b6971888673b64ef60bed8fbcc97bbcbcf5c2ac","subject":"Fix image/tasks API for in-progress tasks","status":"MERGED","created":"2021-04-08 13:29:35.000000000","updated":"2021-09-10 21:23:11.000000000","submitted":"2021-09-10 21:20:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785420","meta_rev_id":"59e2a8f2b47ff897529dafdea7cfe5a68656c6fd","_number":785420,"virtual_id_number":785420,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-09-10 21:20:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2021-09-09 19:57:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2021-09-10 19:35:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2021-09-10 19:35:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-08 15:32:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-09 19:57:20.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2021-09-10 19:35:05.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"}],"messages":[{"id":"744268a7d772dee89f636d7e3aa9d424326c43b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-08 13:29:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9b6cfda6e32d79309d702713ee3a146c4072432","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 15:32:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/04a60a94b4d541baafbc8c2cf1cb54ca : SUCCESS in 1h 15m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/61b656ff35d345ba8d18edee5e47bb34 : FAILURE in 1h 00m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/192ed5807eea44bb8826de64b46ee32b : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e44d24db84544bcea186be8f257827df : SUCCESS in 8m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42b8d31dc6aa4b45bbbfabe60b390637 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7077345ecd64d65a6583f88f1811e78 : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/820606303c0d447f88cc7d5aeadd2e48 : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbe0c25dc79e4d75b38ec809cf7d4ddb : SUCCESS in 7m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1673e414ad744bea9461e7fa57c6267d : SUCCESS in 8m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d646635b84454fedaa454e9a7b1d3a81 : SUCCESS in 9m 29s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9c1f358326c64c3d94a19526edf438bd : SUCCESS in 5m 55s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e91809b8cb7941f1b6d91cf9c22d3ab6 : SUCCESS in 1h 15m 05s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5ec324c8d406484687bab8bdf0182b02 : SUCCESS in 1h 08m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b23653be22ee4cf5a003b388ac9ffa4a : SUCCESS in 1h 05m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a1a2f909886744e0b6a4cdb22e6e2267 : SUCCESS in 1h 24m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bba23c5875c74aefa33c6de0a4491329 : SUCCESS in 58m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dad48058ccd04d2abc43caad64cc8ef4 : FAILURE in 1h 46m 52s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ef8641dbcdc1473eb56ae8045d554892 : SUCCESS in 35m 31s","accounts_in_message":[],"_revision_number":1},{"id":"9c75d006e7178070ea181536f75837a9217bd252","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-08 17:48:59.000000000","message":"Patch Set 1:\n\nrecheck cinder dance","accounts_in_message":[],"_revision_number":1},{"id":"47bc94753c2da7727e106ef01d673a53598c9442","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 19:37:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/1fce92977edb48c88ec7713aa998fe78 : SUCCESS in 57m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c605e18439bb478592aa148d942d5df8 : SUCCESS in 1h 05m 32s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/221802fae9284f6683519b52079d48d4 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0e89b2782ab4485b51cf6edd4008bdc : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42e597c50ced43fa9b8791c420e3198a : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/feac6e1c451446e996d25b8fb6b95b5c : SUCCESS in 6m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f452436428ec415aaf2ccf7482573df9 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06a6c9dce94f426090f64fa9c3c2a5b3 : SUCCESS in 7m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cb081aa242e84103b1ed8f63e2d684cc : SUCCESS in 8m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2f34c174b1084c78b3393753d86febba : SUCCESS in 8m 19s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a6af77b6e64b4b15813e6c39777bced4 : SUCCESS in 5m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ee43b799e17d4636ac7ae3453942ad5a : SUCCESS in 1h 20m 33s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fdb3285d862b4a2aa02d0bcea612f6af : SUCCESS in 1h 44m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d116e04128b84c5db83c07dcbb5bd3cf : SUCCESS in 1h 09m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0096cbfbe766411aadaee280b98fa19e : SUCCESS in 1h 10m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c03e75bdccff4113ba21ce99c531ac98 : SUCCESS in 1h 10m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c23de4c85a1b4620965d6ffa29303ff0 : SUCCESS in 1h 16m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9d121cf76daf48e1b2237fa1df34f82f : SUCCESS in 24m 20s","accounts_in_message":[],"_revision_number":1},{"id":"525553db75456b62d117f1ac8080f867d1da70db","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2021-09-09 19:57:20.000000000","message":"Patch Set 1: Code-Review+2\n\nMerged to master, good thing to backport.","accounts_in_message":[],"_revision_number":1},{"id":"2017418f7a1831bcac7528306aefe95f7014d45e","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-09-10 19:35:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"843886fb5b9eecb1d5ae9930c808d222e95fa1af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 19:35:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"26092383df02a362e37ad01c7838f8391d363a72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 21:20:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/17473689151a4855a5b7c10c74054a38 : SUCCESS in 57m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d273d661e6214e57bd3974c14dc6fbb8 : SUCCESS in 1h 13m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c03f796659324eed803b2ae75497bdea : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7bbd23ad523042e58df22d9463454585 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e858558f260f4e599311b31d04ec0672 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81ecda8a24a04d68a6063a177c2d304a : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80932ac590634588bdf5e7b6614a22eb : SUCCESS in 8m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c1f45cc87f634ce8832bf8401a34c823 : SUCCESS in 9m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a1d106ca5df645a5989b639368e0c353 : SUCCESS in 10m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6ac73c9939b34c0b9abb3592b703b218 : SUCCESS in 1h 09m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b0ba0757e39f49caa2587c6f6ccc1629 : SUCCESS in 1h 07m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1fd21ba186a943e4a45cc462ab420e62 : SUCCESS in 46m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/75168db047a84992b4a563df1417e9a7 : SUCCESS in 1h 42m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/91475ab1cb0c44dbb4c8fb4849227166 : SUCCESS in 30m 54s","accounts_in_message":[],"_revision_number":1},{"id":"82a5233ed20d471cf5b72c7c23f765eb358d58cb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 21:20:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"59e2a8f2b47ff897529dafdea7cfe5a68656c6fd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-10 21:23:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48f95f99edb6476e826585173dc12f72 : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78e8b7506a7d68089cf08ba24085bf99e070e4ce","revisions":{"78e8b7506a7d68089cf08ba24085bf99e070e4ce":{"kind":"REWORK","_number":1,"created":"2021-04-08 13:29:35.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/20/785420/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/20/785420/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/785420/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/785420/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/785420/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/20/785420/1"}}},"commit":{"parents":[{"commit":"b5437773b20db3d6ef20d449a8a43171c8fc7f69","subject":"Update TOX_CONSTRAINTS_FILE for stable/wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b5437773b20db3d6ef20d449a8a43171c8fc7f69"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-07 15:07:05.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-08 13:30:22.000000000","tz":-420},"subject":"Fix image/tasks API for in-progress tasks","message":"Fix image/tasks API for in-progress tasks\n\nA slight error in the tasks_get_by_image() DB API method resulted in\nour excluding in-progress tasks from the returned list. This is\nbecause those tasks have expires_at\u003dNULL, and we were comparing\nthe expires_at\u003e\u003d$NOW to find unexpired tasks. This makes us check for\n\"NULL or not expired\" instead.\n\nWe did have a test asserting the wrong behavior, but it was done to\nincrease coverage and thus was asserting the behavior of the code and\nnot the *desired* behavior. This fixes that as well.\n\nCloses-Bug: #1922928\nChange-Id: I1b6971888673b64ef60bed8fbcc97bbcbcf5c2ac\n(cherry picked from commit 2a0d2303c3346b3e771a40c0fe4d4dceaf16f992)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/78e8b7506a7d68089cf08ba24085bf99e070e4ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/78e8b7506a7d68089cf08ba24085bf99e070e4ce"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
