)]}'
{"id":"openstack%2Fglance~785552","triplet_id":"openstack%2Fglance~stable%2Fussuri~Ib80e9cfb58680f9e8ead5993dc206f4da882dd09","project":"openstack/glance","branch":"stable/ussuri","topic":"ussuri/bug/1913625","hashtags":[],"change_id":"Ib80e9cfb58680f9e8ead5993dc206f4da882dd09","subject":"Add housekeeping module and staging cleaner","status":"MERGED","created":"2021-04-09 02:07:39.000000000","updated":"2021-08-02 05:46:03.000000000","submitted":"2021-08-02 05:44:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785552-ussuri/bug/1913625","meta_rev_id":"89ebab2ac16d93b781e397c0c47db2e53d92b0d0","_number":785552,"virtual_id_number":785552,"owner":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-02 05:44:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-08-02 04:48:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-08-02 04:48:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-09 02:07:39.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2021-04-09 03:07:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-24 16:26:31.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2021-08-02 04:48:36.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"567dcdd4064be20f9b7875f10c3574abe3e70932","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-04-09 02:07:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"67298b59e6fd64014c33ba02c871de5cdff11095","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 03:07:51.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/542ae478c2374c99ac659a78001802b3 : SUCCESS in 58m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c3a4ecd814c54b9ba38b581d09eb16e0 : FAILURE in 27m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5de6f65b9634f62b07445fcb85de3b1 : FAILURE in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1f9d2b5fa8a4cdbb09773d89e58de6b : SUCCESS in 5m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4473ad7eb9d74744b1e6e06ce42f9d63 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/891bc0f7be364dc48b6b96fd1da81697 : SUCCESS in 6m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4f2c2105cba34ec88c32c3a6d7d9b2ba : SUCCESS in 7m 32s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/74c0e455c3874b69b8bba729c7fd9ba3 : SUCCESS in 8m 09s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d9c57d4bfe974d0dbb4aba1bb3557aa4 : SUCCESS in 4m 21s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c5798456b4924447b2f40fb13fc5f888 : FAILURE in 27m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a72ab731ecc43b0be1649d48bdc060f : FAILURE in 26m 39s","accounts_in_message":[],"_revision_number":1},{"id":"4bdc41fac4361c7b45f6d2e4dabf5efb73529012","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-09 13:06:30.000000000","message":"Patch Set 1:\n\nFailure to start glance:\n\nhttps://zuul.opendev.org/t/openstack/build/c3a4ecd814c54b9ba38b581d09eb16e0/log/controller/logs/screen-g-api.txt#191\n\nplus the pep8 issues","accounts_in_message":[],"_revision_number":1},{"id":"e372297d4fa240a113f66831a8ce3a9f23f5d348","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-24 16:26:31.000000000","message":"Patch Set 1:\n\nCyril, could you please look at the pep8 issues on this patch?\nI think it will be better if we get this patch merged before releasing stable/ussuri.\n\nThank you.\n\nAbhishek","accounts_in_message":[],"_revision_number":1},{"id":"20c7cdd369fc1a9dd5fd3e71e6266d4b29dc4d25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-07-26 00:16:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c5168000d5d88e86ab6dcae88fde9d60da9fcba5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 01:34:02.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/09907f75b1604532a54ad8c016d114f6 : FAILURE in 1h 13m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/13b403b677e4485caae2df309ad1f6e0 : FAILURE in 22m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/253c927c489d4f289030e6fc6b6b4309 : FAILURE in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20b89ebb0fc445f0b23e0cc8d8ffdf5f : SUCCESS in 5m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6d88f8309360409596fc2d28edd2478b : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f472309d848f4469988bf7b4d6248392 : SUCCESS in 6m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2bd7dade3d974672911407e46f41c2ce : SUCCESS in 8m 44s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/e40a8f2bd59e49afb8a0c7ca32327d88 : SUCCESS in 9m 14s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9df4afb4bcd54130939c4ce7114d163a : RETRY_LIMIT in 3m 03s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/2e13034417d44584b7027e061cf46c74 : FAILURE in 25m 23s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1907f09e2084e67bf9182ccb6a6a0f8 : FAILURE in 41m 51s","accounts_in_message":[],"_revision_number":2},{"id":"07c3ace705f5698d9351cff0f7a1dccce7489720","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-26 13:53:52.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"9df48e6cb24c9dc73093c337f3d23b0cb3e9cc31","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-07-26 13:54:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9e8fc4824040025f2a00223a5d0acaddf8d639fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-07-26 14:11:04.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ad1274fd9cfa989e87bb91b95210c649053da694","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 15:36:22.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/a63b01fade814b6db8ec75761fef86b4 : SUCCESS in 58m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/97ca1e987a8c4ee78fd5562c1a2abcc7 : SUCCESS in 1h 19m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cdbba43e3ff4bd6b3a4ebb3af69554b : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ee9cce4b0f748c7a835e6aa719635cb : SUCCESS in 4m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe4cc0fb736c47f3b5b2a6ec9c41f7a2 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8470538da234a68b8505c2dc61bad59 : SUCCESS in 7m 09s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2b8ef0fbbac6418b8b7f06277f9630bb : SUCCESS in 7m 33s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/335628e6b07a4228bca4f7598f686599 : SUCCESS in 8m 28s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/2f62b8a4797e4620b9f40481af1481f1 : RETRY_LIMIT in 4m 00s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/ee9984ee59b646a7a45a926013a03b11 : SUCCESS in 1h 14m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1438cdfed622404e926097baf86c0169 : SUCCESS in 51m 36s","accounts_in_message":[],"_revision_number":5},{"id":"e20a6e0bf94e856fead2642e2d0b9905a1bc61a4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-26 15:40:18.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"69cb67f879d02d8901f42ee1ff1d9b9660a859df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 16:56:05.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/5d225e2d4f20453884ebcb0c6f765854 : SUCCESS in 1h 08m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a417f92a92574b15a2bb18360e4d2733 : SUCCESS in 1h 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6a60eb86cca4f39b9ac56c54ec33893 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb741b79da844478b722d9fc48e328f9 : SUCCESS in 6m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29f2c1bb44e74ab6b12ca56b106fa4b6 : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbf243f5f4af49ea875428cda0af9086 : SUCCESS in 7m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ed9055f53e4249e6ae7491314bad703e : SUCCESS in 7m 54s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/d5e84a68e8fd4781b5ee88a0b8c050e0 : SUCCESS in 9m 14s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/fcfaeb27f09d4a3f931bb1328e61f1cd : RETRY_LIMIT in 3m 25s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/60936470d7a6465282829ab95b142caa : SUCCESS in 1h 14m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e2f758c75c842a29a201360f67c84b4 : SUCCESS in 56m 31s","accounts_in_message":[],"_revision_number":5},{"id":"8178a932653760c59a854987a89eb33b871097c0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-26 17:02:45.000000000","message":"Patch Set 5:\n\nrecheck unrelated failure","accounts_in_message":[],"_revision_number":5},{"id":"e45ad5217f1eb632e7b16951ac3db2948aff3630","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-26 18:40:39.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/2638e87fedb44d29bf2528b71bb3d384 : SUCCESS in 1h 32m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/57c54f855b184fa2bc347f55b43fb270 : SUCCESS in 1h 12m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7e4f16d4bf846998bb18cacabdde93d : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c078e64810bc4750936053cceb993957 : SUCCESS in 7m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0de8aae937104483b59ba9cc121de7b1 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcf0f492b1f34fcf88f5dd31c606c3c2 : SUCCESS in 8m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6f20bf4a1f7245cea82a2b8edc21cda9 : SUCCESS in 13m 23s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/34ae8e6e42264245a0daad3cf5be5839 : SUCCESS in 10m 18s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/62e788de84ea4f8f944bc4ff2604086a : RETRY_LIMIT in 3m 54s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/3f65a4a43b2a4042bef53a00be4edb6c : SUCCESS in 1h 11m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ed86f3e9e1a470b98b34312966b5018 : SUCCESS in 51m 25s","accounts_in_message":[],"_revision_number":5},{"id":"32c123cff6e04abf70374757b9583559a6f99150","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-07-27 14:41:24.000000000","message":"Patch Set 5:\n\nrecheck unrelated failure","accounts_in_message":[],"_revision_number":5},{"id":"d9e8b7ceac500f50b2ac7781a20f072b13c5a476","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-27 16:00:46.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/13f17102d6d141f68c696ffebbb43acb : SUCCESS in 1h 07m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/44263fe375d149518d35817c54b7453f : SUCCESS in 1h 08m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b624d268f3647cf9a483b89263e9d22 : SUCCESS in 5m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/507f044a5f2e45e0b58fda2226fb8f36 : SUCCESS in 7m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c2706c762b084d2ab9c34909600d7dce : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/548545d3f848429e8b9d069752703420 : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/67551fe0fb02463bbe23137b76e50cb1 : SUCCESS in 7m 35s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/c05b0575d44f42049bdd2aaeea483f09 : SUCCESS in 8m 00s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0e9dcbcc6c764b458161700a34201b69 : RETRY_LIMIT in 3m 42s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/e1471d09932f4eefbb52776936975178 : SUCCESS in 1h 17m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de7a4cdc622646a29356e40fe1ef7ee6 : SUCCESS in 45m 44s","accounts_in_message":[],"_revision_number":5},{"id":"2d676143f584ae735154ca45007b2b5391bc614a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-29 08:41:42.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"1de1f3730af146293e199e1da440e08337526f12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 10:02:50.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/ce7a30abc3b542c190467a30040529a7 : SUCCESS in 1h 14m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0db49f2e665645e39c8e2f39c85f9c0a : SUCCESS in 1h 16m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/046ba23947794fe9b2fd53846ae45091 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ac1dc3f63ce4d9fb980e395b7225ea2 : SUCCESS in 5m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f764a9a5f0ed480895826a2563417393 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b0a2c30323e45f9bd7a7f52ac2a7e6d : SUCCESS in 7m 02s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f381f2fa1766421d89c97c301dc6af0d : SUCCESS in 8m 03s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/af62c219175246e68fa1141b34b69c67 : SUCCESS in 8m 18s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/22fc30ffbb0b400f87f4261d7397fec0 : RETRY_LIMIT in 4m 37s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/b251d6a789e74aec8a96229fe548ad2d : SUCCESS in 1h 20m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5aa1ffb4ff254074b38f4e97ec912dc6 : SUCCESS in 59m 10s","accounts_in_message":[],"_revision_number":5},{"id":"c0d88dc47154356d395eb2311b5d53992f01414a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-30 08:52:27.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"9a4c502be2ad11da0c0d56670c3231a5a19fdbe0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 08:52:50.000000000","message":"Patch Set 6: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch stable/ussuri.  The error was:\n\n  found undefined alias\n  in \"openstack/glance/.zuul.yaml@stable/ussuri\", line 227, column 31","accounts_in_message":[],"_revision_number":6},{"id":"1f7fd2bb3ca290e5b241a7e36d4061a63be30b3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-30 09:00:30.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"fd64bee6dad94945c31af637ed294905db6d3a10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 09:00:49.000000000","message":"Patch Set 7: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch stable/ussuri.  The error was:\n\n  found undefined alias\n  in \"openstack/glance/.zuul.yaml@stable/ussuri\", line 229, column 31","accounts_in_message":[],"_revision_number":7},{"id":"ac99a7bbce8659b8b1452faaf2bfe0bf1f5d2885","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-30 09:06:02.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"23562a8fb801a8d7465f7dba1aef20916a62cddf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 09:06:22.000000000","message":"Patch Set 8: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/glance on branch stable/ussuri.  The error was:\n\n  The key \"irrelevant-files\" appears more than once; duplicate keys are\n  not permitted.\n\nThe error appears in the following stanza:\n\n  irrelevant-files: \u0026tempest-irrelevant-files\n                - ^(test-|)requirements.txt$\n                - ^.*\\.rst$\n                - ^api-ref/.*$\n                - ^glance/hacking/.*$\n                - ^glance/locale/.*$\n                - ^glance/tests/.*$\n                - ^doc/.*$\n                - ^releasenotes/.*$\n                - ^tools/.*$\n                - ^tox.ini$\n                - ^\\.zuul\\.yaml$\n              irrelevant-files: *tempest-irrelevant-files\n\n  in \"openstack/glance/.zuul.yaml@stable/ussuri\", line 227, column 13","accounts_in_message":[],"_revision_number":8},{"id":"7cd5cc5e57d0668cfae1da1d511c9628da78b278","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-30 09:07:28.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"9724d381ae668d7376b2630d686475ea0e2104d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-30 09:15:11.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"938c9679c224f06fe76079a43a02da84dc07d45f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-30 09:18:13.000000000","message":"Patch Set 10:\n\nThis issue looks related;\n\nhttps://bugs.launchpad.net/nova/+bug/1885825","accounts_in_message":[],"_revision_number":10},{"id":"8f792111fa84ef1559246a59f3633fa67da9e9bd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-30 09:21:35.000000000","message":"Patch Set 10:\n\nI think we need to provide a patch on master to fix this and backport and merge that patch before this one.\n\nhttps://bugs.launchpad.net/nova/+bug/1885825\nExample of where/how to fix - https://review.opendev.org/c/openstack/nova/+/738723/1/tools/test-setup.sh","accounts_in_message":[],"_revision_number":10},{"id":"dd23f1cce3d4f8234c027e5fcbd3aa997f1a63c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 10:44:13.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/d8c9412f0f874654bb607c1451ee4307 : SUCCESS in 54m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7164430e6c704bf6874c46e68e37d4ad : SUCCESS in 1h 10m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69e45a288beb4b13991d575fc399ce90 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc17cfa3ccd34dd1887548897e030cc4 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cf286a63faad4b5d9f37474c87d73e24 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cc3d7af9f144cff8aef898560320a3d : SUCCESS in 6m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1cd1fe2bb9624f7bb8a00a2eb260e289 : SUCCESS in 8m 12s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/caab844c638547c0b8a26bf7e7414ce1 : SUCCESS in 8m 06s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a60688af3f0345ec896b1b3932e2f8b9 : RETRY_LIMIT in 4m 31s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/b58fa8cbe8234497a4088dff6a14288c : SUCCESS in 1h 23m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ad705112c41468d844ce7274aaab81b : SUCCESS in 47m 18s","accounts_in_message":[],"_revision_number":10},{"id":"6109e157a759a3a3b70998ee1384fc7099f72ec7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-30 14:47:42.000000000","message":"Patch Set 10:\n\nrecheck issue solved","accounts_in_message":[],"_revision_number":10},{"id":"5f7829e467b202bec53bfa52b3093c234dd4f290","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-30 16:34:06.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/d233de4e34ac452f82b89ab76c9a255e : SUCCESS in 1h 44m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0b21f1f53c914545837291be961d3c56 : SUCCESS in 1h 06m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80f8a7c1483f48a5a7b2f722b8893fbc : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a432c707a11b4da09b29d56b672df7aa : SUCCESS in 7m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/781a234c4e324d9eb3a5468a0dd06acf : SUCCESS in 8m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/285f3e1addf24cee909fe454444520d3 : SUCCESS in 7m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cf7762b6730549f9ae076894a4840120 : SUCCESS in 7m 41s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/074f491a13ea49bf97563bbae6290953 : SUCCESS in 16m 29s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ce273891155a4d158151e20eb4e40496 : SUCCESS in 5m 54s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/af52b8b13f4344eb921826e94c71fceb : SUCCESS in 1h 37m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/acf8dfba7ac9484582fd1d3d504af4bd : SUCCESS in 59m 51s","accounts_in_message":[],"_revision_number":10},{"id":"901be1d7b5c9ab98d5133a1a194772f2eea41642","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-08-02 04:48:36.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\nThank you Cyril!!","accounts_in_message":[],"_revision_number":10},{"id":"18796542485ed8c6ac020549d4b31e358c887345","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 04:49:04.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"739e4e990cf99c97019883a424ccce9eaad512ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 05:44:46.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/6a50f3cb8f4046c4860ba81e3ebcf5d9 : SUCCESS in 55m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e914713a2a1f41a2ad43466f300e23c1 : SUCCESS in 55m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e80cba7bba854abc96c2a16aabf2aa97 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed430113613a4d1cb46f3f398a6f1dc1 : SUCCESS in 6m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a38e0a1680364a7196d7b205babb9c93 : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bc0ab8300c74c33af97848f6fa3ccf8 : SUCCESS in 8m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cf7eed7306fb4b25bd9aad652e1fb981 : SUCCESS in 7m 56s\n- openstack-tox-functional-py37 https://zuul.opendev.org/t/openstack/build/692fd8f50a9f4bfb83d8bcbdfea00b5b : SUCCESS in 7m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f3418dfcf58410cb55273a51b8632be : SUCCESS in 48m 28s","accounts_in_message":[],"_revision_number":10},{"id":"149db211115d4e83f563253b496b76a07a2d171a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 05:44:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"89ebab2ac16d93b781e397c0c47db2e53d92b0d0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 05:46:03.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1863e113b3d642e8b4dbacfcb8df5c75 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"e87d3fbbe5e0bd404337a0e82b496edc30370e63","revisions":{"dbe7c30381ee064436ac0ba7d6368db0e6352dca":{"kind":"REWORK","_number":1,"created":"2021-04-09 02:07:39.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/52/785552/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/52/785552/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/52/785552/1"}}},"commit":{"parents":[{"commit":"54060e8b72b048d9a6a2ff2ca01a5fc18b6eafec","subject":"Make functional tests set node_staging_uri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/54060e8b72b048d9a6a2ff2ca01a5fc18b6eafec"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 20:52:03.000000000","tz":-480},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2021-04-09 01:59:33.000000000","tz":120},"subject":"Add housekeeping module and staging cleaner","message":"Add housekeeping module and staging cleaner\n\nAs noted in previous discussions, glance should clean its staging\ndirectory on startup. This is important for scenarios where we\nstarted an import operation, but failed in the middle. If, when we\nrecover, the image has already been deleted from the database, then\nwe will never remove the (potentially very large) residue from disk\nin our staging directory.\n\nThis is currently a problem with web-download, but will also occur\nwith glance-direct once we have the non-shared distributed import\nfunctionality merged.\n\nConflicts:\n\tglance/tests/unit/common/test_wsgi_app.py: Does not exist\n\nCloses-Bug: #1913625\nChange-Id: Ib80e9cfb58680f9e8ead5993dc206f4da882dd09\n(cherry picked from commit 232177e68c547a815c31a2f30b6d1f95cdb3098d)\n(cherry picked from commit 94679941368d37520975e0a1dc81d27424167191)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dbe7c30381ee064436ac0ba7d6368db0e6352dca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dbe7c30381ee064436ac0ba7d6368db0e6352dca"}]},"branch":"refs/heads/stable/ussuri"},"ff780791214bb76125e59f2f65c47f0096d9075d":{"kind":"REWORK","_number":2,"created":"2021-07-26 00:16:06.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/52/785552/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/52/785552/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/52/785552/2"}}},"commit":{"parents":[{"commit":"54060e8b72b048d9a6a2ff2ca01a5fc18b6eafec","subject":"Make functional tests set node_staging_uri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/54060e8b72b048d9a6a2ff2ca01a5fc18b6eafec"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 20:52:03.000000000","tz":-480},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2021-07-26 00:15:53.000000000","tz":120},"subject":"Add housekeeping module and staging cleaner","message":"Add housekeeping module and staging cleaner\n\nAs noted in previous discussions, glance should clean its staging\ndirectory on startup. This is important for scenarios where we\nstarted an import operation, but failed in the middle. If, when we\nrecover, the image has already been deleted from the database, then\nwe will never remove the (potentially very large) residue from disk\nin our staging directory.\n\nThis is currently a problem with web-download, but will also occur\nwith glance-direct once we have the non-shared distributed import\nfunctionality merged.\n\nConflicts:\n\tglance/tests/unit/common/test_wsgi_app.py: Does not exist\n\nCloses-Bug: #1913625\nChange-Id: Ib80e9cfb58680f9e8ead5993dc206f4da882dd09\n(cherry picked from commit 232177e68c547a815c31a2f30b6d1f95cdb3098d)\n(cherry picked from commit 94679941368d37520975e0a1dc81d27424167191)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ff780791214bb76125e59f2f65c47f0096d9075d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ff780791214bb76125e59f2f65c47f0096d9075d"}]},"branch":"refs/heads/stable/ussuri"},"5e4bdfdce57711f8cff42b3de5aec694caea98cf":{"kind":"REWORK","_number":3,"created":"2021-07-26 13:53:52.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/52/785552/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/52/785552/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/52/785552/3"}}},"commit":{"parents":[{"commit":"54060e8b72b048d9a6a2ff2ca01a5fc18b6eafec","subject":"Make functional tests set node_staging_uri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/54060e8b72b048d9a6a2ff2ca01a5fc18b6eafec"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 20:52:03.000000000","tz":-480},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-26 13:53:48.000000000","tz":0},"subject":"Add housekeeping module and staging cleaner","message":"Add housekeeping module and staging cleaner\n\nAs noted in previous discussions, glance should clean its staging\ndirectory on startup. This is important for scenarios where we\nstarted an import operation, but failed in the middle. If, when we\nrecover, the image has already been deleted from the database, then\nwe will never remove the (potentially very large) residue from disk\nin our staging directory.\n\nThis is currently a problem with web-download, but will also occur\nwith glance-direct once we have the non-shared distributed import\nfunctionality merged.\n\nConflicts:\n\tglance/tests/unit/common/test_wsgi_app.py: Does not exist\n\nCloses-Bug: #1913625\nChange-Id: Ib80e9cfb58680f9e8ead5993dc206f4da882dd09\n(cherry picked from commit 232177e68c547a815c31a2f30b6d1f95cdb3098d)\n(cherry picked from commit 94679941368d37520975e0a1dc81d27424167191)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5e4bdfdce57711f8cff42b3de5aec694caea98cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5e4bdfdce57711f8cff42b3de5aec694caea98cf"}]},"branch":"refs/heads/stable/ussuri"},"7b0f0c1fb02b0a342aa381820447de6d3f1838e9":{"kind":"REWORK","_number":4,"created":"2021-07-26 13:54:06.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/52/785552/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/52/785552/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/52/785552/4"}}},"commit":{"parents":[{"commit":"54060e8b72b048d9a6a2ff2ca01a5fc18b6eafec","subject":"Make functional tests set node_staging_uri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/54060e8b72b048d9a6a2ff2ca01a5fc18b6eafec"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 20:52:03.000000000","tz":-480},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2021-07-26 13:53:47.000000000","tz":120},"subject":"Add housekeeping module and staging cleaner","message":"Add housekeeping module and staging cleaner\n\nAs noted in previous discussions, glance should clean its staging\ndirectory on startup. This is important for scenarios where we\nstarted an import operation, but failed in the middle. If, when we\nrecover, the image has already been deleted from the database, then\nwe will never remove the (potentially very large) residue from disk\nin our staging directory.\n\nThis is currently a problem with web-download, but will also occur\nwith glance-direct once we have the non-shared distributed import\nfunctionality merged.\n\nConflicts:\n\tglance/tests/unit/common/test_wsgi_app.py: Does not exist\n\nCloses-Bug: #1913625\nChange-Id: Ib80e9cfb58680f9e8ead5993dc206f4da882dd09\n(cherry picked from commit 232177e68c547a815c31a2f30b6d1f95cdb3098d)\n(cherry picked from commit 94679941368d37520975e0a1dc81d27424167191)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7b0f0c1fb02b0a342aa381820447de6d3f1838e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7b0f0c1fb02b0a342aa381820447de6d3f1838e9"}]},"branch":"refs/heads/stable/ussuri"},"0ed8336397f66907c3390a91fa4323859ea6c62d":{"kind":"REWORK","_number":5,"created":"2021-07-26 14:11:04.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/52/785552/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/52/785552/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/52/785552/5"}}},"commit":{"parents":[{"commit":"54060e8b72b048d9a6a2ff2ca01a5fc18b6eafec","subject":"Make functional tests set node_staging_uri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/54060e8b72b048d9a6a2ff2ca01a5fc18b6eafec"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 20:52:03.000000000","tz":-480},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2021-07-26 14:10:47.000000000","tz":120},"subject":"Add housekeeping module and staging cleaner","message":"Add housekeeping module and staging cleaner\n\nAs noted in previous discussions, glance should clean its staging\ndirectory on startup. This is important for scenarios where we\nstarted an import operation, but failed in the middle. If, when we\nrecover, the image has already been deleted from the database, then\nwe will never remove the (potentially very large) residue from disk\nin our staging directory.\n\nThis is currently a problem with web-download, but will also occur\nwith glance-direct once we have the non-shared distributed import\nfunctionality merged.\n\nConflicts:\n\tglance/tests/unit/common/test_wsgi_app.py: Does not exist\n\nCloses-Bug: #1913625\nChange-Id: Ib80e9cfb58680f9e8ead5993dc206f4da882dd09\n(cherry picked from commit 232177e68c547a815c31a2f30b6d1f95cdb3098d)\n(cherry picked from commit 94679941368d37520975e0a1dc81d27424167191)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0ed8336397f66907c3390a91fa4323859ea6c62d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0ed8336397f66907c3390a91fa4323859ea6c62d"}]},"branch":"refs/heads/stable/ussuri"},"96eba1f80e51b9d8c215bccf2fd6a49841c40f70":{"kind":"REWORK","_number":6,"created":"2021-07-30 08:52:27.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/52/785552/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/52/785552/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/52/785552/6"}}},"commit":{"parents":[{"commit":"54060e8b72b048d9a6a2ff2ca01a5fc18b6eafec","subject":"Make functional tests set node_staging_uri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/54060e8b72b048d9a6a2ff2ca01a5fc18b6eafec"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 20:52:03.000000000","tz":-480},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-30 08:52:07.000000000","tz":0},"subject":"DNM - Add housekeeping module and staging cleaner","message":"DNM - Add housekeeping module and staging cleaner\n\nAs noted in previous discussions, glance should clean its staging\ndirectory on startup. This is important for scenarios where we\nstarted an import operation, but failed in the middle. If, when we\nrecover, the image has already been deleted from the database, then\nwe will never remove the (potentially very large) residue from disk\nin our staging directory.\n\nThis is currently a problem with web-download, but will also occur\nwith glance-direct once we have the non-shared distributed import\nfunctionality merged.\n\nConflicts:\n\tglance/tests/unit/common/test_wsgi_app.py: Does not exist\n\nCloses-Bug: #1913625\nChange-Id: Ib80e9cfb58680f9e8ead5993dc206f4da882dd09\n(cherry picked from commit 232177e68c547a815c31a2f30b6d1f95cdb3098d)\n(cherry picked from commit 94679941368d37520975e0a1dc81d27424167191)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/96eba1f80e51b9d8c215bccf2fd6a49841c40f70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/96eba1f80e51b9d8c215bccf2fd6a49841c40f70"}]},"branch":"refs/heads/stable/ussuri"},"bb278acdda7770d5f6bf5aec90d5fee8b8babda6":{"kind":"REWORK","_number":7,"created":"2021-07-30 09:00:30.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/52/785552/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/52/785552/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/52/785552/7"}}},"commit":{"parents":[{"commit":"54060e8b72b048d9a6a2ff2ca01a5fc18b6eafec","subject":"Make functional tests set node_staging_uri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/54060e8b72b048d9a6a2ff2ca01a5fc18b6eafec"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 20:52:03.000000000","tz":-480},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-30 09:00:21.000000000","tz":0},"subject":"DNM - Add housekeeping module and staging cleaner","message":"DNM - Add housekeeping module and staging cleaner\n\nAs noted in previous discussions, glance should clean its staging\ndirectory on startup. This is important for scenarios where we\nstarted an import operation, but failed in the middle. If, when we\nrecover, the image has already been deleted from the database, then\nwe will never remove the (potentially very large) residue from disk\nin our staging directory.\n\nThis is currently a problem with web-download, but will also occur\nwith glance-direct once we have the non-shared distributed import\nfunctionality merged.\n\nConflicts:\n\tglance/tests/unit/common/test_wsgi_app.py: Does not exist\n\nCloses-Bug: #1913625\nChange-Id: Ib80e9cfb58680f9e8ead5993dc206f4da882dd09\n(cherry picked from commit 232177e68c547a815c31a2f30b6d1f95cdb3098d)\n(cherry picked from commit 94679941368d37520975e0a1dc81d27424167191)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bb278acdda7770d5f6bf5aec90d5fee8b8babda6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bb278acdda7770d5f6bf5aec90d5fee8b8babda6"}]},"branch":"refs/heads/stable/ussuri"},"377cc574fcd1f49621c408576b9266a24624b259":{"kind":"REWORK","_number":8,"created":"2021-07-30 09:06:02.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/52/785552/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/52/785552/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/52/785552/8"}}},"commit":{"parents":[{"commit":"54060e8b72b048d9a6a2ff2ca01a5fc18b6eafec","subject":"Make functional tests set node_staging_uri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/54060e8b72b048d9a6a2ff2ca01a5fc18b6eafec"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 20:52:03.000000000","tz":-480},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-30 09:05:54.000000000","tz":0},"subject":"DNM - Add housekeeping module and staging cleaner","message":"DNM - Add housekeeping module and staging cleaner\n\nAs noted in previous discussions, glance should clean its staging\ndirectory on startup. This is important for scenarios where we\nstarted an import operation, but failed in the middle. If, when we\nrecover, the image has already been deleted from the database, then\nwe will never remove the (potentially very large) residue from disk\nin our staging directory.\n\nThis is currently a problem with web-download, but will also occur\nwith glance-direct once we have the non-shared distributed import\nfunctionality merged.\n\nConflicts:\n\tglance/tests/unit/common/test_wsgi_app.py: Does not exist\n\nCloses-Bug: #1913625\nChange-Id: Ib80e9cfb58680f9e8ead5993dc206f4da882dd09\n(cherry picked from commit 232177e68c547a815c31a2f30b6d1f95cdb3098d)\n(cherry picked from commit 94679941368d37520975e0a1dc81d27424167191)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/377cc574fcd1f49621c408576b9266a24624b259"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/377cc574fcd1f49621c408576b9266a24624b259"}]},"branch":"refs/heads/stable/ussuri"},"30a38f49092e4b747b1720f95b3e65b92dc870f1":{"kind":"REWORK","_number":9,"created":"2021-07-30 09:07:28.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/52/785552/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/52/785552/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/52/785552/9"}}},"commit":{"parents":[{"commit":"54060e8b72b048d9a6a2ff2ca01a5fc18b6eafec","subject":"Make functional tests set node_staging_uri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/54060e8b72b048d9a6a2ff2ca01a5fc18b6eafec"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 20:52:03.000000000","tz":-480},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-30 09:07:20.000000000","tz":0},"subject":"DNM - Add housekeeping module and staging cleaner","message":"DNM - Add housekeeping module and staging cleaner\n\nAs noted in previous discussions, glance should clean its staging\ndirectory on startup. This is important for scenarios where we\nstarted an import operation, but failed in the middle. If, when we\nrecover, the image has already been deleted from the database, then\nwe will never remove the (potentially very large) residue from disk\nin our staging directory.\n\nThis is currently a problem with web-download, but will also occur\nwith glance-direct once we have the non-shared distributed import\nfunctionality merged.\n\nConflicts:\n\tglance/tests/unit/common/test_wsgi_app.py: Does not exist\n\nCloses-Bug: #1913625\nChange-Id: Ib80e9cfb58680f9e8ead5993dc206f4da882dd09\n(cherry picked from commit 232177e68c547a815c31a2f30b6d1f95cdb3098d)\n(cherry picked from commit 94679941368d37520975e0a1dc81d27424167191)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/30a38f49092e4b747b1720f95b3e65b92dc870f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/30a38f49092e4b747b1720f95b3e65b92dc870f1"}]},"branch":"refs/heads/stable/ussuri"},"e87d3fbbe5e0bd404337a0e82b496edc30370e63":{"kind":"REWORK","_number":10,"created":"2021-07-30 09:15:11.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/52/785552/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/52/785552/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/52/785552/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/52/785552/10"}}},"commit":{"parents":[{"commit":"54060e8b72b048d9a6a2ff2ca01a5fc18b6eafec","subject":"Make functional tests set node_staging_uri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/54060e8b72b048d9a6a2ff2ca01a5fc18b6eafec"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-02-22 20:52:03.000000000","tz":-480},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-30 09:14:52.000000000","tz":0},"subject":"Add housekeeping module and staging cleaner","message":"Add housekeeping module and staging cleaner\n\nAs noted in previous discussions, glance should clean its staging\ndirectory on startup. This is important for scenarios where we\nstarted an import operation, but failed in the middle. If, when we\nrecover, the image has already been deleted from the database, then\nwe will never remove the (potentially very large) residue from disk\nin our staging directory.\n\nThis is currently a problem with web-download, but will also occur\nwith glance-direct once we have the non-shared distributed import\nfunctionality merged.\n\nConflicts:\n\tglance/tests/unit/common/test_wsgi_app.py: Does not exist\n\nCloses-Bug: #1913625\nChange-Id: Ib80e9cfb58680f9e8ead5993dc206f4da882dd09\n(cherry picked from commit 232177e68c547a815c31a2f30b6d1f95cdb3098d)\n(cherry picked from commit 94679941368d37520975e0a1dc81d27424167191)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e87d3fbbe5e0bd404337a0e82b496edc30370e63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e87d3fbbe5e0bd404337a0e82b496edc30370e63"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
