)]}'
{"id":"openstack%2Fglance~786828","triplet_id":"openstack%2Fglance~master~I64c50cfad194f726fd9e472a688b45e1507aef77","project":"openstack/glance","branch":"master","topic":"bug/1924612","hashtags":[],"change_id":"I64c50cfad194f726fd9e472a688b45e1507aef77","subject":"WIP: Fix image-list filters","status":"ABANDONED","created":"2021-04-19 08:08:27.000000000","updated":"2024-05-27 07:45:27.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b2c98606679ca3a0b584731d41093e391c23d8fb","_number":786828,"virtual_id_number":786828,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-04-19 09:23:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-19 09:23:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e68e08b8960267b5c00c70ade33e65edfa96603b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-04-19 08:08:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b12c1135dc3a49ed9ef2421da2f3be00ea35e57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 09:23:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/dbdead8019d442f1bc767edfe92c2479 : SUCCESS in 1h 02m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/85a01f1273cb4899ab14dfde9b4fd768 : SUCCESS in 1h 00m 13s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8180b34e5a904c4694973b331bb80e83 : FAILURE in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d915c84f35144038a9c86a63fb2011c : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8ff18ec115543409c5c5f6e403f6b4c : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e22c55d50e5475cb6384c87150a4d6d : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9788ae6f3a27448d8da51b0b12bb4068 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1143f25cee649f798c4ed7a6f344bb1 : SUCCESS in 6m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/39cb83ecf7204e36a3d42cc32a0a9f40 : SUCCESS in 9m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3408febdcb0d411bb8dfcc996c605097 : SUCCESS in 9m 50s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9d9e7d493c054768aa52a34b9b7b9165 : SUCCESS in 5m 45s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/253f82f1a9ea45d9aa1230e62f4a6218 : SUCCESS in 1h 12m 37s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d24f87c944b942879e94dcb4bdc91ead : SUCCESS in 1h 06m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/61a216db908d4d03b89c7c14b44e9ad2 : SUCCESS in 1h 03m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9ac2b5dc4bb74e8ea646a8d6f6d1ad52 : FAILURE in 55m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75105dcf15b24d1f9a5974844e445a0f : SUCCESS in 50m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/514f81e38a3e499e98d03d5be3573841 : FAILURE in 52m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9bbaee72144040ef9a0d196b7cc8d035 : SUCCESS in 27m 29s","accounts_in_message":[],"_revision_number":1},{"id":"b2c98606679ca3a0b584731d41093e391c23d8fb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-05-27 07:45:27.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1ec91df398c414b8f6610d3c51cc6175c713c3fb","revisions":{"1ec91df398c414b8f6610d3c51cc6175c713c3fb":{"kind":"REWORK","_number":1,"created":"2021-04-19 08:08:27.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/28/786828/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/28/786828/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/786828/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/786828/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/28/786828/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/28/786828/1"}}},"commit":{"parents":[{"commit":"1157c19a9c1408faad5b845cc9b1873f82fc8f31","subject":"Merge \"Guidelines for core reviewers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1157c19a9c1408faad5b845cc9b1873f82fc8f31"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-04-19 08:07:19.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-04-19 08:07:19.000000000","tz":0},"subject":"WIP: Fix image-list filters","message":"WIP: Fix image-list filters\n\nTodo: Unit tests for coverage\n\nCloses-Bug: #1924612\nChange-Id: I64c50cfad194f726fd9e472a688b45e1507aef77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1ec91df398c414b8f6610d3c51cc6175c713c3fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1ec91df398c414b8f6610d3c51cc6175c713c3fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
