)]}'
{"id":"openstack%2Fglance~787006","triplet_id":"openstack%2Fglance~master~I94cf080753cdb5498110430418cfa573ad4409b5","project":"openstack/glance","branch":"master","hashtags":[],"change_id":"I94cf080753cdb5498110430418cfa573ad4409b5","subject":"Use py3 as the default runtime for tox","status":"ABANDONED","created":"2021-04-20 01:30:57.000000000","updated":"2022-07-07 14:08:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7476d4e4c108b96655174718dd8d9e5d78d6228f","_number":787006,"virtual_id_number":787006,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-20 02:06:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":-1,"date":"2021-06-14 14:39:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-20 02:06:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-14 14:39:08.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"}],"messages":[{"id":"0fb659f96b6ff666fe86a94ee108430e7ccf6261","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-04-20 01:30:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a92e8362130a05a09817fc4bef02bc5790dc38a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 02:06:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ba2f894beea14aa7941d2924b3a27f4b : SUCCESS in 5m 06s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/79b3b3bca654482489bfbc5d29bb1d2e : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c95b1940e32c49b58987b7aa27d8288d : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f250623d10fd45f2b5ecc5583c0f9a79 : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a171dbb449a24148b322d5bf888bf57e : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17cb898418724949ae9fbac36200463c : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4688c3baee14cb494aa31db8ddee59f : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd25a8fad4f04075b7988da817a35229 : SUCCESS in 6m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6a630cb101304f479b59fd2c8062085d : SUCCESS in 8m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/696bb7241abe4ea7abee7fd50ed73222 : SUCCESS in 8m 33s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6366021b244f4559b77fb805184b0780 : SUCCESS in 30m 53s","accounts_in_message":[],"_revision_number":1},{"id":"3af1477e1d3206e36b7ba8a16b55aae476edc379","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-06-14 14:39:08.000000000","message":"Patch Set 1: Code-Review-1\n\nI\u0027d prefer the explicit list and py3.7 is still officially supported, we just test the lowest and the highest. The commit message is conflicting with the change as it would indeed run the tests on py3.7 should that be the default py3 in the environment, which also might be 3.5 or 3.9.","accounts_in_message":[],"_revision_number":1},{"id":"7476d4e4c108b96655174718dd8d9e5d78d6228f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-07 14:08:42.000000000","message":"Abandoned\n\nThis feature is not something we really need/want, and the author has not followed up with Erno\u0027s reviews, so let\u0027s abandon this.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"de7f06d6f511ec2479819f9c55b133264862b4bb","revisions":{"de7f06d6f511ec2479819f9c55b133264862b4bb":{"kind":"REWORK","_number":1,"created":"2021-04-20 01:30:57.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/06/787006/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/06/787006/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/787006/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/787006/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/06/787006/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/06/787006/1"}}},"commit":{"parents":[{"commit":"1157c19a9c1408faad5b845cc9b1873f82fc8f31","subject":"Merge \"Guidelines for core reviewers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1157c19a9c1408faad5b845cc9b1873f82fc8f31"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2021-04-20 01:29:19.000000000","tz":0},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2021-04-20 01:30:55.000000000","tz":0},"subject":"Use py3 as the default runtime for tox","message":"Use py3 as the default runtime for tox\n\nMoving on py3 as the default runtime for tox to avoid to update this at\neach new cycle.\n\nWallaby support officially the following runtimes [1]:\n- Python 3.6\n- Python 3.8\n\nDuring Victoria Python 3.7 was used as the default runtime [2] however this\nversion isn\u0027t longer officially supported.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/wallaby.html#python-runtimes-for-wallaby\n[2] https://governance.openstack.org/tc/reference/runtimes/victoria.html#python-runtimes-for-victoria\n\nChange-Id: I94cf080753cdb5498110430418cfa573ad4409b5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/de7f06d6f511ec2479819f9c55b133264862b4bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/de7f06d6f511ec2479819f9c55b133264862b4bb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
