)]}'
{"id":"openstack%2Fglance~787563","triplet_id":"openstack%2Fglance~master~I7ee77213c0391c9f3a16ca2b6f402a54e0edefad","project":"openstack/glance","branch":"master","topic":"ossn-0088","hashtags":[],"change_id":"I7ee77213c0391c9f3a16ca2b6f402a54e0edefad","subject":"Add example policy.yaml file to restrict metadef API access","status":"ABANDONED","created":"2021-04-22 13:43:00.000000000","updated":"2021-04-28 23:54:09.000000000","total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"5bb801f8d9eaf5854fc084db627554abc3bab341","_number":787563,"virtual_id_number":787563,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-04-22 16:40:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-22 13:54:25.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2021-04-22 16:40:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a257677111f0046572a4bc5418d1124039e8140e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-22 13:43:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"700a65bdbec65d005c995cb6d2ff077cc6fd0c76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 16:40:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/126b5a4338914938981000a8f7825f7e : SUCCESS in 1h 05m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/df003572a97b4865b321a68f1b3413b1 : SUCCESS in 2h 03m 57s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce364d236b4f4248959334bac3216f8b : SUCCESS in 8m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2dbe7a6190f4819a7d4e24848e527fa : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51d432d65730411abdc7eb34ad64d992 : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b3f27f45ee1a41acad2cc4d4d264290e : SUCCESS in 7m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c31ef68026e4e1e8083d9836bbd9a05 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/deb636a5580b45a5b89d8ff957f5c0b6 : SUCCESS in 7m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1c30ea36c3ce495bb096c7a4aa5e83d3 : SUCCESS in 10m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/20e8aa90ef65408faf797115f8bf93a2 : SUCCESS in 10m 13s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2ee2be0608084877b5258c2d9078b074 : SUCCESS in 1h 06m 48s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2b12ce595044480d889787a4ebd76414 : SUCCESS in 1h 54m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ff3eab99722140e7af1c8e930b6e990e : SUCCESS in 1h 13m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/246eccbfc3bc41f48af822a38d727265 : SUCCESS in 1h 06m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94e5384f459e4cb08a916bd94ed369a8 : SUCCESS in 1h 51m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/41f6a2c7336149ff9c3f3b113670972b : SUCCESS in 1h 14m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f6b4529990bc4389bf800788e7608b63 : SUCCESS in 35m 09s","accounts_in_message":[],"_revision_number":1},{"id":"8591b22a3c4eb6af9d3d4584141597b4609397c7","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-04-28 19:07:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f8506ff8dad77c8ad8c5480de3fdba890db69fa3","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-04-28 19:09:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"191154818f97d9119ea8139cde8a71642f37f279","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-28 23:54:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5bb801f8d9eaf5854fc084db627554abc3bab341","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-04-28 23:54:09.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"95e7ad83e7d15cf7b9a5045d8bf6ccbde76abaeb","revisions":{"95e7ad83e7d15cf7b9a5045d8bf6ccbde76abaeb":{"kind":"REWORK","_number":1,"created":"2021-04-22 13:43:00.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/63/787563/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/63/787563/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/787563/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/787563/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/787563/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/63/787563/1"}}},"commit":{"parents":[{"commit":"1157c19a9c1408faad5b845cc9b1873f82fc8f31","subject":"Merge \"Guidelines for core reviewers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1157c19a9c1408faad5b845cc9b1873f82fc8f31"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-22 13:40:14.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-04-22 13:42:19.000000000","tz":540},"subject":"Add example policy.yaml file to restrict metadef API access","message":"Add example policy.yaml file to restrict metadef API access\n\nThis introduces an example policy.yaml file which includes rules to\nrestrict metadef API access to only admin users, to avoid resource\nleak discovred[1].\nThis allows user to more easily consume the recommended rules.\n\n[1] https://wiki.openstack.org/wiki/OSSN/OSSN-0088\n\nChange-Id: I7ee77213c0391c9f3a16ca2b6f402a54e0edefad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/95e7ad83e7d15cf7b9a5045d8bf6ccbde76abaeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/95e7ad83e7d15cf7b9a5045d8bf6ccbde76abaeb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
