)]}'
{"id":"openstack%2Fglance~788055","triplet_id":"openstack%2Fglance~master~Idf5f004b72436df1f9c77bb32d60b9be5ae77a68","project":"openstack/glance","branch":"master","topic":"bp/glance-unified-quotas","hashtags":[],"change_id":"Idf5f004b72436df1f9c77bb32d60b9be5ae77a68","subject":"Enforce keystone limits for image upload","status":"MERGED","created":"2021-04-26 17:35:46.000000000","updated":"2021-06-30 21:36:09.000000000","submitted":"2021-06-30 21:34:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"788055-bp/glance-unified-quotas","meta_rev_id":"2289e7df035000134adc7ddc79f978ec747e0cba","_number":788055,"virtual_id_number":788055,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-30 21:34:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-06-30 21:36:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-06-30 19:59:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2021-06-30 19:59:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-06-30 21:36:09.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-26 19:03:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-29 00:05:28.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-04 02:33:22.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2021-06-04 15:28:28.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-06-30 19:59:52.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"428632f1513478d62523834ef75586371a9e1118","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-26 17:35:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"681904397c4b126e871379444b82357ca4d6adae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 19:03:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/475af6dfe11343ee88dc7eaadea7755e : SUCCESS in 1h 11m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7eef6c119d05422eb261fe6d368a51cf : SUCCESS in 58m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/502ce867cdf140b0805e19d92c22968a : FAILURE in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca27b3d4169a44fbadf5e6b7148893e3 : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7b38bd28b19d49fb9f449d4635b34f18 : SUCCESS in 6m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9e51781eb8445aca6fc9bfef4cf1a4d : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f60a3980c9d4e979b4716c132c21368 : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40c8fd22f01e465a90cb4abeea8eb5af : SUCCESS in 7m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5ba3c45ea933454db5a582d17f3fe98a : SUCCESS in 9m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2043dbeda7e9430abdc1af2f86ac7c5e : SUCCESS in 8m 32s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/c64d44e618d44e93958b1a2cb776ac9c : SUCCESS in 5m 25s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6afed01d88a04c2f99e9fbf4d87ac4ad : SUCCESS in 1h 12m 45s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/af41f2ed2f9d43399bbdb10e155b2479 : SUCCESS in 1h 08m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5a4233f793ab482ea5091a4ddc468db2 : SUCCESS in 1h 07m 52s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fbd740d64f51447ca04efc9ef5351c4d : FAILURE in 1h 21m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58743d16666b4b608c3b840ee5ad2a5e : SUCCESS in 57m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f9e5c3d450f2404bbc297d94fac3e5fd : SUCCESS in 1h 05m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/790471d029d3499ab4dcfc85e9ed8833 : SUCCESS in 37m 05s","accounts_in_message":[],"_revision_number":1},{"id":"1c6755097161d467314f4f585614d934119d512f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-26 19:09:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7fabc0ab4be6a1a130b1296d2217ae8d6f0c3300","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-26 20:30:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2cebd606c68f97368169930a6a8009f55dbb2170","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 21:43:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/9f7d6828ea4b495e9dd7f3da0ac5b59d : SUCCESS in 1h 04m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c60fb1476b8b4d61a9557da08c4ecea6 : SUCCESS in 1h 07m 30s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f9f029a216f4e3d920de564b0e9eb24 : FAILURE in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d4f4899a1524d119fc7843ceb0e9e6c : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5523ba82b376461592cd01a9d244cc75 : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ba8c3dd68b64980b34c101057be1641 : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eec7b055c611447e982d88e15094a422 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50ad3845e6394c8a8f45628befe3f665 : SUCCESS in 7m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/eaf294e7df2242288678f1332a03fa9d : SUCCESS in 9m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b2b894d061034635970707e7f2e5e4eb : SUCCESS in 7m 27s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/015e6e9b42be41eda3ce0eeca661fd05 : SUCCESS in 5m 43s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7e8267be3a0c4bcfb9a0106f8d9ace36 : FAILURE in 53m 49s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d06ca28a174f4d69a6ebd1cec8271557 : SUCCESS in 1h 09m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bcd4b7a5b26f4d13875d89322b4eb945 : SUCCESS in 1h 10m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5efbc2d0ef6d43e6972b26b935ba34a3 : SUCCESS in 1h 02m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00af935262fd4219b559b30dff915873 : SUCCESS in 48m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/74ee86fb08294948a7560f82b7eb7f2d : SUCCESS in 1h 10m 09s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/52f3f04f6bab48b78123438f395cb9be : SUCCESS in 48m 35s","accounts_in_message":[],"_revision_number":3},{"id":"ba152055d0ba0e57c6183eb98da87b4607d0583d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-27 13:15:09.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"907390312092db8da9c9021c4d910086a9495a67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 14:50:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/bba12f290deb45d39921f9738655c015 : SUCCESS in 57m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7d05031bfdcf4a339c69e020530d2391 : SUCCESS in 1h 15m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/711833fc09084fe38ccd088ef303862c : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a42194077784b1ab6a70c54e631eadb : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3db2c8d33164f9d9b52a8a170b94640 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b800b11046e44de83ece5c2e27347e5 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/299aca5878c04430b68c394109d2ebf2 : SUCCESS in 6m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1df6a0fb845f4630bbbdac9fb422293b : SUCCESS in 8m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9ecbe1c6405142aabf0ca075dac01182 : SUCCESS in 9m 18s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/de4e0954cc7b441d83427e4617ae580d : SUCCESS in 6m 04s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/82a05db6ac6e46cfbf65ada1e36c6d40 : SUCCESS in 1h 26m 39s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/76235c2635244130a600a49f492bfd07 : SUCCESS in 1h 05m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0d21f651f0de48e791911cdc7fd9c986 : SUCCESS in 1h 02m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/06eb66f706b84d09b505adaca5ce7319 : FAILURE in 1h 00m 05s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9e864b7ea744b6a94f04503c0c00089 : SUCCESS in 1h 00m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f00ab32fb12144fc85cb8f4a641bb714 : SUCCESS in 1h 01m 09s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/94e5657c9dfa49b89132a68ab9f2d6a6 : SUCCESS in 34m 26s","accounts_in_message":[],"_revision_number":4},{"id":"59cdf1128465074083f9c8e78035a5c86b29fae2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-27 16:58:02.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"e29fe5220046bb2e9e1c2e84a8a7761ccec99a7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 18:22:46.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/a906b3b61df14dea9198509a707f9e38 : SUCCESS in 1h 15m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/649feca848a3494f8404c9a556b1131a : SUCCESS in 1h 00m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/535e489d01644ca69507875d5da6dafd : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/03613c275b3c4940b938173fefe8a414 : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5bcf4fedd9740f79583fef67ea629df : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce49661f4ba4491891b1b6358a393675 : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1bb3750fc9242b29f1a51e35c981fc9 : SUCCESS in 7m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/51224feb314c4e71adafdea157db13c3 : SUCCESS in 10m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a46a64d0838f4a168d325e636693542d : SUCCESS in 8m 22s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/1395fc0d6a72484ab7e76ebfc0803751 : SUCCESS in 5m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/982c1a505c494a2db4a21c017882d809 : FAILURE in 1h 06m 26s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a478e20a20bd4a2d902ca39dae24b01b : SUCCESS in 1h 12m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/17bb9c97d7f24f4b9d4d5ea7c32881d5 : SUCCESS in 1h 19m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/80d81c225ffc426d98af5eca9eb814b6 : SUCCESS in 1h 12m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b56f2bafff543848944a175cec3127f : SUCCESS in 53m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c650acd8f3fc42ec8ce1647b62a5b940 : FAILURE in 44m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a8f7c4dd080d414c9fd27302d3c8ecb5 : SUCCESS in 27m 52s","accounts_in_message":[],"_revision_number":5},{"id":"367efbd38ccefdffa49355a7457bfe9867db5d6c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-28 23:35:07.000000000","message":"Patch Set 5:\n\nrecheck mysql catastrophic fail again","accounts_in_message":[],"_revision_number":5},{"id":"2f65a6e6c8dd7cb71b43044c74a7a08d7b609ae9","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-04-29 00:05:28.000000000","message":"Patch Set 3:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/788055/788055/3/check-not-vote/ext-glance-zuul/cb28c0f : FAILURE in 38m 45s","accounts_in_message":[],"_revision_number":3},{"id":"e5d14d7f4a38966d5d3e8c886349edc649936d69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-29 01:02:15.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/8e7dabac47d140a195aed848738cfa63 : SUCCESS in 1h 04m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9d41395e36ac40eeaddcf3fbe6e5f5cf : SUCCESS in 1h 08m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b577d9dd5ab467abe7e59959084b6e4 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3be1c5b1989c484b99d6eeb2330b31b5 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67a77cbaddc64f28b3e49769cf49192b : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d10beea577c348c7903c316a39161eee : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7df22223d7f42389173580abf7b6982 : SUCCESS in 6m 31s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d93fe3a0f2eb4ee3b0c91bb997edf4fb : SUCCESS in 8m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d088ccfc366c4c53974ca58720ad2718 : SUCCESS in 8m 16s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9bde8fc9cb2346cc971f671aefded576 : SUCCESS in 4m 37s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5dbb4693b00141529a17d083e08ee561 : SUCCESS in 1h 19m 10s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6982c23b6f0c4c59b192963672337fff : SUCCESS in 1h 19m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e66fe08563a34ec4b89daaf3003a4a11 : SUCCESS in 1h 20m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/df7b810e26ab49c7bc1ddeabd210c0d5 : FAILURE in 1h 20m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5c78a955bee4f80b04c7f2bf5a46a8e : SUCCESS in 1h 01m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4f01a2950e8443a7a1720817edf6573a : SUCCESS in 1h 05m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0393749ac1b349ed979356b18588452c : SUCCESS in 36m 01s","accounts_in_message":[],"_revision_number":5},{"id":"db4f9faf0c8949290e3045678188960f4f55d26c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-05-07 16:51:49.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"87bd4c50925ccbab18017b84e2c064cd4d3155a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 18:26:14.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/5d7fc2433f964b2b809ddce9a159eb9f : SUCCESS in 1h 12m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b3dc950fb8e542f28451f3b3af21f278 : SUCCESS in 1h 05m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de5a6f1b900d4974bb73ec7aa6c1feb4 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f8334354a10465e9c81a591ed7ff4ed : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42c538eff85b4b89856a46b69a905914 : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/846128c0228f4f2aaeda1e5112a5b651 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10a610fd02424f338695111f09f7908c : SUCCESS in 8m 41s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e9739513c83d4e3290e55351afe00cec : SUCCESS in 8m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/380582d0b61642c3b74c00a9dbc67ab9 : SUCCESS in 10m 18s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/50621af92dae4c9089b0d7900c3e7894 : SUCCESS in 5m 51s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/df75c7993c6644d599028ab5555d3723 : SUCCESS in 1h 16m 28s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c9c521550f7c4d83882280da6f41e59b : SUCCESS in 1h 30m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d0a5c87a917c41d88a39796b704e0b47 : SUCCESS in 1h 08m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fd204dd53cb74cefad463ef6dbac8d36 : SUCCESS in 1h 24m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71ffbdd417df4bd7b0b0439e43afbbd2 : SUCCESS in 49m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/969af2b7ebe443b6bb85b0e9af1c814a : SUCCESS in 1h 21m 58s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d45641828cbe410ea700c2c63b851b26 : SUCCESS in 47m 07s","accounts_in_message":[],"_revision_number":6},{"id":"50e03240f42d2348f4eb063bb75c8e2f3c50ffa9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-02 15:08:00.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"090e7154d324fd384d226c4fb38f97c801f4c8ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-02 17:48:20.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"03394b280e7b0d1f523dc96a79fe117eed3d89c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 20:06:01.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/66fdb4fea941422fbec0dce174ea8550 : SUCCESS in 1h 05m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/626d7aa13f0b4bb190938a0d556ff782 : SUCCESS in 1h 28m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86037393a0a1482f8a9138e921ada5b9 : SUCCESS in 6m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62c42c60086f41ebb6dbbe15deb50a8d : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94ba2f722bc9487c9cffa1c07a92996c : SUCCESS in 7m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab8ba1127ed34b19b2c286472014ba9f : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f471b76187754ef7a57c6cfdad00ff3a : SUCCESS in 6m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fb15a4041fc642059d1b9ea7b34c7adc : SUCCESS in 10m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6ffa2b5aa7a84333bdec0cf7051a3255 : SUCCESS in 8m 53s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/88e42880a6ed4485acdf3621a4b16145 : SUCCESS in 6m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fc24f7ddeaf24ce1bbb704ffaf008bd8 : SUCCESS in 1h 29m 34s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2fd58448085d4ce1baabf7704f7446b7 : SUCCESS in 1h 19m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/acd8451ac8684da58e0a816e5527088f : SUCCESS in 1h 20m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5969ef289d514a2b82b9ff2705f4443f : SUCCESS in 1h 13m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3c49af4992e7478086b3d9fab04fad11 : SUCCESS in 55m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4434ccca26434891a1b3446715a28745 : SUCCESS in 1h 27m 58s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b64b082721ac4fcbab795c09174ce056 : SUCCESS in 29m 54s","accounts_in_message":[],"_revision_number":8},{"id":"4f1d8e557c236323207c9942234a703968c2c812","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-06-04 02:33:22.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"6b5e96b162e4b134bad3dff0e9e587eb94e93585","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-04 13:45:38.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"0ac8c3b029b24464d52f553f111d9fedce626bd3","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-06-04 13:49:52.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ed012b39a97ec941404a59865699c2ab48c5db6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-04 14:09:35.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"7168e7340b7807095311bc2096c6fe7e4368b190","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-06-04 15:28:28.000000000","message":"Patch Set 9: Code-Review+1 Review-Priority+1\n\n(3 comments)\n\nLooks good to me, just one question inline and I will vote to +2","accounts_in_message":[],"_revision_number":9},{"id":"16d521df29d13240cf216d481755464770fa0439","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 16:21:22.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/93411567ddce4ac6be2a8b69ddae5305 : SUCCESS in 1h 00m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5bc9863560364d339a76ac38b4558548 : SUCCESS in 1h 15m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be215d5aef1a4113947d89dddaa9050a : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/846e2ecf8b634530a258d597e8b972dd : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e219de0348e84bddafc99876c8c9f4bb : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6bd6db18a7a4501ae70aebd0cc017d2 : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0facd41dd564a10bc9018ac8c034204 : SUCCESS in 7m 32s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9da2cf564424477b9b15d810764ce30a : SUCCESS in 9m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ce6448c894eb42e7acab80c3ce27a6de : SUCCESS in 8m 21s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8f8de8f488a344fcb26c5a681c93fe6d : SUCCESS in 5m 26s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/66fddca426a34b92b5e68b317c403e96 : SUCCESS in 1h 18m 39s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/90febbe8a0d94008ba2801572a78eb81 : SUCCESS in 1h 22m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7e27506228084240a14cda394dc04f3a : SUCCESS in 1h 15m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e84d0af63c374029a8444fafaa7a4cc1 : SUCCESS in 1h 18m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cfa5356e730c4e9cb2ca9ae9db455c1a : SUCCESS in 1h 14m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/294cb13bc2c246cdbb942dcebd988dbf : SUCCESS in 1h 24m 04s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/451c446882514429978cde09eef1087b : SUCCESS in 31m 10s","accounts_in_message":[],"_revision_number":9},{"id":"6c95fc816fe11c5385c42129eb7dcde5e6f9fadf","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-04 16:22:23.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"12b313d313ebab8fe1ac7fc070274f4d0ed6451a","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-04 16:46:08.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"a3280f088f08e088d09bf40a1127562d192c8438","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-04 17:08:07.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"1d51434cb4e548a454ca8e01305ec873f58b75a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-04 18:34:23.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"0a02d01a4a33542374a7441b12667c35d379dd24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 19:39:26.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/889cf42cc5c949468e6b94aabc63eaa1 : SUCCESS in 59m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9e705b8812c436fabae3298958eafde : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/96a5a67d87ff42139fa4e7edf7091002 : SUCCESS in 9m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/61054e345adf4a41a92422b93ea0d7fe : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e564e41314fd48779037ef61ecea6e78 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/343bc26f50d9465e844b226e42589d0e : SUCCESS in 6m 19s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/444b9966344d424180c55e5f86b7230a : SUCCESS in 12m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a95bd71a512847c5b10d1ae91b467148 : SUCCESS in 10m 59s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/64dea726f3b94d138ef95b6acccfa0a2 : SUCCESS in 6m 50s","accounts_in_message":[],"_revision_number":11},{"id":"f3d1b6c6d24b7f090473814e4b644cf70f8cc450","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-06-08 14:17:16.000000000","message":"Patch Set 11: Code-Review+2 Review-Priority+1\n\nLooks good to me!!!","accounts_in_message":[],"_revision_number":11},{"id":"4eb32af89899698ef5ce59cfc5c54c904a74790a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-14 14:40:58.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"3aaccc24c65af62ee122b0a7f67a6a9fabdacc30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-14 14:47:30.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"7530599e1fb46858f888a7cba3ebb8d2c6a178b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-14 14:56:36.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"f777b2e6b3bdb0c39cda128cb3aaec9aebbc397a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 16:25:58.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/0d265d70ec8e4880937d9cf6d3d94dab : SUCCESS in 56m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1f6cb59803e4406dbf8b352124607740 : SUCCESS in 1h 05m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b53a1636b114d73930035470a1af103 : SUCCESS in 5m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4074a4f6450348789cfa60b47a033abd : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec72e9894b0b4d33bb37b974bbe16fb2 : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/570f715c26044401aa63a9ae0321210d : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c51d4212b994a5dab8007fb80e5b00d : SUCCESS in 7m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/cf9d0b28b1eb43a9b5bae878df63ad64 : SUCCESS in 8m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b0ecbf26943645a299bae8b090a441cc : SUCCESS in 8m 45s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/5b7fe6e53adb494c9c3a0fcc997d6c13 : SUCCESS in 5m 05s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4db2197e78844ef99c8beb63cad2f4de : SUCCESS in 1h 07m 09s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a999cca7a3394411a580fa9151f6f7fd : SUCCESS in 1h 13m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/40f55af573f14e7b88125375e48b65f7 : SUCCESS in 1h 08m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/937f73b4ed9845328653a25c0501dc06 : SUCCESS in 1h 18m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7fc4dc61ce84a9b97c5de1bb642bd95 : SUCCESS in 51m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7307d207b7344340883f725190ea0936 : SUCCESS in 1h 12m 33s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/59415355ac5e46278eaffed31b0eaf25 : SUCCESS in 30m 58s","accounts_in_message":[],"_revision_number":14},{"id":"b4d62fb7676195220ca423be2a2c55b4bfa9daab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-21 15:46:41.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"c21dd2e993b127c45b25011c96e65f9284fb60dd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-06-21 16:11:58.000000000","message":"Patch Set 15: Code-Review+2 Review-Priority+1\n\nLooks better now!!\nJust to note for other reviewers, Image status cannot be reverted to uploading from importing due to some domain structure reservations, so it is reverted to queued state.","accounts_in_message":[],"_revision_number":15},{"id":"feaff2236942df8c2b49361279e6fe9bd2e5829c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 17:18:18.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/e71f708b2eaa410694619ca3907a7170 : SUCCESS in 52m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/00bd325eac084eb99b4c03da6ab61b19 : SUCCESS in 1h 08m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bb8d64da1a84bc9ada4a96c63655868 : SUCCESS in 6m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/447e0dbd96f94e11895900bcf3cfedad : SUCCESS in 6m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/689d0297dee64ce4acb93508fe2c1adb : SUCCESS in 7m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92bdbbf73dfa4e929f668d0a756c1bbd : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b69b270b394c421db93043c96ee9fa95 : SUCCESS in 7m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e005332ff0c647ad8c52d88d46bf3c5b : SUCCESS in 10m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e17bd5a9943544d69758db949b1b2d0e : SUCCESS in 9m 09s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/556fd9f9b84e4a99969df44aaafab5ac : SUCCESS in 5m 54s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/22f665380be74f838d6a3dae1c3618f3 : SUCCESS in 1h 18m 06s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/490a44e06b584e9991c1dcdf8b7b2173 : SUCCESS in 1h 22m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/630b4f1f0b634e69b4b46d29e91f96c2 : SUCCESS in 1h 11m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c49fc93ed1db4c4084ebee36043245a7 : SUCCESS in 1h 11m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e6887b27c59461097911e21bb12fcd8 : SUCCESS in 1h 03m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7b1566520ebb47d3bf1f5a391981e3a8 : SUCCESS in 1h 14m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/dff1e1efda7e4ec185942fb026fbefab : SUCCESS in 39m 52s","accounts_in_message":[],"_revision_number":15},{"id":"fde8fdf618e424e195b91815aae6379f8cc02e04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-21 17:57:14.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"f5a788ca4a0b5d3416fc58fbf1ae5461e1a9e037","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 19:14:03.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/cf28047cd8c84190bf124c6229f1fca3 : SUCCESS in 49m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5126529e6ef644899f1c1ae2fc9765bd : SUCCESS in 1h 10m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0d0c85b2ad946abac6be718e8d365fc : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d3fcb83ef0e4b35b82aa488ae1df75c : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2345932040e4bdd86639443391e51dd : SUCCESS in 6m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5452d932bdfc4991bf2ba6fa52d7b6a8 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc39b8f1c2864966a25b51d26a5df028 : SUCCESS in 7m 21s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c0d486ff3f134011b965055d616aad4c : SUCCESS in 10m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/58cf64649ec048cfbb6e0edde05a4bb8 : SUCCESS in 10m 11s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/72241593acb844778da99f59a6c129c2 : SUCCESS in 6m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c3fc0ea9c7764140852200035a3230e8 : SUCCESS in 1h 10m 02s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/dbc19888127f4233ae2346e4b435affa : SUCCESS in 1h 08m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0fdfe02d86ac4e48ab1e289e639b2a86 : SUCCESS in 1h 07m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9d23b8c8bfaa46e08dec99062929bb08 : SUCCESS in 1h 12m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6eb2d69c37947079676c6d9bc531c1c : SUCCESS in 47m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/eea04796b9eb44a1b35ddb58599de8a3 : SUCCESS in 1h 12m 00s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b9c83fba40fa4025a800aa292faee34a : SUCCESS in 31m 07s","accounts_in_message":[],"_revision_number":16},{"id":"5f62307330ae5160c5a54646436ca7ed5bc11cc0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-06-22 05:10:49.000000000","message":"Patch Set 16: Code-Review+2 Review-Priority+1\n\nChange in test case name only!!","accounts_in_message":[],"_revision_number":16},{"id":"3d25eb0a070d2b603bed17f4d36db1c3aab48df9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-29 14:51:33.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.","accounts_in_message":[],"_revision_number":17},{"id":"62a4de5063353a0e82fe02cec6057e0ceed5c676","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 16:25:18.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/b53f70f083654dbf87ac441bc93da9aa : SUCCESS in 1h 21m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bff9881b12cf4e2e89a1437dafd68e5f : SUCCESS in 1h 13m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ba4baf72e994ae1bf261333a1cce6d0 : SUCCESS in 7m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/baf6c7d1b7bf476c810d20db1bd56d8c : SUCCESS in 8m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9caa40e35f924d93aa135f2917bf051c : SUCCESS in 8m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55cb6b7d2bb64fa8b5a7b7f40bd61090 : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac312c629d2649019cd8a32eb0ec68f6 : SUCCESS in 8m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5ac50428a3f240b181ec2d80cc6e5552 : SUCCESS in 11m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/41eb3ec62da548d68b9a3238f1ccc31d : SUCCESS in 10m 30s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8a43d257fb3240cfb3c98a273bcf8934 : SUCCESS in 7m 25s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/88d167f12c3740bda0f0f2475c018b9d : SUCCESS in 1h 27m 19s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/664a79d013cd4885a7280aa43249ccfc : SUCCESS in 1h 09m 00s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a8cacef9c3f34fb08003a74318416118 : SUCCESS in 1h 01m 38s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ac564167795e4183b05093e1f060134f : SUCCESS in 1h 17m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8ce3c2cd2894c6291ce8bb586b320b6 : SUCCESS in 53m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/04ba4ec78f424fd4863f2e69a93f28c7 : SUCCESS in 1h 20m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0efa231337e34c998650d5f759be9a7c : SUCCESS in 34m 40s","accounts_in_message":[],"_revision_number":17},{"id":"a79945348bb43a185e5375afa49b1741632d15ad","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-29 23:16:01.000000000","message":"Patch Set 17:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/788055/788055/17/check-not-vote/ext-glance-zuul/b20e502 : FAILURE in 11m 36s","accounts_in_message":[],"_revision_number":17},{"id":"6e567b2cef59af83d007ea93c4d1dc4220183547","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-06-30 19:59:52.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":17},{"id":"d8b4439104255477289273d8e072fae86ea32196","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 20:00:16.000000000","message":"Patch Set 17: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":17},{"id":"013441f3c8e98cc7950c9b29081ea842f72a964d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 21:34:44.000000000","message":"Patch Set 17: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/58c41e5a43f94f9b8ba9ed704a03bafb : SUCCESS in 1h 07m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/256ae578c23b458c9611b74a4d9403e9 : SUCCESS in 1h 29m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2743ab4b23bc47f9aa1926334a968db7 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c13c3cda4c5f44ff9758ce5a2c2d6308 : SUCCESS in 5m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24f6bc8f147647ceb3e88eadb7a2241a : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2cf04cb4d00414f8b8866848930842f : SUCCESS in 6m 40s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d03d9611e2424976955b758ae171fa79 : SUCCESS in 8m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/99b16bbda39a43e6bc11370bae523ec5 : SUCCESS in 9m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3d88ab77b44e4258ac552f9fc1ef36b4 : SUCCESS in 1h 21m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3923a3c60733403ca2277228164f64b0 : SUCCESS in 1h 09m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed5e9fe624664958a2510e7fd4b2f342 : SUCCESS in 53m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/461abbcf19764d67b266de08ba35c9c3 : SUCCESS in 1h 17m 55s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5bfefe86eb404a5c914dfbc1bb00f523 : SUCCESS in 33m 31s","accounts_in_message":[],"_revision_number":17},{"id":"3d1adbf01172d89c3dd84913472c9d4b21a944ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 21:34:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":17},{"id":"5d125681cdc63e18e36b4d3e8edb99f8b71ae070","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 21:36:09.000000000","message":"Patch Set 17:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b1c58ab423d4b2a952bdd965af3fd56 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"76c3011a64ba371f2effabb55c42bfa1eee539a9","revisions":{"02d77977c7cf7f8ce1dd48ee2c976907fed54843":{"kind":"REWORK","_number":1,"created":"2021-04-26 17:35:46.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/55/788055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/788055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/788055/1"}}},"commit":{"parents":[{"commit":"fc282dae38e3667c11f57852bc384db2331ac888","subject":"WIP: Add unified quotas infrastructure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fc282dae38e3667c11f57852bc384db2331ac888"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 14:41:31.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 17:36:46.000000000","tz":-420},"subject":"WIP: Enforce keystone limits for image upload","message":"WIP: Enforce keystone limits for image upload\n\nThis is missing all tests, but putting up so I can get a URL for\nthe devstack depends-on.\n\nChange-Id: Idf5f004b72436df1f9c77bb32d60b9be5ae77a68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02d77977c7cf7f8ce1dd48ee2c976907fed54843"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02d77977c7cf7f8ce1dd48ee2c976907fed54843"}]},"branch":"refs/heads/master"},"029f3b68a2a3047be2e4f125fc634755214f942c":{"kind":"REWORK","_number":2,"created":"2021-04-26 19:09:15.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/55/788055/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/788055/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/788055/2"}}},"commit":{"parents":[{"commit":"9f08356881ab61c9350e30ed5671087c74c4616e","subject":"WIP: Add unified quotas infrastructure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9f08356881ab61c9350e30ed5671087c74c4616e"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 14:41:31.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 19:08:58.000000000","tz":-420},"subject":"WIP: Enforce keystone limits for image upload","message":"WIP: Enforce keystone limits for image upload\n\nThis adds enforcement of the image_size_total keystone limit for\nimage upload and import. We simply check the quota before either of\nthese operations and refuse to proceed further if the user is over\ntheir quota.\n\nRelated to blueprint glance-unified-quotas\nChange-Id: Idf5f004b72436df1f9c77bb32d60b9be5ae77a68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/029f3b68a2a3047be2e4f125fc634755214f942c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/029f3b68a2a3047be2e4f125fc634755214f942c"}]},"branch":"refs/heads/master"},"3f5b14b484ba09516145890d131f222b47a3ab13":{"kind":"REWORK","_number":3,"created":"2021-04-26 20:30:31.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/55/788055/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/788055/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/788055/3"}}},"commit":{"parents":[{"commit":"9f08356881ab61c9350e30ed5671087c74c4616e","subject":"WIP: Add unified quotas infrastructure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9f08356881ab61c9350e30ed5671087c74c4616e"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 14:41:31.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 20:19:04.000000000","tz":-420},"subject":"WIP: Enforce keystone limits for image upload","message":"WIP: Enforce keystone limits for image upload\n\nThis adds enforcement of the image_size_total keystone limit for\nimage upload and import. We simply check the quota before either of\nthese operations and refuse to proceed further if the user is over\ntheir quota.\n\nRelated to blueprint glance-unified-quotas\nChange-Id: Idf5f004b72436df1f9c77bb32d60b9be5ae77a68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3f5b14b484ba09516145890d131f222b47a3ab13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3f5b14b484ba09516145890d131f222b47a3ab13"}]},"branch":"refs/heads/master"},"f49af260800b0bc8d86310f92093a71f34e99c63":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-04-27 13:15:09.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/55/788055/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/788055/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/788055/4"}}},"commit":{"parents":[{"commit":"bb4e392c1050512c50d13db16ebf48ca9e7d5279","subject":"WIP: Add unified quotas infrastructure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bb4e392c1050512c50d13db16ebf48ca9e7d5279"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 14:41:31.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 13:16:26.000000000","tz":-420},"subject":"WIP: Enforce keystone limits for image upload","message":"WIP: Enforce keystone limits for image upload\n\nThis adds enforcement of the image_size_total keystone limit for\nimage upload and import. We simply check the quota before either of\nthese operations and refuse to proceed further if the user is over\ntheir quota.\n\nRelated to blueprint glance-unified-quotas\nChange-Id: Idf5f004b72436df1f9c77bb32d60b9be5ae77a68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f49af260800b0bc8d86310f92093a71f34e99c63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f49af260800b0bc8d86310f92093a71f34e99c63"}]},"branch":"refs/heads/master"},"2cfaa8b7133767c83ffb5d243344ab0c0c5a96de":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-04-27 16:58:02.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/55/788055/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/788055/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/788055/5"}}},"commit":{"parents":[{"commit":"431f4b8b513bf5304b773cbb02ff1bdc67938426","subject":"WIP: Add unified quotas infrastructure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/431f4b8b513bf5304b773cbb02ff1bdc67938426"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 14:41:31.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 15:32:00.000000000","tz":-420},"subject":"WIP: Enforce keystone limits for image upload","message":"WIP: Enforce keystone limits for image upload\n\nThis adds enforcement of the image_size_total keystone limit for\nimage upload and import. We simply check the quota before either of\nthese operations and refuse to proceed further if the user is over\ntheir quota.\n\nRelated to blueprint glance-unified-quotas\nChange-Id: Idf5f004b72436df1f9c77bb32d60b9be5ae77a68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2cfaa8b7133767c83ffb5d243344ab0c0c5a96de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2cfaa8b7133767c83ffb5d243344ab0c0c5a96de"}]},"branch":"refs/heads/master"},"b367dad23140733e127c0a3e5e61c5ae38211d73":{"kind":"REWORK","_number":6,"created":"2021-05-07 16:51:49.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/55/788055/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/788055/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/788055/6"}}},"commit":{"parents":[{"commit":"e50dd05f1143aaf94c07bae06c702b425d2a10ff","subject":"Add unified quotas infrastructure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e50dd05f1143aaf94c07bae06c702b425d2a10ff"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 14:41:31.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-05-07 16:50:22.000000000","tz":-420},"subject":"Enforce keystone limits for image upload","message":"Enforce keystone limits for image upload\n\nThis adds enforcement of the image_size_total keystone limit for\nimage upload and import. We simply check the quota before either of\nthese operations and refuse to proceed further if the user is over\ntheir quota.\n\nRelated to blueprint glance-unified-quotas\nChange-Id: Idf5f004b72436df1f9c77bb32d60b9be5ae77a68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b367dad23140733e127c0a3e5e61c5ae38211d73"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b367dad23140733e127c0a3e5e61c5ae38211d73"}]},"branch":"refs/heads/master"},"25e7316af3d9a9a2d7b9cc005660198b38ef2b70":{"kind":"REWORK","_number":7,"created":"2021-06-02 15:08:00.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/55/788055/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/788055/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/788055/7"}}},"commit":{"parents":[{"commit":"755b8239b6df5880e9a2194f4492abdfb902d857","subject":"Add unified quotas infrastructure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/755b8239b6df5880e9a2194f4492abdfb902d857"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 14:41:31.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-02 15:09:50.000000000","tz":-420},"subject":"Enforce keystone limits for image upload","message":"Enforce keystone limits for image upload\n\nThis adds enforcement of the image_size_total keystone limit for\nimage upload and import. We simply check the quota before either of\nthese operations and refuse to proceed further if the user is over\ntheir quota.\n\nRelated to blueprint glance-unified-quotas\nChange-Id: Idf5f004b72436df1f9c77bb32d60b9be5ae77a68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/25e7316af3d9a9a2d7b9cc005660198b38ef2b70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/25e7316af3d9a9a2d7b9cc005660198b38ef2b70"}]},"branch":"refs/heads/master"},"46be14a6dbd960ff7fffc9b738e44cb9ba072b10":{"kind":"REWORK","_number":8,"created":"2021-06-02 17:48:20.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/55/788055/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/788055/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/788055/8"}}},"commit":{"parents":[{"commit":"126f2c63af09fe2d84540e0b6d472713d56afa7e","subject":"Add unified quotas infrastructure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/126f2c63af09fe2d84540e0b6d472713d56afa7e"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 14:41:31.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-02 17:50:03.000000000","tz":-420},"subject":"Enforce keystone limits for image upload","message":"Enforce keystone limits for image upload\n\nThis adds enforcement of the image_size_total keystone limit for\nimage upload and import. We simply check the quota before either of\nthese operations and refuse to proceed further if the user is over\ntheir quota.\n\nNote that this disables checking of the global size quota if keystone\nquotas are enabled.\n\nRelated to blueprint glance-unified-quotas\nChange-Id: Idf5f004b72436df1f9c77bb32d60b9be5ae77a68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/46be14a6dbd960ff7fffc9b738e44cb9ba072b10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/46be14a6dbd960ff7fffc9b738e44cb9ba072b10"}]},"branch":"refs/heads/master"},"86a8d7b69f560ca878dc2dd3f5e2fe8066fdc02a":{"kind":"REWORK","_number":9,"created":"2021-06-04 14:09:35.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/55/788055/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/788055/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/788055/9"}}},"commit":{"parents":[{"commit":"03a704cee7616a90a98e62d9094a2b238f3fff65","subject":"Add unified quotas infrastructure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/03a704cee7616a90a98e62d9094a2b238f3fff65"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 14:41:31.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-04 14:11:14.000000000","tz":-420},"subject":"Enforce keystone limits for image upload","message":"Enforce keystone limits for image upload\n\nThis adds enforcement of the image_size_total keystone limit for\nimage upload and import. We simply check the quota before either of\nthese operations and refuse to proceed further if the user is over\ntheir quota.\n\nNote that this disables checking of the global size quota if keystone\nquotas are enabled.\n\nRelated to blueprint glance-unified-quotas\nChange-Id: Idf5f004b72436df1f9c77bb32d60b9be5ae77a68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/86a8d7b69f560ca878dc2dd3f5e2fe8066fdc02a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/86a8d7b69f560ca878dc2dd3f5e2fe8066fdc02a"}]},"branch":"refs/heads/master"},"df05fb22b436971757bde20aa4d2d5a03f128ec4":{"kind":"REWORK","_number":10,"created":"2021-06-04 17:08:07.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/55/788055/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/788055/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/788055/10"}}},"commit":{"parents":[{"commit":"07b8bbbbbeb9cc7a337cbdcccb7b6c66b6788023","subject":"Add unified quotas infrastructure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/07b8bbbbbeb9cc7a337cbdcccb7b6c66b6788023"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 14:41:31.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-04 16:48:24.000000000","tz":-420},"subject":"Enforce keystone limits for image upload","message":"Enforce keystone limits for image upload\n\nThis adds enforcement of the image_size_total keystone limit for\nimage upload and import. We simply check the quota before either of\nthese operations and refuse to proceed further if the user is over\ntheir quota.\n\nNote that this disables checking of the global size quota if keystone\nquotas are enabled.\n\nPartially-implements: blueprint glance-unified-quotas\nChange-Id: Idf5f004b72436df1f9c77bb32d60b9be5ae77a68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/df05fb22b436971757bde20aa4d2d5a03f128ec4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/df05fb22b436971757bde20aa4d2d5a03f128ec4"}]},"branch":"refs/heads/master"},"007776e65fea26491666361e2ab4c415721c0bf2":{"kind":"REWORK","_number":11,"created":"2021-06-04 18:34:23.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/55/788055/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/788055/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/788055/11"}}},"commit":{"parents":[{"commit":"17cc6adf1d72351b233f6c4616d171a019b62740","subject":"Add unified quotas infrastructure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/17cc6adf1d72351b233f6c4616d171a019b62740"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 14:41:31.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-04 18:23:02.000000000","tz":-420},"subject":"Enforce keystone limits for image upload","message":"Enforce keystone limits for image upload\n\nThis adds enforcement of the image_size_total keystone limit for\nimage upload and import. We simply check the quota before either of\nthese operations and refuse to proceed further if the user is over\ntheir quota.\n\nNote that this disables checking of the global size quota if keystone\nquotas are enabled.\n\nPartially-implements: blueprint glance-unified-quotas\nChange-Id: Idf5f004b72436df1f9c77bb32d60b9be5ae77a68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/007776e65fea26491666361e2ab4c415721c0bf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/007776e65fea26491666361e2ab4c415721c0bf2"}]},"branch":"refs/heads/master"},"7a449a4b7d389b68fc52698a62eb925f1951c033":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2021-06-14 14:40:58.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/55/788055/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/788055/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/788055/12"}}},"commit":{"parents":[{"commit":"3f4e69c7ccdee041a0b7de69c70e5ab1e2dbe553","subject":"Add unified quotas infrastructure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3f4e69c7ccdee041a0b7de69c70e5ab1e2dbe553"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 14:41:31.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-14 14:42:51.000000000","tz":-420},"subject":"Enforce keystone limits for image upload","message":"Enforce keystone limits for image upload\n\nThis adds enforcement of the image_size_total keystone limit for\nimage upload and import. We simply check the quota before either of\nthese operations and refuse to proceed further if the user is over\ntheir quota.\n\nNote that this disables checking of the global size quota if keystone\nquotas are enabled.\n\nPartially-implements: blueprint glance-unified-quotas\nChange-Id: Idf5f004b72436df1f9c77bb32d60b9be5ae77a68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a449a4b7d389b68fc52698a62eb925f1951c033"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a449a4b7d389b68fc52698a62eb925f1951c033"}]},"branch":"refs/heads/master"},"6bfcbf4be3af93ca0524f43e44ae43a8f545f805":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2021-06-14 14:47:30.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/55/788055/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/788055/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/788055/13"}}},"commit":{"parents":[{"commit":"41e769b431b95314cebed2fc255870e7c22df488","subject":"Add unified quotas infrastructure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/41e769b431b95314cebed2fc255870e7c22df488"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 14:41:31.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-14 14:49:32.000000000","tz":-420},"subject":"Enforce keystone limits for image upload","message":"Enforce keystone limits for image upload\n\nThis adds enforcement of the image_size_total keystone limit for\nimage upload and import. We simply check the quota before either of\nthese operations and refuse to proceed further if the user is over\ntheir quota.\n\nNote that this disables checking of the global size quota if keystone\nquotas are enabled.\n\nPartially-implements: blueprint glance-unified-quotas\nChange-Id: Idf5f004b72436df1f9c77bb32d60b9be5ae77a68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6bfcbf4be3af93ca0524f43e44ae43a8f545f805"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6bfcbf4be3af93ca0524f43e44ae43a8f545f805"}]},"branch":"refs/heads/master"},"7eb5a7376c3111bc501695d56c0d8cd8ae878e00":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2021-06-14 14:56:36.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/55/788055/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/788055/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/788055/14"}}},"commit":{"parents":[{"commit":"744cd977ad796cd99ba537a5d7229949748099e1","subject":"Add unified quotas infrastructure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/744cd977ad796cd99ba537a5d7229949748099e1"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 14:41:31.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-14 14:58:35.000000000","tz":-420},"subject":"Enforce keystone limits for image upload","message":"Enforce keystone limits for image upload\n\nThis adds enforcement of the image_size_total keystone limit for\nimage upload and import. We simply check the quota before either of\nthese operations and refuse to proceed further if the user is over\ntheir quota.\n\nNote that this disables checking of the global size quota if keystone\nquotas are enabled.\n\nPartially-implements: blueprint glance-unified-quotas\nChange-Id: Idf5f004b72436df1f9c77bb32d60b9be5ae77a68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7eb5a7376c3111bc501695d56c0d8cd8ae878e00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7eb5a7376c3111bc501695d56c0d8cd8ae878e00"}]},"branch":"refs/heads/master"},"d31be26af16a055aa6a9b34006ba33d47825f207":{"kind":"REWORK","_number":15,"created":"2021-06-21 15:46:41.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/55/788055/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/788055/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/788055/15"}}},"commit":{"parents":[{"commit":"1b84ff871ed8337aac5a11d97a9a416047245452","subject":"Add unified quotas infrastructure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1b84ff871ed8337aac5a11d97a9a416047245452"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 14:41:31.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-21 15:46:58.000000000","tz":-420},"subject":"Enforce keystone limits for image upload","message":"Enforce keystone limits for image upload\n\nThis adds enforcement of the image_size_total keystone limit for\nimage upload and import. We simply check the quota before either of\nthese operations and refuse to proceed further if the user is over\ntheir quota.\n\nNote that this disables checking of the global size quota if keystone\nquotas are enabled.\n\nNote this includes another fix to couple unit tests that do not\nproperly pass context to the get_flow() method.\n\nPartially-implements: blueprint glance-unified-quotas\nChange-Id: Idf5f004b72436df1f9c77bb32d60b9be5ae77a68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d31be26af16a055aa6a9b34006ba33d47825f207"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d31be26af16a055aa6a9b34006ba33d47825f207"}]},"branch":"refs/heads/master"},"caf105df707f88e7161e607077f50b52e1230b0e":{"kind":"REWORK","_number":16,"created":"2021-06-21 17:57:14.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/55/788055/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/788055/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/788055/16"}}},"commit":{"parents":[{"commit":"06e6542f153a1ba197361f3f8cb0010c7d2d4285","subject":"Add unified quotas infrastructure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/06e6542f153a1ba197361f3f8cb0010c7d2d4285"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 14:41:31.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-21 17:59:17.000000000","tz":-420},"subject":"Enforce keystone limits for image upload","message":"Enforce keystone limits for image upload\n\nThis adds enforcement of the image_size_total keystone limit for\nimage upload and import. We simply check the quota before either of\nthese operations and refuse to proceed further if the user is over\ntheir quota.\n\nNote that this disables checking of the global size quota if keystone\nquotas are enabled.\n\nNote this includes another fix to couple unit tests that do not\nproperly pass context to the get_flow() method.\n\nPartially-implements: blueprint glance-unified-quotas\nChange-Id: Idf5f004b72436df1f9c77bb32d60b9be5ae77a68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/caf105df707f88e7161e607077f50b52e1230b0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/caf105df707f88e7161e607077f50b52e1230b0e"}]},"branch":"refs/heads/master"},"76c3011a64ba371f2effabb55c42bfa1eee539a9":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2021-06-29 14:51:33.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/55/788055/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/788055/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/788055/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/788055/17"}}},"commit":{"parents":[{"commit":"f3eb5214a34eb96fcf5f534723178d46429e5a01","subject":"Merge \"Add unified quotas infrastructure\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f3eb5214a34eb96fcf5f534723178d46429e5a01"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 14:41:31.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-29 15:53:18.000000000","tz":-420},"subject":"Enforce keystone limits for image upload","message":"Enforce keystone limits for image upload\n\nThis adds enforcement of the image_size_total keystone limit for\nimage upload and import. We simply check the quota before either of\nthese operations and refuse to proceed further if the user is over\ntheir quota.\n\nNote that this disables checking of the global size quota if keystone\nquotas are enabled.\n\nNote this includes another fix to couple unit tests that do not\nproperly pass context to the get_flow() method.\n\nPartially-implements: blueprint glance-unified-quotas\nChange-Id: Idf5f004b72436df1f9c77bb32d60b9be5ae77a68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/76c3011a64ba371f2effabb55c42bfa1eee539a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/76c3011a64ba371f2effabb55c42bfa1eee539a9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
