)]}'
{"id":"openstack%2Fglance~788077","triplet_id":"openstack%2Fglance~master~I3cca4e589adc0aec138e5933c311aefd69ccee51","project":"openstack/glance","branch":"master","topic":"bp/glance-unified-quotas","hashtags":[],"change_id":"I3cca4e589adc0aec138e5933c311aefd69ccee51","subject":"Add image_stage_total quota enforcement","status":"MERGED","created":"2021-04-26 20:30:31.000000000","updated":"2021-07-01 05:57:26.000000000","submitted":"2021-07-01 05:56:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"788077-bp/glance-unified-quotas","meta_rev_id":"fb51e3085565e2f5acc770b61d113afa0e0dd1a5","_number":788077,"virtual_id_number":788077,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-01 05:56:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-07-01 05:57:26.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2021-06-30 20:21:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2021-06-30 20:21:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-07-01 05:57:26.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-26 21:55:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-10 12:56:36.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2021-06-04 15:50:27.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-06-05 13:34:34.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-16 13:52:52.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2021-06-30 20:21:41.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"65f3e7b7e28d5a49935d6c3867324c9c65acd709","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-26 20:30:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16a80c589f404e543b14237adf6df9d4dbf638ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 21:55:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/fceb1cc770784ee39f98d2f59d94a57d : SUCCESS in 51m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d8ce97f68d42494da0edb54ccc120123 : SUCCESS in 1h 10m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d918a0f067b4abb86852015fb87de9e : FAILURE in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7007b8348abf40e89c8f25efe35193be : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b74af009aa1840e4a63192b870a183f7 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6bbdf37da84540618a0018acc6cc9f70 : SUCCESS in 5m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e962073ea5c488ba9553458dca7197e : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0db941423a2d4e0a88f917982363fbb9 : SUCCESS in 6m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2401e8bcf335482fb358780e28482f08 : SUCCESS in 8m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e1f4bb81d169489990355ec139d5b6c3 : SUCCESS in 8m 35s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ceef5619ce39410087ca0aa083108859 : SUCCESS in 5m 00s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/83b41b609f2b418e8f7d0674ee393020 : SUCCESS in 1h 10m 03s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fcd12dd5a7df4e72b7820b6723c57092 : FAILURE in 23m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a2976075830448fcbcd94fc54e1270e5 : FAILURE in 19m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/dc4603a383d74453914812974c9fd791 : FAILURE in 29m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77f27478e37240fcad651175d12daed7 : SUCCESS in 1h 04m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5b69879284c44ffa84d183a250e3a79b : FAILURE in 20m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/eb8fca07c45f4d0c9bc90232a32fecf0 : SUCCESS in 31m 51s","accounts_in_message":[],"_revision_number":1},{"id":"633d34bc3a6b198e917e2f19ced710cf2e6db06c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-26 21:58:02.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"73478e940781bf5ee9cf1ef7f1628feeebcb8c54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 23:13:46.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/1c7b160292fc48e7b889dce44cfe6d0c : SUCCESS in 47m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c05d6bce4a654559bd0094f3235ce0da : SUCCESS in 51m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d51026930103415b95d0ebb63dc5383c : FAILURE in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36ace4ded2df4fc7b219a5decbac082f : SUCCESS in 9m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55fb96c680e54620bd201b61b6b4ed15 : FAILURE in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4447fbf7b5164790873f434c292c9576 : FAILURE in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c104dac6ca74a7eb5704f7636938b6d : FAILURE in 8m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02614bf7b6564360b54a56cef6559513 : SUCCESS in 7m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/229841a35fd5410fbdc4364d01086303 : SUCCESS in 9m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a0935d42b4504c22adbe88d333cb9b02 : SUCCESS in 9m 26s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9abfbbebe2c24806945e894ff369371d : SUCCESS in 5m 44s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/10a995e50d30482ca2624ddf6a5ab670 : SUCCESS in 1h 12m 40s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/57ae65bb503c4ec3b578b7f3f66dfc93 : POST_FAILURE in 18m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/aea6be0a59ea4ab492fb0362ec1f559c : POST_FAILURE in 19m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cc71897c8a7b40f29bc4ee0ed0cfb86c : POST_FAILURE in 31m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4d40c7ade5d4afab2bbdad1d4f2bb7d : SUCCESS in 58m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/82c59d0d063148ac9811f17ac59d4398 : FAILURE in 24m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/02a2ada082ac4e788fe1bbe0bf75daff : SUCCESS in 22m 32s","accounts_in_message":[],"_revision_number":2},{"id":"43a3b245b055f90cbfda87f618c804e2b317603e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-27 13:15:09.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"d655f793079e2526b10c1e9efcb965c2489279ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 14:46:31.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/02a40bc954f34bc4aa473cea0ff8307d : SUCCESS in 57m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/eaddce6efd694fc1a6fafaec6907df18 : SUCCESS in 1h 04m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe2cd14ecb0848a28952738e7f3973f2 : SUCCESS in 8m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b77579d2561b440c953310342bb5b1ed : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d87df4100d9b4155b000ee9f83ab6a63 : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c96e18befc3c4962abc15ee7cd93ec79 : SUCCESS in 9m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb375d50c1b2414589d4d3c0f4002190 : SUCCESS in 9m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fc3a5ed3c5bd472b80c648f0157c4f0e : SUCCESS in 12m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/821ec984b22c4cf99d81cef0798b7dd3 : SUCCESS in 8m 11s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/14f4f8aa38b74f1eb11b4e8cea439117 : SUCCESS in 5m 12s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/dd1b567a7a654dafab7a8960765af290 : SUCCESS in 1h 20m 55s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c3af02448bd64dd39b6300125bc7f723 : SUCCESS in 1h 16m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e66d3f8077b349e092fbb19eafb373ac : SUCCESS in 49m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/53868a4cba984133ba6b328e731121c0 : SUCCESS in 59m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d7dafe01b65e4d11a4d8e8f95e53d6c0 : POST_FAILURE in 58m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d519ddb37f11436da343d51c3c12294b : SUCCESS in 1h 10m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f9319f9080334654b6628c3c79cf82ba : SUCCESS in 23m 43s","accounts_in_message":[],"_revision_number":3},{"id":"c8cee166a0857af4bf59073710a6701b1c5e41a4","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-27 14:54:09.000000000","message":"Patch Set 3:\n\nrecheck post fail","accounts_in_message":[],"_revision_number":3},{"id":"f801e2b32923f2e163c50005cf105bc1f04e8b7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 16:55:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/db9fc4f100054df5a3bc361fd16737df : SUCCESS in 1h 03m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f35b4a5d3d69428aa31dcfd6d2d12d0a : SUCCESS in 1h 13m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca31bc3dde4b4683b68d8550c71b5a4d : SUCCESS in 4m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23f9014965734f60973a6f939d3fb4c1 : SUCCESS in 9m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3641e00c98b1473fa39a41dac669c236 : SUCCESS in 9m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/16d880a7516347ff8880c7b4ba2f128d : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c43cafba946e43a7878280ce1f0689c5 : SUCCESS in 8m 10s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/06d0a076c22e473fa4cf7705261d5011 : SUCCESS in 8m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7ca3ec2151e645029c0b52a03b8e2ba8 : SUCCESS in 8m 02s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8bdc730ad27c4496b384b56e5f665817 : SUCCESS in 4m 43s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f8da3d0e9e1243999f738a1afb8e760a : SUCCESS in 1h 21m 52s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/43c6211d584044868ac1c4f65a6ab5ca : SUCCESS in 1h 08m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ad9a6705cac440189a75223eed80b6d0 : SUCCESS in 1h 13m 57s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e0827ca0c034427dbfd5e767a1ea68e3 : SUCCESS in 1h 14m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a637ab8307204293b1117854f6d1be7b : SUCCESS in 57m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f9bf2da3874c4a359521ac32c377ecd6 : SUCCESS in 1h 56m 20s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/06d62de6b34142b88749eba6bc268815 : SUCCESS in 31m 49s","accounts_in_message":[],"_revision_number":3},{"id":"09005a7f0f13da96fe24526a7a987b3602bd6c86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-27 16:58:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"74864b1500aa3a087355b58bdf78b1528cf3eab0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 18:37:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/127afbc47ed9470ab90dad2d8b8e5ffd : SUCCESS in 1h 12m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2c5c04197d6d4ababa14262438d322c4 : SUCCESS in 1h 03m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d57cbb45698244cfa612383df680d9c3 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcf4d5b1b7c945008d658fa827f1b7f2 : SUCCESS in 8m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b2880642a5044e5943475a77582bdab : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de64ec71fedb4113887dbec17feca574 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f31f2844f173456da8799ce299fdc508 : SUCCESS in 6m 20s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/20640f154b1a437698c9d3159b6ff3d1 : SUCCESS in 8m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d8cbf45c83e34300996d887d2add5378 : SUCCESS in 9m 08s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ab7fedfa9f3d4dfca4cfcc418ed991c5 : SUCCESS in 4m 52s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b80c416c28164f19ace50794c70fbd09 : SUCCESS in 1h 09m 40s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3cfdc50aeb6b4f8e9f771924c2a26847 : SUCCESS in 1h 25m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/cab54e87d36d4ede8c41a59febd52787 : SUCCESS in 1h 01m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/97cb8cc0a85e4d399dbf5543ea6818cc : FAILURE in 1h 09m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50552293cd584e209c442322944963be : SUCCESS in 46m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/81e8944c281e4ae9ba7913c4daa0a387 : SUCCESS in 1h 24m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0e74c673ca714cf4bd690213248f63ca : SUCCESS in 25m 01s","accounts_in_message":[],"_revision_number":4},{"id":"d4f07e99c07e3858c9f8c790bb0a52bdb70abe3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-05-07 16:51:49.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"741106cd060f44d3819e5888ef7605b18cf461f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 18:18:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/3487f3ca904f49138569911e635565bf : SUCCESS in 1h 02m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3a87f02e4fad48d48867cff7d548c291 : SUCCESS in 1h 17m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e8e82e5e2a74cd78213472ab00fc872 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d5b12d784d34fb2ba3a3813188e59f7 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5df5740f791d469eae8c922549b341f1 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6972e56c5d44692b2bc8fe549720b5f : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f6ca2905dde4c59993f55d47402f8ca : SUCCESS in 6m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d3ab2870233849de9a7ff963c5a0fbf4 : SUCCESS in 8m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/dc9d0b0d24b845acbc147c85273a6243 : SUCCESS in 9m 12s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e46fc4e688c143ad9a4648a52121212a : SUCCESS in 5m 11s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0b2b24ad482e4fd7aafe02e8efb1bea4 : SUCCESS in 1h 15m 10s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4fbb8ea4b0c7462998c44fb99cc89117 : SUCCESS in 1h 17m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6278cf00cbf64f9287818524e31bb193 : SUCCESS in 1h 18m 28s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7cc57b119e58495d91926cb06cabcd35 : FAILURE in 1h 15m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83cf1e61a3234e3c83c7d3b9cd786f92 : SUCCESS in 1h 03m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d4f98f0a89704dd5a18b08e6400edc80 : SUCCESS in 1h 13m 22s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6daac9a670c240689bef338487008fba : SUCCESS in 37m 12s","accounts_in_message":[],"_revision_number":5},{"id":"76c147a87fca3f8d90c91223672e07eba9e1534f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-05-10 12:56:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4a41d3b7d5707184c55db55ed7501dba85672fda","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-05-10 13:32:20.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"51a040e1353e657c4b8271a3c03eca6d7d2d1aaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-02 15:08:00.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"86bc583bc8795cc8e67e9167965af598a75da35c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-02 17:48:20.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"62fdfef5f455c300b8a919636d969e3c9c0b74b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 20:20:11.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/41654d1df67a4c2aaa37c16613a4f1b4 : SUCCESS in 59m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d4fb1d93497e44bd988a8a6550863b7c : SUCCESS in 1h 15m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4688bbce296d4fca8f7ac735a848e634 : SUCCESS in 7m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/20135b0cc2b3437d92643e01ce32cb0b : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3cc34a88cf2347bf96597a8fb5a595ac : SUCCESS in 7m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bcdc789aa11a42fea91d5679b5f650b6 : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d895daadb1d64bf394d9e5e597610c92 : SUCCESS in 7m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a3ab52592cba47ffa59299e10f7c4c1c : SUCCESS in 10m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fc1fc6fd7af34c3391d4d84d2498416c : SUCCESS in 10m 22s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/daa550dda5194412b66438a79c33fbec : SUCCESS in 6m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/09ab18de00e8403394147f9ec203b937 : SUCCESS in 1h 34m 42s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b908f4819bc64a06b01c3cd32e1a42b5 : SUCCESS in 1h 21m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6d3e18b3ec6240c2b63c8e3b5a5c7879 : SUCCESS in 1h 14m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/562f509e7698471d9b82e8533c765273 : SUCCESS in 1h 35m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cec55e1c49ed4f3da4304ffbd962d269 : SUCCESS in 57m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c8323ffbfb384ad385b4182b6e565166 : SUCCESS in 1h 27m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/139aa963c6ae484892cf078b8f9d8483 : SUCCESS in 33m 06s","accounts_in_message":[],"_revision_number":7},{"id":"b584d9b5266c1fb045d00f3b9c0ab01c379b0c5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-04 14:09:35.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"53e7e708c86a7aca27420968b4fdf76e088c58a2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-06-04 15:50:27.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"5426b8899e68798b895c9408c0c7654b3ba1b380","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 16:42:53.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/a25ce6da9a8148468726cae742a7f068 : SUCCESS in 54m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5c09e56fc4554fcbb0ab0e99c1fd50ce : SUCCESS in 1h 19m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27a58ec1e86b4c5e881252de897705ae : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3211fecc3d1b49eabbafda9db83826be : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a4c8fab14d64c5d984e14ff87385ee6 : SUCCESS in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d711e61ad3b5427dbbc94ea54afa6f1c : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b0c96259b3d4c80a8ed05b481503c97 : SUCCESS in 10m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b44adfab99d645b29c85155b86f4e3db : SUCCESS in 9m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8d51a82adb544c07a34df30450e003f8 : SUCCESS in 8m 54s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0f0b1af30abf48baac8ce88f1ca0bf18 : SUCCESS in 8m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/05de462801384f9b88a0a109e9dc6147 : SUCCESS in 1h 14m 55s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bf3bbd697c7f4e91b65d001dbfcea7e8 : SUCCESS in 1h 13m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/68d32b2dc33e4ab4aafeb265d4af7ea2 : SUCCESS in 1h 36m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/821c50715daf41f1b658895fd8e84743 : SUCCESS in 1h 19m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/925a2dd369b14798b5f23f22398c2443 : SUCCESS in 1h 02m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e87f2720eebc4ca2a747c72939a467de : SUCCESS in 1h 27m 10s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/092855e5521144779c6198ba05adc8e6 : SUCCESS in 31m 35s","accounts_in_message":[],"_revision_number":8},{"id":"528ff7d2e151f028a2fc019724083d9796c72c77","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-04 16:54:51.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"2e24d15cb18ff0978f4c29de5755d5fd59b416e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-04 17:08:07.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"c5a99d3cfabc55f8880d03f74b50ad751ca88db1","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-04 17:23:46.000000000","message":"Patch Set 9: Workflow-1\n\nStill need to address the staging quota on web-download and copy-image","accounts_in_message":[],"_revision_number":9},{"id":"296983af34693a3a70a4c540e197284cf7b1510c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-04 18:34:23.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"af19bfaf3796757d2af610d4ec591459493bbc92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-04 18:56:53.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"7e9c77b4d0ec9d5c222f0737e85035e88686960b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 20:17:22.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/d93be15f8f0f4002a705dfab66fffd76 : SUCCESS in 1h 11m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b520c63cbc54f16acb769e74c1a7357 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/164160c97c68463fb8a99b1a6dc5f80e : SUCCESS in 7m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85b926fe4e2a4359bf1e897dd88470aa : SUCCESS in 7m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2dfb5b21b20e4cc18eda5ccbd99b2fb2 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2347d0dfaa72470db3b66ca02b38d694 : SUCCESS in 7m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/a65f2a92dfc4415eb82a13103c9111d9 : SUCCESS in 8m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5ddc00c387884405bbee009c82816a0f : SUCCESS in 8m 55s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/036541cdf6174fe4aba7c2ed81fc4955 : SUCCESS in 5m 28s","accounts_in_message":[],"_revision_number":11},{"id":"13c78b8aad6a7bf4a48b9f29b94c81c5b1e02837","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-05 13:34:34.000000000","message":"Patch Set 11:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/788077/788077/11/check-not-vote/ext-glance-zuul/b136374 : SUCCESS in 36m 20s","accounts_in_message":[],"_revision_number":11},{"id":"d73c42c95c19a15a622da330b77d42b0577fd804","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-06-08 14:25:09.000000000","message":"Patch Set 11: Code-Review-1 Review-Priority+1\n\n(2 comments)\n\nCouple of comments inline, thank you!!","accounts_in_message":[],"_revision_number":11},{"id":"438ce23b3c946fce460a8e238736a0503808cdad","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-08 14:32:33.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"907593b01eba680f16e8167addd1d667b1ebb7b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-08 14:43:18.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"33fd09b524c8277692ce373a127967f77554e148","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-06-08 15:23:11.000000000","message":"Patch Set 12: Code-Review+2 Review-Priority+1\n\n(2 comments)\n\nLooks good, thank you Dan!!","accounts_in_message":[],"_revision_number":12},{"id":"6eeac779635e465a4b4d2814232230ccd68b8fe1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 17:18:55.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/aeee4ffaad69420bad4c3fb7efd7365a : SUCCESS in 1h 07m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bf146d26c177435598a3777a4d6e6ba3 : SUCCESS in 1h 18m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6fc94af85c3442798248ea442e194233 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67fca98d1df049a899a2f41640aede1e : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/be206d6c20e9464bbdaae87ac6d01f01 : SUCCESS in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbbadbb9b4c14f00bc40203120b4c4cd : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e5f30718dad41c0945de44796cdb33a : SUCCESS in 7m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5f433a0eb919408da66c5627aa5c2f43 : SUCCESS in 11m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/14875ab7b6bb4bafbe30ecc522f824fd : SUCCESS in 10m 55s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/70005f04996243929006450458d28c5b : SUCCESS in 6m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/80c3bdcc204748e184f8fe56345db75f : SUCCESS in 1h 20m 15s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9ba4689e57db46dcaf3f4f1cd6bc2796 : SUCCESS in 1h 19m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b440653bd6724e4db1b7971581171f91 : SUCCESS in 1h 21m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b214e0764307436289d9c5e5119adfb3 : SUCCESS in 1h 37m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f997f46badc428ab988c65b814a46cb : SUCCESS in 1h 01m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/463f826adf9a425fbe9cb1118d356b5c : SUCCESS in 1h 32m 39s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fe99b644b2d349e0b238fa5172ae3d8b : SUCCESS in 40m 37s","accounts_in_message":[],"_revision_number":12},{"id":"6f9d369d69abf824217ed8f24149acbe8ba83091","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-08 17:43:31.000000000","message":"Patch Set 12:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/788077/788077/12/check-not-vote/ext-glance-zuul/2f1a414 : FAILURE in 12m 13s","accounts_in_message":[],"_revision_number":12},{"id":"2378ce440b7c4a2c197b25582ff0b4b2ff6c74dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-14 14:40:58.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"f6da47edc7af577e373bd723de4076027bdd5f3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-14 14:47:30.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"67a2745d28aeb319d4b5358267280286e06fedd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-14 14:56:36.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"73ef9485b2818e31346975dfd52d2def8c94fcbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 16:39:10.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/d452d8d99f4d41c5866ded41296beb2f : FAILURE in 1h 30m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d348bf18ca3a40bbad92ca93674a761f : FAILURE in 1h 06m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0831bbea9ea14159928cc2dff8343e48 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7d61953bb0f3423881804c686521f66c : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0da6b1359554810b2155eb7f4dad6ab : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a277ef92f18d40aa8a16583550040cb5 : SUCCESS in 8m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b59a4d7b08ad4acf9e2d54f6d384af90 : SUCCESS in 19m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e8fd2dae48344e939d8573066c25ead4 : SUCCESS in 10m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5b2ef37c515a416ea05bcbe32d343bc1 : SUCCESS in 8m 57s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0c2e7b6122dc4880a9b8728dabe8d625 : SUCCESS in 6m 06s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0df97d961cce485293947006742aabe8 : SUCCESS in 1h 26m 17s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/17919e2ee88243ad812cc951287db227 : SUCCESS in 1h 14m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/52067bd6ff4d44cab425bc32b33f8121 : FAILURE in 1h 03m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/865c8e5584b14f2990bfb9b9bf0ed9fc : FAILURE in 1h 17m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d8578608a704ba4ab56942e9bce2621 : SUCCESS in 55m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/076f110c3887465bacb69c46c1abd45e : SUCCESS in 1h 25m 58s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/817d0b53bd954da59592a56e5d8f42b4 : SUCCESS in 39m 31s","accounts_in_message":[],"_revision_number":15},{"id":"721f6206fcefc7757c7424fe062e906e5030d809","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-06-16 13:52:52.000000000","message":"Patch Set 15:\n\nrecheck tempest-integrated-storage tempest.api.volume.admin.test_volume_quotas_negative.VolumeQuotasNegativeTestJSON - volume stuck in error_deleting during teardown","accounts_in_message":[],"_revision_number":15},{"id":"fe7f9885203bf44b548473e73d176b6371429a3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-16 16:03:23.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/e44262de8fa140f0b8d1e691cd51a0a0 : SUCCESS in 1h 14m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/09df78d16e394a698e8c1650b1dd34b2 : SUCCESS in 1h 20m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d7c4a7f802243b7afa7801db5af6183 : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/075380f0809e470a984a4aefe92a76c0 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f272a520b3fe438d8288c0f599bcb13c : SUCCESS in 7m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09d7bf8044f24bffbc5257e79e20b001 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12c03581feab4c45b2758cf065111ae4 : SUCCESS in 7m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f4ad2b87f42d48baa39b71b746aaf0a2 : SUCCESS in 13m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bae83b2669184fa4a351456d18eb52a0 : SUCCESS in 9m 30s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/103c63e79449486ab48c05356376d233 : SUCCESS in 5m 40s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/15ad9f01c8184ba1a7f26c023d95f510 : SUCCESS in 1h 23m 38s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/dade9a251a78477b99ccd3c4a22d3dcc : SUCCESS in 1h 17m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/be16da0093044044adbd5cd1cf58168d : SUCCESS in 1h 11m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0b371f60d69e44ad95b4ecc84602a76d : SUCCESS in 1h 12m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53f4dadd07924a3587825fdf90a2a459 : SUCCESS in 54m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/801e2b32c8ed4d7a97929707da316e09 : SUCCESS in 1h 31m 33s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/00460470d15e4cbca92ff8dfec38b81a : SUCCESS in 38m 45s","accounts_in_message":[],"_revision_number":15},{"id":"3d57e46728b448ee35d3fe4f5924bbcdff80062a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-21 15:46:41.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"a50b8affe44b99754149aef6ba748d2982705c35","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-06-21 16:13:41.000000000","message":"Patch Set 16: Code-Review+2 Review-Priority+1\n\nLooks better now!!\nJust to note for other reviewers, Image status cannot be reverted to uploading from importing due to some domain structure reservations, so it is reverted to queued state.","accounts_in_message":[],"_revision_number":16},{"id":"7897c15c39287efa72b15f6535fbaf7e4efcf0ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 17:12:39.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/7b61500955834bdb878fa122474a82f3 : SUCCESS in 1h 08m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c717777dd2cb4e81b1333e7cbb47f4ed : SUCCESS in 1h 07m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/377123f38d534a76a28f2a6e15aa0b1a : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02988df9a84f4ee4904ba65b978d2642 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0e821b9525c4a0991cadef936e13402 : SUCCESS in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe62c1ea6ee64a3189ef09fff12bb3a7 : SUCCESS in 7m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c9b56fdf50c493b80c973c5713412b8 : SUCCESS in 7m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4df32a50730b47a88dbb5f091a5c0de8 : SUCCESS in 9m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bba80309a40b4bc0bf7867cdadae85b4 : SUCCESS in 9m 45s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/93440204c773468c91d720ca63e838a6 : SUCCESS in 6m 43s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/24b16a2417794950919368bee6ad24ea : SUCCESS in 1h 15m 28s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e568a454cc8d48a2a25f3377b187015d : SUCCESS in 1h 12m 12s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fa893d081fb748efb8c50bfa4432a4a7 : SUCCESS in 1h 14m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/be4457e8d4684453a783067c407ad2b1 : SUCCESS in 1h 12m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a24f91273c2e454a9a505d9933be1c3d : SUCCESS in 51m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6ead9c469ccb4eb5840f4cd82a32a04b : SUCCESS in 1h 15m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9b8bf3ecef3f448fbb8af29ba960a442 : SUCCESS in 35m 49s","accounts_in_message":[],"_revision_number":16},{"id":"a02416424948450a7375af84fccdfd8e7d7e6cc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-21 17:57:14.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.","accounts_in_message":[],"_revision_number":17},{"id":"e704b82278b145a0a2c1f897c2deb6819f12456d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 19:26:58.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/8e9981a742a54bd39fa601099cf21d89 : SUCCESS in 1h 01m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9bb68b588a0b43f2a808cc11d9356ae2 : SUCCESS in 59m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6c9e188a11545559ff5e5c225d63fbd : SUCCESS in 9m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c6293bbb1e94a5e9782da19d4bef2dc : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b65d44e96d4f4ef79c2c811d3ab573a2 : SUCCESS in 11m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4307f1162aa64856a9626833bc9c3e1d : SUCCESS in 7m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e1d0fa68adf4c528d521285f9f22fbd : SUCCESS in 7m 36s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b21ec0df2ff345e49db874ef1d0db420 : SUCCESS in 9m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/08e75338232f4fd3a59f2fdaf449bd49 : SUCCESS in 9m 24s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a7e44d4d592844a4b4dfd8190061e18f : SUCCESS in 5m 54s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/83977b70420f48ff8729f80d280533b0 : SUCCESS in 1h 15m 13s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f7ffd821d32d46458569bc8078497733 : SUCCESS in 1h 09m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f0f515e2b99049e0aa626762156f0162 : FAILURE in 1h 21m 08s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/573ce5d01ea74d20935e0b646995e1c1 : SUCCESS in 1h 16m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1ad94e62a27d4d3bbf07d5354171824d : SUCCESS in 46m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0753065c7ffa4920b56e0bff34786881 : SUCCESS in 1h 25m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fc0123b0ad7e45c6955a62510223a0ba : SUCCESS in 32m 46s","accounts_in_message":[],"_revision_number":17},{"id":"16924c7fe7d7feb54f445ebea22dfa90f28bb64a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-06-22 05:16:45.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"0d2137bdaded254859c55eecd0c095b54a40a1f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 06:39:18.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/3106f4b802984de88d6efd3f4e847187 : FAILURE in 42m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/946732345598486d880b3444b0d655ba : SUCCESS in 1h 04m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bffc1ef3611948a9a02a23700283b418 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66fca9f682ba4acba4703d378c27809a : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cee5796365af491cba9f7dcb337d7d28 : SUCCESS in 6m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb731e7e78a74d238f368753fb87f567 : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cff057c24b34652bbf3f4e4ed4019aa : SUCCESS in 7m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1026d1d7df9e4f14bcf9fb6918b51835 : SUCCESS in 9m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8373b98726d74b049875eb3bb0c74747 : SUCCESS in 9m 02s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/87d9d5e60f124bc18f6ac7b1f2688761 : SUCCESS in 5m 28s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/08eae103d3cb46c6bbab656cb40de37c : SUCCESS in 1h 17m 39s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a06363e0a8d940718abd3e66f4bcba5b : SUCCESS in 1h 09m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/cf7b22d997694e3ea160909e52e33d62 : SUCCESS in 1h 05m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/852332809a1a4d56a097ad2562c2cc6f : SUCCESS in 1h 11m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f23906dae23e4cbba32f3103e3b7cf84 : SUCCESS in 46m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/feeede03fe3749f8a96e4fc62548083a : SUCCESS in 1h 08m 33s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fd676588c4f54524be3395f920c8efcb : SUCCESS in 27m 38s","accounts_in_message":[],"_revision_number":17},{"id":"ebda66f3dcdcce1b3639a471f81acdf5003ac52f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-06-22 13:44:30.000000000","message":"Patch Set 17:\n\nrecheck","accounts_in_message":[],"_revision_number":17},{"id":"f27dbc9f9da649f532956c92b9eafc9fb6ca48fc","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-22 15:04:50.000000000","message":"Patch Set 17:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/788077/788077/17/check-not-vote/ext-glance-zuul/963e10d : FAILURE in 1h 17m 20s","accounts_in_message":[],"_revision_number":17},{"id":"c6c23482bc54c7779f3a72ab943a5d5402c3112c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 15:59:17.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/8847c2c7cf564a7ea9d36168786afce5 : POST_FAILURE in 1h 01m 45s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/005a8611ed3848ab9790a3b803255bd4 : SUCCESS in 1h 03m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62bc1f6736b4430d9730f1de3c7cc43a : SUCCESS in 6m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/997abba1c653463d9212e0fe718d98ab : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19347c8b19b94fc3b567cd3341ae6b06 : SUCCESS in 8m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c431a4af10a4559b8d851a23bdd72e6 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efa10cef589940e687c2618b8b32462b : SUCCESS in 6m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f18b5bb15f3c491db4e1fc99d0353dee : SUCCESS in 9m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0adcec974733452ab50f01956c5dc08e : SUCCESS in 9m 01s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a3a9a759d08147b08e10ffe6a3b81b4d : SUCCESS in 5m 29s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/14d21be278504ac88dd51e6481ebba90 : SUCCESS in 1h 34m 12s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d6e446a9671d4b33b93ca8160dc20c54 : SUCCESS in 1h 05m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8cfe6dfdef1447eab688542e1216be1a : SUCCESS in 1h 06m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c33a2463959f46968b4314dbca80958d : SUCCESS in 1h 12m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ecacb6f6fd934d08830972322cb4dc06 : SUCCESS in 1h 00m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d1d1159319634ac580e40d770bc303c1 : SUCCESS in 1h 24m 30s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7a428b53ca1d4921a91f038f0a11852e : SUCCESS in 40m 10s","accounts_in_message":[],"_revision_number":17},{"id":"919e53a2106521f61d8c5b6162deb7d229ce95b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-29 14:51:33.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"70a6fdaa00028901ae37d39c2a73a029235cb557","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 16:51:48.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/d958293b591c41499d5f4e8d098876d1 : SUCCESS in 1h 03m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6cd3c60eb76e4e47af632f025a0f8e5e : SUCCESS in 1h 13m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ce8a4b459c74e5695a60357bfb0b111 : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fdff37a410684383a7c962acb6b326d9 : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f28e30e7bc3a4e26bfd5d8c3abe78229 : SUCCESS in 8m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/728a745ed6164f3bb2db01e279f2fe25 : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a90fef343bb741c2a8572b4bec440a22 : SUCCESS in 8m 15s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3e88667d429b4dda9c08e38a9b65f4a6 : SUCCESS in 11m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f4be5f0a0f92424bb3aeb6268d08c193 : SUCCESS in 10m 20s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8491e85669d14ede950fc56f1b8088bb : SUCCESS in 6m 47s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4f9d332468574210b791da02c24053c0 : SUCCESS in 1h 25m 22s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c2a93182dac84fb68d9d901d0fc5fc23 : SUCCESS in 1h 07m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/002197692e9e460fa18b02927b903e4e : SUCCESS in 1h 08m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f3be12e13bdf41668589adc3c930d6b7 : SUCCESS in 1h 09m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b45d428fd8e46e395562012de32cc8f : SUCCESS in 58m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2e0e214426fd4f4a933688d36c3cfdbc : SUCCESS in 1h 16m 25s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6df87100b99547a9a014170fca53c343 : SUCCESS in 30m 53s","accounts_in_message":[],"_revision_number":18},{"id":"dfa95fe1db817bf5795e2c3e44ba589a92ebc673","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-06-30 20:21:41.000000000","message":"Patch Set 18: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":18},{"id":"9143ecd912d4c0264c24a7da04d36e90307b4b1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 20:22:07.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"30fe95fb9646f803293ea55dca8728323e4bd185","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 02:08:33.000000000","message":"Patch Set 18: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":18},{"id":"793a419b4ead285c5cfa98533a01ad48eef99f04","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-01 02:58:14.000000000","message":"Patch Set 18:\n\nrecheck","accounts_in_message":[],"_revision_number":18},{"id":"444adec8a652e7cd26c45c7018b69300b0511791","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 04:30:15.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/de2ec3b6ce02499baaf45b0593a884ae : SUCCESS in 1h 08m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/450f2de3a331468584863c15d2bee8f4 : SUCCESS in 1h 13m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1a7b147a5a84970b9bfe59d7a9b8f0e : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f18739f94d344aa9f183d7117e5ffc2 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10334ff0f5b24a15b890d172613a53bc : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46f87836f1a24e19a8af9d2f804da0a8 : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0500565dde94062bb5a5f43d733cb1f : SUCCESS in 7m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/057543e6c34d4cba94b0a23c411aeb95 : SUCCESS in 8m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0e035c0053e44aeb83d1bfbe3ec78226 : SUCCESS in 9m 08s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/d0da4e27c11b45ddba5a0ab11db00753 : SUCCESS in 5m 17s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f5a2592fd8854591973f23eb580b7aae : SUCCESS in 1h 09m 54s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e27a833641054eb4b44fed93f48d2b2a : SUCCESS in 1h 23m 21s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/19939fdf70c746bbaefc5ad60ddecac5 : SUCCESS in 1h 06m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cffac4eab45445229a8e171cfffc9e75 : SUCCESS in 1h 21m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee349c2cf0904ccc9055d79de884a872 : SUCCESS in 59m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/960ca841fa8c48048ee0f33d8925fc52 : SUCCESS in 1h 27m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/298eac0bae5641f8b333180d6683f648 : SUCCESS in 36m 36s","accounts_in_message":[],"_revision_number":18},{"id":"02ec6149467a065070aae6c038cf8e57d8eec956","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 04:30:37.000000000","message":"Patch Set 18: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":18},{"id":"da3f5ba69ac37f8a65166f02d669e8a7791d8698","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 05:56:10.000000000","message":"Patch Set 18: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/5a0dc77561574b2ebf4af736cf0165e7 : SUCCESS in 59m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2fc8412536a84e2b92a968339da7dab9 : SUCCESS in 1h 02m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5cfae3c4d9948a286a57d2ce8891eff : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d98eb12e08443509b8ad6a6fbac33bc : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/beade9fe7df34028bc3b040a5a6d5565 : SUCCESS in 6m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54f09ed01c8844afa1ef0acc1e2ced38 : SUCCESS in 6m 49s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b98be6c1e3db4f17ab205e878cc0636b : SUCCESS in 10m 23s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5234644e513a456085aec7ef8713a737 : SUCCESS in 10m 56s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8c7881ca370d4cde8ea3998a1a19ca50 : SUCCESS in 57m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3f5cb3a4cda14e14bd6593cb2edb499b : SUCCESS in 1h 02m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/308e636d084748d7ace8f62c15f93972 : SUCCESS in 47m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ecc0fc32918844ed96ca540f9597f8ea : SUCCESS in 1h 24m 11s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0e6be2cf1616418d9455270719bf0a4d : SUCCESS in 33m 28s","accounts_in_message":[],"_revision_number":18},{"id":"2177aee83730802957f36e813c6088497878892d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 05:56:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":18},{"id":"19ec25fa45deec35dcc3ed442d9ce7a7007a01a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 05:57:26.000000000","message":"Patch Set 18:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bdd9cefa47f4fc98e1c1f9cd921d998 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":18}],"current_revision_number":18,"current_revision":"59990d513a084b811dff00fd6421b9ff92cb2ff0","revisions":{"6b92fcea930aa9d412cef06e5e059233d5676f4f":{"kind":"REWORK","_number":1,"created":"2021-04-26 20:30:31.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/77/788077/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/788077/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/788077/1"}}},"commit":{"parents":[{"commit":"a845f5a896d14db2f7bdab5d436941d0dcda7511","subject":"Add user_get_staging_usage() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a845f5a896d14db2f7bdab5d436941d0dcda7511"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 20:08:48.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 20:31:00.000000000","tz":-420},"subject":"Add image_stage_total quota enforcement","message":"Add image_stage_total quota enforcement\n\nThis makes us enforce a quota on the amount of data a user has in\nstaging.\n\nRelated to blueprint glance-unified-quotas\n\nChange-Id: I3cca4e589adc0aec138e5933c311aefd69ccee51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6b92fcea930aa9d412cef06e5e059233d5676f4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6b92fcea930aa9d412cef06e5e059233d5676f4f"}]},"branch":"refs/heads/master"},"5cf3d77aaf6f5a4e39f4b4ccbbb50dda182c18c8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-04-26 21:58:02.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/77/788077/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/788077/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/788077/2"}}},"commit":{"parents":[{"commit":"e8fdc59ed13e7bde777fc3f059b5762b403afe62","subject":"Add user_get_staging_usage() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e8fdc59ed13e7bde777fc3f059b5762b403afe62"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 20:08:48.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 21:59:19.000000000","tz":-420},"subject":"Add image_stage_total quota enforcement","message":"Add image_stage_total quota enforcement\n\nThis makes us enforce a quota on the amount of data a user has in\nstaging.\n\nRelated to blueprint glance-unified-quotas\n\nChange-Id: I3cca4e589adc0aec138e5933c311aefd69ccee51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5cf3d77aaf6f5a4e39f4b4ccbbb50dda182c18c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5cf3d77aaf6f5a4e39f4b4ccbbb50dda182c18c8"}]},"branch":"refs/heads/master"},"77a2c6a0b06b08b4d2cbbf1d37cdd9960e96cd21":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-04-27 13:15:09.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/77/788077/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/788077/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/788077/3"}}},"commit":{"parents":[{"commit":"f28269a9d6d7c8f03ceb2f4e10a598078c4459a8","subject":"Add user_get_staging_usage() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f28269a9d6d7c8f03ceb2f4e10a598078c4459a8"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 20:08:48.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 13:16:26.000000000","tz":-420},"subject":"Add image_stage_total quota enforcement","message":"Add image_stage_total quota enforcement\n\nThis makes us enforce a quota on the amount of data a user has in\nstaging.\n\nRelated to blueprint glance-unified-quotas\n\nChange-Id: I3cca4e589adc0aec138e5933c311aefd69ccee51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/77a2c6a0b06b08b4d2cbbf1d37cdd9960e96cd21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/77a2c6a0b06b08b4d2cbbf1d37cdd9960e96cd21"}]},"branch":"refs/heads/master"},"1d82edf227ee0e7092f093debac0d89f457bc333":{"kind":"REWORK","_number":4,"created":"2021-04-27 16:58:02.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/77/788077/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/788077/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/788077/4"}}},"commit":{"parents":[{"commit":"aa7f1827eeb71c5342c83e716f1d7201256bb572","subject":"Add user_get_staging_usage() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aa7f1827eeb71c5342c83e716f1d7201256bb572"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 20:08:48.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 15:32:19.000000000","tz":-420},"subject":"Add image_stage_total quota enforcement","message":"Add image_stage_total quota enforcement\n\nThis makes us enforce a quota on the amount of data a user has in\nstaging.\n\nRelated to blueprint glance-unified-quotas\n\nChange-Id: I3cca4e589adc0aec138e5933c311aefd69ccee51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1d82edf227ee0e7092f093debac0d89f457bc333"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1d82edf227ee0e7092f093debac0d89f457bc333"}]},"branch":"refs/heads/master"},"bb42447f453d738ef21b07869da0c0bbdd0222de":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-05-07 16:51:49.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/77/788077/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/788077/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/788077/5"}}},"commit":{"parents":[{"commit":"b76ae0447d8bcb1fcf852674aa19bfc2af52b046","subject":"Add user_get_staging_usage() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b76ae0447d8bcb1fcf852674aa19bfc2af52b046"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 20:08:48.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-05-07 16:53:09.000000000","tz":-420},"subject":"Add image_stage_total quota enforcement","message":"Add image_stage_total quota enforcement\n\nThis makes us enforce a quota on the amount of data a user has in\nstaging.\n\nRelated to blueprint glance-unified-quotas\n\nChange-Id: I3cca4e589adc0aec138e5933c311aefd69ccee51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bb42447f453d738ef21b07869da0c0bbdd0222de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bb42447f453d738ef21b07869da0c0bbdd0222de"}]},"branch":"refs/heads/master"},"b9a2a24947642c534145f23d9203db964dddbd8f":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-06-02 15:08:00.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/77/788077/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/788077/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/788077/6"}}},"commit":{"parents":[{"commit":"bc7245bbf4d7313646c03adedea9477cd8735ddf","subject":"Add user_get_staging_usage() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc7245bbf4d7313646c03adedea9477cd8735ddf"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 20:08:48.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-02 15:09:50.000000000","tz":-420},"subject":"Add image_stage_total quota enforcement","message":"Add image_stage_total quota enforcement\n\nThis makes us enforce a quota on the amount of data a user has in\nstaging.\n\nRelated to blueprint glance-unified-quotas\n\nChange-Id: I3cca4e589adc0aec138e5933c311aefd69ccee51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b9a2a24947642c534145f23d9203db964dddbd8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b9a2a24947642c534145f23d9203db964dddbd8f"}]},"branch":"refs/heads/master"},"8b3492b350ca67a0450b20141c0ab03a9794651a":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-06-02 17:48:20.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/77/788077/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/788077/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/788077/7"}}},"commit":{"parents":[{"commit":"990abca793f737a6e01b907b1dcd7498f25d00b5","subject":"Add user_get_staging_usage() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/990abca793f737a6e01b907b1dcd7498f25d00b5"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 20:08:48.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-02 17:50:03.000000000","tz":-420},"subject":"Add image_stage_total quota enforcement","message":"Add image_stage_total quota enforcement\n\nThis makes us enforce a quota on the amount of data a user has in\nstaging.\n\nRelated to blueprint glance-unified-quotas\n\nChange-Id: I3cca4e589adc0aec138e5933c311aefd69ccee51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8b3492b350ca67a0450b20141c0ab03a9794651a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8b3492b350ca67a0450b20141c0ab03a9794651a"}]},"branch":"refs/heads/master"},"7e069a48322328558ca8eb31fea739ce9d02d3a3":{"kind":"REWORK","_number":8,"created":"2021-06-04 14:09:35.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/77/788077/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/788077/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/788077/8"}}},"commit":{"parents":[{"commit":"b142e23621a93647894782de2b7b89c39579c803","subject":"Add user_get_staging_usage() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b142e23621a93647894782de2b7b89c39579c803"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 20:08:48.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-04 14:11:14.000000000","tz":-420},"subject":"Add image_stage_total quota enforcement","message":"Add image_stage_total quota enforcement\n\nThis makes us enforce a quota on the amount of data a user has in\nstaging.\n\nRelated to blueprint glance-unified-quotas\n\nChange-Id: I3cca4e589adc0aec138e5933c311aefd69ccee51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7e069a48322328558ca8eb31fea739ce9d02d3a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7e069a48322328558ca8eb31fea739ce9d02d3a3"}]},"branch":"refs/heads/master"},"77701374198aa448357975f605c6bf7e5054291c":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":9,"created":"2021-06-04 17:08:07.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/77/788077/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/788077/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/788077/9"}}},"commit":{"parents":[{"commit":"4232b50a95ab84ec064735baa45e4a5530628d90","subject":"Add user_get_staging_usage() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4232b50a95ab84ec064735baa45e4a5530628d90"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 20:08:48.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-04 16:57:39.000000000","tz":-420},"subject":"Add image_stage_total quota enforcement","message":"Add image_stage_total quota enforcement\n\nThis makes us enforce a quota on the amount of data a user has in\nstaging.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I3cca4e589adc0aec138e5933c311aefd69ccee51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/77701374198aa448357975f605c6bf7e5054291c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/77701374198aa448357975f605c6bf7e5054291c"}]},"branch":"refs/heads/master"},"074cb45264327f73e530cb56091f82136b751d01":{"kind":"REWORK","_number":10,"created":"2021-06-04 18:34:23.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/77/788077/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/788077/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/788077/10"}}},"commit":{"parents":[{"commit":"5f175fc79f3a4a8283dac1b94e73159683ab6b6a","subject":"Add user_get_staging_usage() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5f175fc79f3a4a8283dac1b94e73159683ab6b6a"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 20:08:48.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-04 18:23:03.000000000","tz":-420},"subject":"Add image_stage_total quota enforcement","message":"Add image_stage_total quota enforcement\n\nThis makes us enforce a quota on the amount of data a user has in\nstaging.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I3cca4e589adc0aec138e5933c311aefd69ccee51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/074cb45264327f73e530cb56091f82136b751d01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/074cb45264327f73e530cb56091f82136b751d01"}]},"branch":"refs/heads/master"},"f1c216c0281bdf9826b08e95e5f5acea3a006f44":{"kind":"REWORK","_number":11,"created":"2021-06-04 18:56:53.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/77/788077/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/788077/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/788077/11"}}},"commit":{"parents":[{"commit":"5f175fc79f3a4a8283dac1b94e73159683ab6b6a","subject":"Add user_get_staging_usage() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5f175fc79f3a4a8283dac1b94e73159683ab6b6a"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 20:08:48.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-04 18:58:06.000000000","tz":-420},"subject":"Add image_stage_total quota enforcement","message":"Add image_stage_total quota enforcement\n\nThis makes us enforce a quota on the amount of data a user has in\nstaging.\n\nNote this includes another fix to a unit test that doesn\u0027t properly\npass context to the get_flow() method.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I3cca4e589adc0aec138e5933c311aefd69ccee51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f1c216c0281bdf9826b08e95e5f5acea3a006f44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f1c216c0281bdf9826b08e95e5f5acea3a006f44"}]},"branch":"refs/heads/master"},"77369c706b9c8d90c1fd798aced88c6c13ed400b":{"kind":"REWORK","_number":12,"created":"2021-06-08 14:43:18.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/77/788077/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/788077/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/788077/12"}}},"commit":{"parents":[{"commit":"5f175fc79f3a4a8283dac1b94e73159683ab6b6a","subject":"Add user_get_staging_usage() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5f175fc79f3a4a8283dac1b94e73159683ab6b6a"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 20:08:48.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-08 14:44:56.000000000","tz":-420},"subject":"Add image_stage_total quota enforcement","message":"Add image_stage_total quota enforcement\n\nThis makes us enforce a quota on the amount of data a user has in\nstaging.\n\nNote this includes another fix to a unit test that doesn\u0027t properly\npass context to the get_flow() method.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I3cca4e589adc0aec138e5933c311aefd69ccee51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/77369c706b9c8d90c1fd798aced88c6c13ed400b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/77369c706b9c8d90c1fd798aced88c6c13ed400b"}]},"branch":"refs/heads/master"},"5ae4d3955dde75333001572751d55f4da6a3b5e9":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2021-06-14 14:40:58.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/77/788077/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/788077/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/788077/13"}}},"commit":{"parents":[{"commit":"31baf3d44e8d0f2676f8be968667ee0afb7b0bbd","subject":"Add user_get_staging_usage() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/31baf3d44e8d0f2676f8be968667ee0afb7b0bbd"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 20:08:48.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-14 14:42:51.000000000","tz":-420},"subject":"Add image_stage_total quota enforcement","message":"Add image_stage_total quota enforcement\n\nThis makes us enforce a quota on the amount of data a user has in\nstaging.\n\nNote this includes another fix to a unit test that doesn\u0027t properly\npass context to the get_flow() method.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I3cca4e589adc0aec138e5933c311aefd69ccee51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5ae4d3955dde75333001572751d55f4da6a3b5e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5ae4d3955dde75333001572751d55f4da6a3b5e9"}]},"branch":"refs/heads/master"},"2f99ce632d31e174e7bed21ee5ba8c90e9868136":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2021-06-14 14:47:30.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/77/788077/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/788077/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/788077/14"}}},"commit":{"parents":[{"commit":"1fcf616270a7407154c9127f7bfdfdfdc7d90091","subject":"Add user_get_staging_usage() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1fcf616270a7407154c9127f7bfdfdfdc7d90091"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 20:08:48.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-14 14:49:32.000000000","tz":-420},"subject":"Add image_stage_total quota enforcement","message":"Add image_stage_total quota enforcement\n\nThis makes us enforce a quota on the amount of data a user has in\nstaging.\n\nNote this includes another fix to a unit test that doesn\u0027t properly\npass context to the get_flow() method.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I3cca4e589adc0aec138e5933c311aefd69ccee51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f99ce632d31e174e7bed21ee5ba8c90e9868136"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f99ce632d31e174e7bed21ee5ba8c90e9868136"}]},"branch":"refs/heads/master"},"e64db4515d0cbe1fe9c790a0d306d351bff91fda":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2021-06-14 14:56:36.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/77/788077/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/788077/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/788077/15"}}},"commit":{"parents":[{"commit":"36dd16f7f55d2a4e4b43e38286290c5ef5b25480","subject":"Add user_get_staging_usage() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/36dd16f7f55d2a4e4b43e38286290c5ef5b25480"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 20:08:48.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-14 14:58:35.000000000","tz":-420},"subject":"Add image_stage_total quota enforcement","message":"Add image_stage_total quota enforcement\n\nThis makes us enforce a quota on the amount of data a user has in\nstaging.\n\nNote this includes another fix to a unit test that doesn\u0027t properly\npass context to the get_flow() method.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I3cca4e589adc0aec138e5933c311aefd69ccee51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e64db4515d0cbe1fe9c790a0d306d351bff91fda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e64db4515d0cbe1fe9c790a0d306d351bff91fda"}]},"branch":"refs/heads/master"},"622661f0063eef4940cce0a48ae974581ebb0817":{"kind":"REWORK","_number":16,"created":"2021-06-21 15:46:41.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/77/788077/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/788077/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/788077/16"}}},"commit":{"parents":[{"commit":"9e7c2a1b7ead7eba93f3f573865945d497516bab","subject":"Add user_get_staging_usage() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9e7c2a1b7ead7eba93f3f573865945d497516bab"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 20:08:48.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-21 15:47:37.000000000","tz":-420},"subject":"Add image_stage_total quota enforcement","message":"Add image_stage_total quota enforcement\n\nThis makes us enforce a quota on the amount of data a user has in\nstaging.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I3cca4e589adc0aec138e5933c311aefd69ccee51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/622661f0063eef4940cce0a48ae974581ebb0817"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/622661f0063eef4940cce0a48ae974581ebb0817"}]},"branch":"refs/heads/master"},"3125c32ec21ae34073adba775702fac7dfb94e60":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2021-06-21 17:57:14.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/77/788077/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/788077/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/788077/17"}}},"commit":{"parents":[{"commit":"67c3db4e46ec6900e1796226870c1850c852a2e0","subject":"Add user_get_staging_usage() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/67c3db4e46ec6900e1796226870c1850c852a2e0"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 20:08:48.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-21 17:59:19.000000000","tz":-420},"subject":"Add image_stage_total quota enforcement","message":"Add image_stage_total quota enforcement\n\nThis makes us enforce a quota on the amount of data a user has in\nstaging.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I3cca4e589adc0aec138e5933c311aefd69ccee51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3125c32ec21ae34073adba775702fac7dfb94e60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3125c32ec21ae34073adba775702fac7dfb94e60"}]},"branch":"refs/heads/master"},"59990d513a084b811dff00fd6421b9ff92cb2ff0":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2021-06-29 14:51:33.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/77/788077/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/788077/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/788077/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/788077/18"}}},"commit":{"parents":[{"commit":"5261fab3ea9500e895f2ac94c636a9d685fecf8c","subject":"Add user_get_staging_usage() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5261fab3ea9500e895f2ac94c636a9d685fecf8c"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-26 20:08:48.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-29 15:53:18.000000000","tz":-420},"subject":"Add image_stage_total quota enforcement","message":"Add image_stage_total quota enforcement\n\nThis makes us enforce a quota on the amount of data a user has in\nstaging.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I3cca4e589adc0aec138e5933c311aefd69ccee51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/59990d513a084b811dff00fd6421b9ff92cb2ff0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/59990d513a084b811dff00fd6421b9ff92cb2ff0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
