)]}'
{"id":"openstack%2Fglance~788326","triplet_id":"openstack%2Fglance~master~I4d6e071b8d48862032f36c80ee19632a3fe85009","project":"openstack/glance","branch":"master","topic":"bp/glance-unified-quotas","hashtags":[],"change_id":"I4d6e071b8d48862032f36c80ee19632a3fe85009","subject":"Add user_get_image_count() to DB API","status":"MERGED","created":"2021-04-27 16:58:02.000000000","updated":"2021-07-01 06:00:32.000000000","submitted":"2021-07-01 05:58:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"788326-bp/glance-unified-quotas","meta_rev_id":"45ea96bc915eabec32106fc141a6837299c283d1","_number":788326,"virtual_id_number":788326,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-01 05:58:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-07-01 06:00:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-06-30 21:30:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2021-06-30 21:30:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-07-01 06:00:32.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-27 18:27:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-04 15:52:01.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-06-30 21:30:13.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"2d77e1bf13d1a5d6ed93e7282af35bebdac75706","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-27 16:58:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"915982bcda41e0c6a77b3b5039447b1a7030b9ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 18:27:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/5f555a2de3f64061b283f89cfcef8a7d : SUCCESS in 45m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e217a5adfbff45898613b4c86461c27e : SUCCESS in 50m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/add85a119366444a91654d4d933fc9bb : SUCCESS in 8m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad9d0f368d504089a24109fe0c89c432 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ffb47c195cc4ccaac799ce4966fca02 : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f75a8183ee5e4ad1a5279ae69b8f9c63 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44d8b6a4aff147b2a7830922a88ea229 : SUCCESS in 6m 27s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/86c3deacfe2b4049949989153f0746c6 : SUCCESS in 8m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/33a94f8b12de47c9bc34c9e774856639 : SUCCESS in 7m 45s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6f90e2d98ceb49aea0d493b5e015f174 : SUCCESS in 7m 57s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8f0cafb4a39c40838b3bbc20001f8c89 : SUCCESS in 1h 16m 26s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2e63efc6e60547668776c81b61e7260e : FAILURE in 58m 18s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/683738d8e43449ebb402a68ac6105511 : SUCCESS in 1h 19m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/05f8d90d78c541278a8d2c8d5559b053 : SUCCESS in 1h 09m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0aa02b8daf0c45b0b6065d5426c84f37 : SUCCESS in 55m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8dd5cd54c9a44be4a406fa84e52b4a80 : SUCCESS in 1h 08m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/63ab173cfcd94badbdb0d2cf836b64e2 : SUCCESS in 29m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d68d9d21f3f0409d0c040e27435d44f85c8d79e9","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-27 19:11:35.000000000","message":"Patch Set 1:\n\nrecheck cinder dance","accounts_in_message":[],"_revision_number":1},{"id":"5d962c43496a2bfbefca1be2277fbce2ee51d7a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 20:31:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/1fb459bb55154137a90d6efa94520633 : SUCCESS in 1h 14m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7bbc0df52a38451fb16612fc0c2c39eb : SUCCESS in 1h 00m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87894ab69b0c40a093f7357ddd790e8b : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/312fb94bc04349e2af30be292389a826 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab8bf7b4b40c4f818f67b19522c1ce02 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8450bd4cadf34a2f93082d29bd940db6 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0fc2f2fb39c4661acff4c6ed876f940 : SUCCESS in 6m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/81fb723242a540bcb9ff6df70b315ae0 : SUCCESS in 9m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d0d6ec3e224a4275958aefed0a2b4f4f : SUCCESS in 8m 53s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/9e71fbe234664c15ab2c7b3141af4952 : SUCCESS in 7m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1e0e5a49554a48d3a1d0a2d0d094bf0c : SUCCESS in 1h 13m 01s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/88e1605268f54c109c0ffbe37154eeed : SUCCESS in 54m 38s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/74a52fa29f2b4d18a8c57063b65a2185 : SUCCESS in 1h 01m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/634043c11ac8479687165e7e801c437a : SUCCESS in 56m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8e1c9d521914371adc0e9c4afec31d8 : SUCCESS in 1h 00m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9bbe6b522ece4e71865b6cf03334cbe3 : SUCCESS in 1h 11m 16s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2ebaf7476ff1477680940caaec4893f0 : SUCCESS in 20m 37s","accounts_in_message":[],"_revision_number":1},{"id":"e459c3940916f19a7f0afbdc8c0fe16090cbd3ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-05-07 16:51:49.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"88966e8b9d0249d79007f2047d91bb45e13803d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 18:24:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/03646e15c6a34d95928c0a7bac242c2a : SUCCESS in 1h 08m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f36b0f78ecae44da9b23ae0ba87a10a0 : SUCCESS in 1h 08m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d475b2569d9a4cd89144a5eb3c291b36 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d71d920ad33d4c758ab86c87cbd4dca5 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75de3bafeec643819d7ff5c0ea538630 : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/441e5a5560ea4ade84b01d1def46c82a : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83d5a47bf7984230aae5ae940fe6cdc9 : SUCCESS in 6m 55s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9387aaa42ad44f2ca61d1923c27723fe : SUCCESS in 8m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8733656e16d74267916b75cb1c5174be : SUCCESS in 9m 50s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ee01c0c70347400d9ca9972fbd53db04 : SUCCESS in 5m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e7c20651485e4008b2d7e0896278bc70 : SUCCESS in 1h 12m 39s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/322493e82407459da1c740b8d6655e45 : SUCCESS in 1h 23m 50s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3817650a961a43f69169519833808e77 : SUCCESS in 1h 07m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f705a14419a34bbeb02b931f5063a5f5 : FAILURE in 1h 21m 35s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02be3b0a176447e7a083ad792dde8c09 : SUCCESS in 57m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ba5273603b074477ac0c86d960b5160c : SUCCESS in 1h 19m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fb280593e2d04e8f8f2ed6c42984e3c3 : SUCCESS in 35m 38s","accounts_in_message":[],"_revision_number":2},{"id":"21da1e1044a15f521a9a5b9d2150a055d3cdcf63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-02 15:08:00.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"d335ade2cda7f9a2230beb053d947b8486e337f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-02 17:48:20.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"50098d2945eb5b638857fc4bfc98671b3d4674d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 20:27:07.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/e7dbb44428524ec598c6ce7be122165b : SUCCESS in 1h 07m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3dd8736805fe46069145f936b235ac9d : SUCCESS in 1h 29m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88f3a6c4067e4029bdd6f06f90bd0a49 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4698a2059b3742ff9ed0693a23d06013 : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2dbc690ecde94f5aa15507da34c2947a : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d2db3b24e1b400cb58a4b142f912962 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/046116df9f764a0cbeb05ebf1f3c1aff : SUCCESS in 8m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8de4bd742e934a5f8232be11a1adaf87 : SUCCESS in 8m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/79f1856e61814973bc168c617186ec82 : SUCCESS in 8m 33s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/20811c7b9d9a46dd86b3d2d5007e3431 : FAILURE in 10m 34s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2d24b328a5bc446ab95065a403d24668 : SUCCESS in 1h 35m 18s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cddb6698201045b9881c1dc80b57deb8 : SUCCESS in 1h 21m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/16f39f1218b740feb1669d3058925c44 : SUCCESS in 1h 13m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/46bd9d7f7eef46e580696d034cae232d : SUCCESS in 1h 19m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a9c3f4ecf4c43c5928aef8ff4d637da : SUCCESS in 1h 04m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a64c8c8c5cdd4f649db0b54bb15b19cf : SUCCESS in 1h 17m 44s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/52287df65d5c49e18a04ac90c61215ed : SUCCESS in 28m 12s","accounts_in_message":[],"_revision_number":4},{"id":"8e0942fe1e1fbde5af40833354517542e2b2631f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-02 20:43:40.000000000","message":"Patch Set 4:\n\nrecheck infra issue resolved","accounts_in_message":[],"_revision_number":4},{"id":"8bac5ed5bd78cb6a76d392ef7027e65d2fc2846d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 22:16:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/81fe22d008274fcbac86cfc4e7275e20 : SUCCESS in 49m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f2cc78fd07824ef58d70e4bead91c288 : SUCCESS in 1h 07m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/681e1632f0ec4c3ca59cf381b83ed989 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6cfe2358f226407da3b93df61cbeef0b : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e2c256663ef47c18ce68fef7d20c5a2 : SUCCESS in 6m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/acb983d617fa483aa12c205927c43f2b : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e675163428184289ad9f0a275e90e3a1 : SUCCESS in 7m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ad5d88c1770e455eb91bc95f57ad7427 : SUCCESS in 8m 39s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/42ac2f159e0f4a6f98e7cffcb3a7b832 : SUCCESS in 9m 06s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/19cc2240e41245269a82ec9c07bdcb59 : SUCCESS in 5m 14s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/bafedd31af2447ac8fcf0cc308d731c5 : SUCCESS in 1h 21m 43s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f32a381c5ba6457bbc8393c3f11b466c : SUCCESS in 1h 21m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b84583cc01de4aaca9a9a52ea4d844ae : SUCCESS in 1h 25m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8d91e8d0b81a4773909dba62de628a6a : SUCCESS in 1h 17m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca5312eccd4f41b994456f616d0eeb14 : SUCCESS in 59m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fb9e49031f0f4682bef93ca213fe3f56 : SUCCESS in 1h 26m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a4e2f5bea3fe46298ea1273caa9946aa : SUCCESS in 32m 41s","accounts_in_message":[],"_revision_number":4},{"id":"adb5306e51f671d1d2fb03f6690f34b6dd9dc53f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-04 14:09:35.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"7b4cd3e00265eb9248a213fcda13f44cd700b1ae","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-06-04 15:52:01.000000000","message":"Patch Set 5: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"45226c07766dc3ff0b13f53cbbe660c02b2926f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 17:02:02.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/688205157d2c482ba3ea000f0b0448fb : SUCCESS in 1h 09m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/91ea7f92e0194e47819f5ae4e7cf89b6 : SUCCESS in 1h 39m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db963d992e5d492b97cc8265197f2df4 : SUCCESS in 9m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bce41c0316ab4825837f83dd92bca8dd : SUCCESS in 10m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea1bd1debbb74712b75b5afa21f124d7 : SUCCESS in 10m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/783e3c85e72e41bca09335344f513fca : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/771254b87161423b8462ea360b1eedce : SUCCESS in 8m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/213a5a38c0004a96abc2f1b9c1cbe334 : SUCCESS in 8m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bdaaadbd985c4a0aac14960d7599df0f : FAILURE in 11m 42s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/550fb0b179424604b509a0496352c0f0 : SUCCESS in 5m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f1ec609337264a49b805ff10bd76834a : SUCCESS in 1h 31m 23s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3054dc9b34004fd891cba39858a544f2 : SUCCESS in 1h 26m 35s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6d97d9e357a844deaa95bad05962a639 : SUCCESS in 1h 17m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/631c0e385b8d4dea8d27ec0ff80107ed : SUCCESS in 1h 17m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e46db3f700724b03a89a8aea88b64a0f : SUCCESS in 1h 20m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4e4def89a0b04eceb0e29ebde8f7effc : SUCCESS in 1h 54m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/89390a7a171643c2a49ad40c2a9204ea : SUCCESS in 33m 02s","accounts_in_message":[],"_revision_number":5},{"id":"4cfc97efdc3c425c10e325cb4e6d40b6a5a9f378","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-04 17:08:07.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"526ab7b4bdb36261cedf9b03df870344c6f794ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-04 18:34:23.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"3de484389ff61f8947b962b9c75005e22bd73c95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-04 18:56:53.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"b1edae9d34c1111f81359305aef96816aec4d0e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 20:20:17.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/4ce7de8ffccc4f83bd7729e917b7f3fa : SUCCESS in 1h 19m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b70fb4295da488da195d90ffa139bb0 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/542a1febef8442b487f529a2b3f61c35 : SUCCESS in 8m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aaadcc0e025548c3a819b37d17a619dc : SUCCESS in 8m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df4d6bd80f034e2eb5eba7c8eb71fbfd : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5adb21887d6a422eaff6e12967c0b47b : SUCCESS in 6m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fa391cf542b94b6a980a7f05625fc48a : SUCCESS in 8m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4d55e7a180eb49c09882e660aa6aaddc : SUCCESS in 9m 07s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0965b1b14dba47d1b6888e1408e31cb2 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":8},{"id":"9e0c5384432aa9a8d2480a7c1648e2ad2a6479d2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-06-08 14:26:05.000000000","message":"Patch Set 8: Code-Review+2 Review-Priority+1\n\nLooks good, thank you Dan!!","accounts_in_message":[],"_revision_number":8},{"id":"51075d011f3891ade4a2719a57c6e40e2d96e3a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-08 14:43:18.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"b00bf3d9f1a4b0c27253bc7efaa36463a9dde7ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 17:10:24.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/04d883b1b1224a6a9e551b5cc8decffc : SUCCESS in 1h 12m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0108f40bba264b8caa5384445790ce46 : SUCCESS in 1h 17m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84b0e1a7b66a454195933080cb9480de : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23af11f1ef9b43eb9d8fb7478229f6cc : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e692d12739c54929a12d67b153ba9c30 : SUCCESS in 8m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ddaa5ac92066426f85d57eaf5597f2b3 : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8ab8fa48db04dbe87db296f7e240c60 : SUCCESS in 7m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7543d4775f6247b4b41147ba809c41d6 : SUCCESS in 10m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d418a0a1010a4a77b6a1ea993280978c : SUCCESS in 9m 24s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/be1123d18f73461597e082f140bec330 : SUCCESS in 6m 11s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/396cc988ad7c4f63b89b85c951f477f6 : SUCCESS in 1h 27m 04s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/731a0182bb4b4f2092616c0e54886e0d : SUCCESS in 1h 16m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/dc5b4dd22d0b4f4185cc9cc79c62ff20 : SUCCESS in 1h 11m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fee7fa05611e423a97eaa694dd4692a1 : SUCCESS in 1h 30m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8235d13b47b74ed2b37fe548a92bad98 : SUCCESS in 58m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cf6d643faa3e429283462d4eb84ea5ae : SUCCESS in 1h 19m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c7de0b1c36ef465e84199f47762e647a : SUCCESS in 29m 40s","accounts_in_message":[],"_revision_number":9},{"id":"50b2c97b85e88cc10366d472e2a70c2c1da4c9e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-14 14:40:58.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"d93d06f1be554d42c9c70649a528bae00a67ec25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-14 14:47:30.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"0acd41b3650907af933cbb327ff472f3e218a75d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-14 14:56:36.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"da0fc8bc6b297a597145f0b8750da2f8eb513a91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 16:32:36.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/66919f45010d497792b945ad22e74a80 : SUCCESS in 1h 02m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c78a6b0a09a34efab297b37533454b1a : SUCCESS in 1h 23m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16bccd9607df4043a7d69526c46d5741 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/714c07a21cdf40008e6f483ac3150c42 : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91fb976a470241a190d7db64d9a10bdf : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fed48ac40d3c4ed4a7e9d45965fe9b05 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19131bb89e4743a8a25d9c4fadf48914 : SUCCESS in 7m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/78121a5483a54758aa37a3ad0e5b8e48 : SUCCESS in 10m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cd53f4cfdc09470381561f79f186a2af : SUCCESS in 8m 56s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0244efd69af0403a809ca254877551d6 : SUCCESS in 5m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1d58330381234b3e9dcb45a7ab8c4847 : SUCCESS in 1h 15m 09s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6c2d9932dbeb4f4499ecbe88e46b8cb5 : SUCCESS in 1h 22m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3d06defec8eb49a4aa19f5bf01d1be5f : SUCCESS in 1h 04m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/38e62f55da2d43608a0be4c7f8de32b5 : SUCCESS in 1h 22m 19s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b3bd3be6667492188e9c378b0832cfe : SUCCESS in 55m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f67df02b6a1545b1bdb78aa8dc426ead : SUCCESS in 1h 22m 10s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7b8c183386d643fdb353166e49a79218 : SUCCESS in 36m 58s","accounts_in_message":[],"_revision_number":12},{"id":"d2fe6a05411190dac5e50f32d67aca50546f8beb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-21 15:46:41.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"54723f29474de919f8760fce7541960823d4c4a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 17:38:13.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/4e4bf770c3b046398f56f2b9ae1861fa : SUCCESS in 1h 11m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b86d142fdaa948159cc40e5737437923 : SUCCESS in 1h 36m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba6bc019f6f64751af57903ee882a662 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1addb6f06319463088f35963a3c480d4 : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3410d2a94f254acfaa5927c703cc2b01 : SUCCESS in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f698dae3b6bd4624a0a5d30b8679dc91 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7a77a3b0d16401480c51f1f11d45fd1 : SUCCESS in 8m 35s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/2e2af182eb8e4c50b09b20ae3ca57588 : SUCCESS in 10m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1239ebdd83994a96a1d576c5268e9288 : SUCCESS in 9m 51s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/be052dd98a9b48888a4a9e983b38f1ab : SUCCESS in 5m 55s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/361b59ea057240ff9fb096cd1b774e10 : SUCCESS in 1h 14m 20s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4e9a26b6560645a094439315d86c5fa3 : SUCCESS in 1h 16m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/41a53cecedab423f9fe19b28f8fe1ac9 : SUCCESS in 1h 38m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/93ed438308464b0eb69c56d3bf07f452 : FAILURE in 1h 14m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96290380d4ac4a11b8a669b09bd226f9 : SUCCESS in 56m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/38ff30df50fb40f196e8cf26855a718e : SUCCESS in 1h 17m 08s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3b96cd48a62d44fe815a88fa92ccf3b4 : SUCCESS in 38m 58s","accounts_in_message":[],"_revision_number":13},{"id":"a13da8f3a8f76ae362ebfd1e0e5c18408dd842dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-21 17:57:14.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"a11c9ce3c95f29f11bbfd93ac63d369f289b8fe9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 19:31:34.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/7d874b042cff4951aad92d85abce1ed2 : SUCCESS in 59m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d5db206c294f467c973d4fa4057435f5 : SUCCESS in 1h 02m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fd62af9f3d24f5fa29e15d8250258e9 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c10a9f59867417ca7cbe64e297f3374 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae046383d3654aa5a466524cb8df50cd : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab94ea5cd6704f51a572a420909a7f30 : SUCCESS in 8m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/893601e739b447bc8fa48180c6b7293e : SUCCESS in 6m 57s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/b8567b4e6bd04d03818ea888bf5cbdd6 : SUCCESS in 9m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1dbcf7baf4f5432d97d4a82aefdfcd22 : SUCCESS in 8m 39s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/90a16fac2f8b4d5fa02ebebff58f897e : SUCCESS in 7m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d77713ec42f14c56a48c7fcfefac2f30 : SUCCESS in 1h 26m 16s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bbd54981a8f84d35bba5adb19c69fdd4 : SUCCESS in 1h 13m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4864b1d4307e43148c09136128e5bb06 : SUCCESS in 1h 02m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/068bfa2b28404d3cadf4b2f4c57459d2 : SUCCESS in 1h 13m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b2dd00dcae4945fc8595ce2fb9d17dc4 : SUCCESS in 50m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/169867612fcd4dc4b2b9a83f3f02abac : SUCCESS in 1h 08m 00s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/85517153c691431fb233f0a152fc20c0 : SUCCESS in 35m 05s","accounts_in_message":[],"_revision_number":14},{"id":"469e5cc5e76a8fa7ff2549168ac556fcb809ed1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-29 14:51:33.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"277022f255fb3dae90b2a40a69a7e895a525168f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 16:40:07.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/5bc01cb63d094df8a448a5d0467658f5 : SUCCESS in 1h 08m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e02bd92f29884e0c8f047d74f780b4dd : SUCCESS in 1h 08m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/634848139fa84e3a8599cac1cc5a4aee : SUCCESS in 7m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e87d145646440f29cab0d2ff35e1339 : SUCCESS in 8m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/956b250d18174e2da0c73f8c87fca64a : SUCCESS in 8m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17ff2c0de77743948ea43d29b75df73e : SUCCESS in 8m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8980f58ddcc4038b03a2d082dfeec8c : SUCCESS in 8m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/1b1e3288b3494b54ac3fd7236288f456 : SUCCESS in 10m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/03636b040a544b34bd76741cb869a547 : SUCCESS in 16m 02s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7d74724a297f46e6b9b50b6e0caf0129 : SUCCESS in 7m 04s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b8cd76cbfb4f4cf987e8cde662f4258a : SUCCESS in 1h 23m 19s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1978a538aa884edeaac1f86b31586cfd : SUCCESS in 1h 16m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/27bfe9c415e64da9993d34649194373e : SUCCESS in 1h 06m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ff9f6462a70c42858da881a7b282787f : SUCCESS in 1h 41m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bb765df85405495b88c1a1826840ee9b : SUCCESS in 1h 05m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/fae43cbd224e45f7ab871db2637c6875 : SUCCESS in 1h 19m 49s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cb32ef39e4b04182869add5c9f03ff0d : SUCCESS in 35m 49s","accounts_in_message":[],"_revision_number":15},{"id":"f6345c857a60446adcf75f3f5d57b8600c6c594f","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-06-30 21:30:13.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"79afab0d1f1ba5bb45bdb343a76522a13366dfcd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 21:30:37.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"1ae15d7fd3a5aa9fccaae675c1ddcda1cbf44a89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 02:08:38.000000000","message":"Patch Set 15: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":15},{"id":"02eb4727124e7b005240f82a70151af854a56f3b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-01 02:58:33.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"2e59eb32fb6609e3bbc572b33750281e446fc74c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 04:11:02.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/e4dfc6e726b244178128e79a31bc170f : SUCCESS in 57m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/65ad047cb68c41b7b87f415912dfd613 : SUCCESS in 1h 04m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01ed51b6e8884736aaa7c11dd453f46e : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/667c4d6164004645a16f21e1207d871a : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba98ac4b12d44605817c6d376e5180bc : SUCCESS in 7m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2ff2401fdba0412e855a43fd6e024fc2 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35d6c937ff8741eda1df2222c493e53e : SUCCESS in 6m 52s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5be0f08c563a49a4b96e13a416dec1ce : SUCCESS in 11m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/46d7e30e7f9a436788bb8336aad61e55 : SUCCESS in 11m 37s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/be72db0f37a442578a81475c8e344d41 : SUCCESS in 5m 35s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e74e63ee6b804c55822ebe4d05c2c361 : SUCCESS in 1h 10m 55s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/89e5c0aa121442d8b5dc23320ca77dab : SUCCESS in 1h 03m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/23f2dbb98dd142e385527ec8d5808506 : SUCCESS in 58m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a78951f9b2904725927766e98828f713 : SUCCESS in 1h 04m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/30f4496a8ad544f19fdb0e9edad01b20 : SUCCESS in 50m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d0e04b7f87c241eb93bd746a5c7128ff : SUCCESS in 1h 09m 53s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4b211b0bec234d1789121d4a4da08ee4 : SUCCESS in 32m 42s","accounts_in_message":[],"_revision_number":15},{"id":"2a0cdb083d0935c88cf034b2dc0de8e9fc0d8271","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 04:30:39.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"9172355eb3522f93a11b4d7cd862730e36a6caba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 05:58:27.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/412ca14b16db48949a7d9d14ef752bb2 : SUCCESS in 1h 14m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a5ef0cda36244a42b5394913381bd1e9 : SUCCESS in 59m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68c6afb3cc7646919612af9ba36f7000 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c605a1cf17942b0a9461c88313520dd : SUCCESS in 8m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b710557737444f0ba4b3d33f1e8427c : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a25369c157784f9ab2bc2fc7162dfd54 : SUCCESS in 7m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/40ed228390db4ebfa521e8e6c43845cf : SUCCESS in 9m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2a57700cf2c44886b2897581ee3c9749 : SUCCESS in 9m 10s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/68c94f7c71324a28ab999ac2533ca588 : SUCCESS in 1h 20m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2ebf1a0880ef46629095ab91d85d9468 : SUCCESS in 1h 02m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62cf3137d3b24fed8521503af12dcf62 : SUCCESS in 47m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4ebddbada9b846bb965aef7ee684815e : SUCCESS in 1h 22m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/746cb15565cc4f53b05a717c3d28f2b5 : SUCCESS in 32m 42s","accounts_in_message":[],"_revision_number":15},{"id":"beefed560206cee3310eaeddfb9c9fa3cc85a290","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 05:58:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":15},{"id":"3583d629df7d3be1c0aa93054be7e5fc89453ee1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 06:00:32.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df4d71b530674a86859696287f5aa208 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"d3d6a646e32ffe8f59e17fa02b7b5ecf06d7a417","revisions":{"7b82e03a6c3796a6e50f6e73c3c4974cc00889af":{"kind":"REWORK","_number":1,"created":"2021-04-27 16:58:02.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/788326/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/788326/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/788326/1"}}},"commit":{"parents":[{"commit":"1d82edf227ee0e7092f093debac0d89f457bc333","subject":"Add image_stage_total quota enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1d82edf227ee0e7092f093debac0d89f457bc333"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 13:47:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 15:32:20.000000000","tz":-420},"subject":"Add user_get_image_count() to DB API","message":"Add user_get_image_count() to DB API\n\nThis adds an efficient query method for getting a count of a user\u0027s\nnon-deleted images. It can be used by subsequent patches to enforce\nthe image count quota.\n\nRelated to blueprint glance-unified-quotas\n\nChange-Id: I4d6e071b8d48862032f36c80ee19632a3fe85009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7b82e03a6c3796a6e50f6e73c3c4974cc00889af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7b82e03a6c3796a6e50f6e73c3c4974cc00889af"}]},"branch":"refs/heads/master"},"aafce5e416258c76f3c11d0f3cb0ca3c323b03c1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-05-07 16:51:49.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/788326/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/788326/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/788326/2"}}},"commit":{"parents":[{"commit":"bb42447f453d738ef21b07869da0c0bbdd0222de","subject":"Add image_stage_total quota enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bb42447f453d738ef21b07869da0c0bbdd0222de"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 13:47:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-05-07 16:53:09.000000000","tz":-420},"subject":"Add user_get_image_count() to DB API","message":"Add user_get_image_count() to DB API\n\nThis adds an efficient query method for getting a count of a user\u0027s\nnon-deleted images. It can be used by subsequent patches to enforce\nthe image count quota.\n\nRelated to blueprint glance-unified-quotas\n\nChange-Id: I4d6e071b8d48862032f36c80ee19632a3fe85009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aafce5e416258c76f3c11d0f3cb0ca3c323b03c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aafce5e416258c76f3c11d0f3cb0ca3c323b03c1"}]},"branch":"refs/heads/master"},"1a56f91662a1e1b5c192a283ce41b87c009ee033":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-06-02 15:08:00.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/788326/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/788326/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/788326/3"}}},"commit":{"parents":[{"commit":"b9a2a24947642c534145f23d9203db964dddbd8f","subject":"Add image_stage_total quota enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b9a2a24947642c534145f23d9203db964dddbd8f"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 13:47:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-02 15:09:50.000000000","tz":-420},"subject":"Add user_get_image_count() to DB API","message":"Add user_get_image_count() to DB API\n\nThis adds an efficient query method for getting a count of a user\u0027s\nnon-deleted images. It can be used by subsequent patches to enforce\nthe image count quota.\n\nRelated to blueprint glance-unified-quotas\n\nChange-Id: I4d6e071b8d48862032f36c80ee19632a3fe85009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1a56f91662a1e1b5c192a283ce41b87c009ee033"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1a56f91662a1e1b5c192a283ce41b87c009ee033"}]},"branch":"refs/heads/master"},"96690a99c3b6d4afa2796abe336624103e2a0fe0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-06-02 17:48:20.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/788326/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/788326/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/788326/4"}}},"commit":{"parents":[{"commit":"8b3492b350ca67a0450b20141c0ab03a9794651a","subject":"Add image_stage_total quota enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8b3492b350ca67a0450b20141c0ab03a9794651a"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 13:47:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-02 17:50:03.000000000","tz":-420},"subject":"Add user_get_image_count() to DB API","message":"Add user_get_image_count() to DB API\n\nThis adds an efficient query method for getting a count of a user\u0027s\nnon-deleted images. It can be used by subsequent patches to enforce\nthe image count quota.\n\nRelated to blueprint glance-unified-quotas\n\nChange-Id: I4d6e071b8d48862032f36c80ee19632a3fe85009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/96690a99c3b6d4afa2796abe336624103e2a0fe0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/96690a99c3b6d4afa2796abe336624103e2a0fe0"}]},"branch":"refs/heads/master"},"ac13eda8ff9640defd29a7f6a900e79647126fe3":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-06-04 14:09:35.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/788326/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/788326/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/788326/5"}}},"commit":{"parents":[{"commit":"7e069a48322328558ca8eb31fea739ce9d02d3a3","subject":"Add image_stage_total quota enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7e069a48322328558ca8eb31fea739ce9d02d3a3"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 13:47:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-04 14:11:14.000000000","tz":-420},"subject":"Add user_get_image_count() to DB API","message":"Add user_get_image_count() to DB API\n\nThis adds an efficient query method for getting a count of a user\u0027s\nnon-deleted images. It can be used by subsequent patches to enforce\nthe image count quota.\n\nRelated to blueprint glance-unified-quotas\n\nChange-Id: I4d6e071b8d48862032f36c80ee19632a3fe85009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ac13eda8ff9640defd29a7f6a900e79647126fe3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ac13eda8ff9640defd29a7f6a900e79647126fe3"}]},"branch":"refs/heads/master"},"7718bc8d8e14e7e0030b4905c1ca3948ba1bdbeb":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":6,"created":"2021-06-04 17:08:07.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/788326/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/788326/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/788326/6"}}},"commit":{"parents":[{"commit":"77701374198aa448357975f605c6bf7e5054291c","subject":"Add image_stage_total quota enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/77701374198aa448357975f605c6bf7e5054291c"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 13:47:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-04 16:58:07.000000000","tz":-420},"subject":"Add user_get_image_count() to DB API","message":"Add user_get_image_count() to DB API\n\nThis adds an efficient query method for getting a count of a user\u0027s\nnon-deleted images. It can be used by subsequent patches to enforce\nthe image count quota.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I4d6e071b8d48862032f36c80ee19632a3fe85009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7718bc8d8e14e7e0030b4905c1ca3948ba1bdbeb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7718bc8d8e14e7e0030b4905c1ca3948ba1bdbeb"}]},"branch":"refs/heads/master"},"51948283a4ba0ccfe75cbd896dda7f63becac504":{"kind":"REWORK","_number":7,"created":"2021-06-04 18:34:23.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/788326/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/788326/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/788326/7"}}},"commit":{"parents":[{"commit":"074cb45264327f73e530cb56091f82136b751d01","subject":"Add image_stage_total quota enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/074cb45264327f73e530cb56091f82136b751d01"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 13:47:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-04 18:32:12.000000000","tz":-420},"subject":"Add user_get_image_count() to DB API","message":"Add user_get_image_count() to DB API\n\nThis adds an efficient query method for getting a count of a user\u0027s\nnon-deleted images. It can be used by subsequent patches to enforce\nthe image count quota.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I4d6e071b8d48862032f36c80ee19632a3fe85009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/51948283a4ba0ccfe75cbd896dda7f63becac504"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/51948283a4ba0ccfe75cbd896dda7f63becac504"}]},"branch":"refs/heads/master"},"28c5027e9007c632d109f5e5ba9d7d2a0dbe09fa":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-06-04 18:56:53.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/788326/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/788326/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/788326/8"}}},"commit":{"parents":[{"commit":"f1c216c0281bdf9826b08e95e5f5acea3a006f44","subject":"Add image_stage_total quota enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f1c216c0281bdf9826b08e95e5f5acea3a006f44"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 13:47:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-04 18:58:46.000000000","tz":-420},"subject":"Add user_get_image_count() to DB API","message":"Add user_get_image_count() to DB API\n\nThis adds an efficient query method for getting a count of a user\u0027s\nnon-deleted images. It can be used by subsequent patches to enforce\nthe image count quota.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I4d6e071b8d48862032f36c80ee19632a3fe85009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/28c5027e9007c632d109f5e5ba9d7d2a0dbe09fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/28c5027e9007c632d109f5e5ba9d7d2a0dbe09fa"}]},"branch":"refs/heads/master"},"bc1bd374a38765c890dadc739f2416090870e682":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-06-08 14:43:18.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/788326/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/788326/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/788326/9"}}},"commit":{"parents":[{"commit":"77369c706b9c8d90c1fd798aced88c6c13ed400b","subject":"Add image_stage_total quota enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/77369c706b9c8d90c1fd798aced88c6c13ed400b"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 13:47:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-08 14:44:57.000000000","tz":-420},"subject":"Add user_get_image_count() to DB API","message":"Add user_get_image_count() to DB API\n\nThis adds an efficient query method for getting a count of a user\u0027s\nnon-deleted images. It can be used by subsequent patches to enforce\nthe image count quota.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I4d6e071b8d48862032f36c80ee19632a3fe85009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc1bd374a38765c890dadc739f2416090870e682"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc1bd374a38765c890dadc739f2416090870e682"}]},"branch":"refs/heads/master"},"c5f42f99385f9f1eb8c55408281661e0b5107e0e":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2021-06-14 14:40:58.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/788326/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/788326/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/788326/10"}}},"commit":{"parents":[{"commit":"5ae4d3955dde75333001572751d55f4da6a3b5e9","subject":"Add image_stage_total quota enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5ae4d3955dde75333001572751d55f4da6a3b5e9"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 13:47:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-14 14:42:51.000000000","tz":-420},"subject":"Add user_get_image_count() to DB API","message":"Add user_get_image_count() to DB API\n\nThis adds an efficient query method for getting a count of a user\u0027s\nnon-deleted images. It can be used by subsequent patches to enforce\nthe image count quota.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I4d6e071b8d48862032f36c80ee19632a3fe85009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c5f42f99385f9f1eb8c55408281661e0b5107e0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c5f42f99385f9f1eb8c55408281661e0b5107e0e"}]},"branch":"refs/heads/master"},"f4ebff18d5732800b97162c3fc5d3a4fbc7084ea":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2021-06-14 14:47:30.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/788326/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/788326/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/788326/11"}}},"commit":{"parents":[{"commit":"2f99ce632d31e174e7bed21ee5ba8c90e9868136","subject":"Add image_stage_total quota enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2f99ce632d31e174e7bed21ee5ba8c90e9868136"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 13:47:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-14 14:49:32.000000000","tz":-420},"subject":"Add user_get_image_count() to DB API","message":"Add user_get_image_count() to DB API\n\nThis adds an efficient query method for getting a count of a user\u0027s\nnon-deleted images. It can be used by subsequent patches to enforce\nthe image count quota.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I4d6e071b8d48862032f36c80ee19632a3fe85009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f4ebff18d5732800b97162c3fc5d3a4fbc7084ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f4ebff18d5732800b97162c3fc5d3a4fbc7084ea"}]},"branch":"refs/heads/master"},"d18c63cb7de43fa7ddbd46b33aeedf9b97bd4c62":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2021-06-14 14:56:36.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/788326/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/788326/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/788326/12"}}},"commit":{"parents":[{"commit":"e64db4515d0cbe1fe9c790a0d306d351bff91fda","subject":"Add image_stage_total quota enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e64db4515d0cbe1fe9c790a0d306d351bff91fda"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 13:47:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-14 14:58:35.000000000","tz":-420},"subject":"Add user_get_image_count() to DB API","message":"Add user_get_image_count() to DB API\n\nThis adds an efficient query method for getting a count of a user\u0027s\nnon-deleted images. It can be used by subsequent patches to enforce\nthe image count quota.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I4d6e071b8d48862032f36c80ee19632a3fe85009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d18c63cb7de43fa7ddbd46b33aeedf9b97bd4c62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d18c63cb7de43fa7ddbd46b33aeedf9b97bd4c62"}]},"branch":"refs/heads/master"},"a2378dbfa75b29d2f4f0f73b03ea30e72a341216":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2021-06-21 15:46:41.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/788326/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/788326/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/788326/13"}}},"commit":{"parents":[{"commit":"622661f0063eef4940cce0a48ae974581ebb0817","subject":"Add image_stage_total quota enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/622661f0063eef4940cce0a48ae974581ebb0817"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 13:47:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-21 15:47:43.000000000","tz":-420},"subject":"Add user_get_image_count() to DB API","message":"Add user_get_image_count() to DB API\n\nThis adds an efficient query method for getting a count of a user\u0027s\nnon-deleted images. It can be used by subsequent patches to enforce\nthe image count quota.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I4d6e071b8d48862032f36c80ee19632a3fe85009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a2378dbfa75b29d2f4f0f73b03ea30e72a341216"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a2378dbfa75b29d2f4f0f73b03ea30e72a341216"}]},"branch":"refs/heads/master"},"bb04ef466b48216939e48cbc5a6da323dcd8b8bb":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2021-06-21 17:57:14.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/788326/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/788326/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/788326/14"}}},"commit":{"parents":[{"commit":"3125c32ec21ae34073adba775702fac7dfb94e60","subject":"Add image_stage_total quota enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3125c32ec21ae34073adba775702fac7dfb94e60"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 13:47:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-21 17:59:19.000000000","tz":-420},"subject":"Add user_get_image_count() to DB API","message":"Add user_get_image_count() to DB API\n\nThis adds an efficient query method for getting a count of a user\u0027s\nnon-deleted images. It can be used by subsequent patches to enforce\nthe image count quota.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I4d6e071b8d48862032f36c80ee19632a3fe85009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bb04ef466b48216939e48cbc5a6da323dcd8b8bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bb04ef466b48216939e48cbc5a6da323dcd8b8bb"}]},"branch":"refs/heads/master"},"d3d6a646e32ffe8f59e17fa02b7b5ecf06d7a417":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2021-06-29 14:51:33.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/26/788326/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/26/788326/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/26/788326/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/26/788326/15"}}},"commit":{"parents":[{"commit":"59990d513a084b811dff00fd6421b9ff92cb2ff0","subject":"Add image_stage_total quota enforcement","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/59990d513a084b811dff00fd6421b9ff92cb2ff0"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 13:47:58.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-29 15:53:18.000000000","tz":-420},"subject":"Add user_get_image_count() to DB API","message":"Add user_get_image_count() to DB API\n\nThis adds an efficient query method for getting a count of a user\u0027s\nnon-deleted images. It can be used by subsequent patches to enforce\nthe image count quota.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I4d6e071b8d48862032f36c80ee19632a3fe85009\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d3d6a646e32ffe8f59e17fa02b7b5ecf06d7a417"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d3d6a646e32ffe8f59e17fa02b7b5ecf06d7a417"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
