)]}'
{"id":"openstack%2Fglance~788327","triplet_id":"openstack%2Fglance~master~I8af124d9307263cd8289d0701fb9a745d13b1d56","project":"openstack/glance","branch":"master","topic":"bp/glance-unified-quotas","hashtags":[],"change_id":"I8af124d9307263cd8289d0701fb9a745d13b1d56","subject":"Add image_count_total quota enforcement","status":"MERGED","created":"2021-04-27 16:58:02.000000000","updated":"2021-07-01 06:02:46.000000000","submitted":"2021-07-01 05:58:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"788327-bp/glance-unified-quotas","meta_rev_id":"6e59b625b9716f0b8bd53b089cc79ac3d096db3f","_number":788327,"virtual_id_number":788327,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-01 05:58:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2021-07-01 06:02:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-06-30 21:50:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2021-06-30 21:50:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2021-07-01 06:02:46.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-27 18:18:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-04 15:56:49.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-06-14 15:27:31.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-06-30 21:50:32.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"56f855b3df8d9795819e86e5db05dd512d966fdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-27 16:58:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a2c12cc9d1bfed9d030aa003e43e3d6b5fd44da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 18:18:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/068a0ebb7e0a4fb38a0e1ee73ff1cec6 : SUCCESS in 1h 00m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2c1faaf6870345559daf04179b299e8c : SUCCESS in 57m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3b3bbb1cfc044aaba396c185f145739 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6fab8377a8a848738b82a94a832f0678 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e26330f76f364a40a8b142d2b87e3be7 : SUCCESS in 6m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/853a2479b8894fe0ad589cf1a8f57516 : SUCCESS in 9m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d495bae0be7e4710889d0c7d6a5e5399 : SUCCESS in 7m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/ff5b85fe9b3f4b539d36474a4d48e3c7 : SUCCESS in 12m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/aa6b708de75b43428cd075463a8e5f8e : SUCCESS in 9m 09s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/964ddae84c3c41ef99f24df9cca8156f : SUCCESS in 5m 20s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/daa15ef5984d449ba4d0c26aea21f865 : FAILURE in 58m 23s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cf94fa61db4d4d54801a42336a8109ff : SUCCESS in 59m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c1010e2fe674407d80a4a25734092b7f : SUCCESS in 1h 07m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d4f9c5d1b4e04055812cc2fc4c7066ff : SUCCESS in 1h 06m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8fbd221bf08451faa316df5a9373577 : SUCCESS in 52m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e05d077cc7004bceae235faab04cee02 : SUCCESS in 1h 15m 48s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e0bf79f9ae454315aadb8a8f6a03c294 : SUCCESS in 27m 42s","accounts_in_message":[],"_revision_number":1},{"id":"9c62e4c34eb117b8ac1e502925ac83aad009918f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-05-07 16:51:49.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"4330990798003aef2743bb9b6b7254fd72545d3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-07 18:23:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/2e35909b23024c81b7616427e8e334bd : SUCCESS in 1h 13m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c306eb17f0d34be085aa1b2580315097 : SUCCESS in 1h 12m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76623e5a860443c29ab7c32beab0afd9 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18dae32c15fb4a11ba15ed6ca402841e : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d775f39bea9a41bb8f12ea9f5f021e68 : SUCCESS in 7m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef6c026b36c44d4bb9c2b8c24fe3f4a0 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/174f05bf671e4a54845ba41103b984dc : SUCCESS in 7m 01s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f5dfb15099f24e97b606cd2f73ce8575 : SUCCESS in 8m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b656abcf3589455896c2f5256057ad24 : SUCCESS in 8m 10s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/6c29c2162267474f916caaebed21510d : SUCCESS in 5m 37s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/12481228d2044cc1a9b3c113b0605d97 : SUCCESS in 1h 17m 27s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3ab769ed68ad442f9acafe9f9892a2b1 : SUCCESS in 1h 15m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fe0c6cb1731a4e7987403f62794fc435 : SUCCESS in 1h 15m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1c8dd4b34a804dae8d1efc41539aba42 : SUCCESS in 1h 15m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4f9c821cd9f44688fd974ecd5b847eb : SUCCESS in 1h 05m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d3554a52e8fb44d980d72d6798f127fa : SUCCESS in 1h 09m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/91fe1cfb90da4e6080b5b59f19cd1351 : SUCCESS in 31m 18s","accounts_in_message":[],"_revision_number":2},{"id":"2b91af3300f5d7267e7c1761d6ddc17db5d0404c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-02 15:08:00.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"7af7f2f5fe560c9278222b3fb380e692860127ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-02 17:48:20.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"e7521b5ef1a88366b4ddab09e561205b9577ca78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 20:22:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/283bdfa3484f426897a3e8f208df8119 : SUCCESS in 1h 09m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6e7969ef92524e088c4b7bd8a0302956 : SUCCESS in 1h 26m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4d9bc1c47e74165b45bc38bce6e3060 : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e740f5bab2c0458d94b92613a0787e45 : SUCCESS in 8m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ffc2198f8574c2286308b0a71595e2b : SUCCESS in 7m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/489a1b07e40049eeb8695e6b1925b4cf : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17d7a3a8c77f409ab64082611d29c6eb : SUCCESS in 9m 06s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c387df593059424285b8a1190f5d3ae0 : SUCCESS in 11m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/18de450160524d5cabb70c0b03ee407f : SUCCESS in 10m 31s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/502ade4ce62d4be8936c768db38ce3ee : SUCCESS in 6m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/af18d49f95b3481db822906dfd57b90d : SUCCESS in 1h 22m 31s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2abf96c542fa4fe38b254526244e709f : SUCCESS in 1h 17m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9c4a4545f4cf40ce90cbb136d721e190 : SUCCESS in 1h 10m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f3785c61bd014abda0778cd1d4bc662f : FAILURE in 1h 27m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/055c0084c77f483eaeeeae1c56d15862 : SUCCESS in 54m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/44a2cc98c0f6400285a0c03aa3bdb742 : SUCCESS in 1h 33m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a5a1652fa56044e5b99968cbb90e40aa : SUCCESS in 34m 22s","accounts_in_message":[],"_revision_number":4},{"id":"e00917c9a9caee39faf040fa30f450415b8cf526","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-04 14:09:35.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5932aef3b0c012f81524cd7ee9a6300dae62fc34","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-06-04 15:56:49.000000000","message":"Patch Set 5: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"88f0617f7070eda90ed70de6e19b0019facb1f45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 16:50:26.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/3560a1ca7ea94506b0a0729326b422fe : SUCCESS in 1h 00m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f892fb6c791242bc932e3bae7c116221 : SUCCESS in 1h 11m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13a1a5fb43084e99919415936caeb130 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30f8fd13e4a14c929aac59aa9e5badcd : SUCCESS in 9m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b352944d52b04d9caf0c932a8b89f629 : SUCCESS in 7m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7140d4247b6a4547ad480ed7fe207832 : SUCCESS in 7m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04c8a3a62c92491a959038535c714328 : SUCCESS in 8m 13s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/25ee76ef11574acdb27ca2dfaea79bfb : SUCCESS in 11m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f77a1de1b0d44310b1e24ed5091a0ff6 : SUCCESS in 10m 39s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/a662eb5108524040ae099f1a95af1a1e : SUCCESS in 8m 33s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/23062b19451147a2863ce205da818f48 : SUCCESS in 1h 41m 33s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/565a0a5ba4214b63ab9e960e364fc9c5 : SUCCESS in 1h 22m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2f2945e6decc4a2abd8c7219a469d632 : SUCCESS in 1h 14m 46s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e911161f934e4dceaa0953b4f4babda0 : SUCCESS in 1h 41m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47fbb039cf864de0b03c656e29ddfe68 : SUCCESS in 54m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5f7678b165e34871885401f76054e532 : SUCCESS in 1h 28m 06s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/071ca4f74fda40b48e858f659cce1ea8 : SUCCESS in 31m 42s","accounts_in_message":[],"_revision_number":5},{"id":"7bd3daeeb8cae91f79619e541987ad2cfdc98e5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-04 17:08:07.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"286bb0890d4edba01693181123a4b1b5c8ce23d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-04 18:34:23.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b7d71342bf5a4c70e83b2187c4869fe2d5c10183","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-04 18:56:53.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"7d46fa952a9b2e59e599186a773c9214013b4795","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-04 20:10:20.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/56f223250cf345b29b96be680192598a : SUCCESS in 1h 06m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cd5587d78a44cbaa967494a6ae26b75 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09b167de589d4a81b2e8d7db45f31ac9 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/584c17679a694c44b17e92c31827cdb4 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a878e5432fb14d24af573b938ff116d5 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e64b550668ec463fbcc37eb18390297e : SUCCESS in 6m 24s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f0e4620d970348f4a34e1257001e37bd : SUCCESS in 9m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e4cea93889bf4961ba4e8e908ea9297c : SUCCESS in 9m 00s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/729d1152fe394a6db6bcba2214377ebd : SUCCESS in 6m 08s","accounts_in_message":[],"_revision_number":8},{"id":"4c6c0a9875ab820a0f17cb48955b3b8952c9f8fb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-06-08 14:27:02.000000000","message":"Patch Set 8: Code-Review+2 Review-Priority+1\n\nLooks good, thank you Dan!!","accounts_in_message":[],"_revision_number":8},{"id":"04f0f54e0d157374c54d47ee604485d0b3a208a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-08 14:43:18.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"dfdd6da53560ae786bd25d502c21576c8d21a19e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-08 17:18:00.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/dc530dc003614f769ed8c4ed0a1dc784 : SUCCESS in 1h 04m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/66b30c45df144180993735d7d618f063 : SUCCESS in 1h 10m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/650d7e0e0e2d446a9156b3f3af0aef81 : SUCCESS in 7m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e58d18a9cb6f4005988063efc563566e : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4b0779acb424563b5870ba4de2f8b5e : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dca7e8741e4b47a7856f15d426e208c3 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc640369d520421283a485f2ed3b6165 : SUCCESS in 7m 45s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/035e0be6e5da4999bea72af3ab7f94bc : SUCCESS in 9m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ad3ce23c94ae47b8bc936f6114e80b3a : SUCCESS in 9m 53s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8dc21d8489ec4aff9e8567877801a7e6 : SUCCESS in 5m 54s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c85de307d9c34fdcb7fb1e9cb14a61a6 : SUCCESS in 1h 21m 05s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a730f89fea1741ac8e5d779cce8c4fd2 : SUCCESS in 1h 33m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/83e3b4e66d5e4c899e63c3c87ae9d9ee : SUCCESS in 1h 25m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a2545cc8a14e450ab16778011168e0a6 : SUCCESS in 1h 27m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e3cce4f6900425ebffa6e721135f3e2 : SUCCESS in 1h 02m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/58aa97c63b004e20938c838868415e18 : SUCCESS in 1h 30m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c05138664bbf47a38bf6829b0a894efc : SUCCESS in 36m 15s","accounts_in_message":[],"_revision_number":9},{"id":"e45803ea9f8e8c5ab7afb69d4963209c509a70d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-14 14:40:58.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"af1c3719bd3c8948c1a7ae3c611d27843e7d693b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-14 14:47:30.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"443150e77b2f13f699e672341a4997abeb88a613","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-14 14:56:36.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"923f79b860198c331e421bf4e4a8a5c2bf2f8784","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-14 15:24:35.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"aa60f79523fbc7751bdae8176de9dc400c3c2495","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-14 15:27:31.000000000","message":"Patch Set 13:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/788327/788327/13/check-not-vote/ext-glance-zuul/236cbcc : FAILURE in 0s","accounts_in_message":[],"_revision_number":13},{"id":"7171a39699f0289f61dd8a2c05134f87b5c9402d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-14 16:58:40.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/bd83641ffda24dc0b1cda646a1eb81b8 : FAILURE in 11m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d5e18c4209d1444f8b6a76062853d4be : FAILURE in 33m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65ba697043a84e77b391b271bf1468be : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d334e851f53f45989c2b3e3eeb037f28 : SUCCESS in 6m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d89269b76974de8b7d75980ff5bde21 : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0204d333fef4c17929feb6ea0f9c5f1 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1217b3e785c74df291ae83d98be37298 : SUCCESS in 8m 38s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6ba0f1b5c05e4b0592a6b3c5f7ea7cbd : SUCCESS in 11m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cb3ad3dedde1432ca2690316a6265bad : SUCCESS in 11m 14s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/179b79a3900d44f3aca5464988bcabb3 : SUCCESS in 6m 49s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7d67b747fbc940d2a51185fdbfedc139 : FAILURE in 1h 29m 47s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9930438cfd47437f9f508009d87de331 : SUCCESS in 1h 12m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/637788aa08a1479e8fe8fa40b30e05f1 : SUCCESS in 1h 07m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/81912d8fda32456ea0d3a23ee32a1d94 : SUCCESS in 1h 19m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9816a3188edf41e1b8733560444dc604 : SUCCESS in 52m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6a57e6cba77b4f748de96f2c6a8bc3f4 : SUCCESS in 1h 22m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/abc7044c4af1487cafada631c1f8a7fe : SUCCESS in 26m 18s","accounts_in_message":[],"_revision_number":13},{"id":"264c7b4d75a772dca7978f4f1103eb40b2faeeaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-21 15:46:41.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"feacfd7fc9f6f40b899824928d965e28f7dac2b0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-06-21 16:18:40.000000000","message":"Patch Set 14: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":14},{"id":"030845f4e009bef6d928f30fefeabe318f0f06cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 17:22:28.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/4bbcfdc223d34ea78bf67266ce463981 : SUCCESS in 1h 07m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d0286d60d1de446db673ca0b3cf82dba : SUCCESS in 1h 09m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2097066691b445599ac13bd0a6e43d3 : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad791909d6744cd8a3b171e1983e099d : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/13dd5fde06e24501b9b06b1b92ac68a1 : SUCCESS in 7m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ceacce2cc03d4ba9b56ce4c4cc00751a : SUCCESS in 8m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faadbcc38f4c49deae362afbeb2d47bd : SUCCESS in 7m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/db7393bbfb73405db8730d43a82ef948 : SUCCESS in 9m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/742dce1a0944472f8f4bfebb730bb8f6 : SUCCESS in 11m 19s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/75af9c1c8777403d91dfafad9b826f15 : SUCCESS in 6m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/76c40518878a41ce89f422c44c106030 : SUCCESS in 1h 16m 43s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a1e0e31cf3de43fb9a38c753de35c085 : SUCCESS in 1h 17m 26s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0faa75a820f7451fafe8a31d4a598e8b : SUCCESS in 1h 24m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/51000d22af664a5da58c2c157560eae5 : SUCCESS in 1h 11m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94c020a82b654fcbb03620f8698e8a0b : SUCCESS in 52m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/738a284411fa41ddb16e881f9b105253 : SUCCESS in 1h 19m 18s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/eccc5521a38e466aa53be1c11e2ae6c7 : SUCCESS in 30m 17s","accounts_in_message":[],"_revision_number":14},{"id":"4ced9cdd59e3f1994ce81ee39df2b0f2ba9efc5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-21 17:57:14.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"1a8c09541bfb5065a9dfac8cbc67146a03652ac2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-21 19:33:06.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/839d6794cf434fb9a70a39a61cb2ef2d : SUCCESS in 1h 11m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8f07adf7ceaa459b856c1632f34b9213 : SUCCESS in 1h 27m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7235cc09239e4b2aa485843f51e5d103 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12993ec1639c416f8d1bf7e030b87259 : SUCCESS in 8m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4432ba34cdfa4fba896693d97be7c7f9 : SUCCESS in 6m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af31a5d46a9f463d9d292237afbc6037 : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74e97e99aef44ebe9ba02ba0d4e3315d : SUCCESS in 6m 59s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/04fb4ff730a84cc0883d6314a5dcc8c7 : SUCCESS in 10m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/54723b887b2243869e46e536d9b2ee17 : SUCCESS in 8m 45s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8cd0b3a3852d4672a0735ced6cef9923 : SUCCESS in 6m 45s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/79eec063ef7843499098701925136f14 : SUCCESS in 1h 16m 42s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/683c31b7fdb549498858c0229353aa51 : SUCCESS in 1h 02m 38s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3b82e81049a5481c9d6eb7d063ef12d2 : SUCCESS in 1h 11m 22s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9cec5acaa44f482b8b82e586618e0e63 : SUCCESS in 1h 23m 28s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbfb54570e31430fb0efe164b3520330 : SUCCESS in 42m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b94a362b07e74d29836337136a9bac1e : SUCCESS in 1h 27m 48s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9267f6f1538141a1b5aa273c011c4e38 : SUCCESS in 28m 00s","accounts_in_message":[],"_revision_number":15},{"id":"5afa91a4b7a917f813060873c087a9bb10fd1957","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-29 14:51:33.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"019170faf2475608073eb87e2d434b0a34dd7f70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-29 16:23:47.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/f0b990d4f02b4f6ebd553a84c3abb74a : SUCCESS in 1h 04m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f6e9c79e4a9a49b28eb97367e6350488 : SUCCESS in 1h 11m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d116513fe4594027a2f3d7794ed2dfb8 : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6cde77aab1624864a7d07025bcd1d588 : SUCCESS in 8m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc29a958baa84209bcc15a349c226c61 : SUCCESS in 8m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5deefab314e4f6fa59aed916d822d1c : SUCCESS in 8m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e4d32d3f0c24d82bfb268f7cb389446 : SUCCESS in 9m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fd943832a0754b79ab3641dd3f87e98f : SUCCESS in 10m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f434ca10d81c41e5823a8f7fc98d978e : SUCCESS in 11m 31s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/aa17ab0ab4914ec5ba6521215c113c7d : SUCCESS in 7m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e367f13364e145f2a1637b06116aaff7 : SUCCESS in 1h 25m 14s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1023a2f9f9fe49cd9b0a928ed6955285 : SUCCESS in 1h 05m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/66cedf6118584535943db09384b98c19 : SUCCESS in 1h 12m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b8f0067dbdcf4f26975ae4adc1fa0594 : SUCCESS in 1h 19m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/03e5dea444fa4968b8f7ac2ed353d4f0 : SUCCESS in 53m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/513cfe1577ce4c40b53f8ec938c3b467 : SUCCESS in 1h 27m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a590df72abde4b8080a1641aab8cd09a : SUCCESS in 41m 15s","accounts_in_message":[],"_revision_number":16},{"id":"558b2b85c292d7ae10b424027ceb787c3018900e","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-06-30 21:50:32.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"00de17b9fac26ea324c068e9feca659a6669af3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-30 21:50:59.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"ba9644ed23395a9b894fc120bcb926b32c6bab30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 02:08:48.000000000","message":"Patch Set 16: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":16},{"id":"41d6d14bc93a3de2e100a2a0da8fdb0d795d13d9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-01 02:59:00.000000000","message":"Patch Set 16:\n\nrecheck","accounts_in_message":[],"_revision_number":16},{"id":"7a40bf9629dec9c25940d5e102dbe1715e7f103c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 04:24:44.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/647650cb81bb4b30a627bceab736e8dd : SUCCESS in 59m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a0ab002a69dc463e9fc28b082503dc82 : SUCCESS in 1h 11m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79cad1bd046a40d6bebdaace2671987d : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e967cca8f35e4e09b3a95f93dd6e9ebb : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab0957b279b74769b712481d52b5e6ef : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23d213004540461b8292be8280e3df59 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fee0135035614c4b9064ac14869649ca : SUCCESS in 6m 17s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/631ab2393d6747619cd58117779e7366 : SUCCESS in 9m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/293a73fa674e4cf0b70bd2d8df019bde : SUCCESS in 8m 45s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/cde26b381fd443cf80039e83c9c4a5e2 : SUCCESS in 5m 37s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/491ab6db33cf4fef944b35cb1a70e01b : SUCCESS in 1h 08m 25s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/44edc5770c6e4f01bdbc835d24d38fa9 : SUCCESS in 1h 11m 03s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6775e4e5a98d4ce5af9e02af33efbdb2 : SUCCESS in 58m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c2f63b2f63344cf197148398cf2393f4 : SUCCESS in 1h 19m 20s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b1ee1767b3a4e9bbeb33f1e43caca00 : SUCCESS in 47m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f1aae76c12784b8890c462897720425e : SUCCESS in 1h 19m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9d5ebf7e515b465eb9bf5b371fdcf6c0 : SUCCESS in 31m 31s","accounts_in_message":[],"_revision_number":16},{"id":"8960c1b5f75be80108183e3d13cad2cb581f5a66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 04:30:41.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"e8cd072d7d86f7165281bf88c6e9e21d12cda108","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 05:58:34.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/f2e4d23396f94a1d9bd318158bd98da2 : SUCCESS in 1h 03m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8a89c9e478a841a294ddf4726ab19d65 : SUCCESS in 1h 10m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/750a7d111c7d403694bb11841d815bc8 : SUCCESS in 5m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e984e84421854ed6942e66a26606f619 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d8f0147110e4ee18bf22f8488ee703a : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fc6082f08c947f182f45ca72a661131 : SUCCESS in 6m 50s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/6ce4767d3c264d8a8907df73f5393d9a : SUCCESS in 8m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/18ac5e0562514268aebe2a691bdb84bc : SUCCESS in 10m 12s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ac57b3cb31564d20b7dc230e477702d7 : SUCCESS in 58m 09s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7161f41a86b74791ab9cc6c6d4e427bf : SUCCESS in 1h 06m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8067c5ddf8f64f98a7e061f3bd8b121c : SUCCESS in 44m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/68e9b514ae4a42d08747f64637201844 : SUCCESS in 1h 08m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9f56664c519d474ab7550a75e6deea4d : SUCCESS in 33m 35s","accounts_in_message":[],"_revision_number":16},{"id":"0005cfdddc510032f2e681d0305299176c2bbb9a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 05:58:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":16},{"id":"5a1dbfc3d2206299e2b9aa2cf34edd34c9969a66","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-01 06:02:46.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb0ebeda3410453ea93aff628351aaa7 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"a36666e2fe8d61a161daa1836ed84fb2e389dd67","revisions":{"00aaec364119c2163c65838714e871f7b3b8d875":{"kind":"REWORK","_number":1,"created":"2021-04-27 16:58:02.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/788327/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/788327/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/788327/1"}}},"commit":{"parents":[{"commit":"7b82e03a6c3796a6e50f6e73c3c4974cc00889af","subject":"Add user_get_image_count() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7b82e03a6c3796a6e50f6e73c3c4974cc00889af"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 14:09:06.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 15:32:32.000000000","tz":-420},"subject":"Add image_count_total quota enforcement","message":"Add image_count_total quota enforcement\n\nThis makes us enforce a quota on the total number of (non-deleted)\nimages owned by a user.\n\nRelated to blueprint glance-unified-quotas\n\nChange-Id: I8af124d9307263cd8289d0701fb9a745d13b1d56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/00aaec364119c2163c65838714e871f7b3b8d875"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/00aaec364119c2163c65838714e871f7b3b8d875"}]},"branch":"refs/heads/master"},"db8372031aaa556d43da0484ece4ef391f7f44eb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-05-07 16:51:49.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/788327/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/788327/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/788327/2"}}},"commit":{"parents":[{"commit":"aafce5e416258c76f3c11d0f3cb0ca3c323b03c1","subject":"Add user_get_image_count() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aafce5e416258c76f3c11d0f3cb0ca3c323b03c1"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 14:09:06.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-05-07 16:53:09.000000000","tz":-420},"subject":"Add image_count_total quota enforcement","message":"Add image_count_total quota enforcement\n\nThis makes us enforce a quota on the total number of (non-deleted)\nimages owned by a user.\n\nRelated to blueprint glance-unified-quotas\n\nChange-Id: I8af124d9307263cd8289d0701fb9a745d13b1d56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/db8372031aaa556d43da0484ece4ef391f7f44eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/db8372031aaa556d43da0484ece4ef391f7f44eb"}]},"branch":"refs/heads/master"},"9f8e7d17d00fbfcbf9b79279c827f5aa74f0e202":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-06-02 15:08:00.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/788327/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/788327/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/788327/3"}}},"commit":{"parents":[{"commit":"1a56f91662a1e1b5c192a283ce41b87c009ee033","subject":"Add user_get_image_count() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1a56f91662a1e1b5c192a283ce41b87c009ee033"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 14:09:06.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-02 15:09:50.000000000","tz":-420},"subject":"Add image_count_total quota enforcement","message":"Add image_count_total quota enforcement\n\nThis makes us enforce a quota on the total number of (non-deleted)\nimages owned by a user.\n\nRelated to blueprint glance-unified-quotas\n\nChange-Id: I8af124d9307263cd8289d0701fb9a745d13b1d56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9f8e7d17d00fbfcbf9b79279c827f5aa74f0e202"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9f8e7d17d00fbfcbf9b79279c827f5aa74f0e202"}]},"branch":"refs/heads/master"},"3053a76afab20346d0480b8e8825c9236faf53dd":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-06-02 17:48:20.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/788327/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/788327/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/788327/4"}}},"commit":{"parents":[{"commit":"96690a99c3b6d4afa2796abe336624103e2a0fe0","subject":"Add user_get_image_count() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/96690a99c3b6d4afa2796abe336624103e2a0fe0"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 14:09:06.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-02 17:50:03.000000000","tz":-420},"subject":"Add image_count_total quota enforcement","message":"Add image_count_total quota enforcement\n\nThis makes us enforce a quota on the total number of (non-deleted)\nimages owned by a user.\n\nRelated to blueprint glance-unified-quotas\n\nChange-Id: I8af124d9307263cd8289d0701fb9a745d13b1d56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3053a76afab20346d0480b8e8825c9236faf53dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3053a76afab20346d0480b8e8825c9236faf53dd"}]},"branch":"refs/heads/master"},"bba8daa4639e442490a29335a902e98d68d77d37":{"kind":"REWORK","_number":5,"created":"2021-06-04 14:09:35.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/788327/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/788327/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/788327/5"}}},"commit":{"parents":[{"commit":"ac13eda8ff9640defd29a7f6a900e79647126fe3","subject":"Add user_get_image_count() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ac13eda8ff9640defd29a7f6a900e79647126fe3"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 14:09:06.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-04 14:11:14.000000000","tz":-420},"subject":"Add image_count_total quota enforcement","message":"Add image_count_total quota enforcement\n\nThis makes us enforce a quota on the total number of (non-deleted)\nimages owned by a user.\n\nRelated to blueprint glance-unified-quotas\n\nChange-Id: I8af124d9307263cd8289d0701fb9a745d13b1d56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bba8daa4639e442490a29335a902e98d68d77d37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bba8daa4639e442490a29335a902e98d68d77d37"}]},"branch":"refs/heads/master"},"3517be34514a0c9631ae9f899e650b70351ff093":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":6,"created":"2021-06-04 17:08:07.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/788327/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/788327/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/788327/6"}}},"commit":{"parents":[{"commit":"7718bc8d8e14e7e0030b4905c1ca3948ba1bdbeb","subject":"Add user_get_image_count() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7718bc8d8e14e7e0030b4905c1ca3948ba1bdbeb"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 14:09:06.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-04 16:58:45.000000000","tz":-420},"subject":"Add image_count_total quota enforcement","message":"Add image_count_total quota enforcement\n\nThis makes us enforce a quota on the total number of (non-deleted)\nimages owned by a user.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I8af124d9307263cd8289d0701fb9a745d13b1d56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3517be34514a0c9631ae9f899e650b70351ff093"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3517be34514a0c9631ae9f899e650b70351ff093"}]},"branch":"refs/heads/master"},"5941284d85cc5e81f863d6f7b360573cd55829f9":{"kind":"REWORK","_number":7,"created":"2021-06-04 18:34:23.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/788327/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/788327/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/788327/7"}}},"commit":{"parents":[{"commit":"51948283a4ba0ccfe75cbd896dda7f63becac504","subject":"Add user_get_image_count() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/51948283a4ba0ccfe75cbd896dda7f63becac504"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 14:09:06.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-04 18:32:12.000000000","tz":-420},"subject":"Add image_count_total quota enforcement","message":"Add image_count_total quota enforcement\n\nThis makes us enforce a quota on the total number of (non-deleted)\nimages owned by a user.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I8af124d9307263cd8289d0701fb9a745d13b1d56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5941284d85cc5e81f863d6f7b360573cd55829f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5941284d85cc5e81f863d6f7b360573cd55829f9"}]},"branch":"refs/heads/master"},"ad0cedccaaced1d8207faa51c4e1bc2bf88f12c6":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-06-04 18:56:53.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/788327/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/788327/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/788327/8"}}},"commit":{"parents":[{"commit":"28c5027e9007c632d109f5e5ba9d7d2a0dbe09fa","subject":"Add user_get_image_count() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/28c5027e9007c632d109f5e5ba9d7d2a0dbe09fa"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 14:09:06.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-04 18:58:46.000000000","tz":-420},"subject":"Add image_count_total quota enforcement","message":"Add image_count_total quota enforcement\n\nThis makes us enforce a quota on the total number of (non-deleted)\nimages owned by a user.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I8af124d9307263cd8289d0701fb9a745d13b1d56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ad0cedccaaced1d8207faa51c4e1bc2bf88f12c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ad0cedccaaced1d8207faa51c4e1bc2bf88f12c6"}]},"branch":"refs/heads/master"},"aadc24a0b06a293c04802e719edaa61ffd4a69ee":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-06-08 14:43:18.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/788327/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/788327/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/788327/9"}}},"commit":{"parents":[{"commit":"bc1bd374a38765c890dadc739f2416090870e682","subject":"Add user_get_image_count() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc1bd374a38765c890dadc739f2416090870e682"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 14:09:06.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-08 14:44:57.000000000","tz":-420},"subject":"Add image_count_total quota enforcement","message":"Add image_count_total quota enforcement\n\nThis makes us enforce a quota on the total number of (non-deleted)\nimages owned by a user.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I8af124d9307263cd8289d0701fb9a745d13b1d56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aadc24a0b06a293c04802e719edaa61ffd4a69ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aadc24a0b06a293c04802e719edaa61ffd4a69ee"}]},"branch":"refs/heads/master"},"4adad5e85a4d20df7b6bd64b01c52624a82760f2":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2021-06-14 14:40:58.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/788327/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/788327/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/788327/10"}}},"commit":{"parents":[{"commit":"c5f42f99385f9f1eb8c55408281661e0b5107e0e","subject":"Add user_get_image_count() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c5f42f99385f9f1eb8c55408281661e0b5107e0e"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 14:09:06.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-14 14:42:51.000000000","tz":-420},"subject":"Add image_count_total quota enforcement","message":"Add image_count_total quota enforcement\n\nThis makes us enforce a quota on the total number of (non-deleted)\nimages owned by a user.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I8af124d9307263cd8289d0701fb9a745d13b1d56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4adad5e85a4d20df7b6bd64b01c52624a82760f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4adad5e85a4d20df7b6bd64b01c52624a82760f2"}]},"branch":"refs/heads/master"},"88894a11da597d1e6d2be6170ca08c12816e9788":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2021-06-14 14:47:30.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/788327/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/788327/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/788327/11"}}},"commit":{"parents":[{"commit":"f4ebff18d5732800b97162c3fc5d3a4fbc7084ea","subject":"Add user_get_image_count() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f4ebff18d5732800b97162c3fc5d3a4fbc7084ea"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 14:09:06.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-14 14:49:32.000000000","tz":-420},"subject":"Add image_count_total quota enforcement","message":"Add image_count_total quota enforcement\n\nThis makes us enforce a quota on the total number of (non-deleted)\nimages owned by a user.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I8af124d9307263cd8289d0701fb9a745d13b1d56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/88894a11da597d1e6d2be6170ca08c12816e9788"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/88894a11da597d1e6d2be6170ca08c12816e9788"}]},"branch":"refs/heads/master"},"450a36d675735da0f8e2be774cc3ebad05a7410b":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2021-06-14 14:56:36.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/788327/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/788327/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/788327/12"}}},"commit":{"parents":[{"commit":"d18c63cb7de43fa7ddbd46b33aeedf9b97bd4c62","subject":"Add user_get_image_count() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d18c63cb7de43fa7ddbd46b33aeedf9b97bd4c62"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 14:09:06.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-14 14:58:35.000000000","tz":-420},"subject":"Add image_count_total quota enforcement","message":"Add image_count_total quota enforcement\n\nThis makes us enforce a quota on the total number of (non-deleted)\nimages owned by a user.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I8af124d9307263cd8289d0701fb9a745d13b1d56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/450a36d675735da0f8e2be774cc3ebad05a7410b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/450a36d675735da0f8e2be774cc3ebad05a7410b"}]},"branch":"refs/heads/master"},"27a191a4469ee054f3bf2e8c89b62d40979a964d":{"kind":"REWORK","_number":13,"created":"2021-06-14 15:24:35.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/788327/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/788327/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/788327/13"}}},"commit":{"parents":[{"commit":"d18c63cb7de43fa7ddbd46b33aeedf9b97bd4c62","subject":"Add user_get_image_count() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d18c63cb7de43fa7ddbd46b33aeedf9b97bd4c62"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 14:09:06.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-14 15:20:26.000000000","tz":-420},"subject":"Add image_count_total quota enforcement","message":"Add image_count_total quota enforcement\n\nThis makes us enforce a quota on the total number of (non-deleted)\nimages owned by a user.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I8af124d9307263cd8289d0701fb9a745d13b1d56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/27a191a4469ee054f3bf2e8c89b62d40979a964d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/27a191a4469ee054f3bf2e8c89b62d40979a964d"}]},"branch":"refs/heads/master"},"eb6cb4e382523334943b809aab9183a4ed15bc29":{"kind":"REWORK","_number":14,"created":"2021-06-21 15:46:41.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/788327/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/788327/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/788327/14"}}},"commit":{"parents":[{"commit":"a2378dbfa75b29d2f4f0f73b03ea30e72a341216","subject":"Add user_get_image_count() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a2378dbfa75b29d2f4f0f73b03ea30e72a341216"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 14:09:06.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-21 15:47:43.000000000","tz":-420},"subject":"Add image_count_total quota enforcement","message":"Add image_count_total quota enforcement\n\nThis makes us enforce a quota on the total number of (non-deleted)\nimages owned by a user.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I8af124d9307263cd8289d0701fb9a745d13b1d56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eb6cb4e382523334943b809aab9183a4ed15bc29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eb6cb4e382523334943b809aab9183a4ed15bc29"}]},"branch":"refs/heads/master"},"00e0f8bce700fc3148122176694bacfea935fc89":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2021-06-21 17:57:14.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/788327/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/788327/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/788327/15"}}},"commit":{"parents":[{"commit":"bb04ef466b48216939e48cbc5a6da323dcd8b8bb","subject":"Add user_get_image_count() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bb04ef466b48216939e48cbc5a6da323dcd8b8bb"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 14:09:06.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-21 17:59:19.000000000","tz":-420},"subject":"Add image_count_total quota enforcement","message":"Add image_count_total quota enforcement\n\nThis makes us enforce a quota on the total number of (non-deleted)\nimages owned by a user.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I8af124d9307263cd8289d0701fb9a745d13b1d56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/00e0f8bce700fc3148122176694bacfea935fc89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/00e0f8bce700fc3148122176694bacfea935fc89"}]},"branch":"refs/heads/master"},"a36666e2fe8d61a161daa1836ed84fb2e389dd67":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2021-06-29 14:51:33.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/27/788327/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/27/788327/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/27/788327/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/27/788327/16"}}},"commit":{"parents":[{"commit":"d3d6a646e32ffe8f59e17fa02b7b5ecf06d7a417","subject":"Add user_get_image_count() to DB API","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d3d6a646e32ffe8f59e17fa02b7b5ecf06d7a417"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-27 14:09:06.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-29 15:53:18.000000000","tz":-420},"subject":"Add image_count_total quota enforcement","message":"Add image_count_total quota enforcement\n\nThis makes us enforce a quota on the total number of (non-deleted)\nimages owned by a user.\n\nPartially-implements: blueprint glance-unified-quotas\n\nChange-Id: I8af124d9307263cd8289d0701fb9a745d13b1d56\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a36666e2fe8d61a161daa1836ed84fb2e389dd67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a36666e2fe8d61a161daa1836ed84fb2e389dd67"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
