)]}'
{"id":"openstack%2Fglance~788963","triplet_id":"openstack%2Fglance~master~I13597e8145920f9fe7cb9023a1d1c160bc8026a5","project":"openstack/glance","branch":"master","topic":"test-cinder-glance-store","hashtags":[],"change_id":"I13597e8145920f9fe7cb9023a1d1c160bc8026a5","subject":"DNM Test glance_store cinder attach changes","status":"ABANDONED","created":"2021-04-30 14:42:15.000000000","updated":"2022-05-05 14:20:53.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"809a54e5336ce1b5320d8b1f2581394fa20b1aac","_number":788963,"virtual_id_number":788963,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-05-03 15:35:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-30 15:17:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f68b63086a5011bea0a9afd151ce94d2b8ac6a1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-30 14:42:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0fca4e736d9fd8dc96aa2e879fa86d63b4a0f93f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 15:17:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2bd617fc8e3340d6ae3e7f9b7c49aa29 : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/450c575dc4c04aaa8bf768ad085fe6a7 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c3e049dc94640e39d80b0063094dc84 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e883a1e8e0bd4bbab8c93a0f0a3dfdfa : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19d86482d1744d92b0f39a6d6a19b5ce : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f889e873ae5746b49576913073b50ec3 : SUCCESS in 6m 07s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/af68978149e4464ea68ff35cf0d3c458 : SUCCESS in 8m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f90733d6225d412aa67b5d1b2d4dfc2b : SUCCESS in 10m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3d76072f4d62420f91a43810847cde73 : FAILURE in 21m 25s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0a4f1605bf7a4ee5a8f360007f04d031 : SUCCESS in 29m 25s","accounts_in_message":[],"_revision_number":1},{"id":"823e002be0c07b7014dace031cc0f79a16b56317","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-04-30 15:39:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"edf2976329257f91e2d54d95f4f57523814ea90f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-30 17:00:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bc302a1d0f6b42e4aec3013e3f8103d2 : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e142c4a7ea424c93b3818b912aac7deb : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9024fdf122747cba88084b5081edf08 : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84ed59797a2144a68c8eff089eb1a59f : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/217437bbdfac4fa3bbb4bef557ccd7d1 : SUCCESS in 6m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b1c14756c714bfaa8c77550ecce5a75 : SUCCESS in 6m 28s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/97a87bdb3f5043e8a173b24065a131c6 : SUCCESS in 8m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ff1044dba3ff414196d615cb6f08a7ed : SUCCESS in 8m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/74c3d110d12140c9ac39c2524b7a2668 : SUCCESS in 1h 16m 43s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1bd8dc516d6e4c4e9a50c082f128b237 : SUCCESS in 35m 30s","accounts_in_message":[],"_revision_number":2},{"id":"4a8ebda85a147460eed2906a26c755457526957b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-05-03 14:11:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f1fe351cd1220816f88fea9e2cfa9cd4c4fd7902","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 15:35:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/714b041d33464aaea04c80bfa97dee6e : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfc6628251b748e7b2b13196c94b67b5 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/decb75e53b9a4a739ea4904f591d7397 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49bcae479beb452483a6d4ecd9ed24f6 : SUCCESS in 7m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e6753201d734a53857e987fec2fb0e6 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65347ebfcd0c44f7aa3c09391899e18f : SUCCESS in 7m 46s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/591860a067934ccf84d1ae18f7f6db8b : SUCCESS in 10m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2c91cf1b5d5a4c83bfee8925f456282f : SUCCESS in 11m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/22fc7cb2fc6d4db3b7516705251ed247 : FAILURE in 1h 22m 15s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5873c7fab91d48f398eb502c32111c22 : SUCCESS in 25m 17s","accounts_in_message":[],"_revision_number":3},{"id":"809a54e5336ce1b5320d8b1f2581394fa20b1aac","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-05-05 14:20:53.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3054598752c5470a7882c31886caf436f27cffd9","revisions":{"dc4cbb0794a36c6dbb358b6e2f3571f66c5acaef":{"kind":"REWORK","_number":1,"created":"2021-04-30 14:42:15.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/63/788963/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/63/788963/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/788963/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/788963/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/788963/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/63/788963/1"}}},"commit":{"parents":[{"commit":"1157c19a9c1408faad5b845cc9b1873f82fc8f31","subject":"Merge \"Guidelines for core reviewers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1157c19a9c1408faad5b845cc9b1873f82fc8f31"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-30 14:42:46.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-30 14:42:46.000000000","tz":-420},"subject":"DNM Test glance_store cinder attach changes","message":"DNM Test glance_store cinder attach changes\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/782200\nChange-Id: I13597e8145920f9fe7cb9023a1d1c160bc8026a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc4cbb0794a36c6dbb358b6e2f3571f66c5acaef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dc4cbb0794a36c6dbb358b6e2f3571f66c5acaef"}]},"branch":"refs/heads/master"},"b31c51ab9ec3c4f8715d6b8c3d3bf7f54bef9c01":{"kind":"REWORK","_number":2,"created":"2021-04-30 15:39:00.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/63/788963/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/63/788963/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/788963/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/788963/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/788963/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/63/788963/2"}}},"commit":{"parents":[{"commit":"1157c19a9c1408faad5b845cc9b1873f82fc8f31","subject":"Merge \"Guidelines for core reviewers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1157c19a9c1408faad5b845cc9b1873f82fc8f31"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-30 14:42:46.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-30 15:40:24.000000000","tz":-420},"subject":"DNM Test glance_store cinder attach changes","message":"DNM Test glance_store cinder attach changes\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/782200\nChange-Id: I13597e8145920f9fe7cb9023a1d1c160bc8026a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b31c51ab9ec3c4f8715d6b8c3d3bf7f54bef9c01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b31c51ab9ec3c4f8715d6b8c3d3bf7f54bef9c01"}]},"branch":"refs/heads/master"},"3054598752c5470a7882c31886caf436f27cffd9":{"kind":"REWORK","_number":3,"created":"2021-05-03 14:11:27.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/63/788963/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/63/788963/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/788963/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/788963/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/788963/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/63/788963/3"}}},"commit":{"parents":[{"commit":"1157c19a9c1408faad5b845cc9b1873f82fc8f31","subject":"Merge \"Guidelines for core reviewers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1157c19a9c1408faad5b845cc9b1873f82fc8f31"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-04-30 14:42:46.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-05-03 14:12:41.000000000","tz":-420},"subject":"DNM Test glance_store cinder attach changes","message":"DNM Test glance_store cinder attach changes\n\nDepends-On: https://review.opendev.org/c/openstack/glance_store/+/782200\nDepends-On: https://review.opendev.org/c/openstack/os-brick/+/789289\nChange-Id: I13597e8145920f9fe7cb9023a1d1c160bc8026a5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3054598752c5470a7882c31886caf436f27cffd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3054598752c5470a7882c31886caf436f27cffd9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
