)]}'
{"id":"openstack%2Fglance~789281","triplet_id":"openstack%2Fglance~master~I23fbd227274c89735547322cedda565c5987776d","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"9268":{"account":{"_account_id":9268,"name":"Gonéri Le Bouder","email":"goneri@redhat.com","username":"goneri"},"last_update":"2022-07-05 20:15:31.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2022-07-05 20:15:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I23fbd227274c89735547322cedda565c5987776d","subject":"doc: update the kvm kw_vif_model driver list","status":"MERGED","created":"2021-05-03 13:26:53.000000000","updated":"2022-07-05 20:18:09.000000000","submitted":"2022-07-05 20:15:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"789281","meta_rev_id":"d6e25d691959e88d965b83fc992a2ef06f82f8f4","_number":789281,"virtual_id_number":789281,"owner":{"_account_id":9268,"name":"Gonéri Le Bouder","email":"goneri@redhat.com","username":"goneri"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9268,"name":"Gonéri Le Bouder","email":"goneri@redhat.com","username":"goneri"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-05 20:15:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":9268,"name":"Gonéri Le Bouder","email":"goneri@redhat.com","username":"goneri"},"all":[{"value":1,"date":"2022-07-05 14:29:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9268,"name":"Gonéri Le Bouder","email":"goneri@redhat.com","username":"goneri"},{"value":2,"date":"2022-07-05 18:28:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-07-05 17:52:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":9268,"name":"Gonéri Le Bouder","email":"goneri@redhat.com","username":"goneri"},{"value":1,"date":"2022-07-05 18:28:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9268,"name":"Gonéri Le Bouder","email":"goneri@redhat.com","username":"goneri"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9268,"name":"Gonéri Le Bouder","email":"goneri@redhat.com","username":"goneri"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-03 14:55:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-05 17:52:29.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2022-07-05 18:28:22.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"78bf9018d25ce1bebc22bb7f361d8a55d0136375","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9268,"name":"Gonéri Le Bouder","email":"goneri@redhat.com","username":"goneri"},"date":"2021-05-03 13:26:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98fef0f2fdabc2b7f0c9f50b1e900ba218a717bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-03 14:55:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a474c721b4f14c1aadda513bf144a291 : SUCCESS in 4m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/ec139323df964a95adc761bf53ea3590 : SUCCESS in 1h 16m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cd3c0ffbd5e040299682c03b81d81b0e : SUCCESS in 1h 20m 12s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a1c128c94f7f4e5492ec2a0b7178be65 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a26c743b706241b8ac3485e8dd23f2b9 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17b5603597d441c4883f3b0b8cc6de5c : SUCCESS in 5m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/623b88836a934a559061276803bcedaf : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a3a1399d5a64ffe8486babeede56f04 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c1e0a8ccbf84114bdd67f8b360d1fa1 : SUCCESS in 6m 34s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/bd5b14417f214c058718b2592fccf096 : SUCCESS in 7m 56s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f7ac1d0123cb48e785245e00a5cefc9d : SUCCESS in 8m 17s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/92abe2aa17ab44ea9b94a826b4c4d9b9 : SUCCESS in 1h 22m 24s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0f99393e6bf344749058ea179e842b0d : SUCCESS in 58m 12s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/22e21ee10d084165b82ab1f216e82885 : SUCCESS in 1h 17m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/22c0c237764a406995f99d98be8bee78 : SUCCESS in 1h 19m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c9d7b583f6b940ce92567d9389be1c86 : SUCCESS in 1h 03m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2edb0d461bac4573b6ed2f43eac0deb8 : SUCCESS in 1h 06m 36s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4e89a4ba4b8e4fa4bdb14f47a82a0a0b : SUCCESS in 35m 48s","accounts_in_message":[],"_revision_number":1},{"id":"186c3ce3d42c00fb06264ffd8d64210c5270cf11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-04 20:42:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"705091fa5f3feed5d0b99a9d3b6c4eacc07c8361","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 22:11:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fae690ef8ce84f7981f30966222b7d8f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/af0f257059be4f45bd97eeaca0606a87 : SUCCESS in 5m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/73abf73d64d24edea0d39c066c2f950f : SUCCESS in 50m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1bc09881ad3a4e038266ecda50dbba37 : SUCCESS in 1h 20m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b88adf0b0c34158a7161a8764a3866c : SUCCESS in 45m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c8f0115ae5f44cfbfa856201dcf6355 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81b43b4627224bedae9c9e2e91527cc0 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/877e2e1e1bc34b00bb7f7b9149fdad01 : SUCCESS in 6m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ce2109de5d6a4ed8abf56ffd4ea6724d : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72e71b25f1ea4c90907dcf396fd1cfbb : SUCCESS in 6m 33s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/f5df04c562064b7aabffd3f0d0a25f4d : SUCCESS in 11m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fe3499a80bcc43ef96eecac819ebc591 : SUCCESS in 8m 43s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d56764804b464917a567f4d9300dbf63 : SUCCESS in 10m 24s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0f38f08fee814a1dbdb009eb99d67c90 : SUCCESS in 1h 15m 05s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/252914e838464263af8b52edd8deecc0 : SUCCESS in 1h 24m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9516201878ae4ab2af9c7a2c4ced5fd2 : SUCCESS in 1h 18m 18s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/93b413232b1c48aa9e2784de21a745cf : SUCCESS in 1h 17m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16659c5c6a544c09aac67df08fd28238 : SUCCESS in 43m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1a7ff853c36c492b954e274e02a4c527 : SUCCESS in 1h 03m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8be351cee08448d69b288368799fae59 : SUCCESS in 32m 50s","accounts_in_message":[],"_revision_number":2},{"id":"2013b73d04e7426052e0dc4fd4b7342d09c3987e","author":{"_account_id":9268,"name":"Gonéri Le Bouder","email":"goneri@redhat.com","username":"goneri"},"date":"2022-07-05 14:29:00.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"a063b8790e7020bd784ff674ab734e96867cb894","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2022-07-05 17:52:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0741b337f3b8d6ea438facab687183237efb3e4b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-07-05 18:28:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7d3eae4983dc15d730018514d0f0e953978ad8dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 18:29:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"98607266579d24b6b2218964e9a47a87e787f3c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 20:15:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c76d1af7f5a543a19c2707a2cfcfb978\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/67b271fb417c4a5ca813b50233450ffa : SUCCESS in 5m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/c2752a2dc3594399b1adf0f1f48baf23 : SUCCESS in 54m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/39e80ba5531140c2975f17c4bcc7d57c : SUCCESS in 1h 08m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/499ea8260a3c4e56b3c5d9de4d978397 : SUCCESS in 45m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6da84e4fb800490b9bc090b1c61a4369 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d573df5a781d4de5ae4a3bc6a4597dd5 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e250dbc302b46a9879ff6e1907e460d : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5821a037b274fff8908fd08be1ff562 : SUCCESS in 6m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/62ffebbe39bd4c218bf412674737132e : SUCCESS in 8m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/856e4dcb46bf4e0a8c82be64b8942db5 : SUCCESS in 10m 11s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/58f35c99023a46c7931dc47e79e92964 : SUCCESS in 1h 10m 25s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3a9a585cc6a5494b96598dfd7d89520c : SUCCESS in 1h 00m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6cf27447bfe4ba38a3c003681202e90 : SUCCESS in 42m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f29fcd05d0da416dbbee333c64d6734d : SUCCESS in 1h 05m 01s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2d897e83b5d54b379b571e8e1da34b10 : SUCCESS in 27m 13s","accounts_in_message":[],"_revision_number":2},{"id":"bb7f0247c57da20fcbebab8867d20be4a1db21b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 20:15:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d6e25d691959e88d965b83fc992a2ef06f82f8f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 20:18:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95b25aa9890e4b98861f283cb837ab00\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/0c4c0433d9084fab95080986a61956ae : SUCCESS in 51s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fcfa1dc4b964a8c9746377e27cd4247 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d883e4fd5b9bfceb90abdbb53ea6878920f91055","revisions":{"d33b8a00ea99faf61cbfe6b79d1a9028a80af86c":{"kind":"REWORK","_number":1,"created":"2021-05-03 13:26:53.000000000","uploader":{"_account_id":9268,"name":"Gonéri Le Bouder","email":"goneri@redhat.com","username":"goneri"},"ref":"refs/changes/81/789281/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/81/789281/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/789281/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/789281/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/789281/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/81/789281/1"}}},"commit":{"parents":[{"commit":"1157c19a9c1408faad5b845cc9b1873f82fc8f31","subject":"Merge \"Guidelines for core reviewers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1157c19a9c1408faad5b845cc9b1873f82fc8f31"}]}],"author":{"name":"Gonéri Le Bouder","email":"goneri@lebouder.net","date":"2021-05-03 13:22:50.000000000","tz":-240},"committer":{"name":"Gonéri Le Bouder","email":"goneri@lebouder.net","date":"2021-05-03 13:26:52.000000000","tz":-240},"subject":"doc: update the kvm kw_vif_model driver list","message":"doc: update the kvm kw_vif_model driver list\n\nThe kw_vif_model driver list has evolved a bit. This commit adds:\n- e1000e: I12a5d28d75bc32a76a4f3765cb4db4cbc46c0c75\n- vmxnet3 I95203af4ded9a4dbecfb7981b400f0fb988f2d47\n\nChange-Id: I23fbd227274c89735547322cedda565c5987776d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d33b8a00ea99faf61cbfe6b79d1a9028a80af86c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d33b8a00ea99faf61cbfe6b79d1a9028a80af86c"}]},"branch":"refs/heads/master"},"d883e4fd5b9bfceb90abdbb53ea6878920f91055":{"kind":"REWORK","_number":2,"created":"2022-07-04 20:42:15.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/81/789281/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/81/789281/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/789281/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/789281/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/789281/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/81/789281/2"}}},"commit":{"parents":[{"commit":"a71f7ba44e57a7945addb82afd2a19962d5c04cc","subject":"[APIImpact] Correct API response code for DELETE cache APIs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a71f7ba44e57a7945addb82afd2a19962d5c04cc"}]}],"author":{"name":"Gonéri Le Bouder","email":"goneri@lebouder.net","date":"2021-05-03 13:22:50.000000000","tz":-240},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2022-07-04 20:39:25.000000000","tz":120},"subject":"doc: update the kvm kw_vif_model driver list","message":"doc: update the kvm kw_vif_model driver list\n\nThe kw_vif_model driver list has evolved a bit. This commit adds:\n- e1000e: I12a5d28d75bc32a76a4f3765cb4db4cbc46c0c75\n- vmxnet3 I95203af4ded9a4dbecfb7981b400f0fb988f2d47\n\nChange-Id: I23fbd227274c89735547322cedda565c5987776d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d883e4fd5b9bfceb90abdbb53ea6878920f91055"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d883e4fd5b9bfceb90abdbb53ea6878920f91055"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
