)]}'
{"id":"openstack%2Fglance~789914","triplet_id":"openstack%2Fglance~master~Ib6540223468421776f07745010e5d4afe5d58774","project":"openstack/glance","branch":"master","topic":"metadef-policy-poc","hashtags":[],"change_id":"Ib6540223468421776f07745010e5d4afe5d58774","subject":"Make property protection tests use member role","status":"MERGED","created":"2021-05-05 15:59:21.000000000","updated":"2021-08-05 21:53:13.000000000","submitted":"2021-07-16 16:18:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"789914-metadef-policy-poc","meta_rev_id":"bb27be369eaaa12792bf66f88dc85bc846bba194","_number":789914,"virtual_id_number":789914,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-16 16:18:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2021-08-05 21:53:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2021-07-16 14:37:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2021-07-16 14:37:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}],"reviewers":{"REVIEWER":[{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-05 16:30:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-10 11:52:59.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2021-05-10 11:52:59.000000000","updated_by":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"reviewer":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"state":"REVIEWER"},{"updated":"2021-05-11 15:20:44.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2021-05-21 13:57:37.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2021-05-27 14:37:54.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"CC"},{"updated":"2021-06-25 10:47:50.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2021-07-14 16:16:05.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-07-16 14:37:33.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"6611a556c9e70e586f2e3c356b9096c8d4a2159f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-05-05 15:59:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d0a4b98e618cbfe3e9420b19f4714301020f746","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 16:30:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47685544982e4d9fa4c466b2b55d73f3 : SUCCESS in 10m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9eb1c82c918344b89d0de1ea37407c53 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b532d5dab9a747bb90dc4df82edffe95 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/253624b98d374980a452e84bc9f85bb5 : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3cc3d30532274d438de9a0d8945b4395 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0546de02acf34704b03c84d9ae117ae2 : SUCCESS in 6m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4c5e5d6e6ee44d71835c30eb477a9771 : SUCCESS in 8m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/48b74ee0b4c7467482a19c268618a63d : SUCCESS in 8m 15s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/e5844b64a98440f6bd9fd3ae3a625c4c : SUCCESS in 5m 00s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/63a9c6822583480d8a30bd4b81ef322d : SUCCESS in 26m 08s","accounts_in_message":[],"_revision_number":1},{"id":"eadbec2c750b57384c48b4e111dbbbaf78bd5785","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-05-11 15:20:44.000000000","message":"Patch Set 1:\n\nNeed some time to understand in detail why some tests were asserting for Conflict.","accounts_in_message":[],"_revision_number":1},{"id":"582b256c327e1d8d734ef09c48172ed42c41369e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2021-05-21 13:57:37.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)\n\nThis change makes sense, thanks for the detailed explanation in the commit message.\n\nI\u0027ve been looking at this for over an hour and can\u0027t determine a reason why the calls you flagged should be receiving 409 instead of 403.  Forbidden seems correct.","accounts_in_message":[],"_revision_number":1},{"id":"6e449329bc42d832892c0fbde30a49221cd0d7f2","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-05-21 14:24:42.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nWell, glad I stumped ya\u0027ll at least :)","accounts_in_message":[],"_revision_number":1},{"id":"0935cff789ce04d71f701b779a63e051daee20c8","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-05-27 14:37:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1cfdbc15b19e081710ff8e38afebb79844b94503","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-05-27 16:24:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"077ae1955b95d3725d0b56bb101b61a576cd3be3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-02 19:24:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"94da001b6649b4a631fcdf9ed713190a3b462260","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 20:03:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/996b879fad914393ac1616048aaf711b : SUCCESS in 9m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c25106f7b1c41b9b9f99e12a35a97fc : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14cec027332a4e66878c44edf27fdfa9 : SUCCESS in 7m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66fb07944ba8441d8885c86c87c192dc : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5ac87ccd5754117a9f9a1503eec0022 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ae3e63fe29a47389a2fc5547c63de1d : SUCCESS in 6m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f5d0ba38a9ae42688eabb3ee00034f00 : SUCCESS in 8m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b869368ea4c14c3f97e0f25aacb42a70 : SUCCESS in 8m 16s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/00f94805f37742e7b518c48d105b3c07 : SUCCESS in 4m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fb2dc621b4ed400dbea896370de59cce : SUCCESS in 33m 51s","accounts_in_message":[],"_revision_number":2},{"id":"68517b9fd2caae7f74daf91176001516f533fdb3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-03 21:13:55.000000000","message":"Patch Set 2:\n\n(3 comments)\n\nOops, forgot to commit these draft comments.","accounts_in_message":[],"_revision_number":2},{"id":"37511e75166c52ae66487443b54973a4a8749bc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-11 21:39:01.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"37602946765fab806bde50ec7478cd4ae26ed3a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-11 22:18:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/519b3803fce4440d81bc5bc77eaa0b9c : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c4cf7851442477e94c58ee0e30f0eb8 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d881c145e7fd431897e38c7f26bd32f1 : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/726089c1d576493b88d0a829b3586427 : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db1fd80abc4b458290f42e6a660915a0 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9be742d362d2467aa5d232efee6b97db : SUCCESS in 7m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/06aedf31457e4be39f73401583ae04e7 : SUCCESS in 9m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6b58561467d24601a8865c7adbccf959 : SUCCESS in 9m 33s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/53a0fd572631493ebecb579119273242 : SUCCESS in 5m 18s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5afa5ca87a4148dcac33f70d23af8f4f : SUCCESS in 33m 34s","accounts_in_message":[],"_revision_number":3},{"id":"998bad4de7fcfe45c177774b1e6ba1d49213947d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-23 17:07:38.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"f83ef65814f36cea5b0c05512453c762b62a58c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-23 18:33:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5dce8bf73c641589c3e445bf7dad6ab : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc14b117a556410bade249391f2c1e08 : SUCCESS in 6m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0c8db2dc10c4dbd8f52d9ecdc15bcda : SUCCESS in 7m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1cd18f9e0a634812840e100b48e070fe : SUCCESS in 8m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5fedade34c244f97a8cdacf608aebe1f : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3909d3ec454b4f0590c0b3c52e29970a : SUCCESS in 8m 43s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3924a4783e1146fbb8762659c3b63562 : SUCCESS in 8m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4d82812b88e84126ad197e9acc9f2c6e : SUCCESS in 10m 19s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7a51d73f28a8472a8223364b82bc9fec : SUCCESS in 6m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8767f34ab5a34b84953b1a07d5196e23 : SUCCESS in 34m 30s","accounts_in_message":[],"_revision_number":4},{"id":"9fde98e4fe9f7b75001e75cec2be44380b5bff34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-06-25 10:47:50.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"d93522e27fd333a745c2bd54b2880425a120cae4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-25 11:26:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6cfe6501eceb40158e51f67cc59504b7 : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d08de2dfa55f47bdbe47246302115c57 : SUCCESS in 8m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe94a9dca0504585a3f236a88868d706 : SUCCESS in 8m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2a14b4f6f5f47b8b25f7565e8ddf45f : SUCCESS in 9m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3afd851948a445efb863eada91ce4f0e : SUCCESS in 9m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c2adfdfe4c04922abf67b7879996504 : SUCCESS in 9m 23s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/3e488e6932234a83b27a736c5af52a8c : SUCCESS in 11m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/217269c6d94e4051b81bf83b5858073f : SUCCESS in 13m 07s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/0a353087ac564ed48db273b0aef3f8a0 : SUCCESS in 7m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/32c4c0259a7b460796e8bf4156ebf321 : SUCCESS in 36m 16s","accounts_in_message":[],"_revision_number":5},{"id":"8a7fef6d104fb93ebadabe7cbff1241d714a0a84","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-06-28 05:38:59.000000000","message":"Topic delete-policy-poc removed","accounts_in_message":[],"_revision_number":5},{"id":"fb029b461e39dd8257411132321c336d50072c58","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-06-28 05:39:07.000000000","message":"Topic set to policy-poc","accounts_in_message":[],"_revision_number":5},{"id":"e7e5bf848a7546fe026996bfee327d9de5635b1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-06-28 17:48:24.000000000","message":"Uploaded patch set 6: New patch set was added with same tree, parent, and commit message as Patch Set 5.","accounts_in_message":[],"_revision_number":6},{"id":"c99ca0825e8cc1f6ba15fb99b74fb9de2efcdfe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-28 18:27:05.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36a3c5a1729a4dcf80053adf664f67d1 : SUCCESS in 9m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e0880bdcf0c4e7a98a811e1544011ed : SUCCESS in 8m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a18ce87515524bbe99620539ea3a1101 : SUCCESS in 8m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e693083c9799484b9c264aa50acc1b59 : SUCCESS in 9m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23f16b8e22f54fef982c289a5c6505e2 : SUCCESS in 8m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/993068c10c934ba7a41bec03e835d84f : SUCCESS in 10m 48s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/330a5c3e82fe45c28c5df06b20c419e3 : SUCCESS in 11m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c414fcf8832447f7808717cb390107e3 : SUCCESS in 10m 54s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/bf18dab0b4e648acb685c3cd94011e7a : SUCCESS in 9m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f515f2a9a23842078f02073ac688e14d : SUCCESS in 35m 59s","accounts_in_message":[],"_revision_number":6},{"id":"4c04586d2398cc3048a7bd3f8045eb8ca0040386","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-02 14:49:29.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"fefffddb671e709e2e50ac63863e956669975d29","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-02 15:41:17.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c70d6e9e7d74324a72b64f3de4575f0 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47c28a70894a4823bc95643624ff5e5d : SUCCESS in 7m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/362dcc8ba850466b9a8dce53257922e6 : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9266dc86697436eaef3bd2bea3590ac : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/602cd3686954494baeda5e838d474be7 : SUCCESS in 7m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/4ead51d15c174caf97485eaf19524db0 : SUCCESS in 11m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0f963ef26a6040578a931b82456f2cec : SUCCESS in 11m 10s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/46b99155705e44feb2fe6fdd29722e95 : FAILURE in 12m 45s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/8bb96c5add0e4eaa99bb002d513a03bf : SUCCESS in 5m 20s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6174c24490ca41efa1880860b3f47d14 : SUCCESS in 27m 52s","accounts_in_message":[],"_revision_number":7},{"id":"6b5ad32124df461c62e8d411173f6d1499145d8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-06 10:51:18.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"f7b8304e6598c47307c49ce62b33e0dfd6d79d5a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-06 10:54:29.000000000","message":"Topic metadef-policy-poc removed","accounts_in_message":[],"_revision_number":8},{"id":"e843d9a821d08a2df66ce6d49da03a9cfa282245","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-06 10:54:37.000000000","message":"Topic set to policy-poc","accounts_in_message":[],"_revision_number":8},{"id":"b0297f45fc6bba90e2a91db89ce06adda915ec93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 11:33:16.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8142c469b16a49dbade30fad5074ada1 : SUCCESS in 11m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90b6815914564ab69e30ba85c7798590 : SUCCESS in 11m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8991ef5ced224e688ca1b1a351eb1860 : SUCCESS in 12m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52336648a93b46a7856d4dd442f53a9a : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b12e1063f49409c958ef8d7c5d875b5 : SUCCESS in 7m 26s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f196be657cc5476794343b4ca5b083e6 : FAILURE in 9m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2a8a87cc79714c9a9e75d9a116aee38a : SUCCESS in 14m 40s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/2570414f917b4c1a886ab59eec725f50 : FAILURE in 11m 00s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f77b3d471ba14a03b6f1b7fa56c953b4 : SUCCESS in 11m 24s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b65993586a434aedbc5f27201c6f4958 : SUCCESS in 32m 21s","accounts_in_message":[],"_revision_number":8},{"id":"98b416373485f248f04b257284e43d620284af3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-06 18:00:34.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"4cb9f7b0c8d483690ccb6427576adc02af902810","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 18:30:13.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a20eaefaa98a41798fedbcdb18b04afa : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d463b7f3d8214f13b0ba94a21f013f0c : SUCCESS in 8m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/082e0ccd426b4c5ab56ae61bf420d8d7 : SUCCESS in 7m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a59c4a1dd0a043c88d8d7d922afd85bc : SUCCESS in 6m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/865206dcb2f24321862122bac617c132 : SUCCESS in 7m 53s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9460908262374f0d8dd6fb169e653146 : SUCCESS in 11m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c87dacb8307e489abf9a7dc62cc0eda7 : SUCCESS in 10m 07s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/3f45236c93fe44deaf7f85bda1175235 : FAILURE in 13m 05s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/7e5e5d8c7fa04b6bbbcdf98af296b689 : SUCCESS in 6m 24s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8317c197e50b4d2992c00b23bc1f881f : SUCCESS in 26m 46s","accounts_in_message":[],"_revision_number":9},{"id":"cf476d40e4283567d385b49c5950edffa13a8773","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-14 14:03:38.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"79b91f075e7854ad842b7d0ac68c6bce870975ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 14:34:36.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38747f6fe867488d85587b6945f4d551 : SUCCESS in 6m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1893963f3dd4006bafbf1c6ac7db03a : SUCCESS in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9dfa13681c3841779248118ddb4a5e45 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47ddbc6776a143cda1b89c6582732b21 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99d8651c92bf4c1da81e09dd29d8c3cb : SUCCESS in 7m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/e849b25f51b149eda282529c15473f4c : SUCCESS in 10m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c7f7a087a22e48cca3d3d3285b945f71 : SUCCESS in 14m 13s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/fcc8fe481cb34da7a60a00dae2ee9c1a : FAILURE in 8m 56s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/ac094da5491241dab91fd5f984ed1591 : SUCCESS in 5m 57s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d565dd22e2a543c4a470df0537152dac : SUCCESS in 28m 48s","accounts_in_message":[],"_revision_number":10},{"id":"744b773123effa95ea60412ca506515b23834e54","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-14 16:16:05.000000000","message":"Patch Set 10:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/789914/789914/10/check-not-vote/ext-glance-zuul/a46832f : SUCCESS in 36m 52s","accounts_in_message":[],"_revision_number":10},{"id":"dcac2d73532c942c1d9582bcbf6664e8248ba641","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-14 19:32:52.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"dfda114557e8794bc76c7e0cc07017d88663f49f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-14 20:05:46.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea2295c5491b47faa9e97050b91fe31e : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0687e3409726466888c58437738c1b0d : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/57046c0132644473a58d3caa10d54cd7 : SUCCESS in 6m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f59bb1ab7dda4d94aa111116279a344c : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aecc1a5db1ec45e390cabec372e9b6f7 : SUCCESS in 6m 58s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/addef37fa70f46549520b97bfafd6707 : SUCCESS in 8m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d5ed94d0761644fa857547239407e50f : SUCCESS in 9m 40s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/430e7194b35a4d959f0cd00aab1d9a18 : SUCCESS in 8m 21s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/f2d79d0139e74237838f44c804d8b727 : SUCCESS in 5m 49s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4e02799fb16f46f79c1eaa9b173e45b2 : SUCCESS in 30m 04s","accounts_in_message":[],"_revision_number":11},{"id":"981ff970171a52227a7038865687202349b52a69","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-07-14 20:11:07.000000000","message":"Patch Set 11:\n\nBuild succeeded\n\n- dsvm-glance http://207.189.188.190/logs/789914/789914/11/check-not-vote/ext-glance-zuul/90cebdc : SUCCESS in 36m 11s","accounts_in_message":[],"_revision_number":11},{"id":"ba3d59b677f0990baa30efa22516121a56988e9e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2021-07-16 13:41:20.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"866f5b9078b505345854109018d358de4fd19765","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-07-16 13:56:40.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"8fd3b1c6938f31438b8ad7d2791f160cc6b90f3e","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2021-07-16 14:37:33.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"cb63331cf29933f4bc81f72e619ebfdb9ccd72b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 14:44:24.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8122db3e66014cc191a35e7514f46403 : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db069cf8d3d84ff689b857d81e33d2e4 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75051862cac14f93a66d448f3f9b394d : SUCCESS in 7m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc7e55a166c248da9a5166d64768e476 : SUCCESS in 8m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/055d65f29dbf489aa2ea8857e24d3e86 : SUCCESS in 7m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c81bffc2ddec4488933a30df3dc6d752 : SUCCESS in 8m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2287458e27e24b0085bbd1e98c260c4a : SUCCESS in 9m 07s\n- glance-tox-functional-py38-rbac-defaults https://zuul.opendev.org/t/openstack/build/7aed4e71891a4acf8cb00cc5c8e03440 : SUCCESS in 10m 04s (non-voting)\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/61b7389fe987409da8a7daf03a8dfd3c : SUCCESS in 5m 19s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/db709efe945749fe908161e41204bd80 : SUCCESS in 38m 28s","accounts_in_message":[],"_revision_number":12},{"id":"85f8a59ded4dd9bec13faf714a811359604c70e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 14:44:48.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"0bbd508a323b2cb7e4764dd75111b0c400b28b92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 16:18:29.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b75db18795c4ce6a9f0a27de960c871 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/daa9544eb794435aa8e08dbad74d68f1 : SUCCESS in 13m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b66aca91f964deda7e6226a28c9133a : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32f61c4fbb9e4c5fa3d6675ce3b46541 : SUCCESS in 7m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/9fa8ed3d607d42f1a095e38a4380fec7 : SUCCESS in 9m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/eb6f8a378faa40fe826689951d716bdf : SUCCESS in 9m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2f36928e49a14e4d8e2711299751047a : SUCCESS in 32m 43s","accounts_in_message":[],"_revision_number":12},{"id":"4419e57ed60ce5219cbd2b7a9376b5f8290f3178","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 16:18:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"c384f97ea975e67f18b7106ca773d24fd55becdc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-16 16:22:07.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a3d234076d946278a3c24d40f4b5ec3 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":12},{"id":"e569c2f40b4e43dd7d0eda8ca39d91da0d0d6d99","author":{"_account_id":5046,"name":"Lance Bragstad","email":"lbragstad@redhat.com","username":"ldbragst"},"date":"2021-08-05 21:53:13.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"76162a7222c3d2c5eeae03ca7a710cd6401b36da","revisions":{"14451ba55c8679995bc7d23cb23f789806322d82":{"kind":"REWORK","_number":1,"created":"2021-05-05 15:59:21.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/789914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/789914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/789914/1"}}},"commit":{"parents":[{"commit":"bc96b33d014599ae8a1ff0ccd694f1afce370eb2","subject":"WIP: Refactor gateway get_repo auth layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc96b33d014599ae8a1ff0ccd694f1afce370eb2"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-05-05 15:21:08.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-05-05 15:54:28.000000000","tz":-420},"subject":"WIP: Make property protection tests use member","message":"WIP: Make property protection tests use member\n\nIn order to avoid overriding the default policy in tests to be\ncompletely unrestricted (as they are today), the property protection\ntests need to add the member role into the roles being used to\ntest the additional permissions. This is because the actual default\npolicies require \"admin or project member\" permissions to do anything,\nand just using one of the special roles will get blocked by the\nregular policy before property protections are checked.\n\nThis could also be fixed by constructing a very complex policy for\nthese tests which allows those special roles to do their thing in\naddition to member, but I do not believe this is how real deployments\nwould actually implement it.\n\nThese should definitely be reviewed with scrutiny to make sure that\nall the assertions are still valid, and make sure we don\u0027t need more\nthan are already here.\n\nNote that there are a couple assertions that were looking for CONFLICT\ninstead of FORBIDDEN on the special role attempt, but I think that\nis likely a case of \"test the current behavior not the desired one\",\nbut I could be mistaken.\n\nChange-Id: Ib6540223468421776f07745010e5d4afe5d58774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/14451ba55c8679995bc7d23cb23f789806322d82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/14451ba55c8679995bc7d23cb23f789806322d82"}]},"branch":"refs/heads/master"},"f6ac5139087590cd8e0503a4af0f4ae35a878f28":{"kind":"REWORK","_number":2,"created":"2021-06-02 19:24:08.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/789914/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/789914/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/789914/2"}}},"commit":{"parents":[{"commit":"d8fae109dcbbf166f88f2af8061a91c4d50a25ee","subject":"Refactor gateway get_repo auth layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d8fae109dcbbf166f88f2af8061a91c4d50a25ee"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-05-05 15:21:08.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-02 19:22:30.000000000","tz":-420},"subject":"Make property protection tests use member role","message":"Make property protection tests use member role\n\nIn order to avoid overriding the default policy in tests to be\ncompletely unrestricted (as they are today), the property protection\ntests need to add the member role into the roles being used to\ntest the additional permissions. This is because the actual default\npolicies require \"admin or project member\" permissions to do anything,\nand just using one of the special roles will get blocked by the\nregular policy before property protections are checked.\n\nThis could also be fixed by constructing a very complex policy for\nthese tests which allows those special roles to do their thing in\naddition to member, but I do not believe this is how real deployments\nwould actually implement it.\n\nThese should definitely be reviewed with scrutiny to make sure that\nall the assertions are still valid, and make sure we don\u0027t need more\nthan are already here.\n\nNote that there are a couple assertions that were looking for CONFLICT\ninstead of FORBIDDEN on the special role attempt, that seems to be\na case of \"test the current behavior not the desired one\".\n\nChange-Id: Ib6540223468421776f07745010e5d4afe5d58774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f6ac5139087590cd8e0503a4af0f4ae35a878f28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f6ac5139087590cd8e0503a4af0f4ae35a878f28"}]},"branch":"refs/heads/master"},"3fdfdc8e60fa2061f0f19b3f54779b7e177315cf":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-06-11 21:39:01.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/789914/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/789914/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/789914/3"}}},"commit":{"parents":[{"commit":"716bde2845dcdc9302422124f45ab55f047fed11","subject":"Refactor gateway get_repo auth layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/716bde2845dcdc9302422124f45ab55f047fed11"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-05-05 15:21:08.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-11 14:05:01.000000000","tz":-420},"subject":"Make property protection tests use member role","message":"Make property protection tests use member role\n\nIn order to avoid overriding the default policy in tests to be\ncompletely unrestricted (as they are today), the property protection\ntests need to add the member role into the roles being used to\ntest the additional permissions. This is because the actual default\npolicies require \"admin or project member\" permissions to do anything,\nand just using one of the special roles will get blocked by the\nregular policy before property protections are checked.\n\nThis could also be fixed by constructing a very complex policy for\nthese tests which allows those special roles to do their thing in\naddition to member, but I do not believe this is how real deployments\nwould actually implement it.\n\nThese should definitely be reviewed with scrutiny to make sure that\nall the assertions are still valid, and make sure we don\u0027t need more\nthan are already here.\n\nNote that there are a couple assertions that were looking for CONFLICT\ninstead of FORBIDDEN on the special role attempt, that seems to be\na case of \"test the current behavior not the desired one\".\n\nChange-Id: Ib6540223468421776f07745010e5d4afe5d58774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3fdfdc8e60fa2061f0f19b3f54779b7e177315cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3fdfdc8e60fa2061f0f19b3f54779b7e177315cf"}]},"branch":"refs/heads/master"},"439af9f93472a789bff96c572bb6a64fc62dd6fb":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-06-23 17:07:38.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/789914/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/789914/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/789914/4"}}},"commit":{"parents":[{"commit":"19078269a4452ae884409334867a9567da420f96","subject":"Refactor gateway get_repo auth layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/19078269a4452ae884409334867a9567da420f96"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-05-05 15:21:08.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-23 17:09:42.000000000","tz":-420},"subject":"Make property protection tests use member role","message":"Make property protection tests use member role\n\nIn order to avoid overriding the default policy in tests to be\ncompletely unrestricted (as they are today), the property protection\ntests need to add the member role into the roles being used to\ntest the additional permissions. This is because the actual default\npolicies require \"admin or project member\" permissions to do anything,\nand just using one of the special roles will get blocked by the\nregular policy before property protections are checked.\n\nThis could also be fixed by constructing a very complex policy for\nthese tests which allows those special roles to do their thing in\naddition to member, but I do not believe this is how real deployments\nwould actually implement it.\n\nThese should definitely be reviewed with scrutiny to make sure that\nall the assertions are still valid, and make sure we don\u0027t need more\nthan are already here.\n\nNote that there are a couple assertions that were looking for CONFLICT\ninstead of FORBIDDEN on the special role attempt, that seems to be\na case of \"test the current behavior not the desired one\".\n\nChange-Id: Ib6540223468421776f07745010e5d4afe5d58774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/439af9f93472a789bff96c572bb6a64fc62dd6fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/439af9f93472a789bff96c572bb6a64fc62dd6fb"}]},"branch":"refs/heads/master"},"d6c6545ffa84bf13fc625c20a08e21189c16983c":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-06-25 10:47:50.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/14/789914/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/789914/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/789914/5"}}},"commit":{"parents":[{"commit":"44fd0e8eff8b54870dc12d72226798122bb850d9","subject":"Refactor gateway get_repo auth layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/44fd0e8eff8b54870dc12d72226798122bb850d9"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-05-05 15:21:08.000000000","tz":-420},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-06-25 06:54:33.000000000","tz":0},"subject":"Make property protection tests use member role","message":"Make property protection tests use member role\n\nIn order to avoid overriding the default policy in tests to be\ncompletely unrestricted (as they are today), the property protection\ntests need to add the member role into the roles being used to\ntest the additional permissions. This is because the actual default\npolicies require \"admin or project member\" permissions to do anything,\nand just using one of the special roles will get blocked by the\nregular policy before property protections are checked.\n\nThis could also be fixed by constructing a very complex policy for\nthese tests which allows those special roles to do their thing in\naddition to member, but I do not believe this is how real deployments\nwould actually implement it.\n\nThese should definitely be reviewed with scrutiny to make sure that\nall the assertions are still valid, and make sure we don\u0027t need more\nthan are already here.\n\nNote that there are a couple assertions that were looking for CONFLICT\ninstead of FORBIDDEN on the special role attempt, that seems to be\na case of \"test the current behavior not the desired one\".\n\nChange-Id: Ib6540223468421776f07745010e5d4afe5d58774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d6c6545ffa84bf13fc625c20a08e21189c16983c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d6c6545ffa84bf13fc625c20a08e21189c16983c"}]},"branch":"refs/heads/master"},"1f8c317ed87b18198b0e91504e464cda746a26b7":{"kind":"NO_CHANGE","_number":6,"created":"2021-06-28 17:48:24.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/789914/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/789914/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/789914/6"}}},"commit":{"parents":[{"commit":"2b9fe30acd4b7fd92782bbe262eb764d245b4482","subject":"Refactor gateway get_repo auth layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b9fe30acd4b7fd92782bbe262eb764d245b4482"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-05-05 15:21:08.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-06-28 17:50:33.000000000","tz":-420},"subject":"Make property protection tests use member role","message":"Make property protection tests use member role\n\nIn order to avoid overriding the default policy in tests to be\ncompletely unrestricted (as they are today), the property protection\ntests need to add the member role into the roles being used to\ntest the additional permissions. This is because the actual default\npolicies require \"admin or project member\" permissions to do anything,\nand just using one of the special roles will get blocked by the\nregular policy before property protections are checked.\n\nThis could also be fixed by constructing a very complex policy for\nthese tests which allows those special roles to do their thing in\naddition to member, but I do not believe this is how real deployments\nwould actually implement it.\n\nThese should definitely be reviewed with scrutiny to make sure that\nall the assertions are still valid, and make sure we don\u0027t need more\nthan are already here.\n\nNote that there are a couple assertions that were looking for CONFLICT\ninstead of FORBIDDEN on the special role attempt, that seems to be\na case of \"test the current behavior not the desired one\".\n\nChange-Id: Ib6540223468421776f07745010e5d4afe5d58774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1f8c317ed87b18198b0e91504e464cda746a26b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1f8c317ed87b18198b0e91504e464cda746a26b7"}]},"branch":"refs/heads/master"},"784ee8c074f1c36ca6447330dc755631e14ca2bb":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-07-02 14:49:29.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/789914/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/789914/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/789914/7"}}},"commit":{"parents":[{"commit":"7500722d588f4a04c0a2df6ec97147ecf1fdac12","subject":"Refactor gateway get_repo auth layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7500722d588f4a04c0a2df6ec97147ecf1fdac12"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-05-05 15:21:08.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-02 15:32:14.000000000","tz":-420},"subject":"Make property protection tests use member role","message":"Make property protection tests use member role\n\nIn order to avoid overriding the default policy in tests to be\ncompletely unrestricted (as they are today), the property protection\ntests need to add the member role into the roles being used to\ntest the additional permissions. This is because the actual default\npolicies require \"admin or project member\" permissions to do anything,\nand just using one of the special roles will get blocked by the\nregular policy before property protections are checked.\n\nThis could also be fixed by constructing a very complex policy for\nthese tests which allows those special roles to do their thing in\naddition to member, but I do not believe this is how real deployments\nwould actually implement it.\n\nThese should definitely be reviewed with scrutiny to make sure that\nall the assertions are still valid, and make sure we don\u0027t need more\nthan are already here.\n\nNote that there are a couple assertions that were looking for CONFLICT\ninstead of FORBIDDEN on the special role attempt, that seems to be\na case of \"test the current behavior not the desired one\".\n\nChange-Id: Ib6540223468421776f07745010e5d4afe5d58774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/784ee8c074f1c36ca6447330dc755631e14ca2bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/784ee8c074f1c36ca6447330dc755631e14ca2bb"}]},"branch":"refs/heads/master"},"193b6e903686788e7c49f34ef8d97e4c14f88869":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-07-06 10:51:18.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/14/789914/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/789914/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/789914/8"}}},"commit":{"parents":[{"commit":"1d13fed80a753f0f79a1a2eec47f6de3593b9640","subject":"Refactor gateway get_repo auth layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1d13fed80a753f0f79a1a2eec47f6de3593b9640"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-05-05 15:21:08.000000000","tz":-420},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2021-07-05 08:56:54.000000000","tz":0},"subject":"Make property protection tests use member role","message":"Make property protection tests use member role\n\nIn order to avoid overriding the default policy in tests to be\ncompletely unrestricted (as they are today), the property protection\ntests need to add the member role into the roles being used to\ntest the additional permissions. This is because the actual default\npolicies require \"admin or project member\" permissions to do anything,\nand just using one of the special roles will get blocked by the\nregular policy before property protections are checked.\n\nThis could also be fixed by constructing a very complex policy for\nthese tests which allows those special roles to do their thing in\naddition to member, but I do not believe this is how real deployments\nwould actually implement it.\n\nThese should definitely be reviewed with scrutiny to make sure that\nall the assertions are still valid, and make sure we don\u0027t need more\nthan are already here.\n\nNote that there are a couple assertions that were looking for CONFLICT\ninstead of FORBIDDEN on the special role attempt, that seems to be\na case of \"test the current behavior not the desired one\".\n\nChange-Id: Ib6540223468421776f07745010e5d4afe5d58774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/193b6e903686788e7c49f34ef8d97e4c14f88869"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/193b6e903686788e7c49f34ef8d97e4c14f88869"}]},"branch":"refs/heads/master"},"662229ce16af78096c4ed0e101b29ae7d29f5ea5":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-07-06 18:00:34.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/789914/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/789914/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/789914/9"}}},"commit":{"parents":[{"commit":"9c6890c2f0fa9ff81d50342605f7121e55ab5134","subject":"Refactor gateway get_repo auth layer","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9c6890c2f0fa9ff81d50342605f7121e55ab5134"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-05-05 15:21:08.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-06 18:57:36.000000000","tz":-420},"subject":"Make property protection tests use member role","message":"Make property protection tests use member role\n\nIn order to avoid overriding the default policy in tests to be\ncompletely unrestricted (as they are today), the property protection\ntests need to add the member role into the roles being used to\ntest the additional permissions. This is because the actual default\npolicies require \"admin or project member\" permissions to do anything,\nand just using one of the special roles will get blocked by the\nregular policy before property protections are checked.\n\nThis could also be fixed by constructing a very complex policy for\nthese tests which allows those special roles to do their thing in\naddition to member, but I do not believe this is how real deployments\nwould actually implement it.\n\nThese should definitely be reviewed with scrutiny to make sure that\nall the assertions are still valid, and make sure we don\u0027t need more\nthan are already here.\n\nNote that there are a couple assertions that were looking for CONFLICT\ninstead of FORBIDDEN on the special role attempt, that seems to be\na case of \"test the current behavior not the desired one\".\n\nChange-Id: Ib6540223468421776f07745010e5d4afe5d58774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/662229ce16af78096c4ed0e101b29ae7d29f5ea5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/662229ce16af78096c4ed0e101b29ae7d29f5ea5"}]},"branch":"refs/heads/master"},"566d3429109f4ea14de5ada8e54421980b4d7b21":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2021-07-14 14:03:38.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/789914/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/789914/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/789914/10"}}},"commit":{"parents":[{"commit":"2b62197f91cf52566d7acbce6edf68c2e7477a17","subject":"Merge \"Refresh Glance example configs for Xena milestone 2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b62197f91cf52566d7acbce6edf68c2e7477a17"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-05-05 15:21:08.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-14 15:05:43.000000000","tz":-420},"subject":"Make property protection tests use member role","message":"Make property protection tests use member role\n\nIn order to avoid overriding the default policy in tests to be\ncompletely unrestricted (as they are today), the property protection\ntests need to add the member role into the roles being used to\ntest the additional permissions. This is because the actual default\npolicies require \"admin or project member\" permissions to do anything,\nand just using one of the special roles will get blocked by the\nregular policy before property protections are checked.\n\nThis could also be fixed by constructing a very complex policy for\nthese tests which allows those special roles to do their thing in\naddition to member, but I do not believe this is how real deployments\nwould actually implement it.\n\nThese should definitely be reviewed with scrutiny to make sure that\nall the assertions are still valid, and make sure we don\u0027t need more\nthan are already here.\n\nNote that there are a couple assertions that were looking for CONFLICT\ninstead of FORBIDDEN on the special role attempt, that seems to be\na case of \"test the current behavior not the desired one\".\n\nChange-Id: Ib6540223468421776f07745010e5d4afe5d58774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/566d3429109f4ea14de5ada8e54421980b4d7b21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/566d3429109f4ea14de5ada8e54421980b4d7b21"}]},"branch":"refs/heads/master"},"721be05cc578ba97a07ed3b6bedf40d01e795b0d":{"kind":"REWORK","_number":11,"created":"2021-07-14 19:32:52.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/789914/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/789914/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/789914/11"}}},"commit":{"parents":[{"commit":"d47d3a7a4894ba30c9d94aba7fbdcc716de3ee9a","subject":"Make our functional tests compatible with RBAC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d47d3a7a4894ba30c9d94aba7fbdcc716de3ee9a"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-05-05 15:21:08.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-14 20:28:30.000000000","tz":-420},"subject":"Make property protection tests use member role","message":"Make property protection tests use member role\n\nIn order to avoid overriding the default policy in tests to be\ncompletely unrestricted (as they are today), the property protection\ntests need to add the member role into the roles being used to\ntest the additional permissions. This is because the actual default\npolicies require \"admin or project member\" permissions to do anything,\nand just using one of the special roles will get blocked by the\nregular policy before property protections are checked.\n\nThis could also be fixed by constructing a very complex policy for\nthese tests which allows those special roles to do their thing in\naddition to member, but I do not believe this is how real deployments\nwould actually implement it.\n\nThese should definitely be reviewed with scrutiny to make sure that\nall the assertions are still valid, and make sure we don\u0027t need more\nthan are already here.\n\nNote that there are a couple assertions that were looking for CONFLICT\ninstead of FORBIDDEN on the special role attempt, that seems to be\na case of \"test the current behavior not the desired one\".\n\nChange-Id: Ib6540223468421776f07745010e5d4afe5d58774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/721be05cc578ba97a07ed3b6bedf40d01e795b0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/721be05cc578ba97a07ed3b6bedf40d01e795b0d"}]},"branch":"refs/heads/master"},"76162a7222c3d2c5eeae03ca7a710cd6401b36da":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2021-07-16 13:56:40.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/14/789914/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/14/789914/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/14/789914/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/14/789914/12"}}},"commit":{"parents":[{"commit":"525b6addd0bddb9bab8093a15614f286aca94a90","subject":"Make our functional tests compatible with RBAC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/525b6addd0bddb9bab8093a15614f286aca94a90"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-05-05 15:21:08.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2021-07-16 14:58:46.000000000","tz":-420},"subject":"Make property protection tests use member role","message":"Make property protection tests use member role\n\nIn order to avoid overriding the default policy in tests to be\ncompletely unrestricted (as they are today), the property protection\ntests need to add the member role into the roles being used to\ntest the additional permissions. This is because the actual default\npolicies require \"admin or project member\" permissions to do anything,\nand just using one of the special roles will get blocked by the\nregular policy before property protections are checked.\n\nThis could also be fixed by constructing a very complex policy for\nthese tests which allows those special roles to do their thing in\naddition to member, but I do not believe this is how real deployments\nwould actually implement it.\n\nThese should definitely be reviewed with scrutiny to make sure that\nall the assertions are still valid, and make sure we don\u0027t need more\nthan are already here.\n\nNote that there are a couple assertions that were looking for CONFLICT\ninstead of FORBIDDEN on the special role attempt, that seems to be\na case of \"test the current behavior not the desired one\".\n\nChange-Id: Ib6540223468421776f07745010e5d4afe5d58774\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/76162a7222c3d2c5eeae03ca7a710cd6401b36da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/76162a7222c3d2c5eeae03ca7a710cd6401b36da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
