)]}'
{"id":"openstack%2Fglance~791636","triplet_id":"openstack%2Fglance~master~I237daedf88d627286d0c76b1ead3d375a7948b27","project":"openstack/glance","branch":"master","topic":"replace_collections","hashtags":[],"change_id":"I237daedf88d627286d0c76b1ead3d375a7948b27","subject":"Replace collections.Iterable with collections.abc.Iterable","status":"ABANDONED","created":"2021-05-17 05:10:48.000000000","updated":"2021-08-11 11:27:28.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6045ad40567069792e2db37ef9c6b7bb88c31e06","_number":791636,"virtual_id_number":791636,"owner":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-05-17 06:32:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"all":[{"value":-1,"date":"2021-06-14 14:25:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-17 06:32:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-14 14:25:32.000000000","updated_by":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"reviewer":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"state":"REVIEWER"}],"messages":[{"id":"127af3834796bd86a7ca3f02dc7e8b7cc32310df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-05-17 05:10:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6133583ea330f2e349d85317109eb732c26c4c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 06:32:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/7ff4bf0403d24f7694e32884ecc395ae : SUCCESS in 1h 03m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a4e41550ae15446c8fdeeb854188b9ad : SUCCESS in 1h 04m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b29526e390664b55ac26c606ea3853fc : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e43401c5b71e424ba32ae3089997ff0d : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cb26cd4292547099440900a587f1fdc : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58bd95f69cd14989ad419f72ed698ef9 : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0da1eb7289a84f97a44b9f20e602c58f : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7ce850a27aa4f26b8f328feee31fbf0 : SUCCESS in 6m 44s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/fd70c10d2c454e33be179631e5da7892 : SUCCESS in 8m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e463ff357b6844b4a833bed34338a001 : SUCCESS in 8m 14s\n- glance-code-constants-check https://zuul.opendev.org/t/openstack/build/5edd58c29f0e49f782c1fed7946f5692 : SUCCESS in 5m 00s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8724a9ef48a249088fd6f95cb083bd54 : SUCCESS in 1h 17m 40s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c76fc443f71a4bf79d01e6f5827aead4 : SUCCESS in 1h 08m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/48afaeb0c87f405eb13490433b4458f5 : SUCCESS in 1h 08m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/60dd1ba46e1c41a2bed63a85c6d5c994 : SUCCESS in 1h 13m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b516cf68b81450182641a447c4a5f77 : SUCCESS in 52m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/009f8c4ea4a44d619b687a6818e86c71 : SUCCESS in 1h 17m 23s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e0627f8c2ce649739f337704174f2dfe : SUCCESS in 29m 52s","accounts_in_message":[],"_revision_number":1},{"id":"04fda88a2aa86d7cd399aa65fee3ae68de542c7a","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-06-14 14:25:32.000000000","message":"Patch Set 1: Code-Review-1\n\nWwe probably should just smash this together with all collections deprecation violations. There is only couple of them left in the codebase: glance/cmd/cache_manage.py and glance/tests/functional/db/test_migrations.py","accounts_in_message":[],"_revision_number":1},{"id":"62b28d1e7fe16f7241b40201d2e734e0e2fb02b4","author":{"_account_id":5202,"name":"Erno Kuvaja","email":"jokke@usr.fi","username":"jokke"},"date":"2021-06-14 14:43:46.000000000","message":"Patch Set 1:\n\nThis is also duplicate of https://review.opendev.org/c/openstack/glance/+/773357","accounts_in_message":[],"_revision_number":1},{"id":"6045ad40567069792e2db37ef9c6b7bb88c31e06","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2021-08-11 11:27:28.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78eb156f663e125da26c8a3a0ce7eef047e10fef","revisions":{"78eb156f663e125da26c8a3a0ce7eef047e10fef":{"kind":"REWORK","_number":1,"created":"2021-05-17 05:10:48.000000000","uploader":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"ref":"refs/changes/36/791636/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/791636/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/791636/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/791636/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/791636/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/791636/1"}}},"commit":{"parents":[{"commit":"1157c19a9c1408faad5b845cc9b1873f82fc8f31","subject":"Merge \"Guidelines for core reviewers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1157c19a9c1408faad5b845cc9b1873f82fc8f31"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2021-05-17 05:03:45.000000000","tz":480},"committer":{"name":"likui","email":"likui@yovole.com","date":"2021-05-17 05:08:55.000000000","tz":480},"subject":"Replace collections.Iterable with collections.abc.Iterable","message":"Replace collections.Iterable with collections.abc.Iterable\n\nUsing or importing the Iterable from \u0027collections\u0027\ninstead of from \u0027collections.abc\u0027 is deprecated\n\nChange-Id: I237daedf88d627286d0c76b1ead3d375a7948b27\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/78eb156f663e125da26c8a3a0ce7eef047e10fef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/78eb156f663e125da26c8a3a0ce7eef047e10fef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
