)]}'
{"id":"openstack%2Fglance~794356","triplet_id":"openstack%2Fglance~master~Ia5680cc65caf431bde800366eabd95295a44418a","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2022-07-07 14:07:58.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia5680cc65caf431bde800366eabd95295a44418a","subject":"DNM: Testing glance cinder against v2 removal","status":"ABANDONED","created":"2021-06-03 00:40:06.000000000","updated":"2022-07-07 14:07:58.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"26dc59564cb636ab9624321aac0d88950f6bdf99","_number":794356,"virtual_id_number":794356,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-17 15:13:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"all":[{"value":1,"date":"2022-07-07 14:07:58.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"}],"CC":[{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-03 01:18:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-04 18:53:34.000000000","updated_by":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"reviewer":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-17 14:12:46.000000000","updated_by":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"reviewer":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"state":"REVIEWER"}],"messages":[{"id":"65399033c65f394d35605df8f7989414fb40199f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2021-06-03 00:40:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65c1b73820837f4f24d96a16e5acffffc0569b75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-03 01:18:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90f81c5a1e594eeb9f4bfdb8fec5044e : SUCCESS in 7m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/465b6edfaf6b4bcd954987b78b23a04e : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c14a72aefb8540fdb86e6bf4003399f9 : SUCCESS in 6m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3f317a4ac1e4660ae3786801107fb14 : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae51c2eff68f4b32874e01236253f123 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3d64c509ffc4d918c99c31d21a99647 : SUCCESS in 7m 16s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/0d9e65affd9440f1986aa02d925e7bda : SUCCESS in 10m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/20c8b61e42b14560b1f9c03c120f58a9 : SUCCESS in 8m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6dbef625a8e148cd90596334dfb917ff : SUCCESS in 33m 14s","accounts_in_message":[],"_revision_number":1},{"id":"2e5ddf37218052a99bfcbde4dd78d6a9d77f72c5","author":{"_account_id":9008,"name":"VMware NSX CI","username":"vmwareminesweeper","tags":["SERVICE_USER"]},"date":"2021-06-04 18:53:34.000000000","message":"Patch Set 1:\n\nBuild failed\n\n- dsvm-glance http://207.189.188.190/logs/794356/794356/1/check-not-vote/ext-glance-zuul/a0ecf42 : FAILURE in 20m 34s","accounts_in_message":[],"_revision_number":1},{"id":"bcbacf83d62240d2465b0a3eef62acc40fa371b3","author":{"_account_id":30615,"name":"Tushar Trambak Gite","email":"tushargite96@gmail.com","username":"tushargite96"},"date":"2022-05-17 14:12:46.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b3d0a50aeca3bc8d0b173a7cf05b194b1b17c518","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-05-17 14:19:19.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"5138f8003439822aa49b81b0fb6bb2a22997748b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-17 15:13:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9e38a02d5aab415c9f28ae9b674c9090 : SUCCESS in 48m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/504ac58fe762412cb83f3a779b517149 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1414df6edf0046999424b80a670cd528 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff8bddcc6d414540839c75b3d6a60b3a : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7621dc9afbcb48cab3f77f57181454be : SUCCESS in 6m 59s\n- openstack-tox-functional-py36-fips https://zuul.opendev.org/t/openstack/build/dac017bdaac4411e95fda3b944c1e9b2 : SUCCESS in 11m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d7e24515b52b4b59a390c0895300758e : SUCCESS in 9m 22s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fb286d0ca37344f1b8d442dbf7e5933e : SUCCESS in 9m 35s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/917357578d4548068116f94e417d480d : SUCCESS in 45m 33s","accounts_in_message":[],"_revision_number":2},{"id":"7c7d7cd00edba3902755baef97bdf62326cb4db7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2022-07-07 14:07:58.000000000","message":"Abandoned\n\nNo longer required","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4edc2621fd0811aa034195cf66912a127af513ec","revisions":{"f3f156d997be52c21303608ae16704813bb4b293":{"kind":"REWORK","_number":1,"created":"2021-06-03 00:40:06.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/56/794356/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/794356/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/794356/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/794356/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/794356/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/794356/1"}}},"commit":{"parents":[{"commit":"c5c7a2a6342fc3de0b9cc2ec3f0ff7b907cc1fc7","subject":"[Doc] Update Freenode to OFTC as our IRC server","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c5c7a2a6342fc3de0b9cc2ec3f0ff7b907cc1fc7"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-06-02 13:20:22.000000000","tz":-240},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-06-02 13:22:48.000000000","tz":-240},"subject":"DNM: Testing glance cinder against v2 removal","message":"DNM: Testing glance cinder against v2 removal\n\nDepends-On: https://review.opendev.org/c/openstack/python-cinderclient/+/792959\nChange-Id: Ia5680cc65caf431bde800366eabd95295a44418a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f3f156d997be52c21303608ae16704813bb4b293"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f3f156d997be52c21303608ae16704813bb4b293"}]},"branch":"refs/heads/master"},"4edc2621fd0811aa034195cf66912a127af513ec":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-05-17 14:19:19.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/56/794356/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/56/794356/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/794356/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/794356/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/56/794356/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/56/794356/2"}}},"commit":{"parents":[{"commit":"35e6c57bc90365796f45ada9271e532068557b8a","subject":"Fix failing namespace list delete race","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/35e6c57bc90365796f45ada9271e532068557b8a"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2021-06-02 13:20:22.000000000","tz":-240},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2022-05-17 14:19:19.000000000","tz":0},"subject":"DNM: Testing glance cinder against v2 removal","message":"DNM: Testing glance cinder against v2 removal\n\nDepends-On: https://review.opendev.org/c/openstack/python-cinderclient/+/792959\nChange-Id: Ia5680cc65caf431bde800366eabd95295a44418a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4edc2621fd0811aa034195cf66912a127af513ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4edc2621fd0811aa034195cf66912a127af513ec"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
